)]}'
{"id":"openstack%2Fcinder-tempest-plugin~1004866","triplet_id":"openstack%2Fcinder-tempest-plugin~master~If05f28cb73ba2bcb987a465ef148dd8d07b3e8d7","project":"openstack/cinder-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-23 15:34:26.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"If05f28cb73ba2bcb987a465ef148dd8d07b3e8d7","subject":"DNM: Test change I50cf937a39","status":"NEW","created":"2026-09-09 14:49:25.000000000","updated":"2026-09-23 15:34:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b5c9a051428f8c7ec64a445bca6b08827d84c94b","_number":1004866,"virtual_id_number":1004866,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-09 18:14:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 15:01:32.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"CC"},{"updated":"2026-09-09 18:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7df33861068c200d68ec1c178638abf952ff1f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-09 14:49:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1175791d75c6e84376f8315f6de1d2ed93952ccb","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-09-09 15:01:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67b907db454c74b3fa5764868e1ca31a9c98c422","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-09-09 15:02:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e7efe20ed0832f260e3c01f222e2c1ebb4740f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:14:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a95e73ab474400fac2c16169e64264e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2cec4c988b48e6a96ee9ad50167ef9 : FAILURE in 3m 25s\n- cinder-tempest-plugin-multi-ceph-replication https://zuul.opendev.org/t/openstack/build/8fbcb56aa98d4e3a99ae34a53000a51a : FAILURE in 44m 19s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/06036772223b4184b05a0c3e932cea64 : SUCCESS in 1h 59m 39s\n- cinder-tempest-plugin-lvm-concurrency-tests https://zuul.opendev.org/t/openstack/build/67dc2ef8b83f42a7baa65545b6c01a96 : SUCCESS in 27m 41s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c70d9d0c38a04e8ea15cce47ce6e67d8 : SUCCESS in 2h 06m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream-py311 https://zuul.opendev.org/t/openstack/build/dc37ab0b98724b1fa521fe63331f3622 : SUCCESS in 1h 43m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream-2025-1 https://zuul.opendev.org/t/openstack/build/7aaf473a71aa4c68857c5da838a973d3 : FAILURE in 36m 39s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/ec81e29bea6544829bc3e1cd2596a044 : TIMED_OUT in 3h 06m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1 https://zuul.opendev.org/t/openstack/build/a5a87fb4dee241b492b9fa192630d1c0 : FAILURE in 36m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7a90b35d448f437db0a047810bf3404a : SUCCESS in 2h 05m 06s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/163e6b36c120433bb33caf5d48ad4640 : SUCCESS in 1h 58m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/26aa3c2cf4c744f1b3ab79b57184338c : SUCCESS in 1h 33m 16s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/0f3e6160ac2445df8b7d38c7586112fb : SUCCESS in 1h 47m 29s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/d504882a0d1e49d1a0f71a6984ed200f : SUCCESS in 46m 05s\n- cinder-tempest-plugin-basic-2026-1 https://zuul.opendev.org/t/openstack/build/c234166800714397a72b999cfffaedfb : SUCCESS in 42m 04s\n- cinder-tempest-plugin-basic-2025-2 https://zuul.opendev.org/t/openstack/build/381d2dc0dda8450fbe2ffc61d4c9754b : SUCCESS in 35m 58s\n- cinder-tempest-plugin-basic-2025-1 https://zuul.opendev.org/t/openstack/build/933c40c05ac2454eab70d59738f3ed60 : SUCCESS in 43m 04s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d0565fa4b69a4339a3cfcb205a45d438 : SUCCESS in 38m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"b5c9a051428f8c7ec64a445bca6b08827d84c94b","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-23 15:34:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f78f360dc16f306d83d7d398c1a9ee49b537166d","revisions":{"f78f360dc16f306d83d7d398c1a9ee49b537166d":{"kind":"REWORK","_number":1,"created":"2026-09-09 14:49:25.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/66/1004866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-tempest-plugin","ref":"refs/changes/66/1004866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/66/1004866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/66/1004866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/66/1004866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/66/1004866/1"}}},"commit":{"parents":[{"commit":"d4894b4df9916d5f7b1968a4e3093f40af4eed44","subject":"Merge \"Drop python 3.10 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/d4894b4df9916d5f7b1968a4e3093f40af4eed44"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-09 14:48:39.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-09 14:49:05.000000000","tz":-240},"subject":"DNM: Test change I50cf937a39","message":"DNM: Test change I50cf937a39\n\nDepends-On: I50cf937a399d020bf0f5b08c33675d123994f895\nChange-Id: If05f28cb73ba2bcb987a465ef148dd8d07b3e8d7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/f78f360dc16f306d83d7d398c1a9ee49b537166d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/f78f360dc16f306d83d7d398c1a9ee49b537166d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
