)]}'
{"id":"openstack%2Fcinder-tempest-plugin~1005308","triplet_id":"openstack%2Fcinder-tempest-plugin~master~Ibe4810f6304058a7403826f1ef3cca54b9b54b47","project":"openstack/cinder-tempest-plugin","branch":"master","topic":"2026.2-job","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-21 19:25:11.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ibe4810f6304058a7403826f1ef3cca54b9b54b47","subject":"Add stable/2026.2 job","status":"NEW","created":"2026-09-11 17:03:42.000000000","updated":"2026-09-21 19:25:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b494e556bfac7df62c5ee30ec0a5de13c7eb6042","_number":1005308,"virtual_id_number":1005308,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-11 19:26:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2026-09-21 16:16:31.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 19:26:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e34dd88524be64769bd28b9b852021775b5f7d0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-11 17:03:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb94b54710d668544f191d8c6ed15ad6671fe6e5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-11 17:04:41.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e92f034eeedfa3c8ff5181eb02906e2898cd32ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 19:26:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/088d641e67084d3da01ddd2c88e890dd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042c02dc16864f1ab21cb90a729047da : SUCCESS in 3m 38s\n- cinder-tempest-plugin-multi-ceph-replication https://zuul.opendev.org/t/openstack/build/d0d3373c819a4742abbcb7233faa3c4a : FAILURE in 45m 00s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/a68d6671a4bf4cc08dbd6c494d6c640d : SUCCESS in 1h 24m 42s\n- cinder-tempest-plugin-lvm-concurrency-tests https://zuul.opendev.org/t/openstack/build/e35779713ad74dbeaead07db5c6346eb : SUCCESS in 38m 21s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b46b837d7ee54cd58862f1d8a33da025 : SUCCESS in 1h 15m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream-py311 https://zuul.opendev.org/t/openstack/build/813771a889f64423a71f5f062252a910 : SUCCESS in 1h 24m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream-2025-1 https://zuul.opendev.org/t/openstack/build/88cedeb2c350453b9423bbc227c39e83 : FAILURE in 19m 14s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/71d1a38690c549eeb4c80156f1de0509 : FAILURE in 2h 20m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1 https://zuul.opendev.org/t/openstack/build/10817c3f0be947c78441a83cab71376b : FAILURE in 21m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/95ad2541beb043bda3f6a5c0941156f5 : SUCCESS in 1h 42m 59s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/3775bc2494764303b1a5e4c473f7e482 : SUCCESS in 1h 43m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1e689f8dc4d4e3eadb6ec3f066d1004 : SUCCESS in 1h 03m 37s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/c445d967336f481b862f252a95c1d38b : SUCCESS in 1h 24m 50s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/377042dab9674b458f93b04ec593ea90 : FAILURE in 19m 20s\n- cinder-tempest-plugin-basic-2026-2 https://zuul.opendev.org/t/openstack/build/66e8dbe062a9489ba7d7330de35a7494 : SUCCESS in 35m 47s\n- cinder-tempest-plugin-basic-2026-1 https://zuul.opendev.org/t/openstack/build/a525721dfadb4e508c3d87ab7a242200 : SUCCESS in 33m 13s\n- cinder-tempest-plugin-basic-2025-2 https://zuul.opendev.org/t/openstack/build/f7d729cd05de4b3dbd5a0009d963ae74 : SUCCESS in 28m 46s\n- cinder-tempest-plugin-basic-2025-1 https://zuul.opendev.org/t/openstack/build/c95be14879544486a1403c1e6f36a0ef : SUCCESS in 44m 35s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f04a7cb24eb54552bf6fe568c0fa7af2 : SUCCESS in 34m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"1ed4777ff3f9e1d95e9dc790d3992a5a78e58ea1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-17 14:22:55.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"33169f7f39ad53185a16843fb52f93181a1834ad","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-17 14:23:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95d0b70130da7269618388c378efdbd30a9b79c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 17:36:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/904d4c24a7194757abbf8b0fd8994fcc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a87db56c09714c18a2040f171fff90b4 : SUCCESS in 3m 08s\n- cinder-tempest-plugin-multi-ceph-replication https://zuul.opendev.org/t/openstack/build/7532c9af91434c269456980404b3fde6 : FAILURE in 55m 59s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/865545da123d46459fd2e9cf9d7c9a49 : SUCCESS in 1h 56m 10s\n- cinder-tempest-plugin-lvm-concurrency-tests https://zuul.opendev.org/t/openstack/build/97326ccad48541f0bc333743b327fa6b : SUCCESS in 34m 24s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0fad4213b052446bb90ea2a9a7dec54f : SUCCESS in 2h 20m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream-py311 https://zuul.opendev.org/t/openstack/build/59ac820dc6324e169c98f1020acdd9b4 : SUCCESS in 1h 44m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream-2025-1 https://zuul.opendev.org/t/openstack/build/545974c44a36421da1aebed7d5a4babd : FAILURE in 21m 16s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/57252bf1e40d4c1fba9d7fb58dcfeec8 : TIMED_OUT in 3h 07m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1 https://zuul.opendev.org/t/openstack/build/04c8aa61a33f400a9bccc7a3b8b475f0 : FAILURE in 32m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1a138264ab494b12bf94df8e59ae261f : SUCCESS in 1h 50m 03s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/9e2f3f03266e477986fc74a9327c8939 : SUCCESS in 1h 28m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2fae3b6ff43c452c86913e718a21cbfa : SUCCESS in 1h 45m 37s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/efff8ecc448b4031966c34e8629441ee : SUCCESS in 1h 36m 57s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/d8a66ce28b5e458eaae1759774d3213d : FAILURE in 25m 48s\n- cinder-tempest-plugin-basic-2026-2 https://zuul.opendev.org/t/openstack/build/0e5a7ec283264377b46ff02427263d32 : SUCCESS in 44m 02s\n- cinder-tempest-plugin-basic-2026-1 https://zuul.opendev.org/t/openstack/build/83c8d6827787401dbddc2c768d00676b : SUCCESS in 54m 05s\n- cinder-tempest-plugin-basic-2025-2 https://zuul.opendev.org/t/openstack/build/ef86e1f4ed0f4b6b9f485cc41cf03e8e : SUCCESS in 53m 12s\n- cinder-tempest-plugin-basic-2025-1 https://zuul.opendev.org/t/openstack/build/6164d1aca41e45c185d7c7d7250fff25 : SUCCESS in 40m 50s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/826e90e44d104f378ca85cdb0748ff5d : SUCCESS in 26m 12s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"775e06955522533629912e9bae7ceccbec1f6f20","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-21 16:16:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b494e556bfac7df62c5ee30ec0a5de13c7eb6042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:25:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f757b357a38456c9dba135ed9169492\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d46da063abe40f09b41cb8d32942b49 : SUCCESS in 3m 24s\n- cinder-tempest-plugin-multi-ceph-replication https://zuul.opendev.org/t/openstack/build/431efd841d954af895eb3cdc6022c7cd : SUCCESS in 39m 05s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/b78d8c52a17e4865aaf8564b80a0a255 : SUCCESS in 1h 55m 34s\n- cinder-tempest-plugin-lvm-concurrency-tests https://zuul.opendev.org/t/openstack/build/15244b3139544a36a036f094b8a7f463 : SUCCESS in 33m 57s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/176f486d1b9e4536bf0f9315fb0299c5 : SUCCESS in 1h 24m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream-py311 https://zuul.opendev.org/t/openstack/build/6b6a069a63af4f35a43e39c271af5b2e : SUCCESS in 1h 31m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream-2025-1 https://zuul.opendev.org/t/openstack/build/d28de88200ef415da767cbcb24ba3eb5 : FAILURE in 20m 12s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/1f7cc7a4bd3e4518a2cb85555dd6ca24 : FAILURE in 3h 01m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1 https://zuul.opendev.org/t/openstack/build/91b56fd8f5ed4044bb1030ccd223eba8 : FAILURE in 21m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6d6dad1e428449989a90ad2d9b8718b8 : SUCCESS in 1h 24m 57s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/eb225bd336df4692870829f9200259d2 : SUCCESS in 1h 37m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/98bf4166853b4f4fa3e97f5f5a20d082 : SUCCESS in 58m 28s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/ffe520309dd74b1aaf96b937cc48fb94 : SUCCESS in 1h 39m 01s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/e530ab61a856465fad6c69861c1e8bdc : FAILURE in 33m 06s\n- cinder-tempest-plugin-basic-2026-2 https://zuul.opendev.org/t/openstack/build/b15573c9e0e94f2b925e9ef8a7ee7154 : SUCCESS in 42m 58s\n- cinder-tempest-plugin-basic-2026-1 https://zuul.opendev.org/t/openstack/build/4ab7f56cc9fe4162af1dbd63ef492005 : SUCCESS in 46m 27s\n- cinder-tempest-plugin-basic-2025-2 https://zuul.opendev.org/t/openstack/build/0ac22bf4d9834980833d90f141d008b4 : SUCCESS in 43m 55s\n- cinder-tempest-plugin-basic-2025-1 https://zuul.opendev.org/t/openstack/build/30e064c6833041f4a21521d4c9e42f3a : SUCCESS in 44m 41s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0219976a160f4945b24a04050380a75a : SUCCESS in 22m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be6bd9fc7fd5b101eeb24475e898c7f47a5f6d1e","revisions":{"be6bd9fc7fd5b101eeb24475e898c7f47a5f6d1e":{"kind":"REWORK","_number":1,"created":"2026-09-11 17:03:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/1005308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-tempest-plugin","ref":"refs/changes/08/1005308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/08/1005308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/08/1005308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/08/1005308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/08/1005308/1"}}},"commit":{"parents":[{"commit":"d4894b4df9916d5f7b1968a4e3093f40af4eed44","subject":"Merge \"Drop python 3.10 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/d4894b4df9916d5f7b1968a4e3093f40af4eed44"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:03:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:03:32.000000000","tz":540},"subject":"Add stable/2026.2 job","message":"Add stable/2026.2 job\n\nThe branch was created so we should start testing it.\n\nChange-Id: Ibe4810f6304058a7403826f1ef3cca54b9b54b47\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/be6bd9fc7fd5b101eeb24475e898c7f47a5f6d1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/be6bd9fc7fd5b101eeb24475e898c7f47a5f6d1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
