)]}'
{"id":"openstack%2Fcinder-tempest-plugin~524856","triplet_id":"openstack%2Fcinder-tempest-plugin~master~I5046be23703192ed6fad805355c8c50b9b4a71c8","project":"openstack/cinder-tempest-plugin","branch":"master","topic":"bp/goal-split-tempest-plugin","hashtags":[],"change_id":"I5046be23703192ed6fad805355c8c50b9b4a71c8","subject":"Fix wrong links in Cinder","status":"MERGED","created":"2017-12-02 09:58:21.000000000","updated":"2017-12-02 10:11:58.000000000","submitted":"2017-12-02 10:11:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"524856-1512209518312-78624289","meta_rev_id":"9bfaa5fc0cc71386369c9b124d03bec2eac6c713","_number":524856,"virtual_id_number":524856,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25432,"name":"yanghuichan","email":"yanghc@fiberhome.com","username":"yanghuichan"},{"value":2,"date":"2017-12-02 10:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2017-12-02 09:59:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25432,"name":"yanghuichan","email":"yanghc@fiberhome.com","username":"yanghuichan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2017-12-02 09:59:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2017-12-02 09:58:21.000000000","_account_id":25432,"name":"yanghuichan","email":"yanghc@fiberhome.com","username":"yanghuichan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25432,"name":"yanghuichan","email":"yanghc@fiberhome.com","username":"yanghuichan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25432,"name":"yanghuichan","email":"yanghc@fiberhome.com","username":"yanghuichan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-02 09:58:21.000000000","updated_by":{"_account_id":25432,"name":"yanghuichan","email":"yanghc@fiberhome.com","username":"yanghuichan"},"reviewer":{"_account_id":25432,"name":"yanghuichan","email":"yanghc@fiberhome.com","username":"yanghuichan"},"state":"REVIEWER"},{"updated":"2017-12-02 10:11:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b7691c8818254692736665ca478f40d1ec91072","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2017-12-02 09:58:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b0243b0e0966080fb6c748503f58bf4049547e0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2017-12-02 09:59:34.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2\n\nSyncing repo migration.","accounts_in_message":[],"_revision_number":1},{"id":"4afe83bad2758080d8b7c05889048c5230b0af1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-02 10:02:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/56/524856/1/check/openstack-tox-pep8/baf26b5/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"33f4c5236f414548a914a35c3fc733dd4d305810","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-02 10:06:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8b68f5741573fb7e3b5894f940a979b044e328e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-02 10:11:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/56/524856/1/gate/openstack-tox-pep8/be20e89/ : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5be5bc3c733ffbb69c88f2dfcddf6a578eb1c8fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-02 10:11:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7006aa812509c2f99969ef1e3c07f1d64abeb136","revisions":{"7006aa812509c2f99969ef1e3c07f1d64abeb136":{"kind":"REWORK","_number":1,"created":"2017-12-02 09:58:21.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/56/524856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-tempest-plugin","ref":"refs/changes/56/524856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/56/524856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/56/524856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/56/524856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/56/524856/1"}}},"commit":{"parents":[{"commit":"b4e0dab36a21fe7687e718988895290ac2472abe","subject":"Fix tempest test revet_client typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/b4e0dab36a21fe7687e718988895290ac2472abe"}]}],"author":{"name":"yanghuichan","email":"yanghc@fiberhome.com","date":"2017-09-06 07:14:27.000000000","tz":480},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@huawei.com","date":"2017-12-01 22:34:28.000000000","tz":-360},"subject":"Fix wrong links in Cinder","message":"Fix wrong links in Cinder\n\nSome docs links have changed.\nWe should update the wrong links in our codes.\n\nChange-Id: I5046be23703192ed6fad805355c8c50b9b4a71c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/7006aa812509c2f99969ef1e3c07f1d64abeb136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/7006aa812509c2f99969ef1e3c07f1d64abeb136"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b4e0dab36a21fe7687e718988895290ac2472abe","is_merged_in_target_branch":true,"change_id":"Ib4a0c50f699edb1cc2ea669ac711eb2f537be417","change_number":524855,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
