)]}'
{"id":"openstack%2Fcinder-tempest-plugin~918453","triplet_id":"openstack%2Fcinder-tempest-plugin~master~If80008f7fc72fa39c754e03b664d7d5e6e5dfb6d","project":"openstack/cinder-tempest-plugin","branch":"master","topic":"remove-zed","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-05-09 23:25:20.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-05-08 10:11:02.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"If80008f7fc72fa39c754e03b664d7d5e6e5dfb6d","subject":"Remove stable/zed job","status":"MERGED","created":"2024-05-07 14:18:29.000000000","updated":"2024-05-09 23:25:20.000000000","submitted":"2024-05-09 23:25:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918453-remove-zed","meta_rev_id":"b8a5a3263bb4e51e2cfc821bb62ca670c1fc4bb5","_number":918453,"virtual_id_number":918453,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-09 23:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-05-09 01:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":2,"date":"2024-05-08 10:11:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-05-09 01:09:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-05-08 12:21:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-07 14:28:40.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-05-07 16:29:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-08 20:14:34.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-09 01:09:34.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"18c24cceb4a2f3b7ac5e4e405fe0e9608b8dcf74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-07 14:18:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"293d768ca56144c600e8265358db72ecaa87f455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 16:29:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cae16231db564bc58a2bc2c7b08fa472\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31bbc13fa06143949efa4b0acb165093 : SUCCESS in 4m 10s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/dfb1121c61864d71a8fcd34a70a075cf : SUCCESS in 1h 35m 27s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/837b36a4fd854d71b0c8362bd488827f : SUCCESS in 1h 41m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/b64c3822b8ae48ea9f41ee3380792aeb : SUCCESS in 1h 32m 18s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/8fcaefbbc11c4f47b9cdaf88f1650a82 : FAILURE in 1h 35m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b2d88d2d545943f490cd105e3493ec69 : SUCCESS in 2h 04m 45s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/edd02d4771144fb3bcba9592d1e4ddde : FAILURE in 2h 07m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/44970b51f613434ab14e87445bd5ef84 : SUCCESS in 1h 54m 40s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/13149541066945bb8877d783a01565ca : SUCCESS in 2h 01m 51s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/f6a16512d67c409a9e7808768563b914 : SUCCESS in 43m 23s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/676ae4de3e014c67b129e3e6c9bfd1ed : SUCCESS in 58m 52s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/0995c61d15aa4b0595c5f72cfb2bb0cf : SUCCESS in 51m 27s\n- cinder-tempest-plugin-basic-2023-1 https://zuul.opendev.org/t/openstack/build/7423143e9e1d449b8df6368f590b46c5 : SUCCESS in 49m 12s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9af565b6ad7a4e88a3d144cd67254c60 : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8f5aa6878c76d7a2ab76f3de7d5e5f8928bfc4dc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-07 19:06:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abbc3909bcacf0a932ede0a51a97a41d520bcab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 21:30:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/723dbdf0749845e889f62a6be47588f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03211b295d864912bb68c360cb67cc83 : SUCCESS in 3m 01s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/6706456e14c042d78c740f85555610ea : FAILURE in 2h 01m 40s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0c0c0049713942e88641bdaea5402e29 : FAILURE in 1h 53m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/85659a63d07a4e0688c657a275fba703 : SUCCESS in 1h 37m 47s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/6276ac2714914cf180bc95c686ccb936 : FAILURE in 2h 21m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/09665498fde34a48b44d24157708376b : SUCCESS in 1h 46m 22s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/51d6c565bf1b410ba420f0b70dd92bda : SUCCESS in 2h 10m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a22c95729b094cc6ad343e336add7d5f : SUCCESS in 1h 32m 18s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/067f8bfd5fd3455ea2ce6fb7bb5414b4 : SUCCESS in 1h 56m 06s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/660e118708bd448f8983cc5cc2fe885c : SUCCESS in 40m 48s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/2ebafeb7de0a434f9c13890ccbc07980 : SUCCESS in 33m 33s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/91fa6d63aa1c48cd9c6b0ca653716267 : SUCCESS in 32m 24s\n- cinder-tempest-plugin-basic-2023-1 https://zuul.opendev.org/t/openstack/build/80531e99c7904de9a6c6f8162ed941aa : SUCCESS in 33m 29s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b42a0495fbf04e94b9ef1fe8c7cd3e10 : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b9b662589cf91d35b1174a9d800aced8270ef584","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-08 10:11:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b07fd4c5d66a60533e1a9129709d10c3a37e39c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-08 10:11:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6281987b348425d07711f1a35ffe5327919722d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-08 12:17:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb9c600125956237a6e4e8d17a4cd18ce485f0b3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-08 12:21:37.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bea40b9317f2e5177a03efd640bb4e8d78d45d59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 12:50:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca6d33a3478546ca8d3b4809254d6012\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e91d696803245fd8c787ef6e05cac09 : SUCCESS in 2m 50s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/be66332966ec47138821bb7ae32b25c8 : FAILURE in 1h 40m 52s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/26e954ccadd841be9a3ce4dcdabb64a8 : SUCCESS in 2h 16m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/01e23b9b559543079ac5a51dd2ed0746 : FAILURE in 8m 30s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/5c4c92e8ab56472488576e47648ecd95 : FAILURE in 1h 39m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/558c99c562294843b740983b1691eba2 : FAILURE in 9m 35s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/25eac60f5056467c9afe233ca739a0f4 : SUCCESS in 2h 05m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/45a57adc449b40a68977a66d8929c525 : SUCCESS in 1h 43m 27s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/4fa26ad9610849ddba045c8b379dbac7 : SUCCESS in 2h 31m 31s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/17fb5f3b615d424d9651a326844be0cc : SUCCESS in 42m 25s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/944a5c36058c402198b7637c2549299a : SUCCESS in 26m 08s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/77296f00292645e6a5f641afbfab9fac : SUCCESS in 47m 37s\n- cinder-tempest-plugin-basic-2023-1 https://zuul.opendev.org/t/openstack/build/e9eaabe5164946fb9c108a6891b43bd5 : SUCCESS in 42m 28s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b08ff62e573b465e912792d7cb6c6ff6 : SUCCESS in 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9ef0c053dc4e8b421e578e4c9d409313e284a005","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-08 13:46:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2dcafe2e1476c7298cc2ce0780cf0055df65284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 16:57:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/680db68e7c0941bd85097ae236491b0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001b214e00064033a3d5c50b6e6c50f5 : SUCCESS in 2m 48s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/6404dfa6e17045b7aa30d334115cbd8a : FAILURE in 1h 35m 47s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f6890923fc7949cd9d5ed927061c05a2 : SUCCESS in 1h 46m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/b173093e39a241ffafbaecb7e089aeab : FAILURE in 8m 01s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/0bb5b265bcf644efbbc00876ec1cdbb2 : TIMED_OUT in 3h 04m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4ab727e0ae8b46e092f07fc03fe7bb62 : FAILURE in 22m 03s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/054a3007c84d458e8dd1601310deb33e : FAILURE in 2h 17m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/048e215e550c4c41a95be92f60980a17 : SUCCESS in 1h 50m 40s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/51f435ea7d5d4c7fa57e3a9256e340ab : SUCCESS in 1h 36m 16s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/89162c218a034dc8ac04a6a6de9b57e6 : SUCCESS in 31m 17s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/ef4a25fd1d174986a14f93a1e96a2cf3 : SUCCESS in 44m 50s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/5e2bfc94b1514ca39804a939bb1f895e : SUCCESS in 39m 40s\n- cinder-tempest-plugin-basic-2023-1 https://zuul.opendev.org/t/openstack/build/fa4e50e39ced4d249a5b36e6c2e39d3f : SUCCESS in 46m 45s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7c93920986c14c199883e24322380994 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":1},{"id":"250212d079f95446e1d20d77f871aa7a84a62e56","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-08 17:51:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18ed0f9e387987cf7d24e2bbd303fd171893c2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 20:08:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b468c93985d418ea5060ef04ec68b85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02ffc8f2ad524123b181fdb357923ee2 : SUCCESS in 3m 16s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/b14df75be03a453ba18feee3bdf632f7 : SUCCESS in 1h 26m 49s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f3af6515678f48be80b1b11ad78fb867 : SUCCESS in 1h 46m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/95f0fabcbaa747b2917310ffbe982dbb : FAILURE in 18m 34s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/59979cd872f647a3b6736d452124c71f : FAILURE in 1h 50m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/61870a3af8974410b084d7ac3d47b1d3 : FAILURE in 26m 14s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/0a09d7251f724063a03fd03f06940604 : SUCCESS in 1h 52m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a0e7884b2ce74625abde14b25fc4ff33 : SUCCESS in 1h 57m 32s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/d26425ef9ee442c58750175d8ebff3d0 : SUCCESS in 2h 07m 07s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/78bb8fe554ec4d9aa56489b6ed656fd7 : SUCCESS in 31m 58s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/94b4836f5a084eb3855c01fee162c676 : SUCCESS in 31m 03s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/60da3ca1a2844f9984c1179dea012a3d : SUCCESS in 39m 19s\n- cinder-tempest-plugin-basic-2023-1 https://zuul.opendev.org/t/openstack/build/5136af03f10f45acab38a9d2632f645b : SUCCESS in 54m 01s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/da3e268a3cdd40d4a9f4adedc28ac908 : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":1},{"id":"613604f6f7a5de28844882ea94268fff079b12a8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-08 20:14:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/918453/1/upstream-check/cinder-tempest-plugin-ontap-fcp/ff25f3a/ : RETRY_LIMIT in 3m 54s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/918453/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/e9b3921/ : RETRY_LIMIT in 4m 05s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/918453/1/upstream-check/cinder-tempest-plugin-ontap-nfs/547aa4d/ : RETRY_LIMIT in 4m 03s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b3928472f3bc7f800b69563c6225943a591b1791","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-05-09 01:09:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7cc58b9a40b37149e6859f7bcf41ef38d4c7441a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 01:10:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30236bb2ad43f16024e7f85617a4ccd2199f8f63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 03:43:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4f8130736c64714ac054ee1ea507aee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708b5c33906f46f1b6ae101e37b2b5d1 : SUCCESS in 2m 40s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d1cc9eb953984ca9b48da7d6d8e64db8 : FAILURE in 1h 53m 52s\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/173b5a028d2e424ca4ccd2dd25907cf3 : SUCCESS in 2h 02m 01s\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/32c413f5220942188e034ce08fb6aa7d : SUCCESS in 2h 20m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f8c5f19ba1615d87c38b223b7ba66e36ef65afde","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-09 12:12:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a96e47c15c8963897accbc5b429a039147b0e89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 14:16:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ed66f86fced4a8f889ea0646f87dc17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/411a14d382c54b34b316707cb1e22572 : SUCCESS in 2m 52s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/7ff8a2a334af44769df938353aa427f3 : FAILURE in 2h 01m 49s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eccb9f908985409e8707910e2a5f652e : SUCCESS in 1h 39m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/9275f5cce1814fa088f33057cafd1028 : SUCCESS in 1h 39m 25s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/fd1f89761b43442eac192aa5f18117b9 : FAILURE in 1h 53m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3e6ca052e0f54caebf121c8bd7647c2b : SUCCESS in 1h 53m 31s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/8a1fa0151e81493bbd56069004436d6a : FAILURE in 1h 42m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b40ecca244194bfa99d2b7bb76e520e4 : SUCCESS in 1h 42m 33s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/1c94934a0cf74a15a608e1d9f380b54d : SUCCESS in 1h 44m 58s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/ff2e6c211560403db2c1aeb14edd349e : SUCCESS in 32m 09s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/12c749255d914fa6aec3189a57b64b00 : SUCCESS in 49m 21s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/262d02c96cd445bea9f71d26c4a73013 : SUCCESS in 30m 06s\n- cinder-tempest-plugin-basic-2023-1 https://zuul.opendev.org/t/openstack/build/2d5dc0569c5d4e458c5c9353f0fa55fa : SUCCESS in 36m 40s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b92e1eb28d63468d9667c09c57c3a1b4 : SUCCESS in 25m 41s","accounts_in_message":[],"_revision_number":1},{"id":"029a57f667db1924942be2ab6cc75d31295184e3","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-09 17:41:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e905605ea1cd6a0b15276b9114dc242b0b2fa347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 20:12:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e88139bf1c24c969253c9d0bac2751d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f29526c18b64353943891ad00f232bb : SUCCESS in 3m 25s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/f4ca0d1c1be24105aab1c750c9aa063c : SUCCESS in 1h 32m 45s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c67d8301feb84e8196d3afa918c59ced : SUCCESS in 1h 55m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b2e46012fb74548b5613facc9f575b6 : FAILURE in 2h 05m 50s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/3397551439454f049db970960cb012e0 : FAILURE in 1h 41m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2546ab83bb3341fb98f3ae0fdba1d313 : SUCCESS in 1h 54m 23s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/0365f80345ba4ba2ada08a82baa1b211 : SUCCESS in 1h 56m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/17dce67e5af84fbfb7bfb9b56234efc2 : SUCCESS in 2h 04m 28s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/4a982fd76e2949798a774548e548b47d : SUCCESS in 1h 51m 06s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/06acf0afb5a24e5ea2591747f1cae814 : SUCCESS in 36m 16s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/ae78c2f8d5254827bf90a586ca01244b : SUCCESS in 40m 10s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/8d951db65fb64f08b3e5c6f632999dce : SUCCESS in 34m 33s\n- cinder-tempest-plugin-basic-2023-1 https://zuul.opendev.org/t/openstack/build/4c73281d3fbb40a7abfdf47453bb12ea : SUCCESS in 31m 19s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/36a1e625233f4bad8600a90e26f51975 : SUCCESS in 28m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f50374e0b5beecce65d51cfd0287a8c496d6118b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 20:13:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0bb83dc889851cd6fd13cc2eb7c4a537bcc1f0bf","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-09 20:32:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/918453/1/upstream-check/cinder-tempest-plugin-ontap-fcp/d5093df/ : RETRY_LIMIT in 3m 24s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/918453/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/27885f5/ : RETRY_LIMIT in 3m 47s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/918453/1/upstream-check/cinder-tempest-plugin-ontap-nfs/45d1b94/ : RETRY_LIMIT in 3m 38s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"976899cba279af95e9911f18af7f915100e1a59e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 23:25:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c61ba3e5f10438da5936ad8b09ffbd8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de18084706446fe84eaa57b1cea19ad : SUCCESS in 3m 04s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d06c8d75f72a4ead8fd1b648bfa08d4b : SUCCESS in 1h 20m 34s\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/973b5b9e50a5474fbbe391dddb30c74e : SUCCESS in 1h 56m 08s\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/6aa1748f1d1641c69b2ea2c0752fdb7c : SUCCESS in 1h 19m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b8a5a3263bb4e51e2cfc821bb62ca670c1fc4bb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 23:25:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d3f89420447194d677d8577ab6743dfe14c2fa1","revisions":{"7d3f89420447194d677d8577ab6743dfe14c2fa1":{"kind":"REWORK","_number":1,"created":"2024-05-07 14:18:29.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/53/918453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-tempest-plugin","ref":"refs/changes/53/918453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/53/918453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/53/918453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/53/918453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/53/918453/1"}}},"commit":{"parents":[{"commit":"d370d9035fb70797f03e136e1c8f0999884207dc","subject":"Merge \"Update CI for Dalmatian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/d370d9035fb70797f03e136e1c8f0999884207dc"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-05-07 14:17:19.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-05-07 14:17:19.000000000","tz":-240},"subject":"Remove stable/zed job","message":"Remove stable/zed job\n\nThe zed series has gone to Unmaintained status and the stable/zed\nbranch has been deleted, so remove the associated jobs.\n\nChange-Id: If80008f7fc72fa39c754e03b664d7d5e6e5dfb6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/7d3f89420447194d677d8577ab6743dfe14c2fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/7d3f89420447194d677d8577ab6743dfe14c2fa1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
