)]}'
{"id":"openstack%2Fcinder-tempest-plugin~934275","triplet_id":"openstack%2Fcinder-tempest-plugin~master~I00426388dfef92867ea32ee018006fc2dba7d16a","project":"openstack/cinder-tempest-plugin","branch":"master","topic":"2023-1-last","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-11-20 14:23:17.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-11-21 00:39:44.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-11-21 00:39:44.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-21 00:39:44.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-11-21 00:39:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00426388dfef92867ea32ee018006fc2dba7d16a","subject":"Remove stable/2023.1 job from master gate","status":"MERGED","created":"2024-11-06 19:50:57.000000000","updated":"2024-11-21 00:39:44.000000000","submitted":"2024-11-21 00:39:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934275-2023-1-last","meta_rev_id":"725be84fb4adc613db4b49cad3cc4776151bc880","_number":934275,"virtual_id_number":934275,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-21 00:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-11-20 14:23:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-11-19 14:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-11-20 14:23:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-11-19 14:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-06 22:47:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-06 23:04:07.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-11-19 01:51:27.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-11-19 01:51:27.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-11-19 14:51:36.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2024-11-19 14:51:41.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2024-11-19 14:51:48.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-11-19 14:51:58.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"7b1f5f30504cbf7beafd68dea4065844066a1052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-06 19:50:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c1e5a2bb3cf09efac6b9ea0413ef0f805b43883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 22:47:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1867c6a1bc8f4284baa9337c5d7f3a0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/432dfb6449354bb499da03b8dcd1d9a7 : SUCCESS in 3m 55s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/1cc1f970c18f4dc0b4950d042e474d2f : SUCCESS in 1h 33m 30s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/64829ac62ff242018802ab19234795e4 : SUCCESS in 1h 56m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c08f3f53e114225bd178f9924a8404e : SUCCESS in 2h 00m 33s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/31cad3b6c5ab43778a16f48c2ab42aa0 : FAILURE in 2h 54m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/738228f39fba4d378312efc301cbc472 : SUCCESS in 2h 03m 22s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/251336a75b6a4882b825ac6a5bb01b4b : SUCCESS in 1h 51m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f7453df30a214b2091c20158ad754bae : FAILURE in 1h 49m 09s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/65592dac786247c1a6b36330a0be9be0 : SUCCESS in 2h 07m 05s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/6735d6fb4ae94a6aa191a78c770c32be : SUCCESS in 43m 11s\n- cinder-tempest-plugin-basic-2024-2 https://zuul.opendev.org/t/openstack/build/c6e3668ccf6e42819088f764a1040ebf : SUCCESS in 47m 16s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/23f99ea71ccb43a297bc6ce1c13b672b : SUCCESS in 46m 55s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/c195ea9e551c4bc185549fdf2d9a8b13 : SUCCESS in 58m 28s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d3ebf97870084784aceb4b5eb4945b84 : SUCCESS in 32m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f45b317c6f2e195c070b3c6200adfa224cfdee3c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-06 23:04:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/9f179b31c18d4f979e65609ac0d6c3c3 : RETRY_LIMIT in 5m 19s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"129b05ffa6d3c00072b13aef29cc763787a9f345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-07 17:51:05.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce0ce1a183332cf3eb2628dafa809ae59a9c3c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 19:39:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dfe3623ff174cd48069a25d2ee23d91\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01e3e2de955149638b307230bfe78b48 : SUCCESS in 2m 42s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/7919875472ab48d398c9d391619781aa : SUCCESS in 1h 36m 24s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4db186c12fbb460d8bcf411f1f6c5b85 : SUCCESS in 1h 35m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/518a5988d9b74d93a3a6ccd3210fcc67 : SUCCESS in 1h 34m 54s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/7c3287900db64936bb00700459dc4fa3 : FAILURE in 1h 44m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/55067e6954cb4583a277441b2a2a0145 : SUCCESS in 1h 31m 45s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/89fa170760294c72997d5b6913fe7b41 : SUCCESS in 1h 42m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ce48c894d75f47a4ad9ec06970848771 : SUCCESS in 1h 42m 23s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/3ee685ab3b574d11818e2a166c01ba8e : SUCCESS in 1h 27m 32s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/f540433bb82a4b11a6e660ad58d23610 : SUCCESS in 35m 22s\n- cinder-tempest-plugin-basic-2024-2 https://zuul.opendev.org/t/openstack/build/77aae697a9894640b68ca6b8f8d4fa52 : SUCCESS in 43m 21s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/095fb24b199d4ab9bbfa1592f972e399 : SUCCESS in 26m 02s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/56eeff1d3e8a499eb8c3f5d7f5982f1e : SUCCESS in 37m 48s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/854b64fbb6da42b7b8e2b31af1efe98e : SUCCESS in 38m 26s","accounts_in_message":[],"_revision_number":2},{"id":"4b6620cc9d5b30ad28343b653dc26fbc4e2b1bae","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-07 19:46:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/f337ac9c169e4c06ad53c38bd6d13d9d : RETRY_LIMIT in 5m 48s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/0f5139d136a24c4b9ebb6636faafc9df : RETRY_LIMIT in 5m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"594e4b0b69f1cbf556e74a88b1b77aa4024ab865","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2024-11-19 14:51:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"75a155b902ab552dfb2dfae19e44e191d4163475","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2024-11-19 14:51:41.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"790b63a86a79b70abc89dcac28bccef9d02a794e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-19 19:35:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17467fbcff5df00a72356e82410c7bb1782babae","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-11-20 14:23:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c1f1382b83050dde94fee1b8887cf84cc1007ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 14:23:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"43bf0bf3e427c16ffa83836d2bc050c9b2b733f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 16:08:25.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1aa06b4660724c04b0004fa15f3ccc00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a381920d51dd416a963deb881d01d36a : SUCCESS in 2m 37s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2db14de255a14c5da9a0c4cecf78d3ef : SUCCESS in 1h 20m 41s\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/2c0e2f7240cc404ba31822492e5551ba : FAILURE in 1h 44m 25s\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/b8a1e56595cb421db365dfb334d18087 : SUCCESS in 1h 41m 08s","accounts_in_message":[],"_revision_number":2},{"id":"701068f2e4815060dd030a1c9133b76d3a801e33","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-11-20 19:39:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"732cdb25e1f4a0cb2cd6d26f5843b7e922250a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 22:44:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc4299ce2d084e8981e3eeb5354538ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f42cee4524d4cce83e51c337284e06b : SUCCESS in 2m 31s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/045ac00fe30041c99ea2ee1b6a9c2df2 : SUCCESS in 1h 46m 00s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea734a9898094f2c99c04784cfb24809 : SUCCESS in 1h 39m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/f93f9c3944594d899ca6e8b54ba83bf5 : SUCCESS in 1h 49m 47s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/ee72757d1d404f5fad326f8c76c4298f : TIMED_OUT in 3h 03m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/79b70082f7c94380af93562e8ba2477d : SUCCESS in 2h 10m 20s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/09b0af25b5434a4ebc344aa7a3abd235 : SUCCESS in 1h 25m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3e4732cb4c0447e49554c27d2e7f20a6 : SUCCESS in 2h 00m 33s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/1891ea8b27874085be1656eeeb4c29d1 : SUCCESS in 1h 48m 44s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/9003bf3f999b4ac0929512eff832bfcf : SUCCESS in 26m 37s\n- cinder-tempest-plugin-basic-2024-2 https://zuul.opendev.org/t/openstack/build/2fb099640c6a4dbbb5d2a12bcccac444 : SUCCESS in 37m 59s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/ae70d59fe73546059be4e4a4f54119ec : SUCCESS in 42m 43s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/3abac3d907b1425884cb5614bdf07870 : SUCCESS in 29m 41s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8ce632b8f7f94975b5350a6eef6f8fc4 : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":2},{"id":"65c1db9c0d0efcf180b574afff034b8387a53e22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 22:45:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2c28f0a229635de57a5e009965d3406c8e4860ca","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-20 23:01:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/6b837e368f3f4a5396eaded815cbe71b : RETRY_LIMIT in 5m 37s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"86ecb2ca4e6fa6eaab2594d06945b74e467cb786","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 00:39:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f06f99a5541946bd9ed6d4073b0965f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0303ff483fdb429a84e30ae232db46b4 : SUCCESS in 2m 07s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/655268913c4445e6a27ceda60df5d82a : SUCCESS in 1h 54m 16s\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/239642a3395841f0a7cdaa945a310206 : SUCCESS in 1h 51m 54s\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/f4993823114447eaa3f895e3547605e4 : SUCCESS in 1h 10m 55s","accounts_in_message":[],"_revision_number":2},{"id":"725be84fb4adc613db4b49cad3cc4776151bc880","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 00:39:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfde7d5e9f5082a72152ee21aff9c94fe641a31a","revisions":{"34bc574ee170246e463be77e54453465cd3d70c6":{"kind":"REWORK","_number":1,"created":"2024-11-06 19:50:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/75/934275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-tempest-plugin","ref":"refs/changes/75/934275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/75/934275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/75/934275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/75/934275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/75/934275/1"}}},"commit":{"parents":[{"commit":"8033fab49f77abc8517c3212863fba8327aa68fd","subject":"Add stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/8033fab49f77abc8517c3212863fba8327aa68fd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-06 19:39:33.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-06 19:39:33.000000000","tz":-480},"subject":"Remove stable/2023.1 job from master gate","message":"Remove stable/2023.1 job from master gate\n\nstable/2023.1 is planned to be in unmaintained state[1]\nand with python 3.8 drop from Tempest[2], we have to pin the\nTempest for stable/2023.1 testing[3]. This means Tempest and\nits plugin master does not need to support stable/2023.1.\n\nWith that we can remove the stable/2023.1 jobs from tempest\nplugin master jobs.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@\u003e\n[2] https://review.opendev.org/c/openstack/tempest/+/933398\n[3] https://review.opendev.org/c/openstack/devstack/+/933704\n\nChange-Id: I00426388dfef92867ea32ee018006fc2dba7d16a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/34bc574ee170246e463be77e54453465cd3d70c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/34bc574ee170246e463be77e54453465cd3d70c6"}]},"branch":"refs/heads/master"},"dfde7d5e9f5082a72152ee21aff9c94fe641a31a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-07 17:51:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/75/934275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-tempest-plugin","ref":"refs/changes/75/934275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/75/934275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/75/934275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/75/934275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/75/934275/2"}}},"commit":{"parents":[{"commit":"8033fab49f77abc8517c3212863fba8327aa68fd","subject":"Add stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/8033fab49f77abc8517c3212863fba8327aa68fd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-06 19:39:33.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-07 17:51:05.000000000","tz":0},"subject":"Remove stable/2023.1 job from master gate","message":"Remove stable/2023.1 job from master gate\n\nstable/2023.1 is planned to be in unmaintained state[1]\nand with python 3.8 drop from Tempest[2], we have to pin the\nTempest for stable/2023.1 testing[3]. This means Tempest and\nits plugin master does not need to support stable/2023.1.\n\nWith that we can remove the stable/2023.1 jobs from tempest\nplugin master jobs.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/UDQAC7SR5JAQJE5WBAG54A2MTBVBTJ44/\n[2] https://review.opendev.org/c/openstack/tempest/+/933398\n[3] https://review.opendev.org/c/openstack/devstack/+/933704\n\nChange-Id: I00426388dfef92867ea32ee018006fc2dba7d16a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/dfde7d5e9f5082a72152ee21aff9c94fe641a31a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/dfde7d5e9f5082a72152ee21aff9c94fe641a31a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
