)]}'
{"id":"openstack%2Fcinder-tempest-plugin~946703","triplet_id":"openstack%2Fcinder-tempest-plugin~master~Ia760a69a9dddcfc9ce71e5e1e6e78baad6378872","project":"openstack/cinder-tempest-plugin","branch":"master","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-08-08 16:04:23.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2025-04-16 00:53:28.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-08-08 16:04:23.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"Ia760a69a9dddcfc9ce71e5e1e6e78baad6378872","subject":"New snapshot dependency test","status":"NEW","created":"2025-04-08 18:32:10.000000000","updated":"2025-10-09 20:44:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"10b95c7f4a77cb4568f562c42433e0c19213bf7e","_number":946703,"virtual_id_number":946703,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-09 20:44:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-08 18:33:13.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-04-08 18:33:13.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2025-04-08 20:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-08 21:48:40.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"4853d002bae9a94a8a68e933d6a60378e654fdaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-04-08 18:32:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96ba3ce04c630485c67d1491542325b2d63f3c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 20:47:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cf953db399a40e0b25bb18dba9e5b5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c725c3a4342046ffb959c707e1171d1d : SUCCESS in 3m 57s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/3eda85ab5d054486bc91c008a4c0d9a7 : SUCCESS in 1h 47m 51s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/957290032d28457da0665fca78f397ed : SUCCESS in 1h 51m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/99836c5c91fd4c0a935990076dde7b58 : SUCCESS in 1h 33m 16s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/8b21ef9968e14f11bc5f0531e84d08a5 : FAILURE in 1h 57m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/be4be4b306ad4d2fb013cddc792eed45 : SUCCESS in 2h 11m 09s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/e61b167d27dc42b2a8aec58d26efd14c : SUCCESS in 1h 44m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/635e33cde8a64de9b0135d3ee13365a5 : SUCCESS in 1h 00m 18s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/8c613848dc4a4abe8ce796740ff421c6 : SUCCESS in 2h 12m 54s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/ecf12d3f62024d33aa47272fd99a277d : SUCCESS in 49m 14s\n- cinder-tempest-plugin-basic-2024-2 https://zuul.opendev.org/t/openstack/build/e05d0b4f81114f8e8e048c884f7746ae : SUCCESS in 34m 35s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/c61dad611597420c88866a5d3408fd59 : SUCCESS in 20m 09s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/ac14833cb194455ba9f0355095ed24a1 : SUCCESS in 34m 59s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/51c35a10071142c495d4963047fc6db3 : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e16b39e2c8e96ca2ff95e0b5cc9ac475fe0435ea","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-08 21:48:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"a47995ec07886b756883d1d72e11851313097ad4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-04-11 01:11:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4427acf79db647f01a6d7497ae01404c0a870795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-04-11 01:14:19.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"21e73bf9eef5a9feb01baa777998643734d12071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 03:57:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdb5c05fb293443fb4969e62c9902985\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bca2730581149d8b0268ca21000872e : SUCCESS in 2m 03s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/8967a2c3e6d0479ca945938ee4287efc : SUCCESS in 2h 13m 35s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8d09ac44e1134a7ba918850f2fa51b77 : SUCCESS in 1h 43m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/8958585f73144cf8ae0ff879576d92be : SUCCESS in 1h 41m 46s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/b6b3f33e73474c1382d3cfe91d41cba1 : FAILURE in 2h 41m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3b9bdfdaa82a4af984bd2829f15e0d8b : SUCCESS in 1h 42m 04s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/a44ca84b761e4c3dabb45ea68b637429 : SUCCESS in 1h 48m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/633134443d164301bebd7c69de005101 : FAILURE in 1h 27m 36s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/657b00a10d414d8baf6466d8aad91b6a : FAILURE in 1h 46m 08s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/ccb040d08dcc4a7c846a16f8491ddf18 : SUCCESS in 26m 06s\n- cinder-tempest-plugin-basic-2024-2 https://zuul.opendev.org/t/openstack/build/55a6843fae584180b9a58b34a4c12ba4 : SUCCESS in 27m 31s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/71c4dee5425a47bd9756321373ac334b : SUCCESS in 32m 35s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/97a89467697b43c79914f12d8c3d7add : SUCCESS in 19m 41s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4ba5a3decf8a4a4cbced86a7bccf32fe : SUCCESS in 26m 23s","accounts_in_message":[],"_revision_number":3},{"id":"9d1677fdd9d59e73dfebcf1a622c7d87f1384deb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-15 19:33:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"997dd0a3c5c4e586269f6d2160ab5cadb8234610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 21:52:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3082fcb1dd724f4595a683a001f0e426\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1adfb6e96bef4dbb82294f2fde870f0f : SUCCESS in 1m 47s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/3be2c5b6aae2492cbee66cb5b1a3a55a : SUCCESS in 1h 54m 04s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a2fa7c44207e40ddb729fba697782d82 : SUCCESS in 1h 37m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ace118eb5e34a9e973da7841dd14554 : SUCCESS in 1h 47m 10s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/a6a953bd35f841d7813f05098351e40d : FAILURE in 2h 16m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/79b7264593144d38a7fb2a4b3af23e74 : SUCCESS in 1h 44m 17s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/c12aab8edb854bcbaa599132f0e3ffd8 : SUCCESS in 1h 59m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fd433ef181344ae090000e21ba01d7be : SUCCESS in 1h 53m 05s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/51c947bac45f41f9ae4f650d4b8e2254 : SUCCESS in 1h 43m 44s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/cbe43f505e68482bbe8e470d7b626cd9 : SUCCESS in 24m 55s\n- cinder-tempest-plugin-basic-2024-2 https://zuul.opendev.org/t/openstack/build/daf4ae4d60a04596819121642329e018 : SUCCESS in 38m 42s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/dda39b448aea4ca0b1bc72a49f73e1ab : SUCCESS in 23m 13s\n- cinder-tempest-plugin-basic-2023-2 https://zuul.opendev.org/t/openstack/build/4bb138ca7f214f779e65824cc0eabae9 : SUCCESS in 23m 34s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/279b5cd3d30b4a658bcacbf77c4306cd : SUCCESS in 23m 59s","accounts_in_message":[],"_revision_number":3},{"id":"b33b4bff61ad5f60a7aefac36745c88fcf87c236","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-16 00:53:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"f70f3104174d712b079d2b7366c59ed0d2183170","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-21 13:17:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ab21cdbbaf80bc455662e5f6ea5aa7597b02910e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-08-08 16:04:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"999bdaa1f2aef271ea24a49ff12f20e9202e5863","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-09 18:25:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"779698aed649da1429a6294654ee5908aca19e29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-09 18:26:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"10b95c7f4a77cb4568f562c42433e0c19213bf7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 20:44:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b0ebe7d1c274c279bcb211f6d5f73c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/108819b56bb640b58419acefeced6af2 : SUCCESS in 3m 10s\n- cinder-tempest-plugin-lvm-multiattach https://zuul.opendev.org/t/openstack/build/6d0e2fb049264877b017792f9984d629 : SUCCESS in 2h 05m 08s\n- cinder-tempest-plugin-lvm-concurrency-tests https://zuul.opendev.org/t/openstack/build/a0f49abf60fd4b77acde62701b52d284 : SUCCESS in 36m 03s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1f24cc64e6b14ff5a97f458f0046927c : SUCCESS in 2h 13m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream-py311 https://zuul.opendev.org/t/openstack/build/eab4d491a543481bab41b60a3f3fe9e4 : SUCCESS in 1h 57m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-centos-9-stream-2025-1 https://zuul.opendev.org/t/openstack/build/d95918222a884fb2a975464a7ac22de2 : SUCCESS in 1h 53m 38s (non-voting)\n- cinder-tempest-plugin-lvm-tgt-barbican https://zuul.opendev.org/t/openstack/build/b39d799459aa442188295f41a785cf9f : FAILURE in 2h 12m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-2025-1 https://zuul.opendev.org/t/openstack/build/c480f26f64e047ad94b8aae95f7fa7da : SUCCESS in 2h 06m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2f65fd9e67584294991eea67fa3eb502 : SUCCESS in 1h 41m 27s (non-voting)\n- cinder-tempest-plugin-lvm-nvmet-barbican https://zuul.opendev.org/t/openstack/build/b19f2eda58ca41ae967f6efd41a71260 : SUCCESS in 1h 54m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/833b27d495694b6b9488b16399b02995 : FAILURE in 1h 04m 07s (non-voting)\n- cinder-tempest-plugin-cbak-ceph https://zuul.opendev.org/t/openstack/build/7196a364b24f455c99bd590c1cd08f22 : SUCCESS in 1h 15m 06s\n- cinder-tempest-plugin-cbak-s3 https://zuul.opendev.org/t/openstack/build/a57f88b2a4ed4762b5f9c9f36795952f : SUCCESS in 31m 42s\n- cinder-tempest-plugin-basic-2025-1 https://zuul.opendev.org/t/openstack/build/edd3c0858bae44698b6539babad596b2 : SUCCESS in 38m 55s\n- cinder-tempest-plugin-basic-2024-2 https://zuul.opendev.org/t/openstack/build/407842b2060f4190b5f66a3544f18153 : SUCCESS in 25m 39s\n- cinder-tempest-plugin-basic-2024-1 https://zuul.opendev.org/t/openstack/build/3bc3e239745d4d75aab5fd56cc56e0d1 : SUCCESS in 44m 20s\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a6f86f75b35d4fb8956d5a60ae490490 : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"af0d79f3f3056c8e718accd8a45329987cda8e8d","revisions":{"8083841943447e5dd244d6a409151cb52e8ec355":{"kind":"REWORK","_number":1,"created":"2025-04-08 18:32:10.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/946703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-tempest-plugin","ref":"refs/changes/03/946703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/1"}}},"commit":{"parents":[{"commit":"bb5594c061370316bfc881e24d11b3350fb0aef9","subject":"Merge \"Deprecate `volume_image_dep_tests`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/bb5594c061370316bfc881e24d11b3350fb0aef9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-04-08 18:31:12.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-04-08 18:32:05.000000000","tz":-240},"subject":"WIP: New snapshot dependency test","message":"WIP: New snapshot dependency test\n\nCreate a test that tests w/ multiple snapshots on the\nsame volume, similar to the other tests here.\n\nChange-Id: Ia760a69a9dddcfc9ce71e5e1e6e78baad6378872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/8083841943447e5dd244d6a409151cb52e8ec355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/8083841943447e5dd244d6a409151cb52e8ec355"}]},"branch":"refs/heads/master"},"7b562978f31f999511c9bbef580e59b61108e20a":{"kind":"REWORK","_number":2,"created":"2025-04-11 01:11:15.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/03/946703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-tempest-plugin","ref":"refs/changes/03/946703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/2"}}},"commit":{"parents":[{"commit":"bb5594c061370316bfc881e24d11b3350fb0aef9","subject":"Merge \"Deprecate `volume_image_dep_tests`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/bb5594c061370316bfc881e24d11b3350fb0aef9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-04-08 18:31:12.000000000","tz":-240},"committer":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 01:10:51.000000000","tz":-180},"subject":"WIP: New snapshot dependency test","message":"WIP: New snapshot dependency test\n\nCreate a test that tests w/ multiple snapshots on the\nsame volume, similar to the other tests here.\n\nChange-Id: Ia760a69a9dddcfc9ce71e5e1e6e78baad6378872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/7b562978f31f999511c9bbef580e59b61108e20a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/7b562978f31f999511c9bbef580e59b61108e20a"}]},"branch":"refs/heads/master"},"8045f13eaf4e6e1e71d6f27137da6b30119db986":{"kind":"NO_CHANGE","_number":3,"created":"2025-04-11 01:14:19.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/03/946703/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-tempest-plugin","ref":"refs/changes/03/946703/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/3"}}},"commit":{"parents":[{"commit":"bb5594c061370316bfc881e24d11b3350fb0aef9","subject":"Merge \"Deprecate `volume_image_dep_tests`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/bb5594c061370316bfc881e24d11b3350fb0aef9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-04-08 18:31:12.000000000","tz":-240},"committer":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 01:13:54.000000000","tz":-180},"subject":"WIP: New snapshot dependency test","message":"WIP: New snapshot dependency test\n\nCreate a test that tests w/ multiple snapshots on the\nsame volume, similar to the other tests here.\n\nChange-Id: Ia760a69a9dddcfc9ce71e5e1e6e78baad6378872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/8045f13eaf4e6e1e71d6f27137da6b30119db986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/8045f13eaf4e6e1e71d6f27137da6b30119db986"}]},"branch":"refs/heads/master"},"af0d79f3f3056c8e718accd8a45329987cda8e8d":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2025-10-09 18:26:27.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/946703/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-tempest-plugin","ref":"refs/changes/03/946703/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-tempest-plugin refs/changes/03/946703/4"}}},"commit":{"parents":[{"commit":"058e172729cde892e61ee65c931d5defc0d3d81c","subject":"Merge \"Remove support for Pyton 3.8 and older\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/058e172729cde892e61ee65c931d5defc0d3d81c"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-04-08 18:31:12.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-10-09 18:26:03.000000000","tz":-240},"subject":"New snapshot dependency test","message":"New snapshot dependency test\n\nCreate a test that tests w/ multiple snapshots on the\nsame volume, similar to the other tests here.\n\nChange-Id: Ia760a69a9dddcfc9ce71e5e1e6e78baad6378872\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/af0d79f3f3056c8e718accd8a45329987cda8e8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-tempest-plugin/commit/af0d79f3f3056c8e718accd8a45329987cda8e8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
