)]}'
{"id":"openstack%2Fcinder~1000801","triplet_id":"openstack%2Fcinder~master~Ie448e40b8bfd0aa61babc4aafbf866e55cbdeae2","project":"openstack/cinder","branch":"master","topic":"bug/2163150","attention_set":{"38286":{"account":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"last_update":"2026-08-13 17:27:57.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie448e40b8bfd0aa61babc4aafbf866e55cbdeae2","subject":"Dell PowerStore: Fix replicated volume delete and extend","status":"NEW","created":"2026-08-13 07:01:41.000000000","updated":"2026-08-13 22:28:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70c443a679c3ab4f13317ee8c5b06f496e94dc6e","_number":1000801,"virtual_id_number":1000801,"owner":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-13 17:14:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-13 07:31:07.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-08-13 07:58:23.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-13 08:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-13 08:23:23.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-13 08:23:32.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-08-13 08:23:39.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-13 08:58:56.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"},{"updated":"2026-08-13 09:38:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-13 09:41:06.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-08-13 10:16:32.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-13 12:26:57.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-08-13 18:03:19.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"e088aa667d1c02aad1246043b8cedfa3a44ea6fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-08-13 07:01:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fa74bda93fb4af29b746c73912eb5702c3775d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-08-13 07:14:34.000000000","message":"Topic set to bug/2163150","accounts_in_message":[],"_revision_number":1},{"id":"c129aa33696c403176f71909b5235ca3a40aae2f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 07:31:07.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/78b70cd5bca244699988adb60aee73ef/","accounts_in_message":[],"_revision_number":1},{"id":"8f2ab1a797dfd89f0820bc6c9cd2f1f4d563b15c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 07:58:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie448e40b8bfd0aa61babc4aafbf866e55cbdeae2 .","accounts_in_message":[],"_revision_number":1},{"id":"056d1b3b0816486637e6617522b80655e209a7a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 08:03:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14cf9ea20aa94236a2903cafd661f450\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/934b0677b6f8451e878a9c8eb97e182a : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/12c2c19d23704b399192f0a204e0051d : SUCCESS in 17m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d69938b28b1b4de087c7f2aab4c42b8a : SUCCESS in 23m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"176c1616ac07c9c764ee6211905d6a45ec3aea44","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-13 08:23:23.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/800 : SUCCESS in 0h 48m 49s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1000801","accounts_in_message":[],"_revision_number":1},{"id":"6de69c4f44fc6b15205e1b701e1a8bef363eb6fc","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-13 08:23:32.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1161 : SUCCESS in 0h 48m 55s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1000801","accounts_in_message":[],"_revision_number":1},{"id":"2ea0a1274fd417f2d3b2d5fe96224d6ce13541d8","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-13 08:23:39.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/894 : SUCCESS in 0h 49m 1s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1000801","accounts_in_message":[],"_revision_number":1},{"id":"70994086e858689fd17f6fa115f9e8508c1f9b2b","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-08-13 08:58:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/0d87e2db161f45b3b925a6400bbcec26\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/dbb6f9d29e2d472da7b485b4439b4de7 : POST_FAILURE in 1h 56m 45s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/d026aa46a86541929c23ab25e20cbc84 : POST_FAILURE in 1h 22m 02s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/706acacbc825465c9497394ba115f7f7 : POST_FAILURE in 1h 20m 25s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/c0d61c98165843ef98eeda7e25e54a2c : POST_FAILURE in 1h 16m 18s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"2cc3e834da4d4800f7f04c01207b06126261c9df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 09:38:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4c92e70543249baa205accdaf3e0de9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a83e0d9af14226a19cf85be633c635 : SUCCESS in 6m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1ddb7c7f16a4176bb9b15d88645f40c : FAILURE in 13m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8734777ea6184b6f8abb720abfb2a9ef : SUCCESS in 11m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8359e3fa98184b58a65b51f665387875 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2776b7f508984edf8bc8675af9530370 : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/6b3ab63ef681472d8a4d9733a351adee : SUCCESS in 1h 06m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e1642d0551be490c9e9167e4619d066b : SUCCESS in 1h 03m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/79f585e65f2449488fbf880e2b0913eb : SUCCESS in 1h 36m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c2f7a1d786948e796c050bc1ee7b076 : SUCCESS in 1h 00m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccfed1fa8b264fa686f9feccd679609a : SUCCESS in 10m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b228399655a2452c9f2ba881cafa9759 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/64d0990bf8ec4d96b7a1c71acd44c0db : SUCCESS in 4m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/89ba7f1a17024ebba458c6df81ae24e4 : SUCCESS in 6m 46s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/cd33ec10854c4b5495bcee97969e4493 : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e6a8d0e8c0ab48c9960369e0e1a1b257 : SUCCESS in 6m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cef35d4ce3bf4159a76865fb4e5bcf8d : SUCCESS in 6m 33s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/68ecf17ace054c2684cf52bdc10c53ef : SUCCESS in 8m 49s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce11d64a66ae478e96401b04032fc439 : SUCCESS in 35m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f3f7eaa1f6bf48de8c31ff4f2628adcb : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b775c96fc5f485ebd07f6c5e8d5b54a : SUCCESS in 2h 05m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/57b41e1cb22c4353ada61c398cac51d4 : FAILURE in 2h 00m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9372111cfe7742808afd4ced5f3585f4 : SUCCESS in 1h 45m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/5b7b4e93c482471c9bc5404e52e4eb02 : FAILURE in 2h 27m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bacf8c6b79914790bb2aab052b07d214 : SUCCESS in 1h 42m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b2447363f5e6460390306b691ca34cd0 : SUCCESS in 33m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91f080c9e2c84d8780d60a3198775a47 : SUCCESS in 1h 00m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d764b6d3f7e44f9abfc19ac4739bbdd6 : SUCCESS in 1h 33m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6f241ce6552747568513402506ecbe1f : SUCCESS in 50m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a3f040c2ee884ace84210f38430a1b4f : SUCCESS in 1h 55m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1af0464051b740eda2b914cea1a485bc : SUCCESS in 55m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"b26c27af5d94271d374db283788c9510c2011ca5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-13 09:41:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4930/.","accounts_in_message":[],"_revision_number":1},{"id":"5802535433d72ca56cdf590d7f1b791d9dfd73c6","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-08-13 09:55:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c80561f90d4796699592b6b980d7e3e1d620b170","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-13 10:00:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5106/.","accounts_in_message":[],"_revision_number":1},{"id":"71ca7d4c6cf4ebe858c8a25c1fe6ef5221ba5217","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 10:14:24.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/78b70cd5bca244699988adb60aee73ef\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0851db060b6b49afa5655e41a5bf524b : FAILURE in 28m 57s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/91d02b7079f64f89aa76bbba89d4e2f3 : FAILURE in 29m 36s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/19eafcea873941d98f20c2f3b89bf74f : FAILURE in 30m 20s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1938f7ce8e824863911659d05bb2048a : SUCCESS in 38m 39s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9e2c388bebaa41ea8c9423e0da542fc3 : NODE_FAILURE Node request 200-0000790780 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35e6e2f1df4d71300e9d9556be1c0a07c3734f4c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-13 10:16:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/d8036141fbfd47c38608ea7ee452fda7 : SUCCESS in 43m 50s\n- cinder-storpool-tempest-iscsi-multipath build/bdf53aa0c3714d4cb0ccdff4b609055f : FAILURE in 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"aa5b64dcfca9404f7d9aaf86ec57b24fa1967e58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 10:20:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72f1fe8aecb242688f0b758ff84404c7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1931e5d891014b6891a4b4b987f54402 : SUCCESS in 16m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ec1fa931cb654886b0a5b143fea74252 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/731fbb0451d24047aa2203bf5a162924 : SUCCESS in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a18a2b7bfc0da2987bf33427438bfabcceb7592c","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-08-13 12:22:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00a9043de294844cc2029d02319a56f05b48159b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-13 12:26:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-1000801-1-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-1000801-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2319c89096c389abab3577f430a87ecb312ab9f7","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-08-13 12:29:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/6d162888582e40c8b5b44040dd6546c5\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/925b3c1119a440c6a23ef91e55018c2b : POST_FAILURE in 5m 41s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/ba0debbca596422db5f48045d3e5b9b4 : POST_FAILURE in 5m 30s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/bed6604be5724f93bc5504e3abeaf3ea : POST_FAILURE in 5m 21s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/738c573d22a94ec28b0e3ffef1b4698c : POST_FAILURE in 5m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"a66aa01e09c588ed4e6d1ea0622b49c7f8708eaa","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-08-13 12:32:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"906618a00d923782c04d5eef7c0db837fbd759c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 12:37:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87e3a006c7084023a31f5a0b53e0936d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85ebbb1db20a4e9b82c89a52fd508655 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8aad91b54fea47c9826f1bf021fcce30 : SUCCESS in 11m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf51344ccea2495daebec3d81f37d689 : SUCCESS in 10m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/caddb3c621ae4e11aaf2727726a745d8 : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7571633539469780e31c229e4a500f : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/b4330651f62f49a78808fb5489f529e2 : FAILURE in 25m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/924cd8fa01f544bba8f8931ad96d2853 : SUCCESS in 29m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/921eece77b4f4036a99622b0f270648e : SUCCESS in 54m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f144bdc45acf4143bbd3db6ce1e60b8d : SUCCESS in 31m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afecbab1ef584f4ab1a3f11e682c4168 : SUCCESS in 8m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c69ba66909504835830e4c44a2c3fe9c : SUCCESS in 16m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6ce8c882f9454f25857144856e022174 : SUCCESS in 3m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b35ce7222ee4c4f8f89115b0690e150 : SUCCESS in 3m 11s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/edb0fee8cd2d49e48d35337de433313b : FAILURE in 13m 11s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c95f61b5ce5942bab77843dcf3097aca : SUCCESS in 4m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5f35d4d1c57e45bc93bec338f586959d : SUCCESS in 6m 14s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/dc7cffa6634c4160a8d89b8ffd618f6e : SUCCESS in 6m 57s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ed2b596d6bfe45abab8b689b67672b5b : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5993a9901ce4b7bac26875d373dbc7d : SUCCESS in 3m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8585c868a2864b27865363c5eedbfef7 : FAILURE in 31m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1e5ef1679807498496e4724f347a306a : FAILURE in 55m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/165a3ad8e43c44cbbf3a550c4eda8ff9 : SUCCESS in 1h 56m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/76b470b048754cfb81bb720db88bd01e : FAILURE in 2h 40m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fcfd49c195fc463da514af5458c7a1f4 : FAILURE in 2h 07m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0d51da7a49fc46b9ade724075ae9b27f : SUCCESS in 35m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3cce0ca566e9413685f2ec53b370727c : SUCCESS in 1h 09m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/52b4c3d8f5e7484eba2335fe2a788a55 : SUCCESS in 1h 36m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dac312dc49fd4ebf98472690a50cf087 : SUCCESS in 40m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cfef4b45691c476e9777169fed98677e : SUCCESS in 1h 38m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/707b953465f7465aba1960fc299d56f3 : SUCCESS in 28m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"03258130ff3065d2b4999958ac800d365c6b0954","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-08-13 12:39:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/7b7cdb64b3974709a0533505495946cc\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/8ad9c3b70fdb42aeaac7aa30484cc7bf : POST_FAILURE in 6m 11s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/92bf79079c61405f953677f587e567c5 : POST_FAILURE in 6m 03s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/65500d5b704e42f69de51808912a613c : POST_FAILURE in 5m 29s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/fcd213390fe5474880651c72badbccda : POST_FAILURE in 5m 36s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"664a4fb744e72d778667874a87fe620e17ddfbb5","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-08-13 13:02:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62ff4cb48884023cf28d9e13b59ff739950459e3","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-08-13 13:09:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/0858c6ee32c8489ea62596226504da63\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/87b22edd2140407f801f8a731e7505fb : POST_FAILURE in 5m 52s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/bab9de93e488401cb40149c5eeb6fbfa : POST_FAILURE in 6m 05s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/61b30bd28a4d4ab0aa9d0bce80103afb : POST_FAILURE in 5m 58s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/2abf98ca0d074acbb075bf790f0eb414 : POST_FAILURE in 5m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"a58fe46034262b8db7fe367cc8d7f6e1effd4454","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-08-13 14:42:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a3b1bde975b24485955f6ed460f49a2c31538f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:37:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f8f583cf55746e3b36f4ec21516b954\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1ff920cab0224456ad6384fbed465100 : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/185fdfc332454654861fcc88fe6ebef9 : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/277c241c5bee44d5b317485718540c4e : SUCCESS in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1ece162a01d84dadea82d077a0a9224bb43b790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:14:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65faafc8180546b29c650176010f9cf6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4af032af1947279e119cda7f9abcca : SUCCESS in 7m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/facd466bfba9412c978dada27c8c07c4 : SUCCESS in 8m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab5aa0b14050477d993c87bdeaf9e08d : SUCCESS in 10m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e41e415c4b05435cb4aa58ab29e32ce3 : SUCCESS in 14m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5278f51ab2646f5a0fa3217497ec75c : SUCCESS in 12m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/11a6ddc767564005af31d54e9dbcb881 : SUCCESS in 1h 01m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc98b890069a45aa93a8c90413379623 : SUCCESS in 30m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88d1f175b4a0484f9768f136bef7322a : SUCCESS in 1h 15m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2af92fb6dba4fc891270dc301c8846e : SUCCESS in 56m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/535d6f499aae43b49a6978c7816f96f0 : SUCCESS in 6m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/862ea48953c54685ac3144585d56b1a5 : SUCCESS in 26m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/589af83384c049459f8000ad86c9d15e : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c672a723ae41490a8908611d2394519f : SUCCESS in 5m 22s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/18d19d9535fb43de83b0f2d7772eb441 : FAILURE in 11m 33s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dd5b1bd70fc24d62a17a1d5251f4ebbd : SUCCESS in 6m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/87c825494259480ab758cae171ea4ba9 : SUCCESS in 7m 57s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/228cc5836d2f4cc49fb015a47378e1d0 : SUCCESS in 7m 47s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/63386f6713064a718ee426a657f69a06 : FAILURE in 14m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4c5a3f2d87284116b551f264f0e97163 : SUCCESS in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dfba589acc22486eba1df33c90095471 : SUCCESS in 1h 27m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/627fbfc7990e4969a9efeb0bcffdcfda : FAILURE in 1h 57m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/98fe74ae6e41498fa56b9e8989587e18 : SUCCESS in 1h 14m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/9f4400f9be7c49858be719e387f56a25 : FAILURE in 2h 24m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e5be71b2e99a465f98a89cb1ed66fe13 : SUCCESS in 1h 14m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e06518c7816d4f498e16cf372449446a : SUCCESS in 33m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/22e74feb77c3423cbff4849be5a5b3d3 : SUCCESS in 38m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/950c96d5392242c7b9f348862d401c6d : SUCCESS in 1h 28m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c4c6f1aa9e1840c4ab1af1fc6be5cc1a : SUCCESS in 31m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37b24fcb1b604704af03d3b48de37cac : SUCCESS in 40m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30865de5068d452aa2e74cf4b8a063c9 : SUCCESS in 1h 00m 36s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"fe94dbff0e44ab894ba24a3f9b53834b9241e226","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 17:27:57.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/f6c7d84cc68547008b9792afd59baf57/","accounts_in_message":[],"_revision_number":1},{"id":"382e8a6a85fa4981025b875ef44058b808826219","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 18:13:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_08_2026_18_06_35\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"fd10bb1a2c44672f88c56f5946a3b5a87ea5cdac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 20:42:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f6c7d84cc68547008b9792afd59baf57\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/42ee453a63fb4f339e203bd8b5403b98 : SUCCESS in 38m 22s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5b2c1ce75d545e7a8c89076ca040e42 : SUCCESS in 36m 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/343beb38d29f42b38c6494d5a94069fa : SUCCESS in 45m 03s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1947a34d1ebf4b4d8b219f30efbfcc46 : SUCCESS in 33m 36s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2ac584e8011d41f18c7949f0cea6a89a : FAILURE in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70c443a679c3ab4f13317ee8c5b06f496e94dc6e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-13 22:28:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/b804ca7b56b245399026f63f5d284c13 : SUCCESS in 44m 16s\n- cinder-storpool-tempest-iscsi-multipath build/bc5a37e3b40d4b3c82c6e4a32f5efe4e : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5985d29a7fabd7b968b79629ccbef518bae1e6e6","revisions":{"5985d29a7fabd7b968b79629ccbef518bae1e6e6":{"kind":"REWORK","_number":1,"created":"2026-08-13 07:01:41.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/01/1000801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/1000801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/1000801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/1000801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/1000801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/1000801/1"}}},"commit":{"parents":[{"commit":"6fad01356c5950653b3734a762ed2386043579d2","subject":"Merge \"Fix: Unset default type for non-existent project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6fad01356c5950653b3734a762ed2386043579d2"}]}],"author":{"name":"Siddharth Kumar","email":"siddharth.kumar1@dell.com","date":"2026-08-13 06:54:06.000000000","tz":-240},"committer":{"name":"Siddharth Kumar","email":"siddharth.kumar1@dell.com","date":"2026-08-13 06:57:04.000000000","tz":-240},"subject":"Dell PowerStore: Fix replicated volume delete and extend","message":"Dell PowerStore: Fix replicated volume delete and extend\n\nReplicated volumes with a PowerStore protection policy could fail\nduring deletion or extension. Deletion left volumes in error_deleting\nbecause the driver polled the /job endpoint for a completed deletion\njob with too few retries and did not verify the session was gone.\nExtension of standard replicated (including synchronous) volumes\nfailed with error 0xE0A080030036 because the driver did not pause\nthe replication session before extending.\n\nThis change:\n- Polls the replication session endpoint directly in\n  wait_for_replication_session_deletion and increases retries to\n  match the wait_for_end_metro behavior.\n- Adds pause_replication_session and resume_replication_session\n  client methods.\n- Pauses and resumes the replication session around extension for\n  all replicated volumes, not only metro volumes.\n\nCloses-Bug: #2163150\nChange-Id: Ie448e40b8bfd0aa61babc4aafbf866e55cbdeae2\nSigned-off-by: Siddharth Kumar \u003csiddharth.kumar1@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5985d29a7fabd7b968b79629ccbef518bae1e6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5985d29a7fabd7b968b79629ccbef518bae1e6e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
