)]}'
{"id":"openstack%2Fcinder~1000900","triplet_id":"openstack%2Fcinder~master~Ib998143680b41464a7618f4bf2b68d5b8a4e91d2","project":"openstack/cinder","branch":"master","topic":"rbd-group-replication","attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-08-27 18:05:21.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib998143680b41464a7618f4bf2b68d5b8a4e91d2","subject":"WIP: Implement export/import APIs for cross cloud replication","status":"NEW","created":"2026-08-13 20:42:48.000000000","updated":"2026-09-15 10:12:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"86c9ebf1c689fe164a514c9288f533b107f2aa22","_number":1000900,"virtual_id_number":1000900,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-13 21:06:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-13 21:21:19.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-08-13 21:39:19.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-13 23:41:50.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-13 23:48:43.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-13 23:48:50.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-08-13 23:54:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-14 03:22:27.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-14 08:09:56.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-08-19 22:53:54.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"}],"messages":[{"id":"7f602dda4aba6adfd4989074add1dc38f692d864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-08-13 20:42:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75215b65d16203fd81f7f1c8a4dbb0d684a57b9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 21:06:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6738c5eb893a4a10ad36e1a756bd6479\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4745f379f5c644ada7ee21a6203b56c6 : FAILURE in 17m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5394e4e7adaf428597f39ee4385c7148 : FAILURE in 11m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2afcd147e23d4e869037ae7bd6b351b9 : FAILURE in 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01e23d7ea586ed808ed0ef69b8a18a54171f58f9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 21:21:19.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/f387983f7250436cbb31f5ec9cac6554/","accounts_in_message":[],"_revision_number":1},{"id":"c4dcd7fe8523d573b8e44430ce55876cdcb7aa51","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 21:39:19.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib998143680b41464a7618f4bf2b68d5b8a4e91d2 .","accounts_in_message":[],"_revision_number":1},{"id":"4d87f66640908e494b7f9543819c5b694e1c1d02","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 23:15:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f387983f7250436cbb31f5ec9cac6554\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e825635043b4a47ac88f3787bf8202d : SUCCESS in 39m 50s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8840550c27b343e7ac004257fbff964d : SUCCESS in 39m 14s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5a802caf2d1422e91bc1579b32c3c49 : SUCCESS in 47m 58s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f402337d5a0241be93c6e2f83fc3fc7c : SUCCESS in 37m 40s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1ece34710e3e4f7db25a7ce425f78830 : FAILURE in 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11c69910095e734952adcc7fbf151d76e8ffff52","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-13 23:41:50.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/814 : SUCCESS in 0h 46m 40s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1000900","accounts_in_message":[],"_revision_number":1},{"id":"c8ea56211d96a1f6becb3f4f75456de6ec14cc08","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-13 23:48:43.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/908 : SUCCESS in 0h 46m 59s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1000900","accounts_in_message":[],"_revision_number":1},{"id":"821c462110301460f0566d2d9193724a3b1f6ce7","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-13 23:48:50.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1175 : SUCCESS in 0h 46m 48s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1000900","accounts_in_message":[],"_revision_number":1},{"id":"5899fa5a8a132e65d59110e771002e21068450f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 23:54:55.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a29319fc16a94457b41966adbee2726b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f91eb273913245538a9907b41ed545e6 : FAILURE in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf1f23ee449042a1a4e0ad1eba41d118 : FAILURE in 10m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/379e65f78a7846e6b1651bc87870598c : FAILURE in 14m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e14e9db0b7174b7f8081e9309ec48c45 : FAILURE in 13m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b63ba1e0efa24a719a078042c8d23949 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/702395d9ee274f9189a7f9294d5235fd : SUCCESS in 59m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd9bf511e1bf4da3b938b2adb5199a72 : SUCCESS in 1h 04m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a632e125b33a4f909a4e7de0a4c133c6 : SUCCESS in 1h 00m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71ad9999de22437c9a495c822a918379 : SUCCESS in 43m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a3bd77c1bac443780a23e49a3f4272b : SUCCESS in 9m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/60400e9d3d4d4d5487421297e98be51e : FAILURE in 19m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a31b009088bc43c7b27ae37ac59e3952 : SUCCESS in 4m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bd6a18e8eaa04eba978370be1440a37f : SUCCESS in 6m 07s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/3f652df856e34b6897d13af92dc09476 : POST_FAILURE in 9m 21s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0d8e5cdaa54643db837a500995d3b02b : FAILURE in 6m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/112fb0e2f2cf4eb38b7027f4f3403658 : FAILURE in 6m 26s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e4d2042cd91e48a7ae12d63a1bb60cfa : FAILURE in 9m 03s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/89df9519573d4672bae48a6cb187f36f : SUCCESS in 24m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/61719632f4d24b15bc85970f850501e0 : SUCCESS in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a563e8260d9498e8a12d7b155b968f5 : SUCCESS in 1h 17m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c558387be6924c80925e0b05aa8eab57 : FAILURE in 2h 26m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/379b7577a38f43e0a2056871aecc8a30 : SUCCESS in 1h 35m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/356b5411713b49719dcd3d614008203a : FAILURE in 2h 03m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/dfa3882d9be9446ca75ec4b66bab0272 : SUCCESS in 1h 55m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0879e25f0aed45c2a5fb632218ea2f58 : SUCCESS in 35m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ef8287ff40224bfcb19d600e9186c2cc : SUCCESS in 1h 12m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1336c63974b04ddd82c0798e5b9c090b : SUCCESS in 1h 25m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/208e63d5163a442da16eb4accbde5154 : SUCCESS in 45m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f786cfdae9b540a48be37b50accff99d : SUCCESS in 1h 35m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b31ee6a2d7c4d629a38f2e7310078c7 : SUCCESS in 52m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff6996f097954070a7733ec5c778e949 : SUCCESS in 52m 59s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"06f4199955343a591e04ac6fb4a3577a6d4572ca","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-14 03:22:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/d92b3670bd13422ba4e5724fb2c951f1 : SUCCESS in 48m 25s\n- cinder-storpool-tempest-iscsi-multipath build/65e6bd870eca478293ea93ea864f24eb : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":1},{"id":"794f37dd8ce757728ae7202595ff5294e636542c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 08:09:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4945/.","accounts_in_message":[],"_revision_number":1},{"id":"e310219c3f97174ed5511ac02c9f59079d8e19e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 08:54:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5121/.","accounts_in_message":[],"_revision_number":1},{"id":"7cecd3a473fb03940913b40400330e67adf206e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-08-18 17:46:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"38deccd1f02adba7ce43467e5839c82c46dafe66","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-18 18:02:55.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/5201f04f358c4a8ea72aa188db8cd5ea/","accounts_in_message":[],"_revision_number":2},{"id":"0ccaf1080e0dfde4740d7f54dd7ad0286e1c2922","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 18:10:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5aea56e12524c9eadf4ef8b93806236\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c8870ad3eefb4d81bcb354cb1bf7c7ad : FAILURE in 16m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fc32c8c2b3c1471cbfee4e23b85c32cf : FAILURE in 11m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ac042e7af0ed46d5a8cd9fbf07b61b97 : FAILURE in 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"106b83956e93f8b5d624c503d797d34966834d8c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-18 18:38:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5201f04f358c4a8ea72aa188db8cd5ea\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7691d33a473d4b4fb4c83b9a19b203c4 : SUCCESS in 49m 47s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f01497d69e2142e0bd4cc25f448bd07d : SUCCESS in 31m 32s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4c3fcb81073c4310817898c6fb811739 : SUCCESS in 49m 30s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7cf804f4b5c34b7dab521dee605ae7ba : SUCCESS in 31m 44s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/98eee8a3d19d45acb3eeac2c56e22c3f : FAILURE in 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"637f98a842d8efa0e5592fa1f0d59e893bb164ed","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-18 18:42:41.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib998143680b41464a7618f4bf2b68d5b8a4e91d2 .","accounts_in_message":[],"_revision_number":2},{"id":"66ac2097971b2f6501d6f5e60da218b4ded6cea0","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-18 18:48:59.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/974 : SUCCESS in 1h 0m 36s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1000900","accounts_in_message":[],"_revision_number":2},{"id":"5567eae92239bea9ce7e40151ef714b78f1dc02d","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-18 18:49:06.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/880 : SUCCESS in 1h 0m 45s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1000900","accounts_in_message":[],"_revision_number":2},{"id":"f2bef700edb39efdaf83d6f126bfc0774c91d3cb","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-18 18:53:38.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1243 : SUCCESS in 0h 59m 51s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1000900","accounts_in_message":[],"_revision_number":2},{"id":"35370d49fc9a242dd43c6eab0fee9596c7e82776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 20:03:21.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb28ccf520644230abb9931a743482d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a485165920141ae9ddf5210801f838e : FAILURE in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/241db5bb0a724ecf8adfce8483aa16df : FAILURE in 10m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3b34a30d5c54770bccfd23f2d64f91a : FAILURE in 8m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e5b8754419a64c74b8f0958f8c3721bb : FAILURE in 12m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaaea26f4d9946afb82a9cd67481e899 : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/ce2e675b0b2e46d28807682996592e03 : SUCCESS in 53m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8f5a1fa91f6540c0a5f1e0d82c7028e4 : SUCCESS in 1h 00m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e20ec2592aef4d25b60425fa0a31c3ce : SUCCESS in 1h 04m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bbbb204af2e4c67aa714551a37f348e : SUCCESS in 32m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/673a1813c97248189d1a50cff5b90f27 : SUCCESS in 9m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/04bc981afa364f75adb09bf8bc734a67 : FAILURE in 21m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/59eb8cec7920487dad4e5509c260b834 : SUCCESS in 3m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f69991ab97b24c1080011662ea5875e7 : SUCCESS in 4m 14s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/3e4fde39ceb14524b4c40c1306fc616d : FAILURE in 12m 22s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7e1bd4e9827e4a3fb6fd5b9bc9e8a1eb : FAILURE in 5m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/593417647a934357ac985bad5fc7d80f : FAILURE in 7m 35s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/1770dd90a603445baa52ed9d33bfa155 : FAILURE in 8m 53s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5446c18b4e8c4e16b808ec96f91df6bd : SUCCESS in 37m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/320cf7007abe497293ad275f0af03898 : SUCCESS in 7m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6c7927fa2bfb47109b61f8837d37db21 : SUCCESS in 1h 44m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/21854349be1341dc8f6b116ab94d315a : FAILURE in 1h 03m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eb7e592571bb468cbc3c7da862f6dc5d : SUCCESS in 2h 02m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/520921dee3c645eaa425a3adad642a7d : FAILURE in 2h 10m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b63c7951491841b289391a136a999edf : SUCCESS in 1h 25m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4fc7f550cf13403690a128673d80d9e1 : SUCCESS in 34m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b36dd6c82773459d8140c8b62a84811d : SUCCESS in 1h 16m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/700d06baf7f04cc394976a2a0e4c4d1e : SUCCESS in 1h 10m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/66ba96e6595a4eb795ff2dae05fe5653 : SUCCESS in 51m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ddf674ec54144e80b88b44e70e4c6b76 : SUCCESS in 1h 44m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5749332056f44b24841e3cf992895884 : SUCCESS in 43m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c301f6452b0c4a73bc7939ed0fab0497 : SUCCESS in 30m 29s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"5f072902f1e148588bbd419cdb9274bab0990e0b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-18 21:26:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ff8f986882d54f5d96309d9eba4519d8 : SUCCESS in 49m 34s\n- cinder-storpool-tempest-iscsi-multipath build/0f3f18dde2c042389ad3d2228a328a5f : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9bd4f59a9bccc02139dc404d67a938ac1d1211ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-08-19 21:05:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e1ef6e2294b58f6a914880920ce833b3a092da3f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-19 21:57:52.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/5f8716751e8d423bba4928973d34e1b5/","accounts_in_message":[],"_revision_number":3},{"id":"0f3de92e33f90ac0a8af0a55316b7535aeca84bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 22:02:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ec9743d426641fa8081406e7595412e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/50a5b05cab91423aba29a4b026ff7183 : FAILURE in 18m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b3b04c1c685c483da3d24178fb7226e6 : FAILURE in 21m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c92f437ce71d490d9cb013c53558d229 : FAILURE in 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9007407c43eb36d54480463ef9436383ceac6a2e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-19 22:04:08.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib998143680b41464a7618f4bf2b68d5b8a4e91d2 .","accounts_in_message":[],"_revision_number":3},{"id":"1e3c278a853891dc3f8a9305ec421dd89919e0fa","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-19 22:53:54.000000000","message":"Patch Set 3:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1262 : SUCCESS in 0h 46m 47s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1000900","accounts_in_message":[],"_revision_number":3},{"id":"5fe2cae3714d5b5884d28867be32a07a4d2af695","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-19 22:53:54.000000000","message":"Patch Set 3:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/900 : SUCCESS in 0h 46m 54s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1000900","accounts_in_message":[],"_revision_number":3},{"id":"f3cbbe50898d29fe29476cc54c922ac3987b77e9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-19 23:13:32.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5f8716751e8d423bba4928973d34e1b5\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5178f9bbc6f473cb3220bce42252c26 : SUCCESS in 39m 06s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3422b9068f594dc6bc4fc0764b6ff4c4 : NODE_FAILURE Node request 200-0000897778 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/099dc63825d447fc9aa3cf21ade941b7 : SUCCESS in 41m 43s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b225ec81db134925a144440c3f64bc14 : SUCCESS in 29m 39s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0560e697caaa4e9aa3066a53078ecccc : NODE_FAILURE Node request 200-0000897781 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"84c269c225bdf664d65e264a99788d5480bf3004","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-19 23:23:43.000000000","message":"Patch Set 3:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/993 : SUCCESS in 0h 44m 28s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1000900","accounts_in_message":[],"_revision_number":3},{"id":"68bc1898efe7e5c8e801a3d06657280562d6446c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-19 23:37:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/43416752da3c47339dafd45be5133b61 : SUCCESS in 49m 12s\n- cinder-storpool-tempest-iscsi-multipath build/928de2a466804fb29f3a5a08bd622a09 : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":3},{"id":"bde5a6571dcadc404fcfe444f8f022425a691c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 00:23:16.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb365084d6c947f0a689d58899f22eb7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cdb62d1d99c40fa93e538fd23d48a1a : FAILURE in 7m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1add7932e32e4f70bbb1275d7978b5c5 : FAILURE in 10m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2a0d96eb8f745ffaa395cc27dd0b421 : FAILURE in 10m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e91e5ef6387a444d809770af0ef61f75 : FAILURE in 14m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58464f24a04e446b8733d36844e6e463 : SUCCESS in 16m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/4b5a7d1d2d914c2ab4d340b0118a2bb4 : SUCCESS in 1h 07m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4aecf2843a3441bbb8e01e635b31db80 : SUCCESS in 1h 02m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/714cb35701af4cf9a91977b528e7a1e7 : SUCCESS in 1h 36m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b607f3a54954785a6148ba6f5de9785 : SUCCESS in 32m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5fca92dd8cb447392920a7ab49881e5 : SUCCESS in 10m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cbb869afc2a14356beff280a5df109a5 : FAILURE in 24m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/61f0558c0647430da86ec48fd9c3c122 : SUCCESS in 6m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/33e0c314feae47cf91a2f084b2a4f377 : SUCCESS in 7m 29s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/a70fafac77f847de97b6e5e212863157 : FAILURE in 11m 15s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c6ddea096d8e4f82948b6a858afc9b6d : FAILURE in 6m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/389ea5532be64a8f8d18f3e55a8e47b3 : FAILURE in 7m 04s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/bd830c64f57a4958b2ef1e13acf08014 : FAILURE in 8m 41s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/01eb51c19a444d10a55eaa6f57dcbb64 : SUCCESS in 41m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f4f7449a07e43f896cc4cbc6048f3fc : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/16cf00b437a74c2d835c8b5477d06b30 : FAILURE in 32m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8e87ae5dc6b64a53a72d92c6f68a6760 : FAILURE in 31m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b8e3d22f5edf4b54aea446cbddaa6875 : SUCCESS in 2h 18m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/fab4222246874f778bff355c95da5a8f : FAILURE in 2h 17m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ca687da2f7e549fbb7dc3309dcdb4d21 : SUCCESS in 2h 03m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8638c100911444d99df36b2c719530ed : SUCCESS in 39m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d583b9ef107f4fb19847f5e5a7e24745 : SUCCESS in 1h 24m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3106a1d166744019bf8beb370f83f391 : SUCCESS in 1h 33m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3c3fce844ce14d5d95fe5123ac56027f : SUCCESS in 49m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f9da9c56c57643c396e321986c093edc : TIMED_OUT in 3h 11m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cd036bb4247d4277a76b8a1da422a457 : SUCCESS in 1h 39m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20462a747c004da68781b1c74437e5be : SUCCESS in 1h 13m 40s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"66a39f786145781e4a0bf2f2b771e36aa3d0afa9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-20 22:56:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5189/.","accounts_in_message":[],"_revision_number":2},{"id":"dc688a567e9a820f9bb6cc78b104e16cf342cc18","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-21 03:07:57.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5013/.","accounts_in_message":[],"_revision_number":2},{"id":"17df23825dafb35c535f70d1e0f5a00cefcd47f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-08-21 13:33:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"8c54bf6c9e0cfa038eb677cd4c3bb944af13ca7d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-21 13:49:16.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/16cf3690c25f46b594cdd7f688c96b05/","accounts_in_message":[],"_revision_number":4},{"id":"6131f968fd12c548432d42afb12b3c2d8650c46b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 14:10:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9caa0e4b102e4779aeb42d48ec4384df\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2a7e2d00699e4b6ea2dd974080546a2d : FAILURE in 25m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6464d12e91a4790bf7f593c4d5c85de : FAILURE in 26m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c03500058b6f4b8db5023a2875615c2b : FAILURE in 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ab12fbc361b46a1ad47966e6f816a835b43e1139","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-21 14:22:23.000000000","message":"Patch Set 4:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1268 : SUCCESS in 0h 48m 17s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1000900","accounts_in_message":[],"_revision_number":4},{"id":"c06c0c0b51f5c1303c1ca7c69dcac858d8a089d1","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-21 14:22:28.000000000","message":"Patch Set 4:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/999 : SUCCESS in 0h 48m 20s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1000900","accounts_in_message":[],"_revision_number":4},{"id":"429792216deaea724e749ac626b3b1f4d91bd31e","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-21 14:22:31.000000000","message":"Patch Set 4:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/906 : SUCCESS in 0h 48m 24s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1000900","accounts_in_message":[],"_revision_number":4},{"id":"6104f992dbed62d200b1be2db58d1cd89a14c081","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-21 14:23:08.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/16cf3690c25f46b594cdd7f688c96b05\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8b7442c8d8cb429aac386cd68d24fcc1 : SUCCESS in 38m 01s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/77df7a3c1c5047479b6f8580dd0d9f43 : SUCCESS in 32m 44s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/664b0a8b2f3d40e180a484c050dd0212 : SUCCESS in 48m 09s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/380a8fefb26b4eceab49d8cc8d43524e : SUCCESS in 35m 44s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33478425bb0a404ea8165fc2ee038839 : FAILURE in 11m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"388f57a8ec31a3d99c7068f74be964fbf7db1657","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-21 14:35:34.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/b2ec93d05ce94f6ead02dadd154cb2c5 : SUCCESS in 49m 21s\n- cinder-storpool-tempest-iscsi-multipath build/6a70cbd6bc7b4c6f9e6f178690ee2a12 : SUCCESS in 57m 29s","accounts_in_message":[],"_revision_number":4},{"id":"29eeabdc6ee4e3b94f2c36c725cebd0d34a19601","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 14:35:50.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib998143680b41464a7618f4bf2b68d5b8a4e91d2 .","accounts_in_message":[],"_revision_number":4},{"id":"5f448ed0181b3e695362ddc18115676f8159eb58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 16:01:25.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1cf473cf3c44c189a7166a4b8022e49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7357cd763c4746a179ac5c4f5d60a1 : FAILURE in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/840729c9fc12422a8aaca1be01ed5dc0 : FAILURE in 8m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24faf48b638a4cd0917cf32fd4574402 : FAILURE in 8m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02782a3b71024a258bf7aed310ac803e : TIMED_OUT in 41m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc89fb5f9d084d1b8fa1886197d8aff3 : SUCCESS in 14m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/2176ca77011f4ca7962a89b583c4bb3e : SUCCESS in 1h 11m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c82898d838614469a7da69a72a049d2a : SUCCESS in 1h 11m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e53f1e0a26b34ac5999499dba7905f02 : SUCCESS in 1h 36m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9d6d4c8ae28427d91feab8306dadc15 : SUCCESS in 1h 02m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ab784150e6646d2a006511e970a74b2 : SUCCESS in 9m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6920dd7426074fa7acad2ce696ba2fca : FAILURE in 26m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a8bfbb7eba4e4fc2a86a02628d30a175 : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/47e5e832dfa44b85a5b05120aa309d96 : SUCCESS in 5m 26s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/185481afc13f499caedba1ed2f686154 : FAILURE in 10m 07s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7919ddcc1d3c408ab0123a2e8a8398b6 : FAILURE in 6m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/10f7f874a0754fd0bc173344d707e470 : FAILURE in 6m 54s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7cb0f86335a241a38f6f733622428634 : FAILURE in 9m 18s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dca2339e6158428b83ca09bb5c69cb2f : SUCCESS in 38m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/45820c32b2a247b79a3c96677a9e0b1c : SUCCESS in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bf3e0cf6865b418590b1675197c90c31 : SUCCESS in 1h 25m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0ebf9a2ab85b43dea03210c8bce750ec : FAILURE in 2h 21m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3b5a708f2556414cae7208f6d89885cc : SUCCESS in 1h 43m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b6476da4a8154566b95824604f603072 : FAILURE in 2h 08m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3d0510337aff4d5281375afcbfad736b : SUCCESS in 1h 18m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/49eaf3312b56419f90ce54c516990f1f : SUCCESS in 38m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e94fd09f30f14827a0b6c2e738a3a9b9 : SUCCESS in 1h 02m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fe66d3c231d349f5ac927d999f9ee153 : SUCCESS in 1h 12m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c6c47def92e246589d1f41b215feaae1 : SUCCESS in 28m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f0c0dd8bc8d04d93a8ec3a654be62c49 : SUCCESS in 1h 57m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d353a4f273e4bf5a87d2c94ab77f411 : SUCCESS in 1h 34m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd592ea70929481fb1c8b35b25cd452c : SUCCESS in 50m 03s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"190e7b2d74c667467b1f1ea6a1adaf825509a511","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-22 15:20:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5031/.","accounts_in_message":[],"_revision_number":3},{"id":"e598ac19340d4b76c26445ce4d87adccf257f1c4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-22 22:50:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5213/.","accounts_in_message":[],"_revision_number":4},{"id":"1c2b6a4fb67a01791aec9e163abc3d3b4ab77c2d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-22 23:36:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5037/.","accounts_in_message":[],"_revision_number":4},{"id":"8bce2a703cca0039da75dfa60c1bfb73d0c946c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-08-25 04:16:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"3f445c457440253b15b015919d45b560c34c6a76","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-25 04:26:06.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5244/.","accounts_in_message":[],"_revision_number":5},{"id":"3e50405f5b3213bb6f90fcc28846e628de9c41be","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 04:56:06.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/f1e939a364544d37a43adf04eee3e203/","accounts_in_message":[],"_revision_number":5},{"id":"bab8fdc7c633a8c2d9647324b411073ddf86241e","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-25 05:04:03.000000000","message":"Patch Set 5:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/930 : SUCCESS in 0h 47m 18s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1000900","accounts_in_message":[],"_revision_number":5},{"id":"637b079453fcb9ea512f174a040aaccea1506bc6","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-25 05:05:01.000000000","message":"Patch Set 5:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1023 : SUCCESS in 0h 48m 12s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1000900","accounts_in_message":[],"_revision_number":5},{"id":"2b27d9a208fbf3e2ab134ad1a033c7f990f26c70","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-25 05:05:07.000000000","message":"Patch Set 5:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1292 : SUCCESS in 0h 48m 21s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1000900","accounts_in_message":[],"_revision_number":5},{"id":"5d967074ce71ea0c43c48aff86fec94322239068","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-25 05:10:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/4070aca294384766849b2cf8beb456f4 : SUCCESS in 50m 59s\n- cinder-storpool-tempest-iscsi-multipath build/d67a58f923e34afe8a1b855c0cde021b : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":5},{"id":"251e223353ae15f6ef681db8a25e9d420aef2b60","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 05:11:21.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib998143680b41464a7618f4bf2b68d5b8a4e91d2 .","accounts_in_message":[],"_revision_number":5},{"id":"0adce591fee52b1e67038afe858ffff09ed1f864","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 05:26:36.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f1e939a364544d37a43adf04eee3e203\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1727efa18314404999a896ced3359d46 : SUCCESS in 49m 37s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/299b125ed46240368fd41bffa2054430 : SUCCESS in 38m 24s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9d435d2c793041c3b60349af9543c82e : SUCCESS in 46m 51s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c9ee54fbe5d549909ccba2bdbdcd1ff1 : SUCCESS in 39m 13s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b8371fee4a6440dac20c0c5d7cdc9ca : FAILURE in 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d0d9ad1652290ec9575b3ef013501a9144dbc6ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 06:32:26.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fe1a20723e54955a2cae5b1a86b305f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/187c36bac5c84e418a2c7587d60d58db : FAILURE in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1dfbf5c91835483d9d29552d0edc1d22 : FAILURE in 7m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d0e435ba6b142978f2db7897c437b82 : FAILURE in 7m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5ec0631053244a81b5e81cb34443e9e7 : TIMED_OUT in 41m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6428ac4e993451ba88970bae8c6ab95 : SUCCESS in 15m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/a14824bbd29d454a9b357caacbf62d35 : SUCCESS in 1h 08m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4f3f677442c14cb492e764d71d63fe11 : SUCCESS in 1h 01m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b71fe4a36cc4f8bbb48e9593ede1cbd : SUCCESS in 1h 32m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1a7a840cb094fe9b33c1fa5bcaa8e40 : SUCCESS in 39m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cab6a2e9bcbc45d8a93db6e41f12edb8 : SUCCESS in 5m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/05a8318a64684bbb88100e28b25c1888 : TIMED_OUT in 41m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eef1063032ca433a9f6bac44d98f5cd3 : SUCCESS in 4m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/178a6f8e3e2a4310af0399a9d59ddbe0 : SUCCESS in 3m 17s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/a93d0f55415e4bbda2a4a021a473a966 : FAILURE in 11m 38s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/51551401402e43eeb47d3cf7836b7680 : FAILURE in 8m 11s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2b1c57835d0c4d3aa1affeebfa9b8607 : FAILURE in 6m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bc5ae37e930b42eeb5e441481ebf381a : FAILURE in 6m 04s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/41fd7dbf600b4e2591a181782ee221ce : FAILURE in 5m 29s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0707e2550aba4e21907db831d1582a6c : FAILURE in 8m 19s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/083dc9053ac34f9d92c618326b672f03 : SUCCESS in 37m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ef3fa4f9e6a433abcc0afb33cbefceb : SUCCESS in 3m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d2cf53d8e26d456f8f6227f91cba53ec : SUCCESS in 1h 38m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f88ec640a544086a39a6bbdfd54ed02 : FAILURE in 2h 08m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a25cb8c46894dc3b82d51d351bb9767 : SUCCESS in 1h 27m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/490ec3b5d5da4fb88f59c23079e8327d : FAILURE in 2h 10m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b15be1c1479c472bae5edd03e737983f : SUCCESS in 1h 49m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4de5f688406b4f209a7675d931200cf5 : SUCCESS in 19m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c04b22990f0d4f2585dac30c566aa585 : SUCCESS in 43m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d0585be6cb024d1ca776b726ca1ea087 : SUCCESS in 1h 44m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7aa10b134c734a4b9980d3a7e60278ed : SUCCESS in 44m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/70016e5a13a044f294bc0da1f46dac93 : SUCCESS in 1h 37m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/48d721d73b6446a9a3097c6831a3ef87 : FAILURE in 1h 12m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf4d0af7315f4d119ce41bd197275b19 : SUCCESS in 1h 12m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"93a1aa9e58ebb6d916a407af68d6fa1ac811a789","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:17:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de8164fbc5eb47d6ba1b57e16a177308\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d17a52d225e444899109f8b3bbe473f4 : FAILURE in 9m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3699477b39a649a3b898a736acf5e937 : FAILURE in 24m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1eb5012e630c4d15abe29acf0c4713cf : FAILURE in 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ab282c34938afefc8056f3b42da654c310430c09","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-27 18:05:21.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5068/.","accounts_in_message":[],"_revision_number":5},{"id":"697b38ac0b36f1859667ae6f29acc14b791d7daa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-09-15 09:25:56.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"c5e0575fc904b976935d24cc3e54792f079250cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 10:08:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/348ce5dcb55647b5ae24dc6e2b1e4969\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b235bdc07cd34f2ba7bb9179fb89d9d5 : FAILURE in 13m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/271d4824c0c64aa591e54bd7143d67d4 : FAILURE in 16m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f56d47bb8f564b159fd37bc450b26bef : FAILURE in 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"86c9ebf1c689fe164a514c9288f533b107f2aa22","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-15 10:12:29.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ib998143680b41464a7618f4bf2b68d5b8a4e91d2 .","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a0fcff75930ce2ecb105d2e060762d5633c40539","revisions":{"e09062093764509c9e76b53265801c588e274540":{"kind":"REWORK","_number":1,"created":"2026-08-13 20:42:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/1000900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/1000900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/1000900/1"}}},"commit":{"parents":[{"commit":"1257986934a75b2595ceadaa416794556519b347","subject":"WIP: Group failover: Update host and AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1257986934a75b2595ceadaa416794556519b347"}]}],"author":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-13 20:28:11.000000000","tz":330},"committer":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-13 20:29:13.000000000","tz":330},"subject":"WIP: Implement export/import APIs for cross cloud replication","message":"WIP: Implement export/import APIs for cross cloud replication\n\nAssisted-By: Claude (Opus 4.6 for plan, Sonnet 4.5 for implementation)\nChange-Id: Ib998143680b41464a7618f4bf2b68d5b8a4e91d2\nSigned-off-by: rajathere \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e09062093764509c9e76b53265801c588e274540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e09062093764509c9e76b53265801c588e274540"}]},"branch":"refs/heads/master"},"5df57325110ed383d4141d222f4b564469b2bd6d":{"kind":"REWORK","_number":2,"created":"2026-08-18 17:46:52.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/1000900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/1000900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/1000900/2"}}},"commit":{"parents":[{"commit":"1257986934a75b2595ceadaa416794556519b347","subject":"WIP: Group failover: Update host and AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1257986934a75b2595ceadaa416794556519b347"}]}],"author":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-13 20:28:11.000000000","tz":330},"committer":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-18 17:45:36.000000000","tz":330},"subject":"WIP: Implement export/import APIs for cross cloud replication","message":"WIP: Implement export/import APIs for cross cloud replication\n\nAssisted-By: Claude (Opus 4.6 for plan, Sonnet 4.5 for implementation)\nChange-Id: Ib998143680b41464a7618f4bf2b68d5b8a4e91d2\nSigned-off-by: rajathere \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5df57325110ed383d4141d222f4b564469b2bd6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5df57325110ed383d4141d222f4b564469b2bd6d"}]},"branch":"refs/heads/master"},"c12da181319252a913c230664bdc2fc5177d8cec":{"kind":"REWORK","_number":3,"created":"2026-08-19 21:05:06.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/1000900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/1000900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/1000900/3"}}},"commit":{"parents":[{"commit":"1257986934a75b2595ceadaa416794556519b347","subject":"WIP: Group failover: Update host and AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1257986934a75b2595ceadaa416794556519b347"}]}],"author":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-13 20:28:11.000000000","tz":330},"committer":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-19 21:04:36.000000000","tz":330},"subject":"WIP: Implement export/import APIs for cross cloud replication","message":"WIP: Implement export/import APIs for cross cloud replication\n\nAssisted-By: Claude (Opus 4.6 for plan, Sonnet 4.5 for implementation)\nChange-Id: Ib998143680b41464a7618f4bf2b68d5b8a4e91d2\nSigned-off-by: rajathere \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c12da181319252a913c230664bdc2fc5177d8cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c12da181319252a913c230664bdc2fc5177d8cec"}]},"branch":"refs/heads/master"},"70810156d10fcddf30b1f7c6a2d5809aee5e7068":{"kind":"REWORK","_number":4,"created":"2026-08-21 13:33:38.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/1000900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/1000900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/1000900/4"}}},"commit":{"parents":[{"commit":"1257986934a75b2595ceadaa416794556519b347","subject":"WIP: Group failover: Update host and AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1257986934a75b2595ceadaa416794556519b347"}]}],"author":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-13 20:28:11.000000000","tz":330},"committer":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-20 13:51:38.000000000","tz":330},"subject":"WIP: Implement export/import APIs for cross cloud replication","message":"WIP: Implement export/import APIs for cross cloud replication\n\nAssisted-By: Claude (Opus 4.6 for plan, Sonnet 4.5 for implementation)\nChange-Id: Ib998143680b41464a7618f4bf2b68d5b8a4e91d2\nSigned-off-by: rajathere \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70810156d10fcddf30b1f7c6a2d5809aee5e7068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70810156d10fcddf30b1f7c6a2d5809aee5e7068"}]},"branch":"refs/heads/master"},"38845cf60edbcfb79f3871ea171c32324b24f3d9":{"kind":"REWORK","_number":5,"created":"2026-08-25 04:16:26.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/1000900/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/1000900/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/1000900/5"}}},"commit":{"parents":[{"commit":"1257986934a75b2595ceadaa416794556519b347","subject":"WIP: Group failover: Update host and AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1257986934a75b2595ceadaa416794556519b347"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2026-08-25 04:15:13.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2026-08-25 04:15:13.000000000","tz":330},"subject":"WIP: Implement export/import APIs for cross cloud replication","message":"WIP: Implement export/import APIs for cross cloud replication\n\nAssisted-By: Claude (Opus 4.6 for plan, Sonnet 4.5 for implementation)\nChange-Id: Ib998143680b41464a7618f4bf2b68d5b8a4e91d2\nSigned-off-by: rajathere \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38845cf60edbcfb79f3871ea171c32324b24f3d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38845cf60edbcfb79f3871ea171c32324b24f3d9"}]},"branch":"refs/heads/master"},"a0fcff75930ce2ecb105d2e060762d5633c40539":{"kind":"REWORK","_number":6,"created":"2026-09-15 09:25:56.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/1000900/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/1000900/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/1000900/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/1000900/6"}}},"commit":{"parents":[{"commit":"0a1e4581bf8821b040dc5e6dc8c53161c02218d5","subject":"WIP: Group failover: Update host and AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1e4581bf8821b040dc5e6dc8c53161c02218d5"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2026-08-25 04:15:13.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2026-09-15 09:24:16.000000000","tz":330},"subject":"WIP: Implement export/import APIs for cross cloud replication","message":"WIP: Implement export/import APIs for cross cloud replication\n\nAssisted-By: Claude (Opus 4.6 for plan, Sonnet 4.5 for implementation)\nChange-Id: Ib998143680b41464a7618f4bf2b68d5b8a4e91d2\nSigned-off-by: rajathere \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0fcff75930ce2ecb105d2e060762d5633c40539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0fcff75930ce2ecb105d2e060762d5633c40539"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY"},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
