)]}'
{"id":"openstack%2Fcinder~1001087","triplet_id":"openstack%2Fcinder~master~Idcf3ff4449fb6850c26d13a4ea95431b393dba5c","project":"openstack/cinder","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-08-18 13:36:14.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-08-19 16:36:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idcf3ff4449fb6850c26d13a4ea95431b393dba5c","subject":"Add test coverage for utils.tpool_wrap()","status":"MERGED","created":"2026-08-17 07:11:35.000000000","updated":"2026-08-20 17:41:43.000000000","submitted":"2026-08-19 16:36:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001087-eventlet-removal","meta_rev_id":"4b508856d879e409773530aec71d96f7c3426610","_number":1001087,"virtual_id_number":1001087,"owner":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-19 16:36:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-08-18 13:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-08-19 14:17:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-08-19 14:17:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-08-18 13:36:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-17 07:36:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-17 07:58:39.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-08-17 08:02:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-17 08:02:54.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-17 08:07:07.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-17 08:26:42.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-17 08:59:42.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-17 09:01:35.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-17 15:56:55.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-08-17 19:04:32.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-08-19 14:17:47.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-08-20 04:28:47.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"a3d85edac1c73f93fe787bfcadc1f92156428c8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-17 07:11:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"690f3d252bd3fe330bd085e0d808dcd499dfa1de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 07:36:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78e64b780af24c2fbc064ffc06f9ce57\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb417c41411a43e3855eef68227a1b68 : FAILURE in 6m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0344a691a3e43f68c616bbb441e84e1 : FAILURE in 6m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/34e393f0491042d78a7bdbbdde2d04ae : SUCCESS in 22m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c613840ae27f6af6fe17da8fcd5982dccdbd4d4","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-17 07:58:39.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1200 : FAILURE in 0h 3m 17s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1001087","accounts_in_message":[],"_revision_number":1},{"id":"94283d4ba71b3be2e4922a600e79b522b1f589fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 08:02:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/185b9caf81b24534aa09294265f5bf9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f54616a8a5c14137a9f9086446247949 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e80b8c73c520451d8e6ad344b9fa34f3 : SUCCESS in 7m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/148fa9348fca467e89e4b7f028535129 : SUCCESS in 7m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0efab0dc603a4ebb9e924c62fb6ec375 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10337a41474e4a9fb68816a1854be514 : SUCCESS in 7m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aee867e679a34167ae4c5a0053750884 : SUCCESS in 29m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e949380ff4f4559b2e077848195a65c : SUCCESS in 5m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e7c66b7e538545f882a7112872d339d2 : TIMED_OUT in 40m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4b4126c8f1964cad8f15c8146c0f11ad : SUCCESS in 3m 07s\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/6b51ce086c404c6981087a32b453177a : SUCCESS in 9m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ab8440cf6b4323e929f7102b89dd539c7b3be6b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-17 08:07:07.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Idcf3ff4449fb6850c26d13a4ea95431b393dba5c .","accounts_in_message":[],"_revision_number":1},{"id":"b2930e789f146e583ff5dc4e2d10fae91183f19c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-17 08:13:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_08_2026_08_06_31\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"3811bad45b7f3f7d4555a68a0af3fc00ec77952f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-17 08:26:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ebf68652a4bd497393690c924eb81bbb : SUCCESS in 47m 45s\n- cinder-storpool-tempest-iscsi-multipath build/19c30ddc6fe247a09898872e7ec282f2 : SUCCESS in 46m 20s","accounts_in_message":[],"_revision_number":1},{"id":"328730ab215da3ebe80f5299e12cd3558b4a2468","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-17 08:59:42.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/933 : SUCCESS in 0h 58m 52s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1001087","accounts_in_message":[],"_revision_number":1},{"id":"d437ffc2811044506446a93e3111aa83f6f18d56","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-17 09:01:35.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/839 : SUCCESS in 1h 0m 21s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1001087","accounts_in_message":[],"_revision_number":1},{"id":"fc132135193bda00cda81b694dca39e2c7a1d249","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-17 13:09:30.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"3ca54ba06e59a15ab6763644e5dc84acbd210782","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-17 15:56:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d882905930303f59e6a95e5e7e23c1b76845b6de","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-17 19:04:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4970/.","accounts_in_message":[],"_revision_number":1},{"id":"56e1d9306012c7944ba211a3697ce57d5881622b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-17 19:48:24.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5146/.","accounts_in_message":[],"_revision_number":1},{"id":"557f4efebec3439aa4e254da28523aa60048daec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-18 06:37:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7e3b7cbd2f57248f23cfdbc811311a2b6b37d301","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-18 06:40:24.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4f88a03b203d5ba7b29ffa6ea6c4c2870327c436","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-18 06:49:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99a7559aa8047f7758df696de6e4c4266d16d67d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 07:05:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f15359c64489443faae338f3eafce029\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/37c626c2351e4f5f88fc17bad754e400 : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/393e60c8bfe14004b869788ee64eae69 : NODE_FAILURE Node(set) request 67b85b56f42140ca8d9b1089a40bbf94 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a213cecc43b44124bb69c67c78ac2eb1 : SUCCESS in 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9eebf423823df852d83768fe60d19af4d8a86a4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 07:12:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16e4b255e4414df190df5a6904a423bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e1d1146486e4c3681ecbcbed67b0e3a : SUCCESS in 7m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d3096f1613b4903b503a16d024aa952 : SUCCESS in 9m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a227249abf8145759c521e753942c77e : SUCCESS in 10m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0eb3d7c04b0c491ba6b9052e619e98c7 : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16094f6de61947eb8e77329751c52a26 : SUCCESS in 14m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/36f860964f1047eb823059caa9bf4651 : SUCCESS in 33m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d0baf2b4f7d4451b57c0f55ef3585d3 : SUCCESS in 9m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/373054660e934cfcb16afd85c4a5dee8 : SUCCESS in 23m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/68a5b9cd3c844fec8c6763ed43638e0a : SUCCESS in 4m 47s\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/261ae36aa0bd47018bb8e936f71fc7b8 : FAILURE in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1ca75eec09e4b68805e68283c418455a45f9649","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-18 07:34:08.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_08_2026_07_27_08\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"20627ac2052a0c357113d775efa66e5deea0d426","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-18 07:36:47.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Idcf3ff4449fb6850c26d13a4ea95431b393dba5c .","accounts_in_message":[],"_revision_number":2},{"id":"f5810971318bf4c50aa1f7613d755c6b74f1ea8c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-18 07:38:06.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1228 : SUCCESS in 0h 59m 31s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1001087","accounts_in_message":[],"_revision_number":2},{"id":"cfb2d554572e0378c1af3af1d73bfff0a2d6b38c","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-18 07:39:37.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/865 : SUCCESS in 1h 1m 8s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1001087","accounts_in_message":[],"_revision_number":2},{"id":"5031fd3c7910957105214387f4c5667014ebb4b2","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-18 07:40:39.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/959 : SUCCESS in 1h 2m 9s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1001087","accounts_in_message":[],"_revision_number":2},{"id":"716fe66059c57a9c4f9d39055fc8c8b3d5669296","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-18 07:52:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/37b6414c94a14762a935cdbf25a30dc7 : SUCCESS in 45m 59s\n- cinder-storpool-tempest-iscsi-multipath build/f57c90a8cfa04a9980c7b69088ea941a : SUCCESS in 50m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e27a6483eea3b1fee5fda899a318a98ac4404ccf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-18 13:36:14.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9e24cf47f7a5945cee0a0e3d656d99c138dbcb20","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-19 12:30:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5172/.","accounts_in_message":[],"_revision_number":2},{"id":"89299ed4ff577e4c54698287b8b5fed7829598b5","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-19 14:17:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"13f92581bb8766e564f70e40db411abee49ac5f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 14:18:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ee8a3a58b0656f2134552c8439df648434a0b10","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-19 14:22:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4996/.","accounts_in_message":[],"_revision_number":2},{"id":"33199eb0d38b7aead2d3ca515326484565d2eece","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:36:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/009aa2bad72a4135aad2f0c31ccfa902\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790d980d51b6419aa69bfa0dc4a6ce79 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/536cc9a44ec24408ad3d4be382bbb9e7 : SUCCESS in 13m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/270800114b8c447399afd553403b3f50 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15144a0fd4ce4e7f873b2dadc62c54f9 : SUCCESS in 14m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db46de29f73343aeafdfbd4cf2d86ee7 : SUCCESS in 1h 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e70c24492be24c91ac6f3f4fcda8a338 : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":2},{"id":"375d15c4d2c89c82a12523a8abad8110446ba014","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:36:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ee02bdd062e08f5627612a6b28c9b053ead6c618","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:38:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2660af634ea749d599ca622c112b200c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd9b14e343a4bf1841575de35972c35 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9c1a97a11b1ccb62f461c792b9e1e0cee7186dcc","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-20 04:28:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-87-1001087-1-fc : FAILURE in 11m 35s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-87-1001087-1-iscsi : SUCCESS in 1h 1m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4b508856d879e409773530aec71d96f7c3426610","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-20 17:41:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-87-1001087-2-fc : FAILURE in 12m 9s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-87-1001087-2-iscsi : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1639dcd76efe522ccd77c25ca04983ff67fd2a6b","revisions":{"2b8e606674286c2227a4a0d473e5f8806b1c05d2":{"kind":"REWORK","_number":1,"created":"2026-08-17 07:11:35.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/87/1001087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/1001087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/1001087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/1001087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/1001087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/1001087/1"}}},"commit":{"parents":[{"commit":"b5b763129e2bde5077c0cf3a5eb434021abaa6e0","subject":"Merge \"Centralize cooperative yield for eventlet removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b763129e2bde5077c0cf3a5eb434021abaa6e0"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-08-17 07:10:18.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-08-17 07:10:45.000000000","tz":330},"subject":"Add test coverage for utils.tpool_wrap()","message":"Add test coverage for utils.tpool_wrap()\n\nChange-Id: Idcf3ff4449fb6850c26d13a4ea95431b393dba5c\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b8e606674286c2227a4a0d473e5f8806b1c05d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b8e606674286c2227a4a0d473e5f8806b1c05d2"}]},"branch":"refs/heads/master"},"1639dcd76efe522ccd77c25ca04983ff67fd2a6b":{"kind":"REWORK","_number":2,"created":"2026-08-18 06:37:52.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/87/1001087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/1001087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/1001087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/1001087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/1001087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/1001087/2"}}},"commit":{"parents":[{"commit":"b5b763129e2bde5077c0cf3a5eb434021abaa6e0","subject":"Merge \"Centralize cooperative yield for eventlet removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b763129e2bde5077c0cf3a5eb434021abaa6e0"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-08-17 07:10:18.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-08-18 06:12:44.000000000","tz":330},"subject":"Add test coverage for utils.tpool_wrap()","message":"Add test coverage for utils.tpool_wrap()\n\nChange-Id: Idcf3ff4449fb6850c26d13a4ea95431b393dba5c\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1639dcd76efe522ccd77c25ca04983ff67fd2a6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1639dcd76efe522ccd77c25ca04983ff67fd2a6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
