)]}'
{"id":"openstack%2Fcinder~1002451","triplet_id":"openstack%2Fcinder~stable%2F2026.1~Icb81abb1bf216efc744c78eecf18b321235a7400","project":"openstack/cinder","branch":"stable/2026.1","topic":"bug_2074377_qb-stable/2026.1","attention_set":{"13915":{"account":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"last_update":"2026-09-16 11:01:15.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Icb81abb1bf216efc744c78eecf18b321235a7400","subject":"Quobyte: drop psutil requirement","status":"NEW","created":"2026-08-26 08:58:10.000000000","updated":"2026-09-16 12:08:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":992276,"cherry_pick_of_patch_set":8,"meta_rev_id":"293604a10cf1a01481974a91f4fedb1e12d6c245","_number":1002451,"virtual_id_number":1002451,"owner":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 11:01:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"all":[{"value":1,"date":"2026-08-31 15:01:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-26 09:17:29.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-08-26 09:56:31.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-26 10:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-29 11:26:19.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-08-31 15:01:25.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-09-02 11:48:15.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-16 11:01:15.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"379f232754338cbb4df3d189f2541168396f8f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-08-26 08:58:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"84df4059c94d094833bae3395daa5d4321387aed","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-26 09:17:29.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/8fcec166fb54438999b9348e2c5493a6/","accounts_in_message":[],"_revision_number":1},{"id":"c7164b1c6fb67c95d77c4d19d7a9fcbd15bb313f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 09:56:31.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Icb81abb1bf216efc744c78eecf18b321235a7400 .","accounts_in_message":[],"_revision_number":1},{"id":"a5daea1db5080202e122bfdc7b5e8ecbc33d1b1a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-26 10:01:21.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8fcec166fb54438999b9348e2c5493a6\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e02629fcb26a4ea8acc076fca5025502 : FAILURE in 53m 22s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0a6f29986069475eb426a8b739c8db25 : SUCCESS in 38m 09s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4cba7977cb4e455ba1e5dc17f4b63081 : SUCCESS in 1h 02m 00s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f883ac87f2e74bb7b2a222a58c670579 : SUCCESS in 36m 36s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6a53ba7d0d0947e79fcff2385b5e7bf7 : FAILURE in 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24a39851defeaa3abf71a2976d722cb7ecd91dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:49:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec4de1243ad947959c2f8871bee3bffd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f9889dd7bd4bb1b4a68251cb27b126 : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c921ea5d9a34b48a3a7d85f2ecacf29 : SUCCESS in 10m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3275fa3a80a4430bb5b7099d5eb19551 : SUCCESS in 10m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/696ee284da0b405e8054ac4a752a0f2c : FAILURE in 13m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a723c2560a05496486a4dd1077a55c67 : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/0ff1e6d84db74dfc8241bf8a4b75d7f3 : SUCCESS in 50m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3752304eac7f49619281f37fa476fe2d : SUCCESS in 54m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07c13bd5b6924b97a54dcbd7914365cd : SUCCESS in 1h 33m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d075fb29007845b1af4e024b8f61efb8 : SUCCESS in 54m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ea39d13676241adad5a1cd25dc280cc : SUCCESS in 9m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/137a334459764e81b79e2fc86f5db748 : TIMED_OUT in 41m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7966c4ea1b3d41558f59b2a649db3e3a : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8077493adda64fbbbf90560d70d4ff78 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/07001003f5fb440397f74a4bbb6542ee : SUCCESS in 5m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c01444cf9d4e4e28915ab4d350649602 : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/891647df2f974cd19f880c68f49df969 : SUCCESS in 29m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7fa8f99ff23349f1bc1f81fbcef5a0c6 : FAILURE in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f1f84814aa0445ddb7f6d7e14581ffdf : FAILURE in 30m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e76ed705bf2647a3aa3665e41390644a : FAILURE in 31m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/81435f9624654f28935950f6ffe49231 : SUCCESS in 1h 29m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f528ab6dac4040c1a70ffa7285d31894 : FAILURE in 19m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7745406cee034bba86df29b427cf110a : SUCCESS in 44m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d56fe15c5ac946218289c8b0558ea765 : SUCCESS in 1h 16m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/45432884ff114880a40dd5073a8fa002 : SUCCESS in 1h 33m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/55bbfb33f45147cd891cf4f5b9f4fbcc : SUCCESS in 39m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/daa8a574bd954f87b552832ea031a94a : SUCCESS in 1h 45m 34s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6a7d820f992349adbc64b9141dd674a6 : FAILURE in 8m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/024870a7e0644ac9bd7b759a8fb50056 : SUCCESS in 1h 27m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bbdd341e3394ddb9051e748c09ac687 : SUCCESS in 50m 52s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"add832f0539039b67f95c50d29bf13269971e1ad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-29 11:26:19.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5085/.","accounts_in_message":[],"_revision_number":1},{"id":"fe45e914bf97b8ada2c56c36b98d9b40f15d736e","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-31 15:01:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6437e4fae8b4abd6f7e70ddf18fef5e37bf7d95f","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-09-02 09:07:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ba30261c6899f57f4b15807ea46e384481aec3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 11:31:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20fd26b5755d4e12b58d74575752ae67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea28b4e4de674e618022035cd7951281 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d405b3d22391487ca55e9371202c146c : SUCCESS in 11m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b4c03bac76544979a3215813d8542be : SUCCESS in 13m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/33e97bc017534655ad2f5d67f66fa2fb : FAILURE in 14m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b6a4767e15043339e380c9118319b51 : SUCCESS in 14m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/334bd6db9cec49d2a750e39a7bfce3be : SUCCESS in 48m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61fc145252164feca154080edeed0812 : SUCCESS in 1h 18m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd522cecc76a42d792acfbec3e9dec8e : SUCCESS in 1h 30m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7dab70e5669447d894645e2e54f9bea : SUCCESS in 47m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/345449661e3a4404ae5bc5b40cc04fa3 : SUCCESS in 5m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cc4608eb49774f9b994779654e2e15a4 : SUCCESS in 21m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4487c2600f704c1980bc056b3f4414aa : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2dafa7f24816417ca4d6db8534d04a50 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc0f9b7173914dbc81044ec8c27ddacc : SUCCESS in 5m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1e3d6decb631447690bcfaec22458892 : SUCCESS in 6m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/564962d2e78f4df5917b54bcd10d4285 : SUCCESS in 29m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7247d5c915d347898537bfe427e86998 : FAILURE in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5197bab431b54f648d6aa4e52159cf63 : FAILURE in 14m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/73880399b76046bba949597b94ba198f : FAILURE in 33m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/787b7174c4fb479896449574bdd23190 : SUCCESS in 1h 49m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d7e0205e2fc24b8c8e6afe791c4fdd8f : FAILURE in 13m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e6ae0df8198648b1942117ad026e8a98 : SUCCESS in 30m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/49719b9df3c5465fa8a05449fd524b37 : SUCCESS in 37m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/822067b2a105450d8a0bd102ef53ba3d : SUCCESS in 1h 50m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1738c2f9048548a4a943384e6be4e924 : SUCCESS in 30m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f3c81ae218df434f99a5dcc03c524b05 : FAILURE in 2h 07m 21s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2a0bc8876aa34d8796f3cfec577d11b4 : FAILURE in 7m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0afa709bf30d4fbb90ab390f0b0e2565 : SUCCESS in 2h 17m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e012e544dd2241049a2207872325b224 : SUCCESS in 44m 19s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"1d25bdbc182d9f45421c546ee0c1f928c704f441","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-02 11:48:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-1002451-1-fc : FAILURE in 13m 52s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-1002451-1-iscsi : SUCCESS in 1h 36m 4s","accounts_in_message":[],"_revision_number":1},{"id":"adedd4bbbd20730849de0a029425942fcd7c898c","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-09-02 12:07:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84d6c39b24dbb7f0647b729734f068e9b8c9af85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 14:12:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13fe1958c3464550805bfdc50a5c8efa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49dea6de87134327bb1c64d9961217ad : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9a592d156a141ecbb9bc5e8c5d10461 : SUCCESS in 11m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5671c7a1d854dae8f51e80f373afffb : SUCCESS in 10m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/42e0ed342d5b4d898d6dce70ac2a2fc8 : FAILURE in 15m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca473586dca946c4b2e1bc5aea490d24 : SUCCESS in 13m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/467a8db63fae40a783251bf602413715 : SUCCESS in 1h 03m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dea39df67a784fa68a1b1af25a7ecb00 : SUCCESS in 1h 11m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a4322144e9d4c29bab77e1c7148d75d : SUCCESS in 1h 37m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e085b6e3b8864cc68a3686fe39bc936b : SUCCESS in 1h 02m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63629769349646b38158a35e78d6508d : SUCCESS in 8m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b8b57a03981741948830722b5cd248d1 : TIMED_OUT in 41m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/43f46529c27243e5bce07bce31a21b4b : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e23a083e19443fa8d8e8435ceda62ec : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d411d9b18224ae8af073492545188cb : SUCCESS in 6m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4408a02104824ee6b7caac6f5d298d9b : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7bcda9639fc24594b25f62d3791a481e : SUCCESS in 38m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/daed71dab774433d86043e0570035498 : FAILURE in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d7545afaded54a36a318cd334367e3d9 : FAILURE in 28m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8102d244fec04f44837333eefeb278de : FAILURE in 34m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aa892da9c4274abfb02cdddbe65ad29d : SUCCESS in 1h 23m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5b77c0aad8d141309351b23ec230f2ec : FAILURE in 18m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2ba2f78e7a494e0ab266321b2f255fc4 : SUCCESS in 35m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc46c425f6ed4e67b4fb1af0fb9de3c5 : SUCCESS in 1h 13m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6432fceed35b430f897152b137725565 : SUCCESS in 1h 40m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f3cdc6c46c0940c984b3b7c8d2f52e5f : SUCCESS in 45m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/92dd4bcb94d14c258c12600c9197c99d : SUCCESS in 2h 01m 15s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e793d647c7dd4c808b958a69ce63975d : FAILURE in 9m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f2ee7ca084341b08306453a59c5d1db : SUCCESS in 1h 38m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cea5492952647328567cf276de27ba2 : SUCCESS in 37m 22s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"a9be972b0e486190738ab51f1e4d687b61802c4e","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-03 02:20:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-1002451-1-fc : SUCCESS in 46m 50s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-1002451-1-iscsi : FAILURE in 1h 36m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5df665940d9bd8a7d795d670e852cc0ec122e609","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-09-16 08:30:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a4522fce853d081e24760272760ee18bea49c26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 11:01:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aedc51288a8744aa99696790ff08731f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a4b96469894c98af9e2a6bc1aa7f0c : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbbf2dce6c8947519ce6b1db926909aa : SUCCESS in 12m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/22721599c57744cfa3f0a964388555e1 : SUCCESS in 13m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/64b643757cc141029f098c0e048abaf2 : FAILURE in 17m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1923b94dfd4cb6b30cb77e90ed516f : SUCCESS in 14m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/825e22d20390478d8f47e8ebba513568 : SUCCESS in 1h 02m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd44c9edf6054df9afeb009b9c771893 : SUCCESS in 39m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e0e481230b945f885680438676139b0 : SUCCESS in 1h 28m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4be81261bdcc44d5b20e135e07b4a0dd : SUCCESS in 1h 02m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6596007ad6e24176826487e681c775d3 : SUCCESS in 8m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cb38cf7bd1ce4fd9b7826a1a84c1a260 : SUCCESS in 20m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5d117a1504594469acca7f3abca11a39 : SUCCESS in 6m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/df9587d43f164155b9a71c999ff06f71 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a1ff1bc4ca384d67aad9538671d95ee0 : SUCCESS in 4m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ac88a0f3c017487abf14393dcf57d8c4 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e2e3c345e18c49038f58637f12c1ed7d : SUCCESS in 40m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7c88e0fa42b4306bb218b8fcbed2d6b : FAILURE in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2c33540d2a5b47c78b91626b8d43e72b : SUCCESS in 1h 06m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b257a31c58d94cceb8dc4b63c2ed5815 : FAILURE in 2h 20m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3fb92c55086d46f9a3b3fc429c63694a : SUCCESS in 1h 57m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d0fe2827bc6d4c69bd71ca8077912b71 : FAILURE in 20m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ffa89f9ae6764de880d6fee3fa46b296 : SUCCESS in 41m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2c1d21a81e1446f79784bf69912aae27 : SUCCESS in 1h 05m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/176102eae77e4648ae3c8a0cb4357516 : SUCCESS in 1h 38m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/53d782933b2342b0b1cdb2c820f70774 : SUCCESS in 27m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dfc2fabb95ea49bb840c0ffc3b082853 : SUCCESS in 1h 10m 38s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bdf2a984bbec4c1eb585efae35a9d362 : FAILURE in 19m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d325b8ad6e124ee3bf4582067d14177b : SUCCESS in 1h 27m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c72545e423794f29a80f650037aa6237 : SUCCESS in 45m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"119a695900ab2aceeac2261431a045365e6cae5c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-16 11:11:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_09_2026_11_04_45\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"293604a10cf1a01481974a91f4fedb1e12d6c245","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-16 12:08:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-51-1002451-1-fc : SUCCESS in 52m 24s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-51-1002451-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72628f204af736375bc3692574ce711025528220","revisions":{"72628f204af736375bc3692574ce711025528220":{"kind":"REWORK","_number":1,"created":"2026-08-26 08:58:10.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/51/1002451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/1002451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/1002451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/1002451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/1002451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/1002451/1"}}},"commit":{"parents":[{"commit":"1f33a97ae71d6ad95824e98269874a13212ff63c","subject":"Quobyte: Fix metadata for volume from snapshot #2142477","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f33a97ae71d6ad95824e98269874a13212ff63c"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-06-08 14:50:42.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-08-26 08:58:10.000000000","tz":0},"subject":"Quobyte: drop psutil requirement","message":"Quobyte: drop psutil requirement\n\nDrops the psutil dependency from the Quobyte driver, which reads mounts\nfrom /proc directly instead.\n\nChange-Id: Icb81abb1bf216efc744c78eecf18b321235a7400\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n(cherry picked from commit 326830567d3954f020027ceeb3d8214c81e0122e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72628f204af736375bc3692574ce711025528220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72628f204af736375bc3692574ce711025528220"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"b5b763129e2bde5077c0cf3a5eb434021abaa6e0","ours":"1f33a97ae71d6ad95824e98269874a13212ff63c","theirs":"326830567d3954f020027ceeb3d8214c81e0122e","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
