)]}'
{"id":"openstack%2Fcinder~1002975","triplet_id":"openstack%2Fcinder~master~I062efed16a93f630a6fc6901df6e9a2d09e7f312","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-31 19:34:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I062efed16a93f630a6fc6901df6e9a2d09e7f312","subject":"Replace remaining deprecated pyparsing interface","status":"MERGED","created":"2026-08-30 16:48:47.000000000","updated":"2026-09-02 15:37:12.000000000","submitted":"2026-08-31 19:34:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002975","meta_rev_id":"496feca621c3e752b9a2a93a00890a4b53863431","_number":1002975,"virtual_id_number":1002975,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-31 19:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2026-08-31 17:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-08-31 09:06:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-08-31 16:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2026-08-31 17:54:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-30 16:58:24.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-08-30 17:06:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-08-30 17:11:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-30 17:36:57.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-08-30 17:37:28.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-30 17:37:52.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-08-30 17:42:29.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-30 17:44:08.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-30 19:12:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-30 19:12:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-30 19:23:25.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-08-31 09:06:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-08-31 16:56:08.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-08-31 17:54:43.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"43184c4c1ae4dc763fd5b19c9538ecdacb559bb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 16:48:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec22137666f37c3050ce9e1062a3e63cf660fb7a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-30 16:58:24.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5324/.","accounts_in_message":[],"_revision_number":1},{"id":"d4f97f1ed7dbf88dd456dd511772bbb5b38ae13e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-30 17:06:45.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/fe9807d2a32844dda7a27cff0d9faeca/","accounts_in_message":[],"_revision_number":1},{"id":"b546a2860e8a8f82f18eed6a55e50d8f81b47dc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 17:11:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f38634414164b2dbca5fc204b24ad0e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d4ec6b7f1bd64bb7a9426f88f974aa36 : SUCCESS in 14m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ee66846af8347d0a77b7bb41a31de6c : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/394ad869538b42ce9054612aa9317f70 : POST_FAILURE in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc73112b46fd80ac826946422f0ae9d0263f2586","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-30 17:36:57.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1008 : SUCCESS in 0h 47m 47s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1002975","accounts_in_message":[],"_revision_number":1},{"id":"d4390a7626c55372b88b7d679b36593d724bccf6","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-30 17:37:28.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1100 : SUCCESS in 0h 48m 23s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1002975","accounts_in_message":[],"_revision_number":1},{"id":"91c7f5719871ea9d754bc028ea0e2e6ec02653f9","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-30 17:37:52.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1369 : SUCCESS in 0h 48m 45s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1002975","accounts_in_message":[],"_revision_number":1},{"id":"4ca5c823fd0c11b4589d15fa1307b6c2c950821c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-30 17:42:29.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I062efed16a93f630a6fc6901df6e9a2d09e7f312 .","accounts_in_message":[],"_revision_number":1},{"id":"14ed656db6b3d4ad39a858a108940e4f3a922b9f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-30 17:42:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/fe9807d2a32844dda7a27cff0d9faeca\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9fcccac1f44a42298a6b8d619a1cc0aa : SUCCESS in 37m 04s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eb10f4f0a4634d90a970bc1233d5279c : SUCCESS in 35m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/140335cdc63a4293acb0537c2f8153b8 : SUCCESS in 53m 31s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/daaf69c356d843caa6d4a18881ef758d : SUCCESS in 37m 26s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/68ad50bf337f408ba0dda6e447b0d35c : FAILURE in 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c75b750bdca569bff851f146885af0bb4875fbf8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-30 17:44:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/cc1cc74b3b7d4e7ab0e1c3c271df8419 : SUCCESS in 47m 59s\n- cinder-storpool-tempest-iscsi-multipath build/31a9d1a80ceb4b969c3b212aae4557bf : SUCCESS in 53m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3b9cf9014b7357f88280c34dac97a4430ebc038d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 19:12:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6191bb62c9c4aaa9782113e7d9eb30f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06dc7307b60c418597c1c69c9e32c681 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77899d4e548549f5b24006290ae5c64b : SUCCESS in 12m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23cbbe669b014687bbecb72f53bb2d38 : SUCCESS in 9m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7886775347084488a9efe3ff6ab35b7d : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2d314f5fd024481a437e878789c59ea : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/3d96254a287a45b4aa3ddf2567912341 : SUCCESS in 54m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/211c4da5b6604b99becb3db4cbec13cf : SUCCESS in 1h 11m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9eb562705f9a49b5842eb0c70079f3bd : SUCCESS in 1h 09m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d65765b027cc423bbd392662cba83aea : SUCCESS in 27m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d202cd61db84e7fb6d495149c6c07df : SUCCESS in 7m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4b96eceb7b99408bb3b55f525c6c9d55 : TIMED_OUT in 41m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/85b39ef4380d43748952c61a3b88bfb8 : SUCCESS in 6m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c0087d432e846f9902e3f8af78f6100 : SUCCESS in 5m 32s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8f497290c26245e797965a463205e885 : FAILURE in 7m 31s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d9a7d29c14b3488897e75f310d50c3eb : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1ed05aacfd474da2833225103fc5733c : SUCCESS in 6m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cfc623288e984239b41cb6a4770c4da7 : SUCCESS in 5m 48s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b1b5011334c14eb999b2913e8540f3f2 : FAILURE in 3m 47s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0c8898a9aeed41caae56a45cec33c574 : SUCCESS in 8m 37s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/02f0132728c742b39c3d58bf7972927e : SUCCESS in 36m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1edc273aac2f48df967d643a064dbe75 : FAILURE in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1cbaceb61543458ab27d8389e91357bf : SUCCESS in 1h 25m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c71da2973ecc435399132dfac00d01f7 : SUCCESS in 2h 07m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9e84620940324665b8d5bc2c833037bd : SUCCESS in 1h 42m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/3612f3793be34146923ffdeca92b5ac8 : FAILURE in 2h 09m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/41eafb4a9f9847f494202eed42a27547 : SUCCESS in 1h 34m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6ed77a046f9b4860bd068733f2a3f3bf : SUCCESS in 34m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/75efea0591f3437fa85a6449b20c5e1d : SUCCESS in 1h 15m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3f726f0c9fda4f348621818a18abb000 : SUCCESS in 1h 36m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/220a6c54713443e3999ec81599d4ac8e : SUCCESS in 27m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e09434f55ddd40b49aa9c86c158ae25d : SUCCESS in 50m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f42c76053a54ba1b42a4ddf9aca211c : SUCCESS in 29m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"5000c6adce1659afcd921bc8ff25a6a6c6862da1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-30 19:23:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-75-1002975-1-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-75-1002975-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"663497429ea3b635807cb77cb15dafe79e156a8f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-30 19:23:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_08_2026_19_16_14\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"e66598681ac9ce549d5043365c90741e3ddbdbac","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-31 09:06:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"521c802b7864e4731f7edcd331c2bde19d46673f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-31 16:56:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"39ffaf87a91d829951662e835f1286a25850b43f","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-31 17:54:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51a1716c06307681e64d3a74e639de30dbabe7fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 17:55:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6937f2f5791614dcbfb4bade70537fe4a0290c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 19:34:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3b6e49c64334e068b5af1c99871f49a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f590aeb1ff0145c3941f7a4bbde368bd : SUCCESS in 6m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/971e1253c3b545909a3193e8798b33bb : SUCCESS in 7m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd1931e5b84440cbbc115c15cd45a912 : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91bf0963e9244eebb2db2934f522ef3a : SUCCESS in 15m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/db67f48d806f43cf97ded28b6cffffcb : SUCCESS in 47m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49c1dddfec6d42849b1b400b8383e7ad : SUCCESS in 52m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c946b9bec50c48c7bb20523d851c99fa : SUCCESS in 1h 08m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1b51097581347aeae7ab75854d2721e : SUCCESS in 43m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/792c1786d73c43ba93b83e7039003442 : SUCCESS in 10m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a7e480c46dc04f979f45b9c85dbd9278 : SUCCESS in 1h 21m 10s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0415cfdf730646119327dbe3acf7fef7 : SUCCESS in 1h 35m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8959ac597e3049359afcb1c20e4aaae6 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":1},{"id":"87e6e8ccb5379c2a683b20381a84e5a78f970700","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 19:34:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"439a1b9025b7cea41503d84fb8cc68cfcc008e39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 19:35:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52796219d663420696a78d1b11a185ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1250239b4ddc41f5aa46893567f8012a : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bda69db3b05ccbdd42b47947559ff0ccaf0e9d65","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-01 02:35:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-75-1002975-1-fc : FAILURE in 1h 1m 4s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-75-1002975-1-iscsi : SUCCESS in 1h 23m 23s","accounts_in_message":[],"_revision_number":1},{"id":"496feca621c3e752b9a2a93a00890a4b53863431","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 15:37:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5148/.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af7c3f9adb8d2d11d9edb77d7abfee7bdec3f2e5","revisions":{"af7c3f9adb8d2d11d9edb77d7abfee7bdec3f2e5":{"kind":"REWORK","_number":1,"created":"2026-08-30 16:48:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/1002975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/1002975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/1002975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/1002975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/1002975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/1002975/1"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:47:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:48:30.000000000","tz":540},"subject":"Replace remaining deprecated pyparsing interface","message":"Replace remaining deprecated pyparsing interface\n\nIt was not caught in 0602d414f77fbec30dcdc82e0ab5ac96d564628c .\n\nChange-Id: I062efed16a93f630a6fc6901df6e9a2d09e7f312\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af7c3f9adb8d2d11d9edb77d7abfee7bdec3f2e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af7c3f9adb8d2d11d9edb77d7abfee7bdec3f2e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
