)]}'
{"id":"openstack%2Fcinder~1003246","triplet_id":"openstack%2Fcinder~master~I7f69007929270111c260a9aef864b523633f9e67","project":"openstack/cinder","branch":"master","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-01 11:05:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-09-01 11:05:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-18 16:52:08.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-18 21:58:21.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-09-24 05:26:58.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}}},"hashtags":[],"change_id":"I7f69007929270111c260a9aef864b523633f9e67","subject":"Fix typos","status":"NEW","created":"2026-09-01 10:39:56.000000000","updated":"2026-09-24 05:56:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"28a637ecd1751614208a148ce19fa352cdad106e","_number":1003246,"virtual_id_number":1003246,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-18 22:33:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 10:41:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-09-01 10:41:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-09-01 10:50:24.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-01 10:59:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 11:21:37.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-01 11:32:54.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 13:00:37.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-09-01 13:02:21.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-09-01 13:07:42.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-09-01 13:22:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 13:43:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 13:43:51.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-01 20:20:05.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-02 07:02:41.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-09-03 16:25:08.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-18 16:52:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-09-24 05:56:59.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"a3493276016ce308450b76f29701c2702ac21946","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 10:39:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a7e201b0d69367cd53d3d35feffdf5a5e811fd1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 10:40:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"208b593986ff31eeb2205d37393366fa3315fec0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 10:41:05.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"c7ff2734d94117250a4de409628e243b042ec465","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 10:50:24.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5351/.","accounts_in_message":[],"_revision_number":1},{"id":"11de6746ea8756cfb3f34b75967c14d19c0e2422","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 10:59:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a286c8587234c7e83f93cb992faf13d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ce66afd949fe495381b2eb61e76a69a1 : SUCCESS in 14m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d6c2842c30464bf4936e0a0785a9ab2e : SUCCESS in 11m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9436815f5a1a477ca9b32a648aca6750 : SUCCESS in 13m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e569635481827643a91ba838f8a10562ec06311b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 11:00:24.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5352/.","accounts_in_message":[],"_revision_number":2},{"id":"4e499a3ef89a87fd5ebda7a460c37ebb5ff46aec","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 11:01:54.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"85dc5b4143459d7703c88418291aee12731d288e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 11:05:13.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"8937f65b4d86b7d6902b1898cdba5c48d50f3d28","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 11:05:18.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"46af3997c6752c50359ee823414148136412b9a3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 11:16:53.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5353/.","accounts_in_message":[],"_revision_number":3},{"id":"df1eed3a6ecfd2672cc96a38bb79a589486f8d40","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-01 11:21:37.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/ee9430c2082b48669028edf5df91042e/","accounts_in_message":[],"_revision_number":3},{"id":"d70ea97637a22b21a3c77320dbe2d79045a664f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 11:29:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed584cd9e93940bc8affbffcd1488b28\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/be79de634bb54d3f83232a231693c95e : SUCCESS in 16m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3a59a48c8e1240cfba1bc23b2bc9fd6d : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3343b7a3aa924304b1adebd44c9d92ff : SUCCESS in 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4d05e30eb9a9d81b382db1ce1f08f14245a0de53","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 11:32:54.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":1},{"id":"8df445f420fe557a35ac549d1605d7a5b516ecf6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 11:33:41.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":2},{"id":"ee66fbeabc2412765461ea833d1a88a8bf0c16de","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 12:00:05.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":3},{"id":"bfe235568af607e9784b11944da04f66181c1dc2","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-09-01 13:00:37.000000000","message":"Patch Set 3:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1397 : SUCCESS in 0h 46m 12s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1003246","accounts_in_message":[],"_revision_number":3},{"id":"8073ae08e67a3e51dfaaf4713cf00740adb449c7","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-09-01 13:02:21.000000000","message":"Patch Set 3:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1036 : SUCCESS in 0h 45m 52s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1003246","accounts_in_message":[],"_revision_number":3},{"id":"3d1b8dd413a3089de4542a08136c77f23f7026b3","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-09-01 13:07:42.000000000","message":"Patch Set 3:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1128 : SUCCESS in 0h 46m 21s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1003246","accounts_in_message":[],"_revision_number":3},{"id":"b7ec63fd02e206bfab648eac69aabc6f8e7a8989","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-01 13:14:18.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ee9430c2082b48669028edf5df91042e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/373e72682453421298a6065e44137440 : SUCCESS in 45m 41s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/85a7ba3faebd42bd919a19651fc9ecb2 : FAILURE in 32m 02s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/947e04332f864bc4b8c6fe4177eb3e0d : SUCCESS in 52m 38s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ffd641fe699b468881dff40377ccc969 : SUCCESS in 38m 19s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/935edd80a342436f8cb2c49603c0a3c3 : FAILURE in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1d40546ea7ea3864ac7431548d03e8e1d7cea42f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-01 13:22:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/62f59e6f58a547739ea399e017fb5395 : SUCCESS in 41m 57s\n- cinder-storpool-tempest-iscsi-multipath build/ac7c62fdf3d340c1b60ae726b180ced4 : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":3},{"id":"5b6498960fa376f1b5faac92352137528c39e208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:43:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11950278747848eb9f88b5a331685a85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1a8242dd6a4e3fb065125681b919c1 : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e9eba53810649699050d13b8ec2621f : SUCCESS in 15m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/116d820fd3144cdaab32a401dd7a6d5b : SUCCESS in 8m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/90fa833007a44e59bf9156b982285812 : TIMED_OUT in 41m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aecf256e11e47a9b7b4d163615cee76 : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/e6678d50bcc746f9b30142dea990e015 : SUCCESS in 53m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e099ace03554e638281be94a04ed796 : SUCCESS in 1h 05m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69c285dcbcce4168ab72a7ef20c78788 : SUCCESS in 1h 32m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06c2691cbf9f4cbda0df92e13c4f7b81 : SUCCESS in 1h 01m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ab31127fcbf409c9c6df02ff0ff0218 : SUCCESS in 9m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6ddd279d46734718afa71d0a2e9325d5 : SUCCESS in 19m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/75f882d398d8461fa6fb9a56a210ea30 : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e91f08cf53d4d3da530bad2641a248b : SUCCESS in 7m 36s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c41598db2a174bcfae36fb117680c719 : FAILURE in 12m 18s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/32b18da429bd4b0280804ad9c4256ce0 : FAILURE in 10m 34s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/95a68ac352b34fe18a4add7f3ed11dd4 : SUCCESS in 6m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/76aca2dfbbb74d43bc2fb3e41f672ce9 : SUCCESS in 6m 50s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/cb51bba2525541f697a6e7b7f68e690b : FAILURE in 8m 11s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0ba89da1083645c1adc7158078592830 : SUCCESS in 9m 56s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4c07cdcade0f414bb5b4a31eb05fcaa0 : SUCCESS in 40m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc21f6c2252842f2abd6f603af91e3f2 : FAILURE in 6m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5792f1a404ef401295fa4deab364495b : SUCCESS in 1h 25m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f3deb0367e5479da67c654f722f4aaa : FAILURE in 2h 13m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/13959c22ce964ae492e497b65f988549 : SUCCESS in 1h 55m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/566efd22546a47009785ed313a7ce0de : FAILURE in 2h 32m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5d21dcd757bc4191900b4ddd6f50ab4b : SUCCESS in 1h 49m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f5b0eae2209a4112b03e380ff2f9be32 : SUCCESS in 29m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e944e2bdbe854ed09fe64872339da91f : SUCCESS in 1h 22m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5a30b346841042179eaf76cf41b1f3b8 : SUCCESS in 1h 07m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/948639a7b5004b5795262f9b22d4dd19 : SUCCESS in 36m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70ddb3791c6b4e0381c36398a8a90af1 : SUCCESS in 1h 29m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19ff08f066cb42b0aa34bcc1074d96a5 : SUCCESS in 1h 01m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"085a514e500583c0d41c3e563a88e1031d148f95","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-01 13:54:49.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_09_2026_13_47_47\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"acf38753fca63d7b20d7e45e641c2a3536f6fe35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:09:17.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"3e3664aac24c8c1033a579e20c201e78e9a0ce87","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 19:19:11.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5360/.","accounts_in_message":[],"_revision_number":4},{"id":"151d29c31010f192569aa10806215b22bad701a8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-01 19:27:53.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/915c8f1eb79e47a1af96d49fbcc35542/","accounts_in_message":[],"_revision_number":4},{"id":"8a81c91e548ee6420ed2c7263836b66e3a53afe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:49:22.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9348ac2d1992370e6d4d4cc9eb82b1a529812bdc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 19:58:17.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5362/.","accounts_in_message":[],"_revision_number":5},{"id":"ea64e4fd5ae73053b284a039de5a632b80d63d80","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 20:02:56.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":4},{"id":"6b08fc051cf62c452b94054836b41d2b811e28d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 20:12:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7ef38a37bce43a2b137a8599e16b064\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/20db59db5ed74c4d859a66dc76655552 : SUCCESS in 16m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b13e80fd9ea4ad28c5243248d54d4ae : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/36d4c0b7a7d64e5c8af11ea96e03812d : SUCCESS in 18m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bfb8771d8f4c2ccaa75e87c1c56c5656728b54f9","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-01 20:20:05.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-3-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9ce93c032d6c08c7b48f158677b9ca74f8445d63","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-01 20:26:10.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/ca77def9c61140e9b393ab9afd74a35e/","accounts_in_message":[],"_revision_number":5},{"id":"a017b6f293db4b7861a3888c731b9cb252a51179","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 20:44:14.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":5},{"id":"ef358942f774ce696efcc9767cee3d944d6ee6e7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-01 21:30:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ca77def9c61140e9b393ab9afd74a35e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5b8a1d90afee484383e6d653cde02ca5 : SUCCESS in 51m 09s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ba429612c3746f5965a537151a13f71 : SUCCESS in 33m 41s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2d7dcaad933148929ccc9a5e64bad9e7 : SUCCESS in 42m 32s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9cc02fb50b8d4d54832d537d38854cac : SUCCESS in 33m 38s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0207074d656543cda4d51dd699af8226 : FAILURE in 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eb1f696d12b43611d64d25f8b32072d393d067c8","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-09-01 21:32:37.000000000","message":"Patch Set 5:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1403 : SUCCESS in 0h 46m 15s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1003246","accounts_in_message":[],"_revision_number":5},{"id":"20e6006bb26ba50b962f52941ef8ae4b9e2f0efc","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-09-01 21:34:10.000000000","message":"Patch Set 5:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1134 : SUCCESS in 0h 47m 9s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1003246","accounts_in_message":[],"_revision_number":5},{"id":"250559b0d970149a18724f3367683f630528b775","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-09-01 21:34:59.000000000","message":"Patch Set 5:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1042 : SUCCESS in 0h 48m 8s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1003246","accounts_in_message":[],"_revision_number":5},{"id":"d2c0b4639732422f9ab30e8cb433ebc93b02c17d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 22:17:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d445205d3645aca0f717eb7fcb872f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/da329ce5e1ce4ef39693aac22453c247 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e24abc3797c47e5afc02de47e78ba6f : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84eefd06e6da4187b007923932d402e8 : SUCCESS in 12m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15a338a790ff4114a094948f7432a8bd : SUCCESS in 10m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fc8a6311aa6e46ee9199cb8483513e7c : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a6a3b4ccee8452e957596b23420f398 : SUCCESS in 14m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/1946a7b068684399874af32ae3d295f1 : SUCCESS in 57m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/53fbbb1db0d34ec9a89543f29adf7e21 : SUCCESS in 31m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a279285f5f448768aa6501d63e9851f : SUCCESS in 56m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa2fca018a2a4e55a10b65da071a7753 : SUCCESS in 29m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec3b86c67fe04ed48596c630f6cd0150 : SUCCESS in 9m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c3bd1907716d4647bf202aa77996d0c4 : TIMED_OUT in 41m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e9b73bdd2e87427db57b4fb5314d20b3 : SUCCESS in 3m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/350aed237ec144a2a94742adc204d4e3 : SUCCESS in 8m 12s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8f40ebff91874076a966ac718218ddf9 : SUCCESS in 7m 49s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/5a0c7f466b344c84abbdd4a27b41e2d7 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fea58ca736da4204a04eab9c9f9454a9 : SUCCESS in 5m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5eb417d572e44adab3de2db8d448f20e : SUCCESS in 5m 56s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e301bdf551524e5d8e7ea801c89bc354 : FAILURE in 7m 17s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c9dc8acedb6547e88ea3491a2a20b669 : SUCCESS in 8m 49s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e5e2e05b3ea04252af39752fcb0a5313 : SUCCESS in 38m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/84bc15bded744b1e877370f7ca8a56d6 : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/681100b97b054c61bd3a974e97a6a304 : SUCCESS in 1h 41m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ab13b8c1185e44488922d09baec403c2 : FAILURE in 2h 22m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dd665da3b203401d9a95081a19a94fa6 : SUCCESS in 1h 21m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/4b8d74ed30624e1e8973e460f41382cf : FAILURE in 1h 53m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f0dd491938b84dc386e30cd1834349ff : SUCCESS in 1h 46m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7c173c85bfea4388896be713fe5c8408 : SUCCESS in 37m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2430e915e7f84f16a433d0f732ad46b0 : SUCCESS in 1h 24m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e83c376901b74d6e8f3cfce6a50ea8e3 : SUCCESS in 1h 57m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aca23c0fab664a299097189832a921a1 : SUCCESS in 43m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a35033765ee640a8935f6bec492b4838 : SUCCESS in 1h 46m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a4dd0fc7d8e4bb9b837c6f9f7cdd420 : SUCCESS in 54m 54s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"1f6ef7dca155384245fb92bd6751a8f641d37715","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-01 22:28:22.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_09_2026_22_21_10\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"7d785152adc1d56b81fbd27bc3dcf2e330d9325b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-01 22:44:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/aac8f1317f60454a8678ab36b21af113 : SUCCESS in 47m 08s\n- cinder-storpool-tempest-iscsi-multipath build/562369d1d22d47dcbfc235014184589f : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":5},{"id":"ea09a60b91dd136a69696b54acfc2af6680cf4e1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-01 23:47:16.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-5-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-5-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"a3f2d544c762bb139c79f6c64cac6f7876ffc605","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-09-02 07:02:41.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7ed5b83327abf8b54a6664040dc21055880e5261","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-02 14:39:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-5-fc : SUCCESS in 50m 23s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-5-iscsi : SUCCESS in 1h 36m 30s","accounts_in_message":[],"_revision_number":5},{"id":"9e40fbd545826f8da8bcc411c53aad7f42a2b815","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-03 15:38:34.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"cd227399f19df58752d926e3bf881445188fec68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:56:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae328255290943d39682ec23812eec37\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7317a76de39b4f1fb1cb37055ea3c0bd : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9a65a6fcfc648589dc2799f026545e3 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/97213aca0aba4f1fb57fc20d0719a0f0 : SUCCESS in 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5060e38af85c876fe16b2eb612b4c65ec724b909","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-03 15:59:05.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest build/ca40f3b707e943f1852aa112c5fc9399 : RETRY_LIMIT in 17m 52s\n- cinder-storpool-tempest-iscsi-multipath build/c884faed464142afb196d81eb7af8a3f : RETRY_LIMIT in 18m 57s","accounts_in_message":[],"_revision_number":6},{"id":"131876184ef6a56d0531953b141588c7aa79c3e9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 16:10:52.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/1dc2efdbf4654208a68aa02e6486a0aa/","accounts_in_message":[],"_revision_number":6},{"id":"0de5b156451ef4c71a12d83418b006a2c37a8b35","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 16:25:08.000000000","message":"Patch Set 6:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1071 : SUCCESS in 0h 46m 12s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003246","accounts_in_message":[],"_revision_number":6},{"id":"cbbed5d2c1f6f88599d988a0563b68de823f95dc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 16:26:36.000000000","message":"Patch Set 6:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1432 : SUCCESS in 0h 47m 35s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003246","accounts_in_message":[],"_revision_number":6},{"id":"4bedc49917712005aee2b66e097a1b19643539be","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 16:27:36.000000000","message":"Patch Set 6:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1164 : SUCCESS in 0h 48m 38s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003246","accounts_in_message":[],"_revision_number":6},{"id":"22edd30151ff55db8a8a5ee6fe788c75b2b23bb4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-03 16:31:39.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":6},{"id":"849ff32ea125dcf3ab61b677ece49d2daa24b52d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 16:36:08.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1dc2efdbf4654208a68aa02e6486a0aa\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e16c2af0474e46f9a1c517b501e5bd1a : SUCCESS in 40m 18s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8201acc435c847e78d4d8bfd67a49fb4 : SUCCESS in 36m 14s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/94b90d3d735f4ed18e84780585286846 : SUCCESS in 57m 03s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/be3bdf8dd8d5499da18e14512071a5d1 : SUCCESS in 34m 25s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8996f4745a1c4099b62c223c738687bb : FAILURE in 18m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d28da115cbe5b14be9533ad96ae94ac602e5c91d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 17:57:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5175/.","accounts_in_message":[],"_revision_number":1},{"id":"d3e1b4b3aa8c3ab03d5cdd9f50145a21c7e91390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 18:24:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61c4d105467f4f779dee1301c5f66b5b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/362173193e804b39977f3680bef5bdfc : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2879757b8eb74efda36b533189bda0f8 : SUCCESS in 6m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f996a7ec877f47c3b405c250df51c746 : SUCCESS in 11m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14a21b10268e4a968013c0171a9be7ea : SUCCESS in 9m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2df88c94aa3a45d49b243e0689cc4a38 : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0147025476cf46cea9ca9464f3fb345b : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/d82f12e4124842f0a1c32863895084be : SUCCESS in 41m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/60a5e8cdd8a54fdcbbc3c4ef111ed4cd : SUCCESS in 50m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1dafe1e625cc40739daf74a22cf4096f : SUCCESS in 1h 09m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65e7020a91b748dfb342b3da1bbf0680 : SUCCESS in 28m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8eaf77986654828b5d982fdbeeed4cb : SUCCESS in 7m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eda5143beb254f25b18b957d210a70a0 : SUCCESS in 25m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0698496885df42c38a788db4b4d33b65 : SUCCESS in 3m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1742a77f8be74fb393786d550fa44a38 : SUCCESS in 5m 01s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8c1267fa80f2413184da029674cab3a0 : SUCCESS in 8m 04s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a82ab041f3904bd6b41501b98a021a36 : FAILURE in 7m 48s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9ff18eeb512b4b06a084ca0fa113d72b : SUCCESS in 5m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/01c6591ffdba4d43978638e15c40d52d : SUCCESS in 8m 25s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ee0166f719d246e0b89fb30824e36d53 : FAILURE in 5m 53s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/322e61a2c65f4480808e084e7cebc950 : SUCCESS in 6m 41s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5feff25e8ade418c8f48e8ce54e843da : SUCCESS in 38m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e0ab399dd29444c0bb6a2b5b38383b32 : FAILURE in 4m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ae5bf4f1d4bb46d3bd94761208051d09 : SUCCESS in 57m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/680d188f5f614a05a55e89354e8aed93 : FAILURE in 1h 00m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a86334e0f7c47668ed17ec6a5c33ba3 : SUCCESS in 2h 06m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/aea126231cf84b4d98739a47d6ae03f5 : FAILURE in 2h 06m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/00e87b8aa436404baec3c1b4d979cb63 : SUCCESS in 1h 32m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/17a6475ff07645bc8d76bcf8d07f4081 : SUCCESS in 37m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9b7b248dad5345918c9e5b7b460c3dc9 : SUCCESS in 37m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c54e1f972ae743dd9aef5e1c193a9555 : SUCCESS in 1h 12m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2d72a6360e4c444099471c2b31d9aca7 : SUCCESS in 34m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/42d7c153893e464ead1441fa14119ece : SUCCESS in 1h 40m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1056d19f5f8047599dacdbc3fa0feccd : SUCCESS in 59m 56s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"c926913069a80520275fd470062c566ab52237a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-03 18:33:31.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_09_2026_18_27_54\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"4de301e48dfc77b55d661a4c21ea09c671ae1e02","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 18:57:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5399/.","accounts_in_message":[],"_revision_number":6},{"id":"a1c3e90bf1d1c1246fdcd4b3995b124efee448bf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 18:59:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5176/.","accounts_in_message":[],"_revision_number":2},{"id":"484f6979283f78fb6bc246e68aa71a8dcd6671af","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-03 19:30:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-6-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-6-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"409da97d85f4fc03cc27b41c3a10d14ce521f4cc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 19:55:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5177/.","accounts_in_message":[],"_revision_number":3},{"id":"3a601d2ab83edc1051def7f198280deac510423e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 02:38:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5184/.","accounts_in_message":[],"_revision_number":4},{"id":"e39ada325cd348e2d8e859fd338af293a664f7d0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 04:29:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5186/.","accounts_in_message":[],"_revision_number":5},{"id":"ca559dcd10ffe0973b19f234c6a04610bcc0c295","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-04 14:12:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-6-fc : SUCCESS in 52m 50s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-6-iscsi : SUCCESS in 1h 19m 19s","accounts_in_message":[],"_revision_number":6},{"id":"f172a3d319448f9558b3323d80015bf6e27e2177","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 21:54:18.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"52cc58b80a3b7831ebf316e910a054d7c1b22011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 21:54:46.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"6e8dd4132f460319e8adffee7b2f3422e9ebabf6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-06 22:11:54.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/338ac894a036418ea89b0f2e3cb15602/","accounts_in_message":[],"_revision_number":8},{"id":"45f3975438fe800c2d423da642c9207c44133a11","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-06 22:15:00.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest build/9086c06ef8ec45d08f5adebe8e8c01af : RETRY_LIMIT in 17m 47s\n- cinder-storpool-tempest-iscsi-multipath build/b5d342b8cba24e6c915d23cfb44ba39a : RETRY_LIMIT in 18m 39s","accounts_in_message":[],"_revision_number":8},{"id":"40e8eca62b1761463cb217dee9339bc11ef53349","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 22:18:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b49fc22caad847fdb4881878c856c79b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/15ad00e2c3ee42bd944e0277fe67eb90 : SUCCESS in 16m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6b389b0b6c8b4de6a6788820a9f8331f : SUCCESS in 18m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e1b58f7b136b488ca288ef9e61ebdd4d : SUCCESS in 22m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4f6a0e903f9d05986cf45f853538e5ff14e518bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:36:08.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"c953cbfd4eb621e100f59d4054eb43c6ebcfc2a6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-06 22:48:30.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":8},{"id":"2603ac8d8e23847c28b12b5f34cbf8370dd500c8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-06 22:49:48.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":7},{"id":"4863a39d72e15b5072cd91961b0ee490ba8e0c67","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-06 22:52:16.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/f48fe2edd6774cc4869b1d013c754c12/","accounts_in_message":[],"_revision_number":9},{"id":"3a9444b18bf90ee608108276d01cbcd21c2e21af","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-06 22:55:56.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest build/7b8f3ff7c42049418c4ea9955e664b55 : RETRY_LIMIT in 17m 11s\n- cinder-storpool-tempest-iscsi-multipath build/ed8ff156c1f34c36a575816179de42fa : RETRY_LIMIT in 18m 19s","accounts_in_message":[],"_revision_number":9},{"id":"2a12fee00a6912c290bcb276036a78c4bc141164","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-06 23:29:28.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f48fe2edd6774cc4869b1d013c754c12\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/57427bd5252a42d5a887020dfa91b235 : SUCCESS in 44m 57s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6d6a6088033f455aa33b47fc4e103261 : SUCCESS in 39m 09s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a98defba8ab4ba9b887c9b0644f4a58 : SUCCESS in 49m 04s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/50c8dc3d2b974b35816887d6c113cb2f : SUCCESS in 39m 31s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f0e9b2aa4eb04e718be0bca925f59a99 : FAILURE in 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8e058eddb62044c7696972a95c21c8d8d48d9bd4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-06 23:31:43.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":9},{"id":"a2cda0f5c7e9a2260323de164c80ce1967d6bd05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 23:39:03.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b863c0cd154341338f27176f460a8bcc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/099b1958c3cc4eee87d117cb050a748d : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d873dea38576476f911876356eebf423 : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3c2ac5932cd44733bb360ccd67baea4e : TIMED_OUT in 1h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"df671375a53454957562eac133d025772bf73b3f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-06 23:41:23.000000000","message":"Patch Set 9:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 7s\n- FC       : SUCCESS in 0h 46m 36s\n- NVMe/TCP : SUCCESS in 0h 47m 50s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003246","accounts_in_message":[],"_revision_number":9},{"id":"820eb979ffd969a1e3571ae6b2b6fab980938ef3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:43:36.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"1d4269c66cf9ec1cfca08729cc03ed47676db592","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-06 23:58:37.000000000","message":"Patch Set 10:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/127ad9b2f7c14f7cbbaddfabd14bee31/","accounts_in_message":[],"_revision_number":10},{"id":"aaa30f9c2f1feb693f1d68cbb2dc6813df770f43","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 00:03:27.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest build/9112781d1cd54153bb63f32b3c98ab09 : RETRY_LIMIT in 17m 10s\n- cinder-storpool-tempest-iscsi-multipath build/c259195e0f034276bc2404556977ec59 : RETRY_LIMIT in 18m 18s","accounts_in_message":[],"_revision_number":10},{"id":"b6ca5ef068a71bd2991b1c0ec21d79a83a31c19f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 00:32:40.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/127ad9b2f7c14f7cbbaddfabd14bee31\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bf8d6b250804407ab246f9c463a9a805 : SUCCESS in 43m 12s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/47da714d944e48db896affcd23fcb177 : SUCCESS in 42m 09s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c0131f93e5747d39a20cae77c715696 : SUCCESS in 45m 55s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1552a7817cd847f1b6b376161a56542d : SUCCESS in 34m 18s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d7588aa6086d46fab6fbad9a14b1aa78 : FAILURE in 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b0fb93d239985ac5f4bbc30262b038703b0b8af6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-07 00:37:00.000000000","message":"Patch Set 10:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 42s\n- FC       : SUCCESS in 0h 47m 28s\n- NVMe/TCP : SUCCESS in 0h 48m 20s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003246","accounts_in_message":[],"_revision_number":10},{"id":"2d1f5faaffd58e115c04749a6237cd01e3dba336","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 00:39:57.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":10},{"id":"9e7dd4fbd8ae5a11efdb7f406c41b3d3aa2d0ed2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 00:47:16.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1445c62c3caa40e49901fe0d5bb45a24\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b58e69904cfe48f1a1410633381f25e8 : SUCCESS in 17m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9a46bbaf25064b7ea24f684a977efe48 : SUCCESS in 17m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4157d9ce332f494dac299f94d72a4819 : TIMED_OUT in 1h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cc6183001c0de4d9c32632b1370b8d0961b40265","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 01:31:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5259/.","accounts_in_message":[],"_revision_number":7},{"id":"4aaff4f1f508970b6537d976bc81d7d509dcb7d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 01:38:59.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab0a848707be4b9a9f3e238556310f2e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7135fdecd4a14cd5968a8729ca8dcd8a : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b86ff6448ae640ad9d4fd204276977b2 : SUCCESS in 8m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f419c6623414a05b01694e41837127d : SUCCESS in 8m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20da6a78a49d4dcba79a72109701e0ee : SUCCESS in 8m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/78f3cac0d6e04d3fbc03d56dfb50e29a : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c24686a95b74980abadbe6b21a2f344 : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/a9715d6b94cc4358babb91395d729ad2 : SUCCESS in 1h 07m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c72ca66c47b24b8da56391b83d439cde : SUCCESS in 37m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b7d664985a6409b94e2a96908a107fa : SUCCESS in 58m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d8b726a421b470183ed44bb522a3bac : SUCCESS in 39m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93892b4bea5c46f1967baf90d7f6a9de : SUCCESS in 9m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/674554ccee2f467796fcd9b0ec4af5f4 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/99816c09bd274a6595c6eac53727199e : SUCCESS in 3m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/374598834d8d4ff4a3ddd15ad9d75c8f : SUCCESS in 6m 07s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/9ff2da228ace44298606f118fff68762 : FAILURE in 7m 16s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/557dfe29479145cc99273f9e68f671e3 : FAILURE in 10m 02s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a2cc9abff0dd4b9d921ad97e2c4a6318 : SUCCESS in 6m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bd42b42fb40f46e89f2c48b6b21f9893 : SUCCESS in 4m 41s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f49fd8e607164db6bbbf3605622b46f6 : FAILURE in 3m 58s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b3ec5211e8e34c1c9167b9fd7fc029a4 : SUCCESS in 8m 39s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9a14963487634367a6fe8385ed43f561 : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/589470d03f0d4ace86908f7c03d4547e : FAILURE in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/268a7002f9d3450bb9ad6a33c4787d40 : SUCCESS in 51m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1a420e35b53142318b0a0af6baaac201 : FAILURE in 1h 04m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d5978325f9b946d3a358e1ebc3a1770f : SUCCESS in 1h 21m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/222daf4e05f64a839a1ce280aba8bdd9 : FAILURE in 1h 52m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/74a72c01e69b49b79d5a67c18a62a29d : SUCCESS in 1h 18m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/aeec3dcf434049018f67baaf361a29fe : SUCCESS in 22m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2245a60722ee4c32ab003df80a037ff5 : SUCCESS in 1h 16m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/13d17ec97f864af28b772703d5a7750a : SUCCESS in 1h 05m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1a196aaaf1e148bb827173968de9a8ae : SUCCESS in 50m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2769443bcf4b44c4a5ab5547800e3938 : SUCCESS in 1h 41m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c89a6d7ecd548c4962f63ec791310ee : SUCCESS in 31m 07s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"323def3146f0955163054692a351b12808d630e5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 01:42:07.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5436/.","accounts_in_message":[],"_revision_number":7},{"id":"0baed7c8a0f2bb4d87eadfeff8a638b249770a1b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-07 01:48:34.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_09_2026_01_42_39\nLeave a comment with","accounts_in_message":[],"_revision_number":10},{"id":"ddf6559e94b3c526cd5f14034b795ed29c141920","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-07 01:49:48.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-10-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-10-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"b3395e61f958d84f6d56e4553edcfb926644ac3c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 02:33:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5260/.","accounts_in_message":[],"_revision_number":8},{"id":"3f829972a3bf29ca30d3ddf4afcbfd1c07dcda7a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 02:49:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5437/.","accounts_in_message":[],"_revision_number":8},{"id":"faa4950090733a86190c691ea5a7ec6a14ef9ef0","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-07 03:26:09.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-10-fc : SUCCESS in 48m 8s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1003246-10-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"66fc4f3fed63e5866c41d8c2eb1abc0e8445564d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 03:29:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5261/.","accounts_in_message":[],"_revision_number":9},{"id":"89bc992d08aee5d0d12ae5178c91fd7739a4a79a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 03:56:25.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5438/.","accounts_in_message":[],"_revision_number":9},{"id":"5b5068f5286c3f5221d5f6aa23ed727c4685d6b5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 04:26:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5262/.","accounts_in_message":[],"_revision_number":10},{"id":"a189c4f6162b64421a0d272d3a5be140653bd277","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 05:02:27.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5439/.","accounts_in_message":[],"_revision_number":10},{"id":"3c36c6cc40615fece7c2597fb8d9d1c44ec0bdd0","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-09-11 18:37:05.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"57259e7bd508709c3d131778b97dd9415578031d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 16:54:46.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"9cd7ba304b63e9b21fbc0457ff6979f319c5ed5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:19:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5cccf1b057345e69bd7670d7a95aae8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b02903704a5430aaeeb3688fc894639 : SUCCESS in 16m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/effef4862a9a43b8a51e81a2341c6410 : SUCCESS in 16m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3a11cc3122ac4af58601ff29d2ae394d : SUCCESS in 23m 13s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"58507ea4e60b6a6cab1ecd5c60efdb8c7e8e46ac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 17:31:32.000000000","message":"Patch Set 11:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/f10d99665f46474b899232338ca3ad1d/","accounts_in_message":[],"_revision_number":11},{"id":"ad877ed2a22f08c94286cb87873fd97857a24d2e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 17:47:54.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":11},{"id":"e400f663f00fa3afc1ded7a2f0ef5384a671d4c1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-18 18:09:09.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/32caa3e8bbe844a0b3637162601b258c : SUCCESS in 47m 06s\n- cinder-storpool-tempest-iscsi-multipath build/8b46784f9e894b609822dc0d532eff65 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":11},{"id":"a0db51c40b5fed17b2810fdc56f14cfda5fd04b8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 18:13:47.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f10d99665f46474b899232338ca3ad1d\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ed3cf880bf194028ad899d8474076308 : SUCCESS in 38m 41s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/28e0817415bb4549b3a8e2928a0904ac : SUCCESS in 36m 14s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2ba4d7e41357422aa7db4853611e2359 : SUCCESS in 1h 00m 13s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c79c990e0be46858ffa6fda0a45cc0a : SUCCESS in 43m 40s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bd92604f7c7749a3a725f0890e0a9885 : FAILURE in 13m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0b3346241fee363f63831172c3787097d6e9e234","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-18 18:14:10.000000000","message":"Patch Set 11:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 18s\n- FC       : SUCCESS in 0h 46m 53s\n- NVMe/TCP : SUCCESS in 0h 47m 36s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003246","accounts_in_message":[],"_revision_number":11},{"id":"c4f6029faaf98b6588a318a3606fc66505ac3bac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 18:26:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5422/.","accounts_in_message":[],"_revision_number":11},{"id":"6557dc6f63aff2b6edbe61a151e2d045a7496973","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 18:32:19.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5599/.","accounts_in_message":[],"_revision_number":11},{"id":"951bc1dba57a5e8b27a5e1fdabd3e86217070b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:32:42.000000000","message":"Patch Set 11: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97fbc464d8dd4a55b0f404f470a0b846\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/375461fc59264b3b84371eec55951ee6 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b12f0f80e0d5476fa7b6a5a620061044 : FAILURE in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca29ed5847794bcfbf43f29939051457 : SUCCESS in 12m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8824597911594f8686377f7d76006481 : SUCCESS in 11m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a492fcc8639848799aca22b26664a765 : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1262ea906ec64dca9299696fd4111ac9 : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/4b76d208e97742f6a7d80f5c6cda264f : SUCCESS in 1h 05m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e285c5c79a1f4a06ac42c0712c37fbcd : SUCCESS in 1h 02m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4ee8a01ec9b41ad88223e85e925b5e6 : SUCCESS in 54m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19166790e94a4b77bbbeece65d77beb2 : SUCCESS in 47m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cd1160143e14d83864abb9d46b152f1 : SUCCESS in 9m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cfb51d3c8e0d4708b26300844fca9f84 : SUCCESS in 22m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0a29bd45008f410f9646a599e012c573 : SUCCESS in 3m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e51bd4093bc48009688bf010cc6289f : SUCCESS in 4m 08s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/44085f4e20ea4476a92cbb0a07fb2b83 : FAILURE in 10m 54s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/48452ccc565949c8af3be897f7f3c34e : FAILURE in 8m 19s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/86a9e9307b2b4dd1afb30b7278130059 : SUCCESS in 5m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4ab98889ff36491e98e806da675c9ad7 : SUCCESS in 5m 54s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f40386629f794ed1b7eeb8bb7dbd568e : FAILURE in 5m 51s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a8380eccde38493ba2d8d0587a3cfe5a : SUCCESS in 9m 03s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a0a4845da96c45b882016ac1687f786a : SUCCESS in 43m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ed1a3086df2b4b9580acb32998fa0a14 : FAILURE in 4m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/49fa4e9cc61540acb9e6762efbb9d8e8 : SUCCESS in 1h 49m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6c08db3434944467aad3c599cc1adc15 : SUCCESS in 2h 18m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/057e7147b4f24a9fa886a7a23dac57c8 : SUCCESS in 1h 49m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/1a28e96224364d52b2efc88012c179cc : FAILURE in 2h 31m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/af4ba0510c1846d091c4cbb7e33c35dc : SUCCESS in 1h 38m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0d4b605305af42b59933df31bf3e362a : SUCCESS in 36m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0981fd9903b44b4395687b69f1284e60 : SUCCESS in 1h 19m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9ff32f0bb94b4087892338511bfa68ed : SUCCESS in 1h 24m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/119327db679d4dde95715d17336fecdb : SUCCESS in 29m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f7144bbecaed4e1e9e7c0387df27d4ad : SUCCESS in 1h 44m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e32082d824ce4c91a343d834823d941d : SUCCESS in 53m 55s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"27862d41047f3fd590177a506ac9aaf718f7923b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 20:04:42.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"b3efbad5465786f430b7344bbed889cc5e67ea5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 20:05:48.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"385d54ff28077461aae7cfb44cdaf6701d44c5bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 20:30:55.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3fb5dea67a14878a69db79b11706943\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec7e486d8a8c42158e99a842d19362a0 : SUCCESS in 17m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/261af5f3f3ed4ab2973220886074ea83 : SUCCESS in 17m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bca46478abb042bba8a5714b3e07dbab : SUCCESS in 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"47c039441fa43016cc5339e693973ff0e6223382","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 20:45:51.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/1eb198c8ba2d45b1a1782fcc186c62df/","accounts_in_message":[],"_revision_number":13},{"id":"d8d31e773236c9423a4abf6abc3685bb51483d4a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 20:58:03.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":12},{"id":"f9195de36d6e44bd8a445c98f9f8658f6a7c4be2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 20:59:31.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7f69007929270111c260a9aef864b523633f9e67 .","accounts_in_message":[],"_revision_number":13},{"id":"88f320e3def0b28b33bc635f4a5cf3c1e7714de1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-18 21:17:07.000000000","message":"Patch Set 13:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 26s\n- FC       : SUCCESS in 0h 48m 5s\n- NVMe/TCP : SUCCESS in 0h 47m 42s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003246","accounts_in_message":[],"_revision_number":13},{"id":"f0768be8a3acb28625df7581dc08fa43b0938d02","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-18 21:19:20.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/17832a7770204c44936ee369e1a68799 : SUCCESS in 47m 22s\n- cinder-storpool-tempest-iscsi-multipath build/98b2c7d88fee4b9abb4639ba4f6ad764 : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":13},{"id":"73a4fd8b24ae944c150d86fb4939284e9103202b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 21:58:21.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1eb198c8ba2d45b1a1782fcc186c62df\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/363c416762c340bca3757337f686feb5 : SUCCESS in 42m 13s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2f95917052834e7a8252eee6526452dd : SUCCESS in 34m 55s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0afb4d5619c14211aa6b1b1c039e931e : SUCCESS in 45m 03s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0fe5fbe84ad04e3c8cd2bf1fdd6fdde6 : SUCCESS in 34m 53s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/47439eb92d0d4ddf9a99fad736f8ce80 : FAILURE in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"336d66037affdd5dad88a1721b977bc12327299e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:33:58.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54f2cfc290e646c5a6a8eeda63778c98\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3eed0a6eca994a07b91a6b0c91a8b62f : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c8d2625ea074d3bb36d05eaf70b65f7 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ac02f4f6a9f49ffaeb7e6241b7d3bb7 : SUCCESS in 12m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7992d03eddb418daec37b234b853f2d : SUCCESS in 10m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ce73ffc816c148ac9b7c92399cb79b62 : TIMED_OUT in 41m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d67f0346384ed593b408cc55cac215 : SUCCESS in 14m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/0dfaf0231dee4411bc3f6fe3f70e13e2 : SUCCESS in 34m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abd96b7fed7c405a9ec04409c7dcdbd3 : SUCCESS in 31m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6fd00bb0ea04c89b1a54729bfc58008 : SUCCESS in 1h 17m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2eb8f71cc36e4bfc9ca3d133faa1a638 : SUCCESS in 59m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/042823978837458e83de07a33c702de2 : SUCCESS in 10m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6172283d5fc740be85bf7d537515da30 : TIMED_OUT in 41m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d9898cd98e314279a3bc3cacd7f9ef81 : SUCCESS in 5m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf6277ba4929460899fd3032eab4c0de : SUCCESS in 6m 21s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/63ce568ced4346e49a685f29b549bbc0 : FAILURE in 11m 58s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/87c02848034146d4bcea9703770ec2e8 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d762bba6d9f045f08d176e111ab3a44a : SUCCESS in 7m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9b2d95ef170f44d58f3a48aed649aee2 : SUCCESS in 6m 27s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9aa875b479f24ebfae66fed9916d6243 : FAILURE in 5m 58s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/65527b25a76d4b0ebe728a4dd693e3cd : SUCCESS in 10m 43s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d0f54fc832354b62b8ccfa4f69f6499f : SUCCESS in 36m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93b0ae90dac147708c6710ce46766747 : FAILURE in 3m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7d46a7b5fd7f43c6934b57b48d1a9033 : SUCCESS in 1h 16m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ead519187a9e41bb9d769f70e5dec7e3 : FAILURE in 1h 59m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/36de0f646134460591b8e4a4876bfb04 : SUCCESS in 1h 39m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/0b77a3141fa74720bb68913a841942f3 : FAILURE in 2h 23m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/55672e3783724ca39b0ef94a69fcb6cc : SUCCESS in 1h 50m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/62aefe0ddce74463828f7dc90cf7e7e4 : SUCCESS in 35m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b871381c762b47ab80d45e4781f63b28 : SUCCESS in 1h 04m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c7b8345157fb4b5abd03a7469512461c : SUCCESS in 1h 06m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fe783df5e5fa4f3ab41de083473f6df1 : SUCCESS in 43m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d8ee7d2c54f41d184c6fc00384fbc53 : SUCCESS in 1h 33m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdcf0107d03d4d66b93f5fa662285bfd : SUCCESS in 29m 06s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"e72ebb357279a3ef296b75244f5a0275b2d802c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-18 22:43:28.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_09_2026_22_37_37\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"8239ee371e0f5ec2a8d6a54faadf81a4dbd35d40","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-19 00:03:42.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5604/.","accounts_in_message":[],"_revision_number":12},{"id":"b3c67d14e040ad28614db7037410a9494ffa9e4a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-19 00:47:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5427/.","accounts_in_message":[],"_revision_number":12},{"id":"9ce66d2192f9d9c15d040921b26f39b673307707","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-19 01:31:54.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5605/.","accounts_in_message":[],"_revision_number":13},{"id":"63c4d2140a8056d3878ccab75287e6fa5ee8b842","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-19 01:52:50.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5428/.","accounts_in_message":[],"_revision_number":13},{"id":"cfa035f93668b3edafdb9cd074e9e95ab039b1be","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-09-24 05:26:58.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"28a637ecd1751614208a148ce19fa352cdad106e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-09-24 05:56:59.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp cinder-tempest-plugin-ontap-fcp : NODE_FAILURE in 0s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/1003246/13/upstream-check/cinder-tempest-plugin-ontap-iscsi/6f24464/ : SUCCESS in 27m 35s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/1003246/13/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4ab3533/ : SUCCESS in 26m 27s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"68fe2c1ebd81643d7384975e5959553d2ff84577","revisions":{"1814d855f653b43c58fb85f8227e49fca0567acc":{"kind":"REWORK","_number":1,"created":"2026-09-01 10:39:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/1"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1814d855f653b43c58fb85f8227e49fca0567acc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1814d855f653b43c58fb85f8227e49fca0567acc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"08f9c49f7d94fa91befedb7187cbcf13492d4be2":{"kind":"REWORK","_number":2,"created":"2026-09-01 10:40:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/2"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:40:51.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08f9c49f7d94fa91befedb7187cbcf13492d4be2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08f9c49f7d94fa91befedb7187cbcf13492d4be2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"24fd3f3b27e39123e7aa51227ac29d0f487e031a":{"kind":"REWORK","_number":3,"created":"2026-09-01 11:05:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/3"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 11:05:13.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24fd3f3b27e39123e7aa51227ac29d0f487e031a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24fd3f3b27e39123e7aa51227ac29d0f487e031a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bd768a5638fc7cde25ed3300df54761b37aa9ee5":{"kind":"REWORK","_number":4,"created":"2026-09-01 19:09:17.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/4"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:09:16.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd768a5638fc7cde25ed3300df54761b37aa9ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd768a5638fc7cde25ed3300df54761b37aa9ee5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9e426a5845538bb99a49e5b3e01082505efed0f7":{"kind":"REWORK","_number":5,"created":"2026-09-01 19:49:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/5"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:49:22.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9e426a5845538bb99a49e5b3e01082505efed0f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9e426a5845538bb99a49e5b3e01082505efed0f7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f9391e8baaab2c7d815e89236703a6611f193361":{"kind":"REWORK","_number":6,"created":"2026-09-03 15:38:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/6"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-03 15:38:34.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f9391e8baaab2c7d815e89236703a6611f193361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f9391e8baaab2c7d815e89236703a6611f193361"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e851aa34c5e68ad4fb21bcb2f4fd2ad5699c084b":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-09-06 21:54:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/7"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 21:54:18.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e851aa34c5e68ad4fb21bcb2f4fd2ad5699c084b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e851aa34c5e68ad4fb21bcb2f4fd2ad5699c084b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master","description":"Edit commit message"},"2abc5707e11e9edae394eed307d42c4451d3bf41":{"kind":"REWORK","_number":8,"created":"2026-09-06 21:54:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/8"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 21:54:46.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2abc5707e11e9edae394eed307d42c4451d3bf41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2abc5707e11e9edae394eed307d42c4451d3bf41"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ffa5f43f29e9dba3bf57c499b045b61fc736380f":{"kind":"REWORK","_number":9,"created":"2026-09-06 22:36:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/9"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:36:08.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ffa5f43f29e9dba3bf57c499b045b61fc736380f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ffa5f43f29e9dba3bf57c499b045b61fc736380f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f1d6eb7f258d38d52738c3d2ffc5d4ef7775e4ad":{"kind":"REWORK","_number":10,"created":"2026-09-06 23:43:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/10"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:43:36.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1d6eb7f258d38d52738c3d2ffc5d4ef7775e4ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1d6eb7f258d38d52738c3d2ffc5d4ef7775e4ad"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"71ceb77b1dd3c40160054e43d190ff5a4499d32a":{"kind":"REWORK","_number":11,"created":"2026-09-18 16:54:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/11"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-18 16:54:45.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71ceb77b1dd3c40160054e43d190ff5a4499d32a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71ceb77b1dd3c40160054e43d190ff5a4499d32a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"371986ed8c3935b219ef8a3f3477421599e768a2":{"kind":"REWORK","_number":12,"created":"2026-09-18 20:04:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/12"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-18 20:04:42.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/371986ed8c3935b219ef8a3f3477421599e768a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/371986ed8c3935b219ef8a3f3477421599e768a2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"68fe2c1ebd81643d7384975e5959553d2ff84577":{"kind":"REWORK","_number":13,"created":"2026-09-18 20:05:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003246/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1003246/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1003246/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1003246/13"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:39:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-18 20:05:48.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I7f69007929270111c260a9aef864b523633f9e67\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68fe2c1ebd81643d7384975e5959553d2ff84577"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68fe2c1ebd81643d7384975e5959553d2ff84577"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
