)]}'
{"id":"openstack%2Fcinder~1003398","triplet_id":"openstack%2Fcinder~master~I8b01afb034f46fce9a437c30c6ce6161b88dc843","project":"openstack/cinder","branch":"master","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-03 11:31:27.000000000","reason":"\u003cGERRIT_ACCOUNT_36686\u003e replied on the change","reason_account":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"}},"36686":{"account":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"last_update":"2026-09-03 11:36:36.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8b01afb034f46fce9a437c30c6ce6161b88dc843","subject":"Add cluster_name to volume usage notifications","status":"NEW","created":"2026-09-02 02:33:56.000000000","updated":"2026-09-18 15:35:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc3c10bd0a0d8500553fa9b339ce9432097f459f","_number":1003398,"virtual_id_number":1003398,"owner":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-03 16:27:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"all":[{"value":1,"date":"2026-09-18 15:35:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2026-09-09 14:25:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}],"REVIEWER":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 02:43:49.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-02 02:52:43.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-02 03:23:10.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-09-02 03:23:11.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-09-02 03:23:20.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-09-02 03:28:00.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-02 03:33:42.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-02 04:09:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-02 05:22:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 16:46:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2026-09-03 16:27:31.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-03 16:38:23.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-09 14:25:51.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-09-18 15:35:11.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"}],"messages":[{"id":"046a3eebab3f4c9e838783f0959c7c53c981f889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-09-02 02:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c620a93a4b35a008ba424ae9f9cd8c44ba266f96","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 02:43:49.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5365/.","accounts_in_message":[],"_revision_number":1},{"id":"2e81eec22f3c9f9dee49c6c52aedb48db1eb21b9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-02 02:52:43.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/3d89bb59def84f19bc915eb2fef531be/","accounts_in_message":[],"_revision_number":1},{"id":"16e7b3c84e061f05850d9ca5d16432ffe24ab2ca","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-02 03:18:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3d89bb59def84f19bc915eb2fef531be\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/623cff5d94ac471487cc9e5f8e6ab7e0 : SUCCESS in 43m 25s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ebec9882b2d04b53a66495ef6f463740 : SUCCESS in 37m 10s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/86dca65e02254018a1360375f3946024 : SUCCESS in 43m 57s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a69c5fb057c54f1d8feb25214424daed : SUCCESS in 35m 06s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/57af10e883634203b810067a5fdc19d6 : FAILURE in 15m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"275c85685aad4015c617c9ce9f40342994985c53","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-09-02 03:23:10.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1044 : SUCCESS in 0h 48m 54s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/1003398","accounts_in_message":[],"_revision_number":1},{"id":"eb072e8960b24aa040a1c1907ff97ef0f4afd0c5","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-09-02 03:23:11.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1136 : SUCCESS in 0h 48m 54s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/1003398","accounts_in_message":[],"_revision_number":1},{"id":"1fcf1a58e3cab3826e6f65c4ab57c44ba2dba00b","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-09-02 03:23:20.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1405 : SUCCESS in 0h 49m 0s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/1003398","accounts_in_message":[],"_revision_number":1},{"id":"2e9821eaedcfdcf96b018dae8394008b79206de0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-02 03:28:00.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8b01afb034f46fce9a437c30c6ce6161b88dc843 .","accounts_in_message":[],"_revision_number":1},{"id":"83b60444d2afcf61b7132c881eca8b61021ad4cb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-02 03:33:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/f951ea3171f84111938ffb2b2a566ebb : SUCCESS in 52m 00s\n- cinder-storpool-tempest-iscsi-multipath build/986220186c6a43a1a608907825d1553f : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3ee2abe387228afc61f428a8376dc09367b01cc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 04:09:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/741398996e624fa08ec1554f39f75295\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/515f22160d5e4280b2a2e9012a6d81e8 : SUCCESS in 21m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cfd1731dc333438b89c4f8f460fdd6b8 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a253744759b246848c2c0cac3c7720fd : TIMED_OUT in 1h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dcbcb8f151dbdee0536251e2decfdbb0866c1980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 05:22:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36a9c6c36ef044a6a6591662d973f585\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4dfdbfda2b142b2887aba978356b3b9 : SUCCESS in 7m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b12543a565134e29b5ac077406a0322d : SUCCESS in 11m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebce631e883e49df92ff2b85e05bda49 : SUCCESS in 7m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5f275754efb3451d9d61d6c2f257c3b9 : SUCCESS in 14m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa497f80b3b4871b0df338e349e0ab1 : SUCCESS in 14m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/355c422fcbff4c2b98e590a4e5b716a1 : SUCCESS in 31m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/198b1020858d4f6b850367dd4109d21a : SUCCESS in 29m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/454adafc1e564c419752ea36b0f9f76d : SUCCESS in 1h 06m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53652d6218b949b5bbdb5244962c1a0d : SUCCESS in 1h 03m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45ae08ee3725452faee56e44f2bedff2 : SUCCESS in 9m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/30c699632a6e4eb187cd1800abafae7c : SUCCESS in 16m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/727a8a436e204ba297cafcd7c0396358 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d798c03ae494893ba2867ea34ae8020 : SUCCESS in 7m 53s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/4fcb4e9e875443289f2ff0d0acdbe0a9 : FAILURE in 11m 55s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0dc6d53716854c0e85bca881e731f46e : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8cf97ae667334ea299e56b554425982a : SUCCESS in 6m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6bb794900e1648389b99828856b4d316 : SUCCESS in 7m 04s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8a8a0a0a9ac74c7aacfbda864bab7921 : FAILURE in 4m 21s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/df637560a2254512bd72f4d998e10ecf : SUCCESS in 9m 19s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7d63ca0aa4344ce193df5c6ea389574e : SUCCESS in 18m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de4bf7cd6a614afbacb66980656f7b29 : FAILURE in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c0209249e4064bc3a123e1cef172ff24 : SUCCESS in 1h 42m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/351ee9c06673467cb9df13170d490fac : FAILURE in 2h 10m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df02bd2b86204dadae122aa29917bfbe : SUCCESS in 1h 31m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/99964a9cc32f4ba7bd4e0ee9610cfd1b : FAILURE in 2h 21m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3fb7020f070c41c7be4da49a0cf2f879 : SUCCESS in 1h 57m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6a854ffe1f9e4272a7a51a92b7f623cb : SUCCESS in 36m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e61bcbccbf3a48e1a2b2a89d9bd8a925 : SUCCESS in 1h 18m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/94b26f3feda04bcbb3f4f549b400873c : SUCCESS in 1h 29m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cf0019f29c3b48d682204bdb74d15fcc : SUCCESS in 45m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5bcdba07b65545cd9870045863777c19 : SUCCESS in 1h 38m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/90c64b1fe41247c4a5e1a9407cf57b1a : FAILURE in 1h 28m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/591badf0b30e45648eab84578eead576 : SUCCESS in 58m 55s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"bfb4dc3e9fed3e35fe971ca479080c9c642401ed","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-02 16:46:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aadfa9fc519c10b4d688b673d10f1a3fdc0f6d45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-09-03 10:59:21.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"0474ff603c136ca378582dbd43c19b0401012f34","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-09-03 11:31:27.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4d4ff7114b0fcb7ba68ce456a87af80a1c73ae68","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 11:36:36.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/7e373af7d88e483b883de9ff4ede0016/","accounts_in_message":[],"_revision_number":2},{"id":"538f4f170e4658688fb854d518cdd7feb570e166","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-03 11:54:37.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8b01afb034f46fce9a437c30c6ce6161b88dc843 .","accounts_in_message":[],"_revision_number":2},{"id":"d1b35f453f83b08a6719edc3a015328845b4a12a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:13:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4e65a8936f9445aa0f042cf6e4a1d78\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fefb4ce3bbfb472bba53727c6ccd747b : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/631b01ef94674059be07a40cac502f92 : SUCCESS in 22m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e64b73ea35664851869e21afcc6ea6e7 : SUCCESS in 16m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5ed9c1f16306a991fd18bc716e31b7e916d30b95","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-03 12:52:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ad5d7b7c42844b9a9eb298e3856b685e : SUCCESS in 45m 14s\n- cinder-storpool-tempest-iscsi-multipath build/45a05771a6544a898917c75f9192b2e0 : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e02d0157032348397ab9e16bf03db6e140c402ab","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 13:37:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7e373af7d88e483b883de9ff4ede0016\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0dea627566b14c7387eec95b5c91cb99 : SUCCESS in 41m 13s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a3056985786c45f78c8ccade751d9c89 : SUCCESS in 37m 14s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fe218f1f940445c4ab83fb6978e74e9c : SUCCESS in 46m 42s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c7de6194b0b4499b903ce28f579d1c15 : SUCCESS in 36m 44s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f045f5e19e7c4efab10fdcf4b63ac9db : RETRY_LIMIT in 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f14c0a0c82353e9bc4806dbd3e15c5bbbbaa258","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 13:57:39.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5394/.","accounts_in_message":[],"_revision_number":2},{"id":"533b846355174833a34d44bb73c0546e792b8095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:27:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ddaaad04cef402eaeed8171da8e0525\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e629960626674d3fb6921201f32ee4f0 : SUCCESS in 7m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24baea5977d84b30b20546c4324c7490 : SUCCESS in 11m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aec14d5670844cbdac5c5d9325f4fead : SUCCESS in 7m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81ac1b223efa4f98974ba8c32dd08b8d : TIMED_OUT in 42m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c222ad28fb4cbdbac1753a2a30f924 : SUCCESS in 15m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/d33adf6e8523470382e72380a8317192 : SUCCESS in 1h 17m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/490c58176df74446a0309195ad313fc6 : SUCCESS in 1h 06m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb529584ee2941239d5a302d5cda801e : SUCCESS in 56m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14270c0bedbf435fa939bca9c1efc607 : SUCCESS in 47m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5f1d33194ea4cb2ae37145f9d34d3ca : SUCCESS in 10m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8c26392de35a4ee5b8cd5971ba8defca : TIMED_OUT in 41m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/670a17654c1d4014ac7865c12faac3ef : SUCCESS in 5m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b86bc93a228c4ffcb45667ddfeabb7d2 : SUCCESS in 7m 28s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/958c8724aa5948d29ad44a8f389096f5 : SUCCESS in 15m 43s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ce2ace87f43841c5b43cf2be4c8c00b3 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/58f05ebee8d34c8abfa43919aa6c1ecc : SUCCESS in 6m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cfffb6fba06b4b498d47753ddeaa302b : SUCCESS in 7m 09s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/fe11068045fa42a6a1262eb89d0732fa : FAILURE in 5m 27s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/bf82e94274d34c95b78a43fcf56bfa29 : SUCCESS in 8m 51s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cd38a081f7564923bbe4d1b845aeb3dd : SUCCESS in 38m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/34934f4635a646f59fefedc6e8d559ac : FAILURE in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/822c83ed08b946cc832908e6cc3ca841 : SUCCESS in 1h 37m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/09b37e9c3b5f4cf2ae187e435497657c : FAILURE in 2h 07m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2fc6eed7ccae40ef89249a687560dc10 : SUCCESS in 1h 49m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/f1c2d7bb0be14061b0d43f9330f5fb28 : FAILURE in 2h 11m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7602128ced8d4fac967ebb94084d9efe : SUCCESS in 1h 33m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8a77e9d9c1cc4a67a95d5a1ee3572e56 : SUCCESS in 25m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c11f620842e4bb3a3e2ea9703d93752 : SUCCESS in 1h 13m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/245602decfd84314a02a9a2e171b2932 : SUCCESS in 1h 24m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7b37084809f044389602aad29eacc5ee : SUCCESS in 39m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/459f7cb8e92e4e93a03bba89c9396038 : SUCCESS in 1h 39m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f5fd93ae4dc849a7a559ca2bcecb18c2 : SUCCESS in 1h 35m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36847ff2a7d04caaafad5c74b8eb47cf : SUCCESS in 1h 05m 53s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"141ee3ce3ca30a06db53e267815dc8e6c53bb62c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-03 16:36:37.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_09_2026_16_31_06\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"d4d98f92c871aaaa693dfc7e3cf269653659109b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-03 16:38:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-98-1003398-2-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-98-1003398-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"2f03e837a3f6931f6df6e19537532264d693cf7b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 07:21:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5189/.","accounts_in_message":[],"_revision_number":1},{"id":"4141af27063e3aaab97fd1cafc957e247ba93c17","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-04 18:07:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-98-1003398-2-fc : SUCCESS in 1h 3m 8s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-98-1003398-2-iscsi : FAILURE in 1h 57m 14s","accounts_in_message":[],"_revision_number":2},{"id":"bdaf310379af3e78d379b94050c42c7415de8f54","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-09-09 14:25:51.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc3c10bd0a0d8500553fa9b339ce9432097f459f","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-18 15:35:11.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"484fc1da53bd2368aca93fb40b3f356375a194aa","revisions":{"5dd030cdd98372e890baaf1203086d8076ff6979":{"kind":"REWORK","_number":1,"created":"2026-09-02 02:33:56.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/98/1003398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/1003398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/1003398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/1003398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/1003398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/1003398/1"}}},"commit":{"parents":[{"commit":"f4c563acfd95790394f2a1d2277fa0724bd8736d","subject":"Merge \"Everpure: lock attached hosts on force detach\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4c563acfd95790394f2a1d2277fa0724bd8736d"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-08-21 03:04:58.000000000","tz":540},"committer":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-09-02 02:28:13.000000000","tz":540},"subject":"Add cluster_name to volume usage notifications","message":"Add cluster_name to volume usage notifications\n\nThe volume usage payload does not report the volume\u0027s ``cluster_name``.\nAdd it alongside the existing ``host`` field.\n\nChange-Id: I8b01afb034f46fce9a437c30c6ce6161b88dc843\nSigned-off-by: hongp \u003cinyong.hong@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5dd030cdd98372e890baaf1203086d8076ff6979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5dd030cdd98372e890baaf1203086d8076ff6979"}]},"branch":"refs/heads/master"},"484fc1da53bd2368aca93fb40b3f356375a194aa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-03 10:59:21.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/98/1003398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/1003398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/1003398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/1003398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/1003398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/1003398/2"}}},"commit":{"parents":[{"commit":"f4c563acfd95790394f2a1d2277fa0724bd8736d","subject":"Merge \"Everpure: lock attached hosts on force detach\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4c563acfd95790394f2a1d2277fa0724bd8736d"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-08-21 03:04:58.000000000","tz":540},"committer":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-09-03 10:49:56.000000000","tz":540},"subject":"Add cluster_name to volume usage notifications","message":"Add cluster_name to volume usage notifications\n\nThe volume usage payload does not report the volume\u0027s ``cluster_name``.\nAdd it alongside the existing ``host`` field.\n\nCloses-Bug: #2166337\nSigned-off-by: hongp \u003cinyong.hong@samsung.com\u003e\nChange-Id: I8b01afb034f46fce9a437c30c6ce6161b88dc843\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/484fc1da53bd2368aca93fb40b3f356375a194aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/484fc1da53bd2368aca93fb40b3f356375a194aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
