)]}'
{"id":"openstack%2Fcinder~1003531","triplet_id":"openstack%2Fcinder~master~I25fb8ae7e14e58968f003459c16e2ebb34896e51","project":"openstack/cinder","branch":"master","topic":"policies","attention_set":{},"removed_from_attention_set":{"39279":{"account":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"last_update":"2026-09-04 20:19:15.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_39279\u003e by clicking the attention icon","reason_account":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"}}},"hashtags":[],"change_id":"I25fb8ae7e14e58968f003459c16e2ebb34896e51","subject":"Fix multiple policies DocumentedRuleDefault","status":"NEW","created":"2026-09-02 13:46:07.000000000","updated":"2026-09-04 20:19:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9e2a99a5ce39ca777a6cd076f2a8cd61fc940b7e","_number":1003531,"virtual_id_number":1003531,"owner":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-02 19:40:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 14:38:50.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-02 15:18:41.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-02 15:28:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-02 15:46:37.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-02 15:50:03.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-02 19:38:05.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-02 19:40:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 19:40:56.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-02 22:01:05.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"a47f2732c4ebff8e4c06b1fcd9a32cef116b17de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"date":"2026-09-02 13:46:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"165c61ba0d08faa358ac6101fe6ab20ad2b6be9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"date":"2026-09-02 14:38:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7621c675f5f08abfc73c8f62d4591181ca1ec3a","author":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"date":"2026-09-02 14:38:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"735f0b6a481069941efa8f1585f02f0e28e8f0f1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-02 14:38:50.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I25fb8ae7e14e58968f003459c16e2ebb34896e51 .","accounts_in_message":[],"_revision_number":1},{"id":"453ab74cbfa9002634d14082bc86d9f189cf1e56","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-02 14:50:45.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I25fb8ae7e14e58968f003459c16e2ebb34896e51 .","accounts_in_message":[],"_revision_number":2},{"id":"ae180f4b198872fe59a902fe885dd8ae3e3a6b19","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-02 15:18:41.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/558552fa7d1d412cabb4e0b7e2b42855/","accounts_in_message":[],"_revision_number":2},{"id":"6b6a8f791eff24388e321bc69c241f0544473f4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:28:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e6d04f3c95f47fa8391de27e178d51d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/412dfa05ff8541219894e462421ba455 : SUCCESS in 17m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0c4c4e3b8b2846dc8fdcc5e2d1089ac6 : SUCCESS in 15m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3374710484884a4690ed8d8fdc436673 : SUCCESS in 31m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19688c7e80b7c2e15477fa6c3cc397ce3196075b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-02 15:46:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/7659c931d74f45fab5beea9f284433e6 : RETRY_LIMIT in 8m 59s\n- cinder-storpool-tempest-iscsi-multipath build/2cd4ff32680546548bad88b728a35534 : RETRY_LIMIT in 9m 50s","accounts_in_message":[],"_revision_number":2},{"id":"dec8a18dc9bf511fbce874cd60088457a2720584","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-02 15:50:03.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1147 : SUCCESS in 0h 44m 8s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003531","accounts_in_message":[],"_revision_number":2},{"id":"81fa214b4fb10b20781bedfaa31214fe0637ff98","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-02 15:57:37.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1415 : SUCCESS in 0h 43m 56s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003531","accounts_in_message":[],"_revision_number":2},{"id":"524a7ba3cdc2f0ebf842df242800bc653fb7743a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-02 16:03:38.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1054 : SUCCESS in 0h 46m 48s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003531","accounts_in_message":[],"_revision_number":2},{"id":"02b72e0df545b82f6288cfb2291c45a45099a854","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-02 16:04:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/558552fa7d1d412cabb4e0b7e2b42855\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e0893471e64a49d682beaacfa94563c6 : SUCCESS in 41m 35s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a70c52df8dae40499378cb8fe23979c9 : SUCCESS in 40m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4bd6080ba2b244c8b35f7e0565319fb9 : SUCCESS in 49m 52s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e491bc8b4aba4463871592969f8a732c : SUCCESS in 40m 01s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/411f0852ca6542cabe0976d94c0ef386 : FAILURE in 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba278b806070ad598f73891ae80a9f252051cc5e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 19:38:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5380/.","accounts_in_message":[],"_revision_number":1},{"id":"7a4386d3cc4c15943e4c82455bf28c260d357161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:40:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d1932e3aeb94427a04e18c64de77a7a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddb74e496d184ac68db5e4221752a251 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af61b826ef474765be0d9b78eb193063 : SUCCESS in 11m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64d22f08d79746899df7c8aff74abc22 : SUCCESS in 12m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b07e2d28b1584670b9a786012c18cc3a : SUCCESS in 14m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f457620fbcfa4ae684f8d77619d6445e : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/282f3c986cfb41369e69aea6c35ba568 : SUCCESS in 53m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b212eee2915a4d25b637b994dbd70678 : SUCCESS in 1h 07m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d58b8cf6c5a1401daa2317b56d43df60 : SUCCESS in 1h 27m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0647f53a3cc410eb9c801f1cdf45f02 : SUCCESS in 59m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98bbdf81016a487b954d32c4a2c52aec : SUCCESS in 6m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5e4e1c5cab8e4a4bb8293242cabca48e : SUCCESS in 20m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1bfff3fdce1e408789e1c5174493f07d : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/38fca97023604847a694d1b8001ccd88 : SUCCESS in 4m 11s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/57ab47fe49ae4827ae41f294d6b2fb7b : FAILURE in 11m 04s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/95ae4f7bb1234f57bce52ed59bfc790f : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0693805467944f35a134157706bbd139 : SUCCESS in 5m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/033a9a0d2ff6474eb02e51b75d4a7249 : SUCCESS in 5m 45s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b02a630d2f8643adb5482c0b95ad8a31 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4566553fa34146c2a92616008c5dad59 : SUCCESS in 6m 22s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f512888675f446cd8e55141b6b7dda6b : SUCCESS in 30m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e344f60d5ffa4c59a6ba12edbc67f9a8 : FAILURE in 4m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/834be6344f094ffba59e73e8bc6cf431 : SUCCESS in 1h 19m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/65011ae039d5449f8a09db6a8fe22f9b : FAILURE in 2h 01m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ec290eaf9058445ea4439f24f7e92261 : SUCCESS in 1h 50m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/4474d023238c48b1b926645570fd3035 : FAILURE in 2h 23m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/df46fc2ff5bc43ef87efb7ca9381bc27 : SUCCESS in 1h 20m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ad22b11574fa4a7e92a53ccaeddc07af : SUCCESS in 28m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b3b5cf66bbde4a53809d62cb72b0dab7 : SUCCESS in 1h 21m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91532419b7694f11bda30cf8cf492b7c : SUCCESS in 1h 16m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/18d904be140746a69a0884205fe68038 : SUCCESS in 45m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/82ff5fd707034462a79fbb2ec21083a9 : SUCCESS in 1h 08m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d5bf5dfd21e4cae86911a593294b277 : SUCCESS in 27m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"d684ca2200575e998a73f314acbeba8be48d4075","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 20:35:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5381/.","accounts_in_message":[],"_revision_number":2},{"id":"225186f2416875c2d65fcd8cde602e170dfdf6b9","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-02 22:01:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-1003531-2-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-1003531-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"83afc42c403ead4fafc7bb23a1686600930401a3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 19:15:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5203/.","accounts_in_message":[],"_revision_number":1},{"id":"82e235ada9d80f9479dd7dd41f65d2a863207ae1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 20:18:17.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5204/.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"407b0e816c7c4f30a7a91ff596aede09ae4084c1","revisions":{"70f344a06438a57c1ab3171d33875000da51b80c":{"kind":"REWORK","_number":1,"created":"2026-09-02 13:46:07.000000000","uploader":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"ref":"refs/changes/31/1003531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/1003531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/1003531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/1003531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/1003531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/1003531/1"}}},"commit":{"parents":[{"commit":"b475eafb370358e13999bbc1a8d6c7b4fe664500","subject":"Merge \"netapp: Fix clone_image() crash when ZAPI fallback is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b475eafb370358e13999bbc1a8d6c7b4fe664500"}]}],"author":{"name":"Alexandre Engloner","email":"alexandre.engloner@ovhcloud.com","date":"2026-09-02 13:45:55.000000000","tz":-240},"committer":{"name":"Alexandre Engloner","email":"alexandre.engloner@ovhcloud.com","date":"2026-09-02 13:46:06.000000000","tz":-240},"subject":"Fix multiple policies DocumentedRuleDefault","message":"Fix multiple policies DocumentedRuleDefault\n\nAdds operations or fixes paths and/or methods of existing DocumentedRuleDefault\n\nThis is used only for documentation purposes in\noslopolicy-policy-generator generated policy files and it does not\naffect policy enforcement. The objective is to align the documentation\nwith the API.\n\nChange-Id: I25fb8ae7e14e58968f003459c16e2ebb34896e51\nSigned-off-by: Alexandre Engloner \u003calexandre.engloner@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70f344a06438a57c1ab3171d33875000da51b80c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70f344a06438a57c1ab3171d33875000da51b80c"}]},"branch":"refs/heads/master"},"407b0e816c7c4f30a7a91ff596aede09ae4084c1":{"kind":"REWORK","_number":2,"created":"2026-09-02 14:38:10.000000000","uploader":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"ref":"refs/changes/31/1003531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/1003531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/1003531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/1003531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/1003531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/1003531/2"}}},"commit":{"parents":[{"commit":"b475eafb370358e13999bbc1a8d6c7b4fe664500","subject":"Merge \"netapp: Fix clone_image() crash when ZAPI fallback is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b475eafb370358e13999bbc1a8d6c7b4fe664500"}]}],"author":{"name":"Alexandre Engloner","email":"alexandre.engloner@ovhcloud.com","date":"2026-09-02 14:38:03.000000000","tz":-240},"committer":{"name":"Alexandre Engloner","email":"alexandre.engloner@ovhcloud.com","date":"2026-09-02 14:38:03.000000000","tz":-240},"subject":"Fix multiple policies DocumentedRuleDefault","message":"Fix multiple policies DocumentedRuleDefault\n\nAdds operations or fixes paths and/or methods of existing DocumentedRuleDefault\n\nThis is used only for documentation purposes in\noslopolicy-policy-generator generated policy files and it does not\naffect policy enforcement. The objective is to align the documentation\nwith the API.\n\nChange-Id: I25fb8ae7e14e58968f003459c16e2ebb34896e51\nSigned-off-by: Alexandre Engloner \u003calexandre.engloner@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/407b0e816c7c4f30a7a91ff596aede09ae4084c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/407b0e816c7c4f30a7a91ff596aede09ae4084c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
