)]}'
{"id":"openstack%2Fcinder~1003765","triplet_id":"openstack%2Fcinder~master~Iba119184cc5ff0ea25360f87e5ec9430d83b53fd","project":"openstack/cinder","branch":"master","attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-09-09 10:50:10.000000000","reason":"\u003cGERRIT_ACCOUNT_38614\u003e replied on the change","reason_account":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-09 10:50:10.000000000","reason":"\u003cGERRIT_ACCOUNT_38614\u003e replied on the change","reason_account":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"}},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-09-07 18:33:49.000000000","reason":"Change was marked ready for review"},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2026-09-07 18:33:49.000000000","reason":"Change was marked ready for review"},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-09-07 18:33:49.000000000","reason":"Change was marked ready for review"},"38373":{"account":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"last_update":"2026-09-07 18:33:49.000000000","reason":"Change was marked ready for review"},"38614":{"account":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"last_update":"2026-09-09 11:01:10.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-07 18:33:49.000000000","reason":"Change was marked ready for review"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-09-09 10:50:10.000000000","reason":"\u003cGERRIT_ACCOUNT_38614\u003e replied on the change","reason_account":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"}},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-09-07 18:33:49.000000000","reason":"Change was marked ready for review"},"38479":{"account":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"last_update":"2026-09-07 18:33:49.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-09-09 14:14:05.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}}},"hashtags":[],"change_id":"Iba119184cc5ff0ea25360f87e5ec9430d83b53fd","subject":"NetApp ONTAP: validate replication configuration at init","status":"NEW","created":"2026-09-03 13:05:12.000000000","updated":"2026-09-09 14:14:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"62c76ef56cbffc2a40f764c41f7e6e46a6a4e81e","_number":1003765,"virtual_id_number":1003765,"owner":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 12:17:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"value":1,"date":"2026-09-09 14:14:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 13:42:24.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-03 13:54:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 13:59:27.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 14:13:25.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 15:32:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 16:56:53.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-03 17:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-06 14:00:12.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-09-06 14:00:12.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-09-06 14:00:12.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-09-06 14:00:12.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-09-06 14:00:12.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-09-06 14:00:12.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-09-06 14:00:12.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-09-06 14:00:12.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-09-06 14:24:04.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-06 23:08:03.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-09 10:50:10.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-09-09 10:50:10.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-09-09 10:50:10.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-09-09 11:12:41.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"7dcf0d306cdbf2ce97eaa916b113efdd6d5b649e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-03 13:05:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0fc0cf847f751ea63f0b7af471482f16d0834d4","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-03 13:05:45.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"c9795e8e8b5f54a9e8497694b076fa21a42f6ba9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 13:42:24.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/44e4eaef3a834137aec41742c3b4a122/","accounts_in_message":[],"_revision_number":1},{"id":"5cc6b6ca1cde225d38629de1851d655ac6bc8959","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:54:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fb02c49ff4c4b51a1baba2912c04201\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e5905ff1af874fc08766c5e3e8c3747d : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4e1b5e2204c413691efe39eaefa2d90 : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fdfa972077e54a1fa994093fb11ed87c : SUCCESS in 29m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfe7289a6aad8daa141d4017ca8901f30c7c5eb7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-03 13:59:27.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":1},{"id":"b156fd58c109ab971a0a9d501ee0be318ee659d7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 14:13:25.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1069 : SUCCESS in 0h 45m 20s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":1},{"id":"9862da53a7e115cfe4324bfeacce58be313d32bd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 14:14:08.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1162 : SUCCESS in 0h 45m 16s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":1},{"id":"1eb9baa8442643c7a161cfaefe179b223305b605","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 14:20:28.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1430 : SUCCESS in 0h 47m 48s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":1},{"id":"2a3a97763b493bc85f4534cc5bbeabacfc890a72","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-03 14:29:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42811aabde39d8171b3c9739cc2c092e15e23be7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:01:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a82a2834d5742388de66486193f48f9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/393d7b9b95a34730bfe42aec0a3bb6fd : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9bb17437fa6642f9beed9395f5171715 : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7b16f2a209724b6fa4853dc15dc6662c : SUCCESS in 27m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81c26d585c431da7b0d437fee3a272e23fde42ee","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 15:06:00.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/8be39d213b474ea38d584c4edefe38b1/","accounts_in_message":[],"_revision_number":2},{"id":"eac0cd7a33080b30aa943ca0294f7e0e149e92f7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 15:16:10.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1070 : SUCCESS in 0h 46m 27s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":2},{"id":"8b5a1f327d25c613be55a9368ae90aa714cda397","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 15:17:44.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1431 : SUCCESS in 0h 48m 4s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":2},{"id":"7b070fddb6f1f458212fbb38b6efd52192924b65","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 15:18:55.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1163 : SUCCESS in 0h 49m 11s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":2},{"id":"e6ab41ff39ce95297aa4fa3e038828e6c3f9d592","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-03 15:24:37.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":2},{"id":"c8c279e63a626896626ec92b379bbd3aacbaf125","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-03 15:32:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/d1a3f5230f95442db08046f389789ad0 : SUCCESS in 43m 38s\n- cinder-storpool-tempest-iscsi-multipath build/c47ced2d5a444dbbb458e5558e7b6fa0 : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":2},{"id":"eb1eb26c3cd7403ac65272efff1853d5151d0eda","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 15:48:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8be39d213b474ea38d584c4edefe38b1\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bd01157fad4b4912b1b2b4c9c0746acf : SUCCESS in 44m 40s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/925a01bc490a4d25a3237773272b48b0 : SUCCESS in 38m 56s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1cf0c5e2971c4db693b6ddd8eddb3405 : SUCCESS in 49m 32s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e3de3348328a4ce983b7a3aa7f66c15a : SUCCESS in 36m 08s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cd76d9cd75f145d7bc7abb3464a05bf6 : FAILURE in 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"abf049f888520aa50c66fda32dd7cfa6b57b0193","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 16:56:53.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5397/.","accounts_in_message":[],"_revision_number":1},{"id":"08ac0320a84f38d776891d6cbc6a82632e86e366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 17:03:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a140079bc7c84882898a2f67dd826743\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4f4e59b1c1497d8261104d79eb99ef : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e72922f8575f4750a3c4b26d292a4a5b : SUCCESS in 8m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98d9924b546e47fab82cae5e3915be66 : SUCCESS in 11m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/97939c70863b426aafa8fe0014a5cf5e : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82cf7a558d3f474cbdc6fbd8cd418917 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/38c659cb939744818dcd0c32169c695c : SUCCESS in 57m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/28bbfac5bc7a42aa86af905d07da883d : SUCCESS in 1h 14m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0561bb7a0b60477bb142d1c78d9b15cd : SUCCESS in 54m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93cfcaa6a92f4e64a0bc8ebdd224dcee : SUCCESS in 52m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89ec5be4b4724c7ba17a801b6a8d1d98 : SUCCESS in 9m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/68399423cd3e40389ac18afe51d2d175 : TIMED_OUT in 41m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fae14096ff1e4cc28be1b5fc1ec7bd95 : SUCCESS in 4m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/386d4b7f84ba4d9eb3498804da750ea4 : SUCCESS in 6m 33s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/a2e081e8846846b1bff2b6ec93c818b9 : SUCCESS in 11m 07s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2022a9806f48454cbc235c9de319a139 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/275fbaaa3dca4080b4aee54926775d13 : SUCCESS in 6m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/172aa1969963418ca60ba0c1826674d5 : SUCCESS in 6m 26s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3c13f46f38d54787b696c86954d381f0 : FAILURE in 6m 09s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9e79c433ad1c49f19e47dc4479b4c143 : SUCCESS in 9m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a6a19025dc8a4ef38d3cdaee99ac0883 : SUCCESS in 36m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c17e18a8c29f42c696b090ca8d40a68d : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9498f3ee4db04cfc9ba7f322507835a0 : SUCCESS in 1h 41m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2719964d138447b299078e03ac13fc57 : SUCCESS in 1h 04m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5954d7f96da84efa8d4d8cdfe9aad038 : SUCCESS in 2h 22m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/52e68a1bb3d04b6e93bf3bcb68ffb569 : FAILURE in 2h 08m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8e425d92596646c4abb2a137e63814de : SUCCESS in 1h 55m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/04070ce5bab84031a1dea3aecfe7e378 : SUCCESS in 40m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2da44882987c48328cf2f487b4446044 : SUCCESS in 1h 21m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/17094cb786124ba2b58d828ae363af21 : SUCCESS in 1h 20m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/19a759d7342d430bbd0b72fc9d366041 : SUCCESS in 42m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16c7712f4e3647f3b89d527f33c8b9a9 : FAILURE in 55m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0173a825b62c49b5af4f05e5f9980df1 : SUCCESS in 41m 19s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"fe2f6bcac4cb2aa8cf00401aabcdaa97313bed6d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 17:53:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5398/.","accounts_in_message":[],"_revision_number":2},{"id":"9a4951418ac6b61936e2993dc4eea09aab0e1d3f","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-06 14:04:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"490adbceb2564ddb534832c864a1d47bf2bdd530","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-06 14:24:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-2-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"05a0d43e935e80d4e0cda078ccf18eb5bb1d85aa","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-06 15:53:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-2-fc : SUCCESS in 54m 31s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"464393e0cd8e140521e4321a3a37d8ca24702e3d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-06 20:42:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b4b3b18f9346c6eb2c8abdd4a942721f97af247a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-06 21:00:36.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/9219e2bcb41844e2a003df72d8e6a146/","accounts_in_message":[],"_revision_number":3},{"id":"39ff67623039dd390d733d26afc24e61be84c420","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-06 21:01:53.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/87a3ff200b184e84838c66f7ee84d37c : RETRY_LIMIT in 17m 04s\n- cinder-storpool-tempest-iscsi-multipath build/5206530a8cc94515b12e155e0542ede1 : RETRY_LIMIT in 17m 59s","accounts_in_message":[],"_revision_number":3},{"id":"7134b7578bdd69a5e6e2e83b853babbbdc6c3ef3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 21:09:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74518ec61b6f49a7bcab2692b8e3c321\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1f9ddcf218d14c6bbc0ce73ff2af4789 : SUCCESS in 17m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9f3b16ea846b499681d4b6744ecdc503 : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ecc45d455f5b41838673c0d668a6f64f : SUCCESS in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ce1e3e310ce3073450a10c464be54e4508b6d99","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-06 21:37:59.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":3},{"id":"21fed7f74b2aa9dd7a2a595b6a5d1901fe177a9c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-06 21:57:13.000000000","message":"Patch Set 3:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 31s\n- FC       : SUCCESS in 0h 48m 49s\n- NVMe/TCP : SUCCESS in 0h 48m 36s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":3},{"id":"2bc70aad8a1f15ee04cf282145c81c905fbf51f7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-06 22:19:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5256/.","accounts_in_message":[],"_revision_number":3},{"id":"e2ab12c90fa8f21dd017006274b053dd105a05a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-06 22:23:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5433/.","accounts_in_message":[],"_revision_number":3},{"id":"6a5b23d38d54a3a8de6330ee724f88835f45b79e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 23:07:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9415233bdb34d4b8c8da01e771f24cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a2dc0269e234f13bebbbc854ee26d5d : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69782732d0b14b6c99cc98af324fb197 : SUCCESS in 12m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8939872eb8442f29e2dd7197a251a54 : SUCCESS in 12m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a76f512fa1f04c8c93be574dc49ac51e : SUCCESS in 15m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab2f1e280044c1084cb908dce18da6f : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/e64e52ae422e4922ac8b4a3251ebec5f : SUCCESS in 56m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9675268b17c14c53b2ef8dba1d9b64b6 : SUCCESS in 48m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44587c3f352e434090bea98d92d24be2 : SUCCESS in 1h 26m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/637fc346b8cd4368b1899af10701520d : SUCCESS in 1h 00m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ee72cb92d3749c895d70c3246cd5a46 : SUCCESS in 8m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84cd5ce5a91f4d62adad1f620acf84a9 : FAILURE in 26m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fc8bb5ece967489bb9310d75b15836e9 : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6459e80d45fd4b5097a3e5d29655dfb8 : SUCCESS in 6m 09s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/7339bb24948f4a7ebc486afcb89cc32e : SUCCESS in 12m 23s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/39d163134dc74420bcf41a844167d39f : FAILURE in 8m 07s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bf723ba984634531929b742fc85e7a85 : SUCCESS in 5m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0bbed3214e4a4f1cbc9fdb6c2415474e : SUCCESS in 6m 32s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8c6b6f377c344fa0a563b3069e9c7f51 : FAILURE in 7m 06s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/118d03cfd9cc4974866bbb874b6b4264 : SUCCESS in 7m 30s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc6086f6583748b892407b4972bd809d : SUCCESS in 36m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d960799095e740e8973d0cf3f26f36a8 : FAILURE in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/137c384a114142f9891be1728bba2171 : SUCCESS in 1h 43m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/537cff0a5de744909e2116212d970e8b : FAILURE in 2h 13m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/38ad59c0e4cc486f9a832c794c1f268c : SUCCESS in 1h 29m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/3269f42b948942c0af2d3a049aec029a : FAILURE in 2h 23m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f9904f2484e54bb8b9ad9a26c19eaa7c : SUCCESS in 1h 25m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b60484d4b2244aac99969de61cf254dc : SUCCESS in 33m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/64f03e8fad9c49488a97c5cb5d20d49a : SUCCESS in 1h 10m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8f7c87c86ec8484291cccd063328f5df : SUCCESS in 1h 42m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/97911709188342e498784c22e474c7b0 : SUCCESS in 30m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/10b919faab204868b535dbe141016363 : SUCCESS in 1h 26m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4e37333230c4257aef0ea79b912054a : SUCCESS in 54m 33s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"86a27823f8cb68955a47cc6d1f01b0142d0c1f91","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-06 23:17:09.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/06_09_2026_23_11_32\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"d9b00409015abe8ae29d25e1a12bfcccafb07f3b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-06 23:18:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-3-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"5d1c805809c35a430b35a5d8210331ca60a20115","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-06 23:25:11.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9219e2bcb41844e2a003df72d8e6a146\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b1c663d688884cff8df2f53734aa52df : SUCCESS in 46m 21s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c195310b3c3f4db8a3b977ebb8b4a92f : SUCCESS in 37m 09s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/25bc19ae6702496faf11a6f1772fc131 : FAILURE in 2h 42m 20s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ba649e2aea8245e988f8dd8432cea579 : SUCCESS in 36m 16s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ab986c0cd8c14ce4a1fc63d541002d66 : FAILURE in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5f778752ca3fe36d5e389fdbd863b2dd277f47d4","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-07 00:46:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-3-fc : SUCCESS in 53m 47s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2f896679d0d09cd59b49cddad32878bc87502c80","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 04:40:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aace97ff43af5fca692c1964115c9dbf8b7bc158","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 05:34:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"abb9a3d2bb6b2ef83ceb70dcf5a29f5a0c4e2748","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 05:36:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1796a6dd85fa95657f8a967574ebb731f3f235f4","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 05:37:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"58e27d98771a3c4d619a9bfef8152f3a8f1b92c4","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-07 05:48:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-5-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-5-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"4b5e4bfc057527e65625286dab01e230ccdab405","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 05:52:41.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/992191547f444e9a99151021fa7b84d1/","accounts_in_message":[],"_revision_number":5},{"id":"17615fa609ff909dd7e95ae2b12e4e4b6c204158","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 05:56:04.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest build/58c57ba1c6834d7e885e8aa34321b7f5 : RETRY_LIMIT in 16m 48s\n- cinder-storpool-tempest-iscsi-multipath build/8b7e67e8a30748978774c932cd2777cf : RETRY_LIMIT in 17m 54s","accounts_in_message":[],"_revision_number":5},{"id":"bfdf7c7c324dc8b13a8c37c8fd8fa85ca7bfd352","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 06:04:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b96e553ac88434f94b104a8b992770e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a478ac4387a48e0b1e740b6737d355f : SUCCESS in 16m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f2ec3c1aaa8c4e828a3aee717b32fe24 : SUCCESS in 17m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ec0d821780f242909756b41464a8ea2a : SUCCESS in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6c47171cc0b277ae61d3b14ad30c011adf84e535","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 06:28:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/992191547f444e9a99151021fa7b84d1\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/35e67d89268347ae9e8b6eab4fbbc01a : SUCCESS in 41m 11s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a3c56207341c4c21a19aa5a95a858151 : SUCCESS in 33m 42s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8e305d5324dd40aca53ce1679c81aebb : SUCCESS in 50m 40s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/321b6bdb3176433a844a7975880e477d : SUCCESS in 36m 02s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/38d67502a88e4111a79155b3a72f0d22 : FAILURE in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"abcbd95159db64385d107e917703e01734751f82","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 06:28:42.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":4},{"id":"85c3de9f087525dab88a98653259735eecd67049","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 06:30:40.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":5},{"id":"dbf0224cddb00250603040faa7802d0782cc763f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-07 06:33:39.000000000","message":"Patch Set 5:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 49m 19s\n- FC       : SUCCESS in 0h 49m 57s\n- NVMe/TCP : SUCCESS in 0h 48m 27s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":5},{"id":"1e679151b5651e425cba522a4ce2f91ae86d0e9e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 06:34:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5264/.","accounts_in_message":[],"_revision_number":4},{"id":"cf604d6e92af524f051d18fed968156393b50ec9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 06:56:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5441/.","accounts_in_message":[],"_revision_number":4},{"id":"01486e1bf9522946c3747720b6ade9cd14d6ea3b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-07 07:09:17.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-5-fc : SUCCESS in 46m 56s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-5-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"b874623684f51138ca4af2c03914ff0d2f0d5d07","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 07:35:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5265/.","accounts_in_message":[],"_revision_number":5},{"id":"32fa44a9bae97ddc02f7686443b48c45789d02a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 07:43:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69c8ef62a6b64ece854d86430cafd429\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f597b86429414f3c879d1ee95957feef : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb7fb2b66bdf49bc907a93aaa6ca6fdc : SUCCESS in 8m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb72547a65f14d67baffd1a3be61f7d2 : SUCCESS in 9m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b8b84bda7c3b4942ad8befa33166576a : SUCCESS in 14m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aadcc57b95e40f1a7fdd99274f3d7f1 : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/3d7e2864c5354ad99f6871798f92f4c2 : SUCCESS in 33m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84476396475e4b58aa8c780505f26a3f : SUCCESS in 1h 04m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35dad0ade3e34a29a9ebb67e9142f832 : SUCCESS in 1h 04m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41f15df9118a4a89958b5fcf0f273417 : SUCCESS in 44m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ab830d9aa14408397f4560a1015ce69 : SUCCESS in 9m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2a116962f08146869cf1f6b537627791 : TIMED_OUT in 41m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/441f98e509e64c09aa102510195df26b : SUCCESS in 7m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/71bdf58fb05241069a1cec06e16b5db6 : SUCCESS in 6m 12s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8f765dc9f3ce4a308355d4e6fdcd16f1 : FAILURE in 12m 20s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/1706dd1c3f2b452c8975ee6e528c78e8 : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bbd8870357d343afb92ea172f87c9701 : SUCCESS in 8m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/32f7e47e096d4d2fae19f9a092d682b9 : SUCCESS in 5m 55s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/47b3f5fc7df94147b3dca35bc18f2dee : FAILURE in 6m 07s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5f81f7c2eba74116885a9437a6dc4b1d : SUCCESS in 7m 57s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5f2da37680f24aed92581a323c5a8a07 : SUCCESS in 25m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/281c490beba94dea8203e68f05c7d479 : FAILURE in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/48a6998e1b5f4fecb110a3bdc64accc7 : SUCCESS in 1h 48m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/34c396b69ed84eb2a0c90234fffce7c2 : FAILURE in 1h 57m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/514ab6f3d6d64db5a7c2b77259afda5c : SUCCESS in 1h 28m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/49417fa1ff084b22b575dc0bb2c409a7 : FAILURE in 1h 29m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e72110bfddb04ff89ab4afd5a50b4f91 : SUCCESS in 1h 48m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8b45c2111c694cfc907e9b29e7f8ada6 : SUCCESS in 35m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/24e9dfd965194290a79aa9bcfad58bef : SUCCESS in 1h 18m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/361d015d4795423bbf90256cd2302679 : SUCCESS in 1h 12m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f21c4387d3d7446685c558a4ea6a52e4 : SUCCESS in 43m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d2ec8d0f540a42a0a59b905556ea4b6c : SUCCESS in 45m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ed396df6f844efb88b6b22d34cacc37 : SUCCESS in 28m 03s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"4fcbc2dd01863c782acedebb00aa7ac56bd2b7b5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-07 07:52:31.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_09_2026_07_46_50\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"b78a4c83b37888aecfdb39910d716bae55b686d1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 07:53:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5442/.","accounts_in_message":[],"_revision_number":5},{"id":"9c554268152c0b3da806be668f45bab576cd5309","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 10:04:50.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"aad950387d9045e86c87c71c4261352f4aaa18a9","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 10:06:39.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e6c9fa66ffc3665d1e00e129c33841e91044578e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 10:09:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8097ffc831cf5f3fb555124497331161f64e785e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 10:30:01.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest build/65156c4089174c66944138f45ef90539 : RETRY_LIMIT in 17m 01s\n- cinder-storpool-tempest-iscsi-multipath build/f47983f369bd4970b0ba00cb5697c0c1 : RETRY_LIMIT in 18m 06s","accounts_in_message":[],"_revision_number":7},{"id":"4bd5b12611e4d69a76627c9e0434155cf5edc6a9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 10:43:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"794229f92d8c1408f76e7ba0561a57e5eb899f21","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 11:02:24.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":6},{"id":"6a8e62cc6eb9ea5bff05f7109770ef7f4f872b5f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 11:02:39.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":7},{"id":"46260b6b7c899ab41c0fd860c38bb634ae03204c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 11:07:07.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest build/699c0602e4b5458ea3177b9b42505a0e : RETRY_LIMIT in 16m 12s\n- cinder-storpool-tempest-iscsi-multipath build/dba501f3812546d2a2c3581e400b6e54 : RETRY_LIMIT in 17m 18s","accounts_in_message":[],"_revision_number":8},{"id":"9af1f4c484a654c34843bd4c93591e006362ccc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:09:58.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77b6be30a80343a9a2dc4c1903b176f2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d17a73900c7048c7ac0e730179512cdc : SUCCESS in 17m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/29f07d30feb449f8b8791a8c7aa7053d : SUCCESS in 17m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/95614f5c45e541eea673328d6a6e3e9a : SUCCESS in 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"67abcb7171a967aa38a2bf3f0275650f8bcdcd7c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 11:10:08.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fc84016f56f53ae4d616a73e3d5417d1cc980780","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:28:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00fa74dc2b364c3c81446626c59919d7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/42717dd22a1b4a61a9acaaa9fc9480a6 : SUCCESS in 15m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fa1095d086b04b12a0e3b056f4132b67 : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/94e11655151b4428b4b8f6697e67d1f9 : SUCCESS in 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9bbf94b68b2b0b8db93514aa59698e11c8ea44f9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 11:29:44.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest build/cdcdbb8a72714f45942677258355998b : RETRY_LIMIT in 17m 38s\n- cinder-storpool-tempest-iscsi-multipath build/ff624fc823654c748604eb2db69e9d53 : RETRY_LIMIT in 18m 07s","accounts_in_message":[],"_revision_number":9},{"id":"a697612a2dd1f3f3035e976673bdb19311f00591","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 11:37:04.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":8},{"id":"3ecece8ad693770e6cda19e294f81952637502c1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 11:57:55.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/0b6e003ee8b3422f890373d7d13d1f96/","accounts_in_message":[],"_revision_number":9},{"id":"d5f848846cbc86c1e69a7ac4cdd475e80e82ff03","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 12:06:15.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":9},{"id":"8e7b98e810efc09de5e15e07ea369e7e814aa40c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 12:31:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5270/.","accounts_in_message":[],"_revision_number":6},{"id":"a924983f01791e103ed6f47b795e310bd2da28c1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-07 12:38:57.000000000","message":"Patch Set 9:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 2m 15s\n- FC       : FAILURE in 0h 2m 11s\n- NVMe/TCP : FAILURE in 0h 51m 9s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":9},{"id":"6e4dc8a5edc709a3d4df36b511201d3b40c3f55a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 13:00:52.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0b6e003ee8b3422f890373d7d13d1f96\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/00ef2dd21e234073ac87428f2889e36e : SUCCESS in 47m 07s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/532438e435e04209b1dfc48f5d0c3fc0 : SUCCESS in 47m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5379634d8847408e96b91a86f862a33a : SUCCESS in 44m 38s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/daea713b83d74123a54d054f1ee6052f : SUCCESS in 38m 17s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0e9a3380ac6e4f36bd5c7631a6aec49c : FAILURE in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6d7b27240ea7f4f16b329ca2890978edda714b7f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 13:37:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5271/.","accounts_in_message":[],"_revision_number":7},{"id":"bd2f77050ed895987f7cde1564a50096b00ec12a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 13:37:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5447/.","accounts_in_message":[],"_revision_number":6},{"id":"25165cf7317419e2d27d3bf76610ef155fda6380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:45:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e93d0f2f26904884a6f395fe079929aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dba591faf89743beb37e41c0bfb6f5ee : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8a9a6aa115c4499be370a83aaef8dd4 : SUCCESS in 11m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7da3425ec7a044f3a12596b5a3d2f947 : SUCCESS in 11m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b23b99a472764ffcb0af9bf16c5b33a8 : TIMED_OUT in 41m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37315e5d13ac465b9d8ad9a90af1f2b3 : SUCCESS in 15m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/edc4c0a0c6c542308db5b7d4d10e24b5 : SUCCESS in 1h 12m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/81ac7c1070784f3f8ea9776eda61122e : SUCCESS in 1h 04m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e709f94de95c4454ac289db4cd7a10fa : SUCCESS in 1h 18m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bba343fc5f144ec0a27d7f3a9ecfd7b0 : SUCCESS in 47m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99946c7f26e9441d935b81a1aeaf303f : SUCCESS in 10m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5ea6a3cb9f4743f69fa05531c6fc045f : TIMED_OUT in 41m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b6e2f0be6b05443492e09ca286d9cd75 : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4dc9c2e31d824e308b1f394fc858cc80 : SUCCESS in 7m 32s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/10c5b8e38d5e478aaa65aab8e1fb2ebf : FAILURE in 11m 23s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/06a4d114c2714321b32067dc9dac540a : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7bfb3834e49544c6870a5bcca3e5fce4 : SUCCESS in 6m 29s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1ff809626e5b488094f5ab5ab5afa173 : SUCCESS in 5m 57s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5c845090ea364ea1a8302b642bb73bf4 : FAILURE in 7m 47s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/38f256b7877547a4a3b0ea3beb55e54f : SUCCESS in 8m 59s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ebd884b35bbc492386d46bb27b619325 : SUCCESS in 34m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1c197e4617b941fe9efe738ffdc8d8ea : FAILURE in 7m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c9bb8710b6714ef09389a4bdceeae7d9 : SUCCESS in 1h 31m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d32364b496eb4fa6ba84663b51f8ea3e : FAILURE in 1h 54m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d1637226fcdb4ae59e39724f6b288c22 : SUCCESS in 1h 38m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/c2e74ef9f2344aaf877e82f29828b7ac : FAILURE in 1h 41m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6c8ab2bf2c4a44dc99f48b945cb6c4ca : SUCCESS in 1h 50m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8523661f7c224c78bf7c4c295d533e65 : SUCCESS in 41m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/49a33cc40e054b30aab00d4cf49380e9 : SUCCESS in 1h 22m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a3a853ef0c5a4fbfa522f1e988374102 : SUCCESS in 1h 31m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cce6bc4c49b44ec79e99eca770932fbb : SUCCESS in 46m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b60fa25425a453eb98b63576362ac1e : SUCCESS in 1h 32m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/beb2265c9cd04e79b8691aa76f4a1383 : SUCCESS in 1h 01m 06s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"ebd47e41476f3618674950d02c007d79f58a91f2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-07 13:55:17.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-9-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-9-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"288eda8e9f4dcc62f6f2f514e2ff4c5fb02f0a3a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-07 13:56:05.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_09_2026_13_50_03\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"6fed1ca2f0cff7d9d249ba0e8b9578032a44797d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-07 13:56:35.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-9-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-9-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"c563940c3b42e00a62788a8f6c3e6a8042e33b6e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 14:50:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5448/.","accounts_in_message":[],"_revision_number":7},{"id":"e3f78c2ebab0a4f81a95846f4e4cab5574037a34","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 15:33:02.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5273/.","accounts_in_message":[],"_revision_number":8},{"id":"57fa53be4c22a8dc756b06e2688e6b209f3a7c71","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 16:34:48.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5274/.","accounts_in_message":[],"_revision_number":9},{"id":"cdf17047b54b4f244b3544ef0434b04d00a4d5f4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 16:42:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5450/.","accounts_in_message":[],"_revision_number":8},{"id":"f7d72a18d560cbd1f6f9ccd1298943ca9ef2b622","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 17:22:22.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"b8677e20b9671d055950e938e4d2c33f60d7789f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 17:41:58.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest build/46f63606990e44e7b77ae6a0348a02e6 : RETRY_LIMIT in 17m 12s\n- cinder-storpool-tempest-iscsi-multipath build/deeade0134c04cd697c9917f76117318 : RETRY_LIMIT in 18m 07s","accounts_in_message":[],"_revision_number":10},{"id":"04f2641f79e423f7c1baa150eefb05b70ab723fd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 17:45:16.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9781618dda644936f73f10d2b446e5616da50f7a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 18:04:56.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest build/b5a53f5a5af14337a62f4d735ab2d623 : RETRY_LIMIT in 16m 53s\n- cinder-storpool-tempest-iscsi-multipath build/ab01ed5cc34a4033bf5a0b90eb3852ba : RETRY_LIMIT in 18m 05s","accounts_in_message":[],"_revision_number":11},{"id":"e34a109192ddb0b53163f4a0230730b12ae53909","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 18:10:38.000000000","message":"Patch Set 11:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/84b761d3480d4bbd93967d3c9b757147/","accounts_in_message":[],"_revision_number":11},{"id":"98783983a8e290ea7975dec87dbe9ef9bbec4629","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 18:11:20.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e120094e91044b9999fd31f88e55da6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/69f575bcba6c4ff79ebd8c31ec460328 : SUCCESS in 18m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c162eb81287a499e91808f75cb934df6 : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/41aaf5f1c7734ab1aadeb11c42472653 : SUCCESS in 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3027c032d4cca2a852f25747d17ed002967750c1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 18:17:10.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":10},{"id":"4134a85e49a7bec9eacc9bb2b5516c3495033a1a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 18:33:12.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"95c83465884c13a5838b8e159ddb7300a4da71e4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-07 18:33:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":12},{"id":"b303cb4ad31b211cd7cd20b59da8f72366e2a9df","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 18:38:09.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":11},{"id":"7b6e07f74347508cdee27f0e5853e781d4536ec8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 18:53:06.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest build/9241da6fe21a40eb966cd0b8bd57f941 : RETRY_LIMIT in 17m 23s\n- cinder-storpool-tempest-iscsi-multipath build/60b7d66e40af40bfb343ee85725a32a9 : RETRY_LIMIT in 18m 22s","accounts_in_message":[],"_revision_number":12},{"id":"0ca5fe436ac66cacfd1b59bbdc336b82816f29ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 18:58:49.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df6e081f63e749249b184e9df560b298\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/600b7462c3094bae811d5bd7a181b85d : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c2f1c79639c149ef86ec25e3596a697d : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/855ed70dd9774bc68c4d33f0790dfe74 : SUCCESS in 23m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"01805b9d8f67d7be27ec983dd806a24b7d82fcc3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 19:07:03.000000000","message":"Patch Set 12:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/0051d79d7da54379b4ceee79d2e8fab5/","accounts_in_message":[],"_revision_number":12},{"id":"53fb40cb1a3bb04cab690d1078ffbb311d0d669a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 19:25:35.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":12},{"id":"2f2070191b10bb6769eede6473fa62fe0a178f4e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-07 19:25:45.000000000","message":"Patch Set 12:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 4 min 29 sec\n- FC       : FAILURE in 2 min 14 sec\n- NVMe/TCP : SUCCESS in 0h 43m 43s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":12},{"id":"d7c20f5d756b632ea22376653f252a98c2d489e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 20:38:12.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/185e98ebc46c409195a1a29bbbac2891\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df6a278637a4f3b939d655150db57cf : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/00e1efba8a3b4f6ea15e7e2895ae76a4 : SUCCESS in 10m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd6186e88e1c4eadbf443a0dff72ee08 : SUCCESS in 12m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/df7574c1eb75456586b6181d3700867a : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a280d4fe6774479c9f1f48488f03cf84 : SUCCESS in 13m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/b62dc0697a904314b28139b61f2a6582 : SUCCESS in 42m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b0f2f6c40eb4f248c3c6897bd628f01 : SUCCESS in 34m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e07ce64839ef43d89fc9721b4847baf9 : SUCCESS in 1h 25m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4a6ad2282384883b0d18a95d77f7291 : SUCCESS in 1h 01m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/537a158cabac424390d947ee6edb5ae0 : SUCCESS in 10m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8caa9069ac5f4ea3b281083c99d455a1 : SUCCESS in 16m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4c5f2d436b004ef49166bcde9f989b79 : SUCCESS in 6m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/475399d2aa0d40779d64eb64ba974c3b : SUCCESS in 3m 48s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/b07a3aecce254762855defa38309f1f7 : SUCCESS in 7m 44s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d4eb22ec3451411d82315a0090a35f18 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e889e7ddb64b41b5833bc2c1dc16f6ee : SUCCESS in 7m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/04acf8d999dd4e9fbfbff3bd5e651764 : SUCCESS in 6m 32s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/cce5d22d2a214d65b114ae9670b9126e : FAILURE in 6m 07s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/27b1b03b04eb4f95a255c179cba02227 : SUCCESS in 7m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a498fcf9ad2c49ee92c810f30e9039b6 : SUCCESS in 34m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/155953fffabd4b92b29b909a555ce8b0 : FAILURE in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/337102f9c77f4afb94c127a23445a344 : SUCCESS in 1h 29m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3b94681466c34dc4a64057c4e4db563f : SUCCESS in 1h 41m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/874af8a155f8430ea0a01ac94f568a23 : SUCCESS in 1h 24m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/8616e3cde46242de856f76c5df731df8 : FAILURE in 1h 59m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8c664a2a912941ec835d1a068e5cda3a : SUCCESS in 1h 39m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8d354c4bde79435da242dc30607edc73 : SUCCESS in 19m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0c2fcf35e71c4be98e8ded8bc586fd36 : SUCCESS in 41m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ea9902d56ec14e8d8187f0cda19319bc : SUCCESS in 1h 45m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e3ce61ba731b4cee99d3eb1c72a9fc95 : SUCCESS in 48m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0273ebc97b554e21bead4c6c67d83899 : FAILURE in 1h 37m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfd4ee0fd11f4405a75a93747cec74ab : SUCCESS in 31m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"50226134279e24708e4fcd8c5c984b4640ec578b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 21:33:52.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0051d79d7da54379b4ceee79d2e8fab5\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/41659004998a4e1c850be91268771117 : SUCCESS in 40m 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6ae86c545aa64d4d91e8efc5ae40f50d : FAILURE in 1h 43m 11s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/11774facde474474b406aeeba55550c6 : FAILURE in 3h 00m 08s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2e91b89af1434c2492bc69ce90a7bea7 : SUCCESS in 33m 21s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/23b70f08d9084210b35bdf09624b1c36 : NODE_FAILURE Node request 200-0001215667 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0c136890b84a85cdadc10423006694a7592f10a6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 23:26:29.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5281/.","accounts_in_message":[],"_revision_number":10},{"id":"34455b1fccd73ddc58e468bd6b3acbffbc466ddd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-08 00:10:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5458/.","accounts_in_message":[],"_revision_number":10},{"id":"c8d7a0e0d7d4924a073df4ddea09a6bf5fe449c9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-08 00:20:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5282/.","accounts_in_message":[],"_revision_number":11},{"id":"d621444864666b4d90e91a244163c6f804214d8b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-08 01:16:26.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5283/.","accounts_in_message":[],"_revision_number":12},{"id":"98541bd63329040286f2b9c9d4ff1b19399bad73","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-08 01:20:46.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5459/.","accounts_in_message":[],"_revision_number":11},{"id":"f2a1832ab7b91df9d412c14b0b22a395413b09dc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-08 02:20:34.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5460/.","accounts_in_message":[],"_revision_number":12},{"id":"954dfc0bc262fb531f7e995f5a2cd899b41e895e","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-09-08 11:53:59.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b1d47c069fd27f4643c42e09cf211b8d25055b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-08 16:29:29.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"acb1aa2b17477e6dcbdc749e5bc6ca9224288194","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-08 16:37:37.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"b1bb6e9748f0f698951d7c88995c0186d4746f7d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-08 16:49:01.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest build/c42550fc5ea94d2886a7d0ba98968b93 : RETRY_LIMIT in 16m 56s\n- cinder-storpool-tempest-iscsi-multipath build/941b8ea0d38d4cc2890519aeab9d0f45 : RETRY_LIMIT in 18m 08s","accounts_in_message":[],"_revision_number":13},{"id":"c8429156050cfcbbfa5190407b56002bd7798833","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-08 16:49:03.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/8f50a9e912db4006933e2985f3094346/","accounts_in_message":[],"_revision_number":13},{"id":"72e2ce992f1b3033d6287958d64554de7daf59a4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-08 17:17:43.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8f50a9e912db4006933e2985f3094346\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f896c584a1e44b4aa7663341695096f7 : SUCCESS in 37m 04s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1819511b142e44aa8e0487a72855fe59 : SUCCESS in 34m 22s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/da9cf0eb21374b7da68c88edeeb92b34 : SUCCESS in 47m 43s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6cd8b477f7744a8799b08ce5d565e433 : SUCCESS in 37m 21s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/97d2e29a005b47d7953c87d0aaa2525d : FAILURE in 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"37753d496eb42205331a28686d2ed95e08b7c40b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-08 17:22:45.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":13},{"id":"51d6510952e32d3a65f4af9fdeef6858a6a5cd6c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-08 17:22:59.000000000","message":"Patch Set 13:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 8s\n- FC       : SUCCESS in 0h 48m 50s\n- NVMe/TCP : SUCCESS in 0h 48m 40s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":13},{"id":"011faa7a743fe6367d501ce7b60ed8b501b9cc26","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-08 17:29:56.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5467/.","accounts_in_message":[],"_revision_number":13},{"id":"801fa35de8b74a76fd81f9324f0b9645a73990d7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-08 17:34:21.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5290/.","accounts_in_message":[],"_revision_number":13},{"id":"93871419128856067a5a9d9c2f835f0ecd1ecb5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:07:26.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b692e1715bfd4f87aedfb6748f860f8a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/475a0324386e4b4790b1fe99835fd2b4 : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/721183afe9384b82b11396d80a9c3108 : TIMED_OUT in 1h 01m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/228953fad2ad4f0cba189743ac8a9d5b : SUCCESS in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"31491c99411118eae099c15e33d07f65c41a2afa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:49:08.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8805fef36751494f9aa36f7be956f617\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/009b27455ed947feae6142c811ce9d1e : SUCCESS in 7m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c120ca2e8964fcfa423ba16d3fc5546 : SUCCESS in 12m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c5add4978c4428a961a244b96eca2a1 : SUCCESS in 11m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/969026558570446a9aac68eceec8714d : SUCCESS in 12m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6771597b00646f4bcd5b00dc5f7fb41 : SUCCESS in 14m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c9cc7abfdcdb495cbb3054b865bbfe84 : SUCCESS in 1h 01m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3fef2dc9e9f7422689f507328e714ec8 : SUCCESS in 1h 03m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72b0920a637a49abbc026097d184b83d : SUCCESS in 1h 21m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9886293d2b849e28893089202a709b8 : SUCCESS in 1h 04m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4db7f07403fe40028c66d7d719c72ef5 : SUCCESS in 5m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bfa8e148909d454db5dc6b90a7d517ce : SUCCESS in 21m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/91a5b376012d414cb5d1712bc487f6f0 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6a0d62c7a5734f168a54d7dcf6748269 : SUCCESS in 6m 38s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/a2bdc5aebcb44bb68927ab96e7344d7d : SUCCESS in 11m 07s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/af39692b488a4e00a10a33d60fb0d09c : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/444868327596468e8a9445cba5ff570a : SUCCESS in 6m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9d6fa8c79f5a43799a465f7fda0549e5 : SUCCESS in 7m 27s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/dc1252388b1c43c08694f5c5d4d92a47 : FAILURE in 4m 37s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e6b2f45090ff4ad18ccb77b4d0354b05 : SUCCESS in 9m 52s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9c58f9210a534e21a4a14237efa1011d : SUCCESS in 35m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ab22a4056a2f4405affd43575ca5d8f6 : FAILURE in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/68c968471123479fa88f31fc66600d80 : SUCCESS in 1h 32m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/11008f23ecdf4c99a45b6073195e3eb1 : SUCCESS in 2h 06m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ccb2e2116db3453683a6309671a192c7 : SUCCESS in 1h 11m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/a5d6b4d0e6944d0684aae37209743038 : FAILURE in 2h 12m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7bf775ee8a914e5e96ad2b65195b4d96 : SUCCESS in 1h 37m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a97366e463f942909d2c1b2f5f93b5a5 : SUCCESS in 38m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e52eab6ac02745abb526f2fb3a46fbf2 : SUCCESS in 1h 16m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/be8e15bad0944ed0bb63118c71bae4aa : SUCCESS in 1h 43m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0a5840718e7d4aa08332a8b49f1e3159 : SUCCESS in 41m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a93d31bedae48f2a4df5aba88914286 : FAILURE in 1h 28m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ab7fabbf8d241668c7363ee4a67a263 : FAILURE in 23m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"842822a34b9974799e81a7d7e04f7ed617823245","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-08 18:53:13.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"fc46baf17fa76c79466028bc3bef4cbbe57f3ecc","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-08 19:04:07.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-13-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-13-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"969bbadfc2a928cf653dcc31f8912ca5f6f1ad0f","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-09-09 06:37:42.000000000","message":"Patch Set 13: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"2e9efb4962658e6247efc8ac984aafa9d058401e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-09 09:55:29.000000000","message":"Uploaded patch set 14: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"760e717029df6ab4754d46d477f5fc309696b819","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-09 10:01:27.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f4a0d647e9c2fbe62779ec1efff52e4b71dec48a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 10:22:51.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d4f91659a34483384ffdc82909b5cb0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3017f09502754c66b29c4816a14c81b9 : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/51b72ba4aa614a29b264314408f56779 : SUCCESS in 18m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/05fb269e552d49c5ba9a6ece0a99d6f0 : SUCCESS in 25m 11s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7af1b4b19a051949020a678ac445a78a44f378ce","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-09 10:33:12.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest build/6d59bff23e7349ee83d27d5cbd1ea661 : RETRY_LIMIT in 16m 22s\n- cinder-storpool-tempest-iscsi-multipath build/aa8609e9e4df4856bac9c95eb107d181 : RETRY_LIMIT in 17m 24s","accounts_in_message":[],"_revision_number":14},{"id":"885430d4bfb9c0cfdbdb24ddaddaf34b87b90d45","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-09-09 10:44:11.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"14497a94d89ae02589a31b99a1031515227344ba","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-09 10:49:08.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iba119184cc5ff0ea25360f87e5ec9430d83b53fd .","accounts_in_message":[],"_revision_number":14},{"id":"93b72f2ce097e8ab34384f5a23fc1195569e08b7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-09 11:01:10.000000000","message":"Patch Set 14:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/02008fa49c064c6291a4054628423d81/","accounts_in_message":[],"_revision_number":14},{"id":"adb7bbb3090b53fbf2d2aefcea330c3cd68820bc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-09-09 11:12:41.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/1003765/14/upstream-check/cinder-tempest-plugin-ontap-fcp/a2d1818/ : FAILURE in 24m 51s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/1003765/14/upstream-check/cinder-tempest-plugin-ontap-iscsi/c0b6b0b/ : SUCCESS in 26m 16s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/1003765/14/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/cf0fca3/ : SUCCESS in 26m 22s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"1b72d21e6772a008bf8996456c31c4553e13f8af","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-09 11:47:38.000000000","message":"Patch Set 14:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/02008fa49c064c6291a4054628423d81\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2c7193faf2c9445a83eece3a73ae39e3 : SUCCESS in 37m 51s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f765bfa2714e4a9bbb8dd15483dd10c8 : SUCCESS in 40m 08s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0787c07a2ef4666b6155a8fb84abf81 : SUCCESS in 45m 56s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/73c8a5d9764d49fe94e4d186a077d690 : SUCCESS in 33m 57s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b15a27f1a524c41b3a477eac3a9c3c1 : FAILURE in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"61460e1774d418ede5fdd492fd87a53ab0bc64d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 12:17:35.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae9645dde1245d995e184b9f2c03780\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc30bb4e52b24152994ea0fc0fb7c33f : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ef248da08ba40c3885c66edf7ae7494 : SUCCESS in 13m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbe0499171714519a88d2a13d52b320f : SUCCESS in 11m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7c71bdbe0804e80b935ee1eebc6b625 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea6ce384b52465294f882a53bf610e3 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/467244c366734f8ca7b4099e4b3a0f4e : SUCCESS in 1h 08m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/005f65748d124af38db9787c678b2750 : SUCCESS in 54m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9394fc210e1e49bba81eeaa0307926f4 : SUCCESS in 1h 20m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ea1073c4cbf4324ba62b93930388939 : SUCCESS in 59m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de0dbc9114eb4ecebc2230db090f07b0 : SUCCESS in 7m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a8d2586d294f43cc826fcad16574c570 : TIMED_OUT in 41m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bda265071ae04c4d9b4b26491a98a502 : SUCCESS in 5m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f3d78e7b63a467d8b6653433e4a4a28 : SUCCESS in 6m 35s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/ad757bb5ca6d4a23bd4b61c3ea2a33bc : SUCCESS in 9m 59s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3359efb82b574d34a26e9a9542bbddea : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6a863408f8f740a1bfde94bcf688f5d5 : SUCCESS in 8m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fc93965ad16448049ba58eb48a768135 : SUCCESS in 7m 43s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/80ce6ae87a7047efae646aa887ba739e : FAILURE in 7m 27s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/76f3eedfeca34a8383c52154ab28228d : SUCCESS in 8m 57s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/25a6e97f183e477f8ec9187bfbf37a7b : SUCCESS in 26m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fd53f765e62a407a84b7f5b89f30b352 : FAILURE in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dc0c5c20848348c98420dfa543b0f095 : SUCCESS in 1h 30m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9af6f71e19984a8ea8c289ee8f7cb8aa : FAILURE in 1h 20m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4f074e0e53664bf994fb678a3c3aa028 : SUCCESS in 1h 50m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b93097f4c397402da28a7fdd273ebaa0 : FAILURE in 2h 16m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d37425d7e6614fc29d43b076b8c1fc19 : SUCCESS in 2h 02m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9ac16eed54e04efe9c8a45846ede3999 : SUCCESS in 41m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2b5dbcda11eb4cff890a596c62a221a5 : SUCCESS in 1h 15m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b45dfb080e1e4341a377cb7c18353fb3 : SUCCESS in 1h 30m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/25b0c59e89ac489c9a5417e180a0d88a : SUCCESS in 53m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a575359988a34066932cc0bdbe7240fc : SUCCESS in 1h 14m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb91d5de0b65479d9af7e6485638ec2a : SUCCESS in 1h 04m 29s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"c3f0d86d010d95762b077d99771facff8c9dd3bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-09 12:26:53.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_09_2026_12_21_18\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"dfefe5d50cc609b8d0afc6bd6d8d23e9930332ec","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-09 12:28:10.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-14-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-1003765-14-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"234f6b1d50d284eba18e6a87a8681407d2d72b96","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-09 12:35:53.000000000","message":"Patch Set 14:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 46m 39s\n- FC       : SUCCESS in 0h 45m 56s\n- NVMe/TCP : SUCCESS in 0h 47m 24s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003765","accounts_in_message":[],"_revision_number":14},{"id":"62c76ef56cbffc2a40f764c41f7e6e46a6a4e81e","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-09-09 14:14:05.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"8d96b6af76f0c4b90328cc4231fdb0c625ea5b6b","revisions":{"d592dfdcc6c02e9b7085654f8c596e8327ec727f":{"kind":"REWORK","_number":1,"created":"2026-09-03 13:05:12.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/1"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 13:04:31.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d592dfdcc6c02e9b7085654f8c596e8327ec727f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d592dfdcc6c02e9b7085654f8c596e8327ec727f"}]},"branch":"refs/heads/master"},"45a8c7de98e43a85e32469879db02c64bf3ed505":{"kind":"REWORK","_number":2,"created":"2026-09-03 14:29:23.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/2"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 14:29:06.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45a8c7de98e43a85e32469879db02c64bf3ed505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45a8c7de98e43a85e32469879db02c64bf3ed505"}]},"branch":"refs/heads/master"},"a7e5459eea9a611ec7de78b015e20e86ea300160":{"kind":"REWORK","_number":3,"created":"2026-09-06 20:42:19.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/3"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-06 20:41:55.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7e5459eea9a611ec7de78b015e20e86ea300160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7e5459eea9a611ec7de78b015e20e86ea300160"}]},"branch":"refs/heads/master"},"22377a47c87d005b8024c0a091f0d2571dbc640b":{"kind":"REWORK","_number":4,"created":"2026-09-07 05:34:07.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/4"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-07 05:33:50.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22377a47c87d005b8024c0a091f0d2571dbc640b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22377a47c87d005b8024c0a091f0d2571dbc640b"}]},"branch":"refs/heads/master"},"08849afbd7e4c32873c27a7a281bbb188cc4417a":{"kind":"REWORK","_number":5,"created":"2026-09-07 05:36:42.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/5"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-07 05:36:27.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08849afbd7e4c32873c27a7a281bbb188cc4417a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08849afbd7e4c32873c27a7a281bbb188cc4417a"}]},"branch":"refs/heads/master"},"f5125eb0070945546594b07a65b94acfc48b1e12":{"kind":"REWORK","_number":6,"created":"2026-09-07 10:04:50.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/6"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-07 10:04:30.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5125eb0070945546594b07a65b94acfc48b1e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5125eb0070945546594b07a65b94acfc48b1e12"}]},"branch":"refs/heads/master"},"aaf20a6ff0885e250572bb287e3753144744f60a":{"kind":"REWORK","_number":7,"created":"2026-09-07 10:09:51.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/7"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-07 10:09:36.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aaf20a6ff0885e250572bb287e3753144744f60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aaf20a6ff0885e250572bb287e3753144744f60a"}]},"branch":"refs/heads/master"},"5615e108c05b48e8071c857021265f76325c7a5b":{"kind":"REWORK","_number":8,"created":"2026-09-07 10:43:37.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/8"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-07 10:43:19.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5615e108c05b48e8071c857021265f76325c7a5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5615e108c05b48e8071c857021265f76325c7a5b"}]},"branch":"refs/heads/master"},"549bfd2a5b9d872f81c1b95ee3132ab54c478058":{"kind":"REWORK","_number":9,"created":"2026-09-07 11:10:08.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/9"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-07 11:09:52.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/549bfd2a5b9d872f81c1b95ee3132ab54c478058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/549bfd2a5b9d872f81c1b95ee3132ab54c478058"}]},"branch":"refs/heads/master"},"488b90601256ed350895eb2bf231226bfd289906":{"kind":"REWORK","_number":10,"created":"2026-09-07 17:22:22.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/10"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-07 17:22:07.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/488b90601256ed350895eb2bf231226bfd289906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/488b90601256ed350895eb2bf231226bfd289906"}]},"branch":"refs/heads/master"},"29a6f4239d857aa59766679c6cbbe70a3479bb19":{"kind":"REWORK","_number":11,"created":"2026-09-07 17:45:16.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/11"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-07 17:44:58.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29a6f4239d857aa59766679c6cbbe70a3479bb19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29a6f4239d857aa59766679c6cbbe70a3479bb19"}]},"branch":"refs/heads/master"},"44218f9aa31d23e73d31a88a337f1f55fa961171":{"kind":"REWORK","_number":12,"created":"2026-09-07 18:33:12.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/12"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-07 18:32:55.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44218f9aa31d23e73d31a88a337f1f55fa961171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44218f9aa31d23e73d31a88a337f1f55fa961171"}]},"branch":"refs/heads/master"},"5abbb8325765b127e7567b434223da7474816a29":{"kind":"REWORK","_number":13,"created":"2026-09-08 16:29:29.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/13"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-08 16:28:56.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), which has no\n  per-FlexVol pool model for SnapMirror to provision against\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5abbb8325765b127e7567b434223da7474816a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5abbb8325765b127e7567b434223da7474816a29"}]},"branch":"refs/heads/master"},"8d96b6af76f0c4b90328cc4231fdb0c625ea5b6b":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2026-09-09 09:55:29.000000000","uploader":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"ref":"refs/changes/65/1003765/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/1003765/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/1003765/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/1003765/14"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-03 12:00:28.000000000","tz":330},"committer":{"name":"Mada, Sailaja","email":"sailajam@netapp.com","date":"2026-09-09 09:54:58.000000000","tz":330},"subject":"NetApp ONTAP: validate replication configuration at init","message":"NetApp ONTAP: validate replication configuration at init\n\nThe driver reported replication_status\u003denabled for backend\nconfigurations that cannot support replication, deferring the failure\nto the first SnapMirror operation where it surfaced as an opaque\nONTAP error.\n\nValidate the configuration during check_for_setup_error() and fail\nfast with a descriptive message when a replication_device is\nconfigured together with any of:\n\n* netapp_disaggregated_platform\u003dTrue (ASA r2), does not have\n  replication implementation yet.\n* a netapp_replication_policy outside the ONTAP-recognized set\n* AutomatedFailOverDuplex without netapp_consistent_replication\u003dTrue\n* netapp_consistent_replication\u003dTrue on a pool containing FlexGroup\n  volumes, which cannot be SnapMirror Consistency Group members\n* an ONTAP cluster older than 9.16.1\n\nCloses-Bug: #2166341\nCloses-Bug: #2166342\nCloses-Bug: #2166343\nCloses-Bug: #2166345\nAssisted-By: Cursor\nChange-Id: Iba119184cc5ff0ea25360f87e5ec9430d83b53fd\nSigned-off-by: Mada, Sailaja \u003csailajam@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d96b6af76f0c4b90328cc4231fdb0c625ea5b6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d96b6af76f0c4b90328cc4231fdb0c625ea5b6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
