)]}'
{"id":"openstack%2Fcinder~1003955","triplet_id":"openstack%2Fcinder~master~If49244dfd0bd5fafa385b6a240322fed3e4d9587","project":"openstack/cinder","branch":"master","attention_set":{"36686":{"account":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"last_update":"2026-09-04 13:30:59.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If49244dfd0bd5fafa385b6a240322fed3e4d9587","subject":"Log exceptions in BackupManager.continue_backup","status":"NEW","created":"2026-09-04 13:16:36.000000000","updated":"2026-09-05 01:57:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2ccd97201fe0e5edfade6bfa20b3f09abf4b234e","_number":1003955,"virtual_id_number":1003955,"owner":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-04 15:58:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 13:30:59.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-04 13:39:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 13:52:22.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 14:08:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 14:12:57.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 15:13:43.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-04 15:58:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 15:58:20.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-04 18:41:37.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"0cee320c3d29e9cb801bae4c01b53ac31efb4dc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-09-04 13:16:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fd9d6412f7231ca21cf36edc1b3d10a936312fa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-04 13:30:59.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/67dfafeb03c34e90876030f3f1c6f3d9/","accounts_in_message":[],"_revision_number":1},{"id":"5e1f1364baf8b7792e6d36b64331102830cea194","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 13:39:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daad6b1a487a4b98bebf79799640b2ca\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8dc0885c002147c0aa84bee529c8bff5 : SUCCESS in 19m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f5bdf5b709ec437488a1d2d149ba1a73 : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/55da788443ed46be8e30be550c31b57f : SUCCESS in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9f5833dec872b12f085e0ac1a8a4f77718e5f16","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-04 13:52:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/605dc5e0203e44b59a3772c03d58fd8e : RETRY_LIMIT in 16m 07s\n- cinder-storpool-tempest-iscsi-multipath build/5347a51ea5994b4294a9a1c1ed2a4f24 : RETRY_LIMIT in 17m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ad449d229721f63dce0491b62aa4912108eac621","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-04 13:58:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/67dfafeb03c34e90876030f3f1c6f3d9\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/02d7794a3ee741779a2ce8384824fd02 : SUCCESS in 37m 16s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fafa993c798d47e8b71378600adcf48d : SUCCESS in 35m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/43c496bc706546b489a2f46b365aed43 : SUCCESS in 41m 33s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8b6b26bf22414450af8b1fb45725176e : SUCCESS in 28m 37s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/468dacdacac84fd9a636031f42905830 : FAILURE in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62e56087a2af0715603ff8474e3ee6d281e049b4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-04 14:08:45.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 51 min\n- FC       : SUCCESS in 50 min\n- NVMe/TCP : SUCCESS in 51 min\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1003955","accounts_in_message":[],"_revision_number":1},{"id":"6920a3a7175db72a795e29bf6f7cb1b40ff1df6e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 14:12:57.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/If49244dfd0bd5fafa385b6a240322fed3e4d9587 .","accounts_in_message":[],"_revision_number":1},{"id":"06e407c1e73cde1825435fe25d94f3bac03c40f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 15:13:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5412/.","accounts_in_message":[],"_revision_number":1},{"id":"1ad47dea1f04ccdfe598c00b524c8e61e267b26a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:58:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0064c0a63a9b45be9f64b296462c2986\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a21101382bb7492ab3439fba672870b4 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e42370fceed4f1cac510c05463dfa3f : SUCCESS in 11m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1dbd74d649e54adb9d48ecced3a0ac36 : SUCCESS in 10m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/99857c7823ab4074ae0c77d1189086e4 : TIMED_OUT in 41m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed716200adab43118f7ce779b5928b48 : SUCCESS in 15m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/929d836cccc648bab21b4b9dda41bad1 : SUCCESS in 1h 04m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d9b9784109444fd79576b55cbb986d53 : SUCCESS in 1h 15m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab566d14478546dbb685b22edf77392e : SUCCESS in 1h 05m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b202e484ff84796903de05a98622c01 : SUCCESS in 53m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a40dc93050842c9bb782bcae7f2d6a0 : SUCCESS in 7m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/07e5213ef3aa40a489b1035061ea9bba : SUCCESS in 17m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/12e8163a34394c1d835ad42adf78cf41 : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/05c134f990da45d6a35b19ba80935741 : SUCCESS in 6m 59s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/b3583794b98949fcb3ac8343ab6ad7dc : SUCCESS in 12m 25s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/94292931ba454bf48b3ff8c0966f9364 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4ddb1fd200ab4e85976dc3ae6e8371a9 : SUCCESS in 6m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6b7248135f7044a69e92b26eaa8108eb : SUCCESS in 6m 30s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/267991d3c1684fe4965bb766c56d5a0e : FAILURE in 4m 21s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e7da662902434ff3b48ce1496498090f : SUCCESS in 7m 32s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/adbd958156bc4934b2a9146d1b6ca324 : SUCCESS in 30m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0dd431149b8b4f34aa267a864df16cbe : FAILURE in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4bd96d1dabc6477a80b3bf11ec5db679 : SUCCESS in 50m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b7481598542c4d0485b53ab0504b4217 : FAILURE in 2h 10m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/97017da4c56645aebf3933a45f82823a : SUCCESS in 1h 45m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/8d23dd0ba3a948b0a91500023245deb4 : FAILURE in 2h 37m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/13a9abaf1c3e44639ba8123353f56ce0 : SUCCESS in 1h 59m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dd72716687cf494a95f06d484fe7930f : SUCCESS in 35m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/462696d06160474985acf68dd5d7567c : SUCCESS in 1h 18m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d400c63f10544d09b3e398c1047537fd : SUCCESS in 1h 31m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6d61e973c42a48399ba93584428f97bc : SUCCESS in 54m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b68b90008ec41898ec0a7f1030a21fe : SUCCESS in 1h 28m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e54b9f5720fa4fb2b8770c28f51e28df : SUCCESS in 58m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"e5bebdfb6fe1b3d43cfd8dfd29ad09b2a4176203","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-04 16:07:25.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_09_2026_16_01_49\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"ea29a46439cddcd07f9ceb95d6c3acc01d9aa90d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-04 18:41:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-55-1003955-1-fc : SUCCESS in 59m 9s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-55-1003955-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2ccd97201fe0e5edfade6bfa20b3f09abf4b234e","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-05 01:57:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-55-1003955-1-fc : SUCCESS in 55m 9s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-55-1003955-1-iscsi : SUCCESS in 1h 14m 5s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0f165334b8371d29e1d878bb98aae2facffa035","revisions":{"e0f165334b8371d29e1d878bb98aae2facffa035":{"kind":"REWORK","_number":1,"created":"2026-09-04 13:16:36.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/55/1003955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/1003955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/1003955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/1003955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/1003955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/1003955/1"}}},"commit":{"parents":[{"commit":"e3b9582a60a58d6ea630e845bbe5d96c8780fd8c","subject":"Merge \"mypy: annotate lvm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3b9582a60a58d6ea630e845bbe5d96c8780fd8c"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-09-04 13:04:27.000000000","tz":540},"committer":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-09-04 13:16:32.000000000","tz":540},"subject":"Log exceptions in BackupManager.continue_backup","message":"Log exceptions in BackupManager.continue_backup\n\nThe exception handlers in continue_backup update the backup state and\nre-raise the original exception, but do not log it. This makes failures\nduring backup device attachment, driver execution, device detachment, or\ntemporary resource cleanup difficult to diagnose from the backup service\nlogs.\n\nLog each failure with the backup ID while preserving the existing error\nhandling and user-facing message creation.\n\nSigned-off-by: hongp \u003cinyong.hong@samsung.com\u003e\nChange-Id: If49244dfd0bd5fafa385b6a240322fed3e4d9587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0f165334b8371d29e1d878bb98aae2facffa035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0f165334b8371d29e1d878bb98aae2facffa035"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
