)]}'
{"id":"openstack%2Fcinder~1004176","triplet_id":"openstack%2Fcinder~master~Ie5114caab4d57681261c053a5a525d0892f13f76","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-06 14:32:58.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ie5114caab4d57681261c053a5a525d0892f13f76","subject":"Use common environment classifier","status":"NEW","created":"2026-09-05 13:27:35.000000000","updated":"2026-09-06 14:32:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"05cb7da730b70afc12ce941c290983ee109e90b5","_number":1004176,"virtual_id_number":1004176,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-06 06:08:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 13:48:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-05 13:50:09.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-05 13:54:13.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-05 14:22:16.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-05 14:24:35.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-05 14:30:32.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-05 15:49:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-06 06:08:47.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-06 06:19:33.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"57e90dbf977098bf2e883965d7b0d1074851f5fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 13:27:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d3fa09ce00a4cdf80bbaef0b85ede3226a14701","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 13:48:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63f7e6326d57445a9039f015dea9b298\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b0c53a7d8ddf4242bf6c01c529c2e75e : SUCCESS in 14m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4266bf20e2bc4bb48df5b8f4040532ba : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d47223a4a23f4eb0974c3cfdb73ef96f : SUCCESS in 18m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3316d72006e7bc8fb3e3c50833461614e539e4ff","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-05 13:50:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/86dff672e46d425a926fefd4bd329a59 : RETRY_LIMIT in 21m 01s\n- cinder-storpool-tempest-iscsi-multipath build/71c4cd818c6545a19af4897fffe6c7e2 : RETRY_LIMIT in 19m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d835ed690bc5750ffb77286a2ebef82c93452b8f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-05 13:54:13.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/9ce66c950a28477c859ea6d80c6cca88/","accounts_in_message":[],"_revision_number":1},{"id":"3620b10f336ea59d422a727cf485a905a9ad6780","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-05 14:22:16.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 49m 46s\n- FC       : SUCCESS in 0h 48m 49s\n- NVMe/TCP : SUCCESS in 0h 49m 15s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004176","accounts_in_message":[],"_revision_number":1},{"id":"3de2e119db10ecdede0487e00ad0c1212d1c4800","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-05 14:24:35.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie5114caab4d57681261c053a5a525d0892f13f76 .","accounts_in_message":[],"_revision_number":1},{"id":"1f66e87ddf74061723c3af537149f06c617a94a0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-05 14:26:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9ce66c950a28477c859ea6d80c6cca88\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93116a0272314b819e49048f634b28d3 : SUCCESS in 48m 50s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4459f41dcf05440bb6f96adaeb927135 : SUCCESS in 46m 44s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f5f96ea66da644deac540ead1e98b6ac : SUCCESS in 58m 03s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90a2ac108c2e47e2ba9c2431734db6f5 : SUCCESS in 38m 31s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0bdfae0c87f647e4985f7bdc908b141b : FAILURE in 23m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd630f4a121087bb000afc27c201c267946ab7de","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-05 14:30:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5420/.","accounts_in_message":[],"_revision_number":1},{"id":"961f1c3c8fb1f95d5c83f95b754347838230bb2a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-05 14:34:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5243/.","accounts_in_message":[],"_revision_number":1},{"id":"f9ddac0020cbf3535eeb321ed1bbb06acc12cf36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 15:49:45.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c30edd8232314a14a67c12ee0740c7b1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/67ddf1956ff64d30b76721e91246f1d5 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55aa11451ba94902ae24054ec2dc7ed0 : FAILURE in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/afc923da6f174c1a9bd965819f245c77 : SUCCESS in 10m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3954ea2ed03045b2b518981cecfc2f19 : SUCCESS in 8m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/acd4c0649e54483c92e77602acb5545f : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd886152fc043868939254c06db3741 : SUCCESS in 13m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/60d0b54ff9ca47c4848b7427c53c35fe : SUCCESS in 3m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/be17f17ad2774b99be66cd6243332bf9 : SUCCESS in 1h 05m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/42b56465232548e4834cfef6fa020044 : SUCCESS in 1h 04m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f98e849bc08b4b25b207723957797363 : SUCCESS in 1h 07m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46c3a577d0f4455c98dc4b24495b58c3 : SUCCESS in 32m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0abfce451f6a4a60969bb3ab964c9925 : SUCCESS in 6m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/68eb6c686f48467e8144bc185bc51f2d : SUCCESS in 21m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1d157efe886d474c92a82cca06179e7c : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7744be171ce440a3b1cbcabecc54a527 : SUCCESS in 7m 28s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/aecf5f52fbe242f8abfa41a2e7bbe4c7 : FAILURE in 11m 19s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7197878b6e074b4d80119bf466e006e4 : FAILURE in 11m 41s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/59b5c8abfbf54e65bbe06d87b375da83 : FAILURE in 5m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9fa4f4494c2a48b58e51d75d12dfffc7 : FAILURE in 4m 14s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/88e7484b39284e22a88ad3bdb4e746c1 : FAILURE in 5m 33s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d3b03d07a44444ffac5d62075b4f0077 : FAILURE in 7m 14s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0cb4aef813dd4a98b98cf58920b399d7 : SUCCESS in 40m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/231c6f2e8fca43f58c72d3ed5e4deabd : FAILURE in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0239fa147629482aa4c1f789b2153b3c : SUCCESS in 1h 36m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/21780f90b7de4b5fbedc5f78228986ad : FAILURE in 1h 20m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/05bc65cd1d9b4551afe42c80abddea71 : SUCCESS in 1h 54m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/9a97063b0f0049369997616abcc37af4 : FAILURE in 2h 16m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5a3dd912d84a4028b893bbe3d3f08112 : SUCCESS in 1h 20m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b9b715149b804ed28570af26f4a4bdbc : SUCCESS in 17m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1a4076c969584316be130846acab6cab : SUCCESS in 1h 21m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/804429e59d7243c88cff91d797cc550b : SUCCESS in 1h 06m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2d7d179403934ebda41c7774fdeab4bc : SUCCESS in 45m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f940ad4c183047c698c1f02c36e6c845 : SUCCESS in 45m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c365fb1de0a14909a97e2269e8e8d9dc : SUCCESS in 1h 11m 31s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"691ac9419484b29ba590ec5fde4ad883ce9f7bc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 04:00:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1de066e01d655dc506009aa038342022b79eaf85","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-06 04:17:43.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/615116c7c25d4ca4b1d43ffae1724228/","accounts_in_message":[],"_revision_number":2},{"id":"438c2b89499962955d1893a26326076ad7254d63","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-06 04:19:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/0559d31a47c94bf5b6920f0734b5a00f : RETRY_LIMIT in 17m 11s\n- cinder-storpool-tempest-iscsi-multipath build/672449e960bb48feaa6a461ee2283829 : RETRY_LIMIT in 17m 57s","accounts_in_message":[],"_revision_number":2},{"id":"e4b228f928e16815d27741c5a5f01e81f4238fed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 04:31:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb33781fc1fc49a8a6c5e87065a378d9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1a5389e9df674de790bd65f2ced82e0b : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e1d90d0126444f038224bf4b7a5caed3 : SUCCESS in 15m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9012650b80e24e1db9ba5d3241579358 : SUCCESS in 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"620f5e6900dc578257cf79f8952f0621335cb48e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-06 04:53:37.000000000","message":"Patch Set 2:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 33s\n- FC       : SUCCESS in 0h 48m 26s\n- NVMe/TCP : SUCCESS in 0h 48m 23s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004176","accounts_in_message":[],"_revision_number":2},{"id":"c884875ecbfcd6a602812184e31e4f6df8ed280b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-06 04:54:51.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie5114caab4d57681261c053a5a525d0892f13f76 .","accounts_in_message":[],"_revision_number":2},{"id":"496d8ca0bf3cba66aca2f95b2f899ac9339b5a48","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-06 05:01:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5427/.","accounts_in_message":[],"_revision_number":2},{"id":"2d86bbb6b2bed0fc6bb4e188e559e61b44f8d1b6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-06 05:01:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5250/.","accounts_in_message":[],"_revision_number":2},{"id":"9443cab978413ac70912dfe85541a654bfb2c270","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-06 05:45:30.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/615116c7c25d4ca4b1d43ffae1724228\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eaf482a886d94316a0684e7f2e5db1c5 : SUCCESS in 44m 00s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b97ab6b2de4e4dfa82ee0bb6acc17a89 : FAILURE in 1h 44m 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/749ad5cdbe4545a39a61b64bf62543fc : SUCCESS in 53m 13s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/405079b9ab3141d6b01ff2abdd38ee46 : SUCCESS in 35m 00s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b9af3f02c3cf41c19cd5c28c71e40207 : FAILURE in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d47dd24e57750e08725efef475588a9198d453e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 06:08:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b21b3ca30204c68bb65b5966ecf87d5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c6f04626cd804c609a28d48166e18e10 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ca88fbec3649278fdf190611ae663e : SUCCESS in 7m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2db32a4322a54e978b232420f55ee53f : SUCCESS in 10m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02c05e2fd5d64cc9a03639c7b277b557 : SUCCESS in 11m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/abdb1a8aeb794ad4aaa398f4c7b97760 : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/661786a73dc4459abb59bd5a35a0c3d8 : SUCCESS in 12m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a52a7e8b93884cba8bf5540d7cc1c507 : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/6c4134a3504844369887758d693cf425 : SUCCESS in 1h 04m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a808bc26db334f59af1a2f20c9076579 : SUCCESS in 1h 06m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/41bf114839b048db9f76d21535f1dbb4 : SUCCESS in 1h 04m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3717ec444d5c45c48f8220901e0647a8 : SUCCESS in 28m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a82a3051d5ca465996ced9b2809f265a : SUCCESS in 10m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a22ae3a274d845b7982e243d44b60906 : SUCCESS in 23m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/510b2e87ed4941b98dc19779d907c062 : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b4bf34d544ea437b9531678b99837576 : SUCCESS in 7m 49s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/0f4da82747ad460d96bfd28d343f75c1 : SUCCESS in 9m 18s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/eba8811a66324a31908d78efae7e6112 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7eda7d2eb5bc44e5b82446bb6fc072b3 : SUCCESS in 6m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/08e8489d76d24d03a3d586d56692ab63 : SUCCESS in 4m 46s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f621a073a1b44b81a4a555cb2a7fdbe9 : FAILURE in 4m 56s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ff4a73507859495badede60ba694232d : SUCCESS in 11m 07s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/54ea89b9b8e84d75b3bc3eb8a0ca666b : SUCCESS in 20m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83e76d8477194feba831f6e620766993 : FAILURE in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/67e08153a0fa4cf29454b6dffbef6946 : SUCCESS in 1h 36m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/efdc6e81bc4e4ad78903431dba0d76ff : FAILURE in 1h 58m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f13ff395dfa34b8db882263c7ffe08ae : SUCCESS in 1h 33m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/29b0a18a6ef944798e9408f3674775fb : FAILURE in 2h 06m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5665a20e34844c809c9e6e3b152467a7 : SUCCESS in 1h 31m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f0de983d10c540ae8b30d76da245ee13 : SUCCESS in 18m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/64e5cf18fff64ee0a5f25e5d22325fe3 : SUCCESS in 43m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d9b9a41186e44b75857f2863f86d8d7d : SUCCESS in 1h 11m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c8a50bc6ac6647f4b9ab787139f9bab4 : SUCCESS in 25m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1d1f67d9af0b475daee19f201ea7643a : SUCCESS in 1h 30m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/958dfb13e7e3425c86b67728c2be6e98 : SUCCESS in 49m 42s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"d266db585c37ca126d9b9de5ed0686c672221848","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-06 06:19:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-76-1004176-2-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-76-1004176-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"04a070561705d1e1e6f5c4fdc4664e9972c590c2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-06 08:04:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-76-1004176-2-fc : SUCCESS in 51m 35s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-76-1004176-2-iscsi : FAILURE in 10m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca1765e948a702080e754dd6df7586410fcae276","revisions":{"50666cbd1e0ca1658a63743221e9124a9ff9eeb6":{"kind":"REWORK","_number":1,"created":"2026-09-05 13:27:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/1004176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/76/1004176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/1004176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/1004176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/1004176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/76/1004176/1"}}},"commit":{"parents":[{"commit":"ca4708870cf0516274263c6c45b44705bd593cc0","subject":"Merge \"NetApp:Fix to handle error during consistency group expand operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca4708870cf0516274263c6c45b44705bd593cc0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 13:27:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 13:27:26.000000000","tz":540},"subject":"Use common environment classifier","message":"Use common environment classifier\n\nThe \"Environment :: OpenStack\" classifier is commonly used in most of\nthe OpenStack projects.\n\nChange-Id: Ie5114caab4d57681261c053a5a525d0892f13f76\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50666cbd1e0ca1658a63743221e9124a9ff9eeb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50666cbd1e0ca1658a63743221e9124a9ff9eeb6"}]},"branch":"refs/heads/master"},"ca1765e948a702080e754dd6df7586410fcae276":{"kind":"REWORK","_number":2,"created":"2026-09-06 04:00:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/1004176/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/76/1004176/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/1004176/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/1004176/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/1004176/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/76/1004176/2"}}},"commit":{"parents":[{"commit":"5c5f6d8452f2b333dbd9a2d85d33b5924099a006","subject":"Merge \"[Pure Storage] Implement Tiramisu group replication actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c5f6d8452f2b333dbd9a2d85d33b5924099a006"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 03:59:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 03:59:55.000000000","tz":540},"subject":"Use common environment classifier","message":"Use common environment classifier\n\nThe \"Environment :: OpenStack\" classifier is commonly used in most of\nthe OpenStack projects.\n\nChange-Id: Ie5114caab4d57681261c053a5a525d0892f13f76\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca1765e948a702080e754dd6df7586410fcae276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca1765e948a702080e754dd6df7586410fcae276"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
