)]}'
{"id":"openstack%2Fcinder~1004441","triplet_id":"openstack%2Fcinder~master~I6cda5d2c9d41156250cd46c7f7864e8448d2c5b9","project":"openstack/cinder","branch":"master","topic":"powerstore-nfs-dellfcopy-fallback","attention_set":{"37173":{"account":{"_account_id":37173,"name":"Alex Fernandes Figueirêdo","display_name":"Alex Figueiredo","email":"alex.fernandesfigueiredo@windriver.com","username":"afernand"},"last_update":"2026-09-07 15:29:31.000000000","reason":"\u003cGERRIT_ACCOUNT_38749\u003e replied on the change","reason_account":{"_account_id":38749,"name":"Luis Henrique Silva Cazuriaga","display_name":"Luis Cazuriaga","email":"LuisHenrique.SilvaCazuriaga@windriver.com","username":"lsilvac1"}},"38749":{"account":{"_account_id":38749,"name":"Luis Henrique Silva Cazuriaga","display_name":"Luis Cazuriaga","email":"LuisHenrique.SilvaCazuriaga@windriver.com","username":"lsilvac1"},"last_update":"2026-09-07 15:48:48.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"33594":{"account":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"last_update":"2026-09-08 13:10:40.000000000","reason":"\u003cGERRIT_ACCOUNT_33594\u003e replied on the change","reason_account":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"}}},"hashtags":[],"change_id":"I6cda5d2c9d41156250cd46c7f7864e8448d2c5b9","subject":"Make dellfcopy optional for PowerStore NFS","status":"NEW","created":"2026-09-07 15:28:42.000000000","updated":"2026-09-09 07:37:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f27fe4bd2198d0caabed6f62789576d50055283","_number":1004441,"virtual_id_number":1004441,"owner":{"_account_id":38749,"name":"Luis Henrique Silva Cazuriaga","display_name":"Luis Cazuriaga","email":"LuisHenrique.SilvaCazuriaga@windriver.com","username":"lsilvac1"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37173,"name":"Alex Fernandes Figueirêdo","display_name":"Alex Figueiredo","email":"alex.fernandesfigueiredo@windriver.com","username":"afernand"},{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-07 18:37:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37173,"name":"Alex Fernandes Figueirêdo","display_name":"Alex Figueiredo","email":"alex.fernandesfigueiredo@windriver.com","username":"afernand"},{"value":1,"date":"2026-09-08 13:10:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37173,"name":"Alex Fernandes Figueirêdo","display_name":"Alex Figueiredo","email":"alex.fernandesfigueiredo@windriver.com","username":"afernand"},{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":37173,"name":"Alex Fernandes Figueirêdo","display_name":"Alex Figueiredo","email":"alex.fernandesfigueiredo@windriver.com","username":"afernand"},{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},{"_account_id":37173,"name":"Alex Fernandes Figueirêdo","display_name":"Alex Figueiredo","email":"alex.fernandesfigueiredo@windriver.com","username":"afernand"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 15:29:31.000000000","updated_by":{"_account_id":38749,"name":"Luis Henrique Silva Cazuriaga","display_name":"Luis Cazuriaga","email":"LuisHenrique.SilvaCazuriaga@windriver.com","username":"lsilvac1"},"reviewer":{"_account_id":37173,"name":"Alex Fernandes Figueirêdo","display_name":"Alex Figueiredo","email":"alex.fernandesfigueiredo@windriver.com","username":"afernand"},"state":"REVIEWER"},{"updated":"2026-09-07 15:29:31.000000000","updated_by":{"_account_id":38749,"name":"Luis Henrique Silva Cazuriaga","display_name":"Luis Cazuriaga","email":"LuisHenrique.SilvaCazuriaga@windriver.com","username":"lsilvac1"},"reviewer":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"state":"REVIEWER"},{"updated":"2026-09-07 15:48:23.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 15:48:48.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-07 16:21:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 16:22:25.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 16:33:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 16:42:20.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"},{"updated":"2026-09-07 18:37:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 18:37:06.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-07 21:26:34.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-09 06:28:21.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"}],"messages":[{"id":"064d576be57e69e709851a7bc5f5a978512e69f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38749,"name":"Luis Henrique Silva Cazuriaga","display_name":"Luis Cazuriaga","email":"LuisHenrique.SilvaCazuriaga@windriver.com","username":"lsilvac1"},"date":"2026-09-07 15:28:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49363b55acd9165079e5733defef3c6119623fa9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 15:48:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/3703f8be102d45819714eb34f41091a9 : RETRY_LIMIT in 17m 05s\n- cinder-storpool-tempest-iscsi-multipath build/4223592d4d64498b9b30e21b3674111f : RETRY_LIMIT in 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d310e83fe06e01c03a44be3bf8d3f78375b6742f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 15:48:48.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d7976773915c4aae946db7e798fceced/","accounts_in_message":[],"_revision_number":1},{"id":"26e6b136545a2819ebcfe70cdc6a3ed72ca531e7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-07 16:21:45.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : SUCCESS in 0h 47m 34s\n- FC       : FAILURE in 0h 12m 8s\n- NVMe/TCP : SUCCESS in 0h 48m 14s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004441","accounts_in_message":[],"_revision_number":1},{"id":"21713389a8bac2edb440b243d6e36e6726abf838","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 16:22:25.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6cda5d2c9d41156250cd46c7f7864e8448d2c5b9 .","accounts_in_message":[],"_revision_number":1},{"id":"cc7cddce8bf241404cbbd5ec4ff4ad71040c9571","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 16:22:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d7976773915c4aae946db7e798fceced\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8995d716927a45fdba148cba81096830 : SUCCESS in 43m 52s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8c7852a46cbe4a7ead284eedb3754d27 : SUCCESS in 34m 11s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/efd82060b3e1481f8fd2f7a8980d1aad : SUCCESS in 49m 27s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/462478bbc9bd4b29a161109bc532a7e2 : SUCCESS in 35m 24s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e05c5af9f4c643a8a8148547de066f60 : FAILURE in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18fcb8c4e696194d704e1bc2ffd0c15b7b51454b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 16:33:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36de755ba0a94b018aa678aea75ed9f3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0aefee7cad8742b0b27655062b60db89 : SUCCESS in 17m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9e9ca6294364b8dbb80f3dc887bfdd4 : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c25067949f6e45dc9c9afc5841685763 : TIMED_OUT in 1h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce33975351d3e1834c86dfe71c6d6c7308dc4b77","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-07 16:42:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/2e6af113401a4fadb473151f6c8a29d0\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/a777bc4851d6433d8c6c60012fb77580 : POST_FAILURE in 1h 05m 05s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/234c4671de5a42b7b8c4d759db43421c : SUCCESS in 1h 08m 18s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/50e6ace9ce4646509c73ea386c070dd5 : SUCCESS in 1h 12m 23s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/b1ac08ad2ca042939684ef04d57caa5c : SUCCESS in 1h 03m 43s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"87e58729a5eed47d3cec8c26283eac0bf38c70a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 18:37:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79358760cf3c42d0bd65cdb8edbc17b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4818dd375d47c2871e34f1ce26df54 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23fdee2ac57d4407adfd582a7eefab20 : SUCCESS in 12m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89fb0c85dafe4f93a9349fd222eec51a : SUCCESS in 8m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b88265a7148b4d6298f9b893946a35c4 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da7835cccd4a4769af40ab90840729f2 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/8de8a91baa1f4a8b8f3c1f945dff145d : SUCCESS in 37m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/146d12f8e40143948f85b709658dc9fc : SUCCESS in 1h 08m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84bdcd7d324745dd938d46df7a20af3d : SUCCESS in 1h 28m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02e4e7ec3b904b05a91b4680ea374fc1 : SUCCESS in 44m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e606dd6201d14d0aa66cf4f6faf2e22c : SUCCESS in 8m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/97d76f414b844c30bced4f8f81571c90 : SUCCESS in 20m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bd100f0c438246418c4d897fad153ef6 : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4596902834dd4007976f6d37dae20d54 : SUCCESS in 6m 28s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/39d75f14bf804aa6ae5e93b2eb304a5e : SUCCESS in 9m 08s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7f772e1f772d4c4d8dca9deb3d61c561 : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0798353bb7194c03af51cb592bbd5916 : SUCCESS in 5m 54s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bdc4c9a5edaa4c5bbe139fd541f743ac : SUCCESS in 6m 00s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4b87e6fbea42430187b5949b9aeb5b5e : FAILURE in 6m 29s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7c6fe4616aed411d96ed7064c6dd6de1 : SUCCESS in 7m 26s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c40ef7c65c20483483aa4e1b9efde34d : SUCCESS in 23m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0f5aace7fee4389b96d6a13406fca89 : FAILURE in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fd06cc0bf6184a84b9fb8cc1b394b92c : SUCCESS in 1h 46m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dc458a7558434d14b2631e4d8b64486a : TIMED_OUT in 3h 04m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0697b21d971c4f689ef790ac4953f0d8 : SUCCESS in 1h 19m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/8478549ef54f4667a8fe6e09b8d8441d : FAILURE in 2h 13m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7349c9c8fa21438d904d9ebc15ffbb80 : SUCCESS in 1h 57m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/addff0fc99ba44fab8c142909c21cfb0 : SUCCESS in 21m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3a3847609fce41e782f602d1fd90f4be : SUCCESS in 1h 17m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0c06a70661f24dafb6f782a98f9004c2 : SUCCESS in 1h 06m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/941604cf9df64674bb3757150e64054a : SUCCESS in 26m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3a4d545026d4d358a828bffc6374435 : SUCCESS in 1h 15m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bd909e5a1034a4d861748264096e639 : SUCCESS in 36m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"4f5aabadadefd4cb989d709524deda84c6a32faa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-07 18:46:06.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_09_2026_18_40_32\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"1007aa0be2f88481c3a3bf334698b90651d68a15","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 21:26:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5279/.","accounts_in_message":[],"_revision_number":1},{"id":"1281c22fd00d8c10ca9cc7998b626b8ce65d7379","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-07 21:59:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5456/.","accounts_in_message":[],"_revision_number":1},{"id":"efe7ca4bf9fc54776789b6707af6d4ce3b32f560","author":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"date":"2026-09-08 13:10:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"29e28aaf685975ddc9863d6809df82a391fe1a5b","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-09 06:28:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74771618d76a7fc6d2c79950996258566b0ceafd","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-09 06:29:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f27fe4bd2198d0caabed6f62789576d50055283","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-09 07:37:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/68a792ac7aa7471498f2975f05f4c10f\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/486b4215102c4a769f0b3959f9041cc9 : SUCCESS in 59m 10s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/9a5255491670456b88d8b644b2b24bed : SUCCESS in 1h 07m 39s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/6a52e7ffdd7e4ee4ad6cd1bb2df510c7 : SUCCESS in 1h 03m 05s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/fe16de00efdd437d8bfbc902cccb1def : SUCCESS in 1h 01m 20s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39cbef0c6560486392d62adbbc03f345a200b0a1","revisions":{"39cbef0c6560486392d62adbbc03f345a200b0a1":{"kind":"REWORK","_number":1,"created":"2026-09-07 15:28:42.000000000","uploader":{"_account_id":38749,"name":"Luis Henrique Silva Cazuriaga","display_name":"Luis Cazuriaga","email":"LuisHenrique.SilvaCazuriaga@windriver.com","username":"lsilvac1"},"ref":"refs/changes/41/1004441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/41/1004441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/1004441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/1004441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/1004441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/41/1004441/1"}}},"commit":{"parents":[{"commit":"a4767351ce74612281497cef7cd7a36f8d784f56","subject":"Merge \"[Hitachi] Update version for Hibiscus release / small doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4767351ce74612281497cef7cd7a36f8d784f56"}]}],"author":{"name":"Luis Henrique Silva Cazuriaga","email":"LuisHenrique.SilvaCazuriaga@windriver.com","date":"2026-09-07 15:14:25.000000000","tz":-180},"committer":{"name":"Luis Henrique Silva Cazuriaga","email":"LuisHenrique.SilvaCazuriaga@windriver.com","date":"2026-09-07 15:22:48.000000000","tz":-180},"subject":"Make dellfcopy optional for PowerStore NFS","message":"Make dellfcopy optional for PowerStore NFS\n\nThe PowerStore NFS driver fails to initialize when the proprietary\ndellfcopy utility is not installed, even though only clone and\ncreate-volume-from-snapshot operations require it.\n\nDetect the utility during setup and fall back to cp when unavailable.\nThis preserves the source image format required by qcow2 rebasing and\nvolume metadata. Continue using fast-clone offload when installed.\n\nCloses-Bug: #2166699\n\nChange-Id: I6cda5d2c9d41156250cd46c7f7864e8448d2c5b9\nSigned-off-by: Luis Henrique Silva Cazuriaga \u003cLuisHenrique.SilvaCazuriaga@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39cbef0c6560486392d62adbbc03f345a200b0a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39cbef0c6560486392d62adbbc03f345a200b0a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
