)]}'
{"id":"openstack%2Fcinder~1004728","triplet_id":"openstack%2Fcinder~master~I58f9126028956b07440225f95e93a533bbc02359","project":"openstack/cinder","branch":"master","topic":"fix-readme-test","attention_set":{},"removed_from_attention_set":{"38554":{"account":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"last_update":"2026-09-14 23:53:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I58f9126028956b07440225f95e93a533bbc02359","subject":"fix PEP 8 error in eternus_sx_fc.py","status":"ABANDONED","created":"2026-09-09 01:18:40.000000000","updated":"2026-09-14 23:53:51.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2ebce810a2713a7b8eb3ab7bcb939be5ba22e0be","_number":1004728,"virtual_id_number":1004728,"owner":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 03:37:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39269,"name":"Takahiro Yamauchi","email":"t.yamauchi_07@jp.fujitsu.com"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 01:44:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 02:13:32.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 03:30:28.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 03:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 03:37:21.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-09 03:48:23.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-09 04:13:33.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-09 04:28:12.000000000","updated_by":{"_account_id":39269,"name":"Takahiro Yamauchi","email":"t.yamauchi_07@jp.fujitsu.com"},"reviewer":{"_account_id":39269,"name":"Takahiro Yamauchi","email":"t.yamauchi_07@jp.fujitsu.com"},"state":"CC"},{"updated":"2026-09-09 05:13:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 06:24:58.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"bcf01ee5f5499c8fb1f9a680290c7aac0091af81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"date":"2026-09-09 01:18:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"761acf1c65f4d4206077d34acf270bb1ba67dded","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 01:44:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50d80aa86137439b9f0a6df8c3ddc2d2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec2aa2c2fc064ffe81409d1092b41f78 : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e2047814169c44f5afd27c6b512dca16 : SUCCESS in 18m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b22a1d27112744b38ffcd2c54953c3f7 : SUCCESS in 24m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8e022a4ba02c669022e8efa5a340cd617c54903","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-09 02:13:32.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I58f9126028956b07440225f95e93a533bbc02359 .","accounts_in_message":[],"_revision_number":1},{"id":"beabde1b9d0c4a83c52b2a4757157d301dc74e34","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-09 03:30:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/8154ba8d3fe1414ba6084bc04b0233bd : RETRY_LIMIT in 16m 23s\n- cinder-storpool-tempest-iscsi-multipath build/362ddf3b3efa47038b76cff46e9b6cb9 : RETRY_LIMIT in 17m 19s","accounts_in_message":[],"_revision_number":1},{"id":"42f983a4360d2676b5d96ecddc8e233f7c6539c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 03:37:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96d303d1356f48b5997aa8d237102f66\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2275d01af35949ca80bd4cefa3542281 : SUCCESS in 7m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d542f60e23364bc58fb88635fc517e68 : SUCCESS in 11m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18166c63ca0343e3a8cb9b3f80d28549 : SUCCESS in 10m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/61a77d561e5c4d189aee41cf023b988b : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33461024eb06448cb99cf3331a81be2d : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/3c02e0d88b8c4fc39154c7f05d9a1ef9 : SUCCESS in 1h 05m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/911b22b28fc34c349c67b0489d8df726 : SUCCESS in 51m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91eb380ec63248c090de0895ea58d1b5 : SUCCESS in 1h 30m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5176067a23664b06b9b364b3078cce0b : SUCCESS in 46m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ac5e24e3bb14f50a6ffbc7cfccbee08 : SUCCESS in 9m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9db83404079941fcb8a4dad059a9c003 : TIMED_OUT in 41m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/76b1e0fec97946c38e75b5bcaefd1d80 : SUCCESS in 6m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c06bbe7b6fde48b18ce61b8da168982c : SUCCESS in 3m 59s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/6cc30d152dc547a78dd80defd739bbce : SUCCESS in 11m 24s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/035455be6c914364bb9f30e59b71feb4 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7fe1ea2959c248dcac97a98bd483db14 : SUCCESS in 4m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e23fe9896517477d977f9ede2335da46 : SUCCESS in 7m 21s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b1df364c64e04b0cbaddf0ad401ef5d6 : FAILURE in 4m 56s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/497198f44694446ebf511bf3c3b76367 : SUCCESS in 8m 51s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/07c53cf1cef145bcacdb13e480439017 : SUCCESS in 34m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ffa9fe7b4e7437d8de1f13c225985ff : FAILURE in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/81ef5e42def44693a748fb920ce82094 : SUCCESS in 1h 33m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ff7d6df82d7141b5812ba17f052c5eb5 : FAILURE in 59m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/916bf045b4a2419eb1fa6f885b370388 : SUCCESS in 2h 00m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/76fcd18e917f49408e82e97f3350fc00 : FAILURE in 2h 16m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ca993e0e2cf34026baebe38b843d1c9a : SUCCESS in 1h 27m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9e384d523eb0466f8077221908f1623c : SUCCESS in 25m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/49840fad26694f45a57b6370028c3462 : SUCCESS in 1h 12m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3ca507b895d74f65bc1873bca3bdda84 : SUCCESS in 1h 53m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/eda24fc2d5c049558f43e55ab461287f : SUCCESS in 47m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a391dfc5cb31466f82cde4788f9f4349 : SUCCESS in 2h 08m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2b0b0a879334790a280737a8648944f : SUCCESS in 57m 06s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"17971e70a792bdbb4ac0853c663d2bd47d78959b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-09 03:46:45.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_09_2026_03_40_49\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"487dbf631399ce7cc02c60b8821e17d1b9ddce5b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-09 03:48:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-28-1004728-1-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-28-1004728-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"07f02d94317ab5a8e3b30f462f5d1bbc7e8c05d0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-09 04:13:33.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/a2777f517f5248a29d8a4a91a0e116f9/","accounts_in_message":[],"_revision_number":1},{"id":"3216476a0aea36c29085682be0b3a5c9207f1739","author":{"_account_id":39269,"name":"Takahiro Yamauchi","email":"t.yamauchi_07@jp.fujitsu.com"},"date":"2026-09-09 04:28:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3636457c0778981deecc475a0a39d73e3da57c72","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-09 05:13:56.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 1m 13s\n- FC       : FAILURE in 0h 1m 32s\n- NVMe/TCP : FAILURE in 0h 1m 28s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004728","accounts_in_message":[],"_revision_number":1},{"id":"39085d81859363a1e33d1244de63bcfbdf5e1503","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-09 06:24:58.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5482/.","accounts_in_message":[],"_revision_number":1},{"id":"f19d99d55406d54e7efd91415d549bbdc5451ba2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-09 06:26:59.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5305/.","accounts_in_message":[],"_revision_number":1},{"id":"f354355a9d2c5a2f79306c7303a360801ba097fe","author":{"_account_id":39269,"name":"Takahiro Yamauchi","email":"t.yamauchi_07@jp.fujitsu.com"},"date":"2026-09-09 07:02:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ac6bbc4ea5343e55a351d1e6a21fb97f2db6554","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-09 08:55:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a2777f517f5248a29d8a4a91a0e116f9\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c3ff0cb5da804d4eb4a53587ee2855c1 : SUCCESS in 45m 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b3fa3e6a5a6f4d23ac17bd61fdfaa2d0 : SUCCESS in 34m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f028fa71c25940f788f6048518a10877 : SUCCESS in 49m 45s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ae22432f3c954c5eaf52a60cfd394882 : SUCCESS in 35m 11s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8d1a20fdc206412f9947d9626582764c : FAILURE in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"929ee19c30829279b167566a0b0cf45094487430","author":{"_account_id":39269,"name":"Takahiro Yamauchi","email":"t.yamauchi_07@jp.fujitsu.com"},"date":"2026-09-10 00:00:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ebce810a2713a7b8eb3ab7bcb939be5ba22e0be","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"date":"2026-09-14 23:53:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96a48fe031ebb80dec92ff8d422b6094badd34ec","revisions":{"96a48fe031ebb80dec92ff8d422b6094badd34ec":{"kind":"REWORK","_number":1,"created":"2026-09-09 01:18:40.000000000","uploader":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"ref":"refs/changes/28/1004728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/1004728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/1004728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/1004728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/1004728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/1004728/1"}}},"commit":{"parents":[{"commit":"74a8c3256d4d8e945d1e70bdf9e33959f1a8e0e1","subject":"Fujitsu Driver: Added line for CI test.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74a8c3256d4d8e945d1e70bdf9e33959f1a8e0e1"}]}],"author":{"name":"ftilucas","email":"l.lafontaine@jp.fujitsu.com","date":"2026-09-09 01:15:38.000000000","tz":540},"committer":{"name":"ftilucas","email":"l.lafontaine@jp.fujitsu.com","date":"2026-09-09 01:19:31.000000000","tz":540},"subject":"fix PEP 8 error in eternus_sx_fc.py","message":"fix PEP 8 error in eternus_sx_fc.py\n\nChange-Id: I58f9126028956b07440225f95e93a533bbc02359\nSigned-off-by: ftilucas \u003cl.lafontaine@jp.fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96a48fe031ebb80dec92ff8d422b6094badd34ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96a48fe031ebb80dec92ff8d422b6094badd34ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
