)]}'
{"id":"openstack%2Fcinder~1004816","triplet_id":"openstack%2Fcinder~master~I40bfec6e6c57f084cb6875421345061b2fd08599","project":"openstack/cinder","branch":"master","attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-09-10 11:14:44.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I40bfec6e6c57f084cb6875421345061b2fd08599","subject":"Add volume type name to volume usage notification","status":"NEW","created":"2026-09-09 14:12:27.000000000","updated":"2026-09-10 11:14:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9aedb2c87759b6cd51421cba5f87849066b62e2f","_number":1004816,"virtual_id_number":1004816,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-09 17:18:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 14:27:28.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-09 14:44:17.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 14:55:53.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-09 15:06:23.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 16:21:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 16:53:59.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 17:18:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 08:50:31.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"7374f97edb96af032e6e31db90e18a4e36328fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-09-09 14:12:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"898c8af244f45bc6580bbdd3b05af172354bd2c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-09-09 14:13:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"988c5ff9b51f0099b3bc1ec44f8f43123ba47b7b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-09 14:27:28.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5312/.","accounts_in_message":[],"_revision_number":1},{"id":"4a1fe36a7f176b4c6f37a4ee54c8bbe4121ba5fb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-09 14:28:15.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5489/.","accounts_in_message":[],"_revision_number":1},{"id":"a35baa157dabdd720637c9d2be5c6ec8f4bb81a6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-09 14:38:02.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5313/.","accounts_in_message":[],"_revision_number":2},{"id":"866b73609183fbdded9e269ef495101f00152a47","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-09 14:39:05.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5490/.","accounts_in_message":[],"_revision_number":2},{"id":"844827d4ac14b8b586c68724997708e9e6afa2fa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-09 14:44:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/0e42f13d941e4c5aa263915caed9e511 : RETRY_LIMIT in 16m 51s\n- cinder-storpool-tempest-iscsi-multipath build/afcdb25594f64922ac4f469d836998b4 : RETRY_LIMIT in 17m 34s","accounts_in_message":[],"_revision_number":2},{"id":"1ee34cd4fd7b6ee41c2b4e5421750459ebef22c0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-09 14:55:53.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d01a3558aff649b2a152b3118efd8802/","accounts_in_message":[],"_revision_number":2},{"id":"90503a804c2f309f2f2af13a579a3b824a4351b9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-09 15:06:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I40bfec6e6c57f084cb6875421345061b2fd08599 .","accounts_in_message":[],"_revision_number":1},{"id":"4c05ec416ee8745e2c25f681458ef16b97a4d491","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-09 15:08:40.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I40bfec6e6c57f084cb6875421345061b2fd08599 .","accounts_in_message":[],"_revision_number":2},{"id":"67e2e732ec4752827a8d33811dd3742c221789ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:21:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05d6bc0ac6b6469a92cdd375bd43672b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1dc7c86e592d418c94dbd46f4723ed3a : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/84cc6e0a07024c569a14030d1ee2976e : SUCCESS in 22m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/14f8688adbc04605bb15c9fdc7ece817 : SUCCESS in 30m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a7c6bf7b962052ff541f62704abe420245d1ff3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-09 16:53:59.000000000","message":"Patch Set 2:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 46m 52s\n- FC       : SUCCESS in 0h 46m 17s\n- NVMe/TCP : SUCCESS in 0h 47m 0s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004816","accounts_in_message":[],"_revision_number":2},{"id":"0c60d1c28f78ab9787d9c4f58868d84c9565dbe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 17:18:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33150be6abdf459d945826a84e5a2e7d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63c17a52cf6b4c3c9b15897c03be79c8 : SUCCESS in 6m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd0b47fcddb44840bc45acb9ce234006 : SUCCESS in 11m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27bad366f94447b090410556434c9471 : SUCCESS in 11m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/41708f4a754442dabf880db24d80414b : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c54620da8254172b98df41600a52037 : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/2f05ce7ab52b4fd08cd3dcd13da4921a : SUCCESS in 1h 14m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d40ec5e8d12344bd95511a04e98681b6 : SUCCESS in 1h 13m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75ea83f86ec14c8b9b3f9062e4f2a76d : FAILURE in 1h 28m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67d042a81703403eba48962359de79e5 : SUCCESS in 56m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8547e79687fb40ee9ad74a8f5f22eba6 : SUCCESS in 10m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/97abf4e74a11476a9393f7ca1d8a3dac : TIMED_OUT in 41m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/219b8503a70f47cca8ecad16f701fb04 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee4865d766794922a1e7255b5e177012 : SUCCESS in 7m 39s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/438c01e62eb242278a495ea4084da294 : SUCCESS in 13m 38s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/58ae18bfb0034cb8bbc7c756fd9b538d : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/504d6c8f8f7b4ef3b928d290aa03dad2 : SUCCESS in 6m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/53b51f124a384f2094fcfd80b5504385 : SUCCESS in 6m 32s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/001cecbaffa54a66a39d5a3b6bc16f1e : FAILURE in 5m 38s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/2b4e32e803aa423391d0813471cae24e : SUCCESS in 8m 55s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eca62bae7f8047cca77b565e4c94dbe7 : SUCCESS in 35m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d6c4274281184723950109cc4912ce84 : FAILURE in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e91fd071d0648bc87df675607390e7f : SUCCESS in 57m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5ca4854c61794f2f980fd73f56f3fd5d : SUCCESS in 1h 25m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e16c6bd781f04fb8b6a0b4cab04a4c89 : SUCCESS in 2h 29m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/bb10f5b1d6c648dda60c69d1a105456f : FAILURE in 2h 31m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fd775da12e224a019586015623691686 : POST_FAILURE in 1h 52m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/78c7c5702367417fac03c8bbb01e553d : SUCCESS in 45m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2df6212cef7d4db190203cf25d068935 : SUCCESS in 1h 05m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8cb962d0befe420296788e54487df6d2 : SUCCESS in 1h 36m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9de6c4b25aea4167a95e00270343f672 : SUCCESS in 47m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c15c4c03173d4f87a87fb2ce03534297 : SUCCESS in 1h 35m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/17a80ed0c57a4011a38eca884ecb22ba : FAILURE in 1h 39m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6927f254251c4e93a7a4cd4751e3c00e : SUCCESS in 57m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"b106263d45ec66b18bb0870107c6e2fbe8b9292d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-09 18:06:39.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d01a3558aff649b2a152b3118efd8802\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/555da984ee7c4e5c8889da78e11444b0 : SUCCESS in 42m 45s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ae4d2a1d26f44f5aec2a14c3b5f1f78 : FAILURE in 1h 14m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/384e0ab9148343b6bcd4f243cedd75ab : SUCCESS in 48m 04s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/838f7440659c4d3483a50943c3fcad24 : SUCCESS in 35m 16s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0a9b01e1655049b9aed4291eee82e855 : FAILURE in 16m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4954daaf639fdce611e3b34d98e75705d8b000fc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-09-10 08:39:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f24674068be611d500d0b9b6a9744f5db1d4471b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-10 08:50:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-16-1004816-2-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-16-1004816-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1048b821bb5f79e2524e518da020aecf2b472a2f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-10 09:30:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/5c18438c59904a1da262f62b6fce8984 : SUCCESS in 49m 39s\n- cinder-storpool-tempest-iscsi-multipath build/a0681ab1d00e42b99892623e49b6662a : SUCCESS in 49m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c656b605bba3b384201634eb97b0886330a756f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 09:46:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73b52c57eac4a7ba7656d3988340dbc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/443528f46ac64acbbcd895c786e2da5a : SUCCESS in 18m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/09e15a0d77fd4fc0b1d752f496155d3d : SUCCESS in 21m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/51d8d61e94cb4e5cbf4599157d76cc5f : TIMED_OUT in 1h 01m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7ec77dece2d8dc4608187a49e5fec0dd8cddb22","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-10 10:35:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-16-1004816-2-fc : FAILURE in 5m 36s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-16-1004816-2-iscsi : SUCCESS in 1h 9m 57s","accounts_in_message":[],"_revision_number":2},{"id":"9aedb2c87759b6cd51421cba5f87849066b62e2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 11:14:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46655554f14e4c649cb88a0a67d059d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624490f93908462eb1df92a1105049e4 : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0941e6c6ecbe40b38640cacf5524580b : SUCCESS in 10m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6bfb2691384495f98dc97f4798a8705 : SUCCESS in 8m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0d0da1991d794bb6ad8a01e146222985 : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65c52fd639b4d94b79ac2b8484d1202 : SUCCESS in 14m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/e3de58b4582043f6bbc870a3d6fad848 : SUCCESS in 1h 05m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c35fa9b1c5f438698a54cf163bbb9a9 : SUCCESS in 1h 01m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ddcca04b413842208c14d597967a7763 : SUCCESS in 1h 00m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffb02775994148de9477a915b4713b59 : SUCCESS in 58m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5bb56d96a674c59b688bd18e7672c9f : SUCCESS in 8m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/32e4ccd27e6244fe98ec94a0ab103c07 : SUCCESS in 21m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d8f51627d28648aa8c98fa834d047be5 : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b89dfc61c154ceda11dbc7522783d4f : SUCCESS in 4m 51s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/beab188c21db4fd4b6e5238ebd3e1793 : SUCCESS in 12m 16s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/24de021aebee49e0bcfd14b7b316163c : FAILURE in 7m 34s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dfa645f8b47749ca85a400f5d2595420 : SUCCESS in 6m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2f6409448fba486e9d3f6c3d5a4d2870 : SUCCESS in 6m 22s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d823653364b547618d30260b92b4032c : FAILURE in 5m 58s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9bff9fd1c6d9443b99e842cd2539fc4a : SUCCESS in 6m 36s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/04de7a95a28b4580b22e5ad437e484a9 : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/101c33b067254f19bc5d9d8f7750c092 : FAILURE in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/631b5fb1a290453f9da1e11cc188f17c : SUCCESS in 1h 27m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/87b71fc911c844d38fab9ff185f4d3e7 : FAILURE in 2h 01m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/62ef82d043c840c0b801a7f7e1f267ef : SUCCESS in 1h 38m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/a550be958f5b439898f6bbf960091685 : FAILURE in 2h 05m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b3bae53566944df3bdf7ba4febbd6bcb : SUCCESS in 1h 52m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0547df27bf324545a8fb4e8c95fafe8f : SUCCESS in 37m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6f5d75ebb4034be7910a4d0209a9daf8 : SUCCESS in 1h 15m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b1e078380e3349db83e429fe8f595099 : SUCCESS in 1h 17m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f082f7b9fac9496596ae9b0558795680 : SUCCESS in 26m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c5d42fe9b37d4cdb9159107f910a7de6 : POST_FAILURE in 2h 24m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3a020e589dba440bb6ecea248cdbbe8f : SUCCESS in 1h 32m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abb2dbb42ebe4ac3896fb65ae5a5a052 : SUCCESS in 59m 27s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9d5989d5507d207e1ddca05b3f9b7f30ae95fd09","revisions":{"56667f7e3bb6d84ac65b09b3063bd53387fff2d7":{"kind":"REWORK","_number":1,"created":"2026-09-09 14:12:27.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/1004816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/1004816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/1004816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/1004816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/1004816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/1004816/1"}}},"commit":{"parents":[{"commit":"8e68009860b632fdaf9aaa3a3b7d2181c10d273b","subject":"Merge \"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e68009860b632fdaf9aaa3a3b7d2181c10d273b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-09-09 14:11:29.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-09-09 14:11:29.000000000","tz":120},"subject":"Add volume type name to volume usage notification","message":"Add volume type name to volume usage notification\n\nThis adds the volume type name as volume_type_name in\nthe volume usage notification payload so that one can\ncheck the volume type by name instead of UUID from the\nvolume_type_id when parsing notifications.\n\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nChange-Id: I40bfec6e6c57f084cb6875421345061b2fd08599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56667f7e3bb6d84ac65b09b3063bd53387fff2d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56667f7e3bb6d84ac65b09b3063bd53387fff2d7"}]},"branch":"refs/heads/master"},"9d5989d5507d207e1ddca05b3f9b7f30ae95fd09":{"kind":"REWORK","_number":2,"created":"2026-09-09 14:13:49.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/1004816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/1004816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/1004816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/1004816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/1004816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/1004816/2"}}},"commit":{"parents":[{"commit":"8e68009860b632fdaf9aaa3a3b7d2181c10d273b","subject":"Merge \"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e68009860b632fdaf9aaa3a3b7d2181c10d273b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-09-09 14:11:29.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-09-09 14:13:42.000000000","tz":120},"subject":"Add volume type name to volume usage notification","message":"Add volume type name to volume usage notification\n\nThis adds the volume type name as volume_type_name in\nthe volume usage notification payload so that one can\ncheck the volume type by name instead of UUID from the\nvolume_type_id when parsing notifications.\n\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nChange-Id: I40bfec6e6c57f084cb6875421345061b2fd08599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d5989d5507d207e1ddca05b3f9b7f30ae95fd09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d5989d5507d207e1ddca05b3f9b7f30ae95fd09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
