)]}'
{"id":"openstack%2Fcinder~1004920","triplet_id":"openstack%2Fcinder~master~I5550c0e3233141643a03cee287a70043c9dce3ae","project":"openstack/cinder","branch":"master","topic":"eventlet-removal","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-21 13:31:29.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5550c0e3233141643a03cee287a70043c9dce3ae","subject":"WIP: Remove eventlet from ssh_utils","status":"NEW","created":"2026-09-09 19:03:15.000000000","updated":"2026-09-21 16:05:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c9f539dc75f287732807e675b5675edc8ea6b2f8","_number":1004920,"virtual_id_number":1004920,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-21 15:56:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 19:14:25.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-09 19:21:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-09 19:22:54.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 19:38:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 19:57:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 19:58:00.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 22:04:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 15:56:40.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"87f1ee520500010d23a9d619245cd7e477f6dbad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-09 19:03:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acae9ed3e24289911a208b667ec3651309474355","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-09 19:03:34.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"b6145df24699caa090fae93b581730bd147fa5ea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-09 19:14:25.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5494/.","accounts_in_message":[],"_revision_number":1},{"id":"9d825d3276b4a090be7b48926c1e71ebdc01a344","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-09 19:14:28.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5317/.","accounts_in_message":[],"_revision_number":1},{"id":"9f89fd752ee1f484b05c9aa3a294b998419e180a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-09 19:21:36.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/97139893ab6e40caaf030e82f28fa085/","accounts_in_message":[],"_revision_number":1},{"id":"4b0c263d80e3e3ca59dd904662f460d990aa96cc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-09 19:22:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/ee24854e13c848e18ddd720a3eed9a46 : RETRY_LIMIT in 17m 13s\n- cinder-storpool-tempest-iscsi-multipath build/72529d63c155486cade1a23bae2b4679 : RETRY_LIMIT in 18m 10s","accounts_in_message":[],"_revision_number":1},{"id":"82d09dfaa9bf6ca7d0ea3ba68856cc41a44c73da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 19:38:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9de634d134c4b9c8e06e591d766bdbb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bd54de3fbc744395acdfe7be78247848 : SUCCESS in 17m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9845a6d2c5c4c99b7e87f17f26ee857 : SUCCESS in 19m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ec6fcfa57fc7404fa0c3cf86f09e9832 : SUCCESS in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e99ad7d22d693f3bd9bce2a787b3187e7c5debb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-09 19:51:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/97139893ab6e40caaf030e82f28fa085\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/459c10eaab594c538241c95dee69c936 : SUCCESS in 39m 36s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/65e8f83b71ad43e6aaf7191936653b14 : SUCCESS in 34m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/99cc000953864838b60aac3b649ee234 : SUCCESS in 47m 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/79db7375f232452a9efc35763231729a : SUCCESS in 31m 47s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7931fe0a5f27466190122f3243c97a5b : FAILURE in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7f0d72e5eae87848338ce8cfcef87c3af25fa07","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-09 19:57:45.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 49m 29s\n- FC       : SUCCESS in 0h 48m 26s\n- NVMe/TCP : SUCCESS in 0h 49m 43s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004920","accounts_in_message":[],"_revision_number":1},{"id":"d1aa96798fe3b1559a170a2c856543a31fe55435","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-09 19:58:00.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I5550c0e3233141643a03cee287a70043c9dce3ae .","accounts_in_message":[],"_revision_number":1},{"id":"80bd7f2959663c6e08a666b55978aac0446ea132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 22:04:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ef9271744f04a4b907b20e2ad2cab35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00dcb8a77a914d4581be83f5d8cfa7a2 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/624aa94d9e444010bf31d647ee7fe42f : SUCCESS in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b061db867e4545e0a855adb69a3f1908 : SUCCESS in 10m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8cd41521641426f8b41821b6c2f7011 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18b0415253a944da9840f2d40e637aab : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/09b52256c0584bc2a8cac03fa1f5f837 : SUCCESS in 35m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b98fd549f281464ebb0ce2a99a69a260 : SUCCESS in 1h 07m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35d05a32c157498cb930089ac410b3fa : SUCCESS in 1h 28m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/339a2963341b4b0099aa3c35106b9e03 : SUCCESS in 1h 01m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f39fb2936e949749d485e5d72cdc19a : SUCCESS in 10m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e3769a5dce424a59971f971c055a3ade : FAILURE in 27m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2255371c537448a7bcb49cb23f87701d : SUCCESS in 6m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8574a3e6481949e68da2ecbc623f073a : SUCCESS in 6m 35s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8f0f5a9b50a249a0899ce7b4cf9bcc7c : SUCCESS in 9m 01s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/05eb9238cbb948ac8a30aae5406aca91 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/58ce1d7531254cb2930b563dcfd3db02 : SUCCESS in 5m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7f0bd54a5d1041fcba3efc2129e5de46 : SUCCESS in 5m 54s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/2da7cd91e11f4c7facf05dc77710798d : FAILURE in 5m 27s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d73e2107dd7d431e8684c8365d28eee6 : SUCCESS in 10m 00s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/995f87868d784271aa4f1a310c0b2bf8 : SUCCESS in 39m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aea51cc49d564f97b1609646cdfff0b1 : FAILURE in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b67ed1f224124028b4c285bf77e1ee3c : SUCCESS in 1h 12m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f5b36083b8064e4f96f5ad83c11ee249 : FAILURE in 1h 56m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/31009e5c31584a639f033419179db75e : SUCCESS in 2h 17m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e082045fa2ab40bfadb6e3c5475ff394 : FAILURE in 2h 50m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e234b9a076494046b5b1567c20467e2c : SUCCESS in 2h 06m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ee0c41f25ce04cbda653b6add9f2ccf8 : SUCCESS in 36m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ed01b51987584f64962a3e0ffb550941 : SUCCESS in 1h 16m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d77f074c8a8e4e898f22b4b35a8c49e7 : SUCCESS in 1h 32m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9f27e9c477a94ef18f46307bd993c46d : SUCCESS in 50m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/06de79b154744917b29fc1e915c3f208 : FAILURE in 52m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc15236f74f4478781197783b0bf1043 : SUCCESS in 57m 25s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"1ab8c027c01297709603947eab6e61d450862708","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-21 13:14:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2987dc4d8d0d21b793789c286a1be13f78666d0a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-21 13:31:29.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/6db0a22d0cde4def80e5ef267d351773/","accounts_in_message":[],"_revision_number":2},{"id":"227a12d9cd0f82bee81c2fcf6e5ea1f7fbec087e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:40:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab89e16adb664e3bb987240d5f1a3efb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/474e7da04cfe4e70bf92b0d1dae87984 : SUCCESS in 16m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/68ee1df45d124facb4d87e4fae2d9c6c : SUCCESS in 17m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/164f41a6f92e4847aa0abe9e3853d37a : SUCCESS in 23m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9359c3d8f6612e208b3fc28afa5e463256e81482","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-21 13:58:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6db0a22d0cde4def80e5ef267d351773\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ccdf19ac219242b8a4d0c4832a123941 : SUCCESS in 42m 11s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/772986d649764c309b1be83113615c16 : SUCCESS in 38m 08s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a3937a6931c4b71b307bc26f5c9f984 : SUCCESS in 43m 08s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0043f249a93c458bafa85aaef944ac72 : SUCCESS in 35m 19s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3127f5de4cd3482a9b62f6df991e3a86 : FAILURE in 12m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"101f73bd72b546160b519747cb522c8d75b37104","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-21 14:09:32.000000000","message":"Patch Set 2:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 49m 36s\n- FC       : SUCCESS in 0h 49m 28s\n- NVMe/TCP : SUCCESS in 0h 48m 33s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004920","accounts_in_message":[],"_revision_number":2},{"id":"39151108e7a2b23887223ad26eb697cfcfc8a8aa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-21 14:09:45.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I5550c0e3233141643a03cee287a70043c9dce3ae .","accounts_in_message":[],"_revision_number":2},{"id":"ca05c4a5360443bf9d599fbff5fa27cb35cca37f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-21 14:12:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/7923df47d0e44f1a87ad6a062335da5c : SUCCESS in 44m 54s\n- cinder-storpool-tempest-iscsi-multipath build/c892d7e92fc146d792884d07d076d8e3 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":2},{"id":"74dcb5de7fae496f582ec9c4eeaaa709f2bfcb44","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-21 14:15:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5615/.","accounts_in_message":[],"_revision_number":2},{"id":"fe30270acdb986e5cf65df46e41c19ac6b43a08e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-21 14:19:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5438/.","accounts_in_message":[],"_revision_number":2},{"id":"50b021c7db9167883834399dd9f653d44b0bd882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:56:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cb81ab631764ee590d0144aba3c9ee1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239446dad53a4df886aa4e18202b6b1f : SUCCESS in 6m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fffe7f2fad494e21ad1fef5e7afec66f : SUCCESS in 9m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8af2f3ba8161473f86728fd823a4f08a : SUCCESS in 10m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fd556c4bf0ee43f889fedc70475a2522 : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8271eb3872264f2d9ddb9b1c7861380b : SUCCESS in 14m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/e1e18737b98f41eb87c7841f90d43f6c : SUCCESS in 1h 02m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1fa809c6eda74d8eba7c72a83de8c399 : SUCCESS in 1h 06m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4eb59ff64419410ea9a3ef606ea1be3e : SUCCESS in 57m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7eb2a764905474f996e5e3c802742f4 : SUCCESS in 1h 07m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/477613d706324391976cfcb0b69c4b4d : SUCCESS in 9m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7a5c256980aa45379d627d14543a09b5 : SUCCESS in 18m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/95f045bacc8845128f45f04e41f508e3 : SUCCESS in 5m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e9440dfa15df4d6e8ec343b2c581a529 : SUCCESS in 6m 33s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c5861afced2b4d4b87f48f007bd613fb : SUCCESS in 9m 23s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/16aa13d7189a4907be2717e8efafe014 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e663a3b3ef024ea49e737883a4dcff83 : SUCCESS in 6m 54s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/42f93850c4124d76892b7f2103e54042 : SUCCESS in 5m 40s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9bc94bf68083492587ae69cff7e30fcd : FAILURE in 4m 06s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/77ee3562494745548236ae65475a5879 : SUCCESS in 7m 46s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5f56ce8b52bc42efb7ca6f0dca571a5a : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de74787dc7ea424d808ebcd41df828d8 : FAILURE in 4m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2d218b291b14859b90ff6780b77b7ea : SUCCESS in 1h 35m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7a16d10636644f4e8a9523863b69dec5 : FAILURE in 2h 27m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/508cb8b26aac40e9a8b8dc20c697fea1 : SUCCESS in 1h 59m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/bf80c49fcff64c9494b82178af90dc33 : FAILURE in 2h 05m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b39d3a238a694afeabe21d025e6ae114 : SUCCESS in 1h 50m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/78cfb057ac5a47ca9c19b7c41f9a0110 : SUCCESS in 37m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6b89c0794100410484d92d056348bc98 : SUCCESS in 1h 21m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d960a0cb2ea748b3a9fe396b880340c6 : SUCCESS in 1h 07m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8e4a3c91c9c045068e8202f3664a523d : SUCCESS in 31m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d93bfeef7a08413fafac2269e017bce5 : SUCCESS in 1h 42m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/886f029d15254b97aeddb3e79f96c947 : SUCCESS in 55m 55s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"c9f539dc75f287732807e675b5675edc8ea6b2f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-21 16:05:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_09_2026_16_00_12\nLeave a comment with","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"814c01c8cec818216ba05ba3e3854029fb06c293","revisions":{"c53dcaaa2c60478237bb8ea0b61c80147344de36":{"kind":"REWORK","_number":1,"created":"2026-09-09 19:03:15.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/20/1004920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/1004920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/1004920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/1004920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/1004920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/1004920/1"}}},"commit":{"parents":[{"commit":"31ec30d44f086bcd2e60fad0c4cbc461a1044518","subject":"Merge \"[SVf]: Enable support to extend a volume with snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31ec30d44f086bcd2e60fad0c4cbc461a1044518"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-09 19:00:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-09 19:02:20.000000000","tz":-240},"subject":"WIP: Remove eventlet from ssh_utils","message":"WIP: Remove eventlet from ssh_utils\n\nCreate a new Pool class that replaces eventlet\u0027s\nPool with similar semantics. This allows using our\nSSHPool in threaded mode w/o eventlet.\n\nGenerated-By: Claude Code Opus 4.8\nChange-Id: I5550c0e3233141643a03cee287a70043c9dce3ae\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c53dcaaa2c60478237bb8ea0b61c80147344de36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c53dcaaa2c60478237bb8ea0b61c80147344de36"}]},"branch":"refs/heads/master"},"814c01c8cec818216ba05ba3e3854029fb06c293":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-21 13:14:57.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/20/1004920/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/1004920/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/1004920/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/1004920/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/1004920/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/1004920/2"}}},"commit":{"parents":[{"commit":"6c8eb67a8574c459f4edbcb94a524f1ea6129395","subject":"Merge \"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c8eb67a8574c459f4edbcb94a524f1ea6129395"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-09 19:00:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-21 13:14:57.000000000","tz":0},"subject":"WIP: Remove eventlet from ssh_utils","message":"WIP: Remove eventlet from ssh_utils\n\nCreate a new Pool class that replaces eventlet\u0027s\nPool with similar semantics. This allows using our\nSSHPool in threaded mode w/o eventlet.\n\nGenerated-By: Claude Code Opus 4.8\nChange-Id: I5550c0e3233141643a03cee287a70043c9dce3ae\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/814c01c8cec818216ba05ba3e3854029fb06c293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/814c01c8cec818216ba05ba3e3854029fb06c293"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"31ec30d44f086bcd2e60fad0c4cbc461a1044518","ours":"c53dcaaa2c60478237bb8ea0b61c80147344de36","theirs":"6c8eb67a8574c459f4edbcb94a524f1ea6129395","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
