)]}'
{"id":"openstack%2Fcinder~1004996","triplet_id":"openstack%2Fcinder~master~I1d482bf01ffd1489da9adb1c00de528f02707a86","project":"openstack/cinder","branch":"master","attention_set":{"37888":{"account":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"last_update":"2026-09-22 12:38:36.000000000","reason":"Someone else replied on the change"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-09-10 11:48:01.000000000","reason":"\u003cGERRIT_ACCOUNT_37888\u003e replied on the change","reason_account":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"}},"38319":{"account":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"last_update":"2026-09-10 11:48:01.000000000","reason":"\u003cGERRIT_ACCOUNT_37888\u003e replied on the change","reason_account":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1d482bf01ffd1489da9adb1c00de528f02707a86","subject":"[IBM SVf] Add NVMe/TCP host attach support for Storwize/SVC driver","status":"NEW","created":"2026-09-10 11:46:08.000000000","updated":"2026-09-22 12:38:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"332e8bd56fdcc2bc97a39bbde5f4cc29ccfdcd95","_number":1004996,"virtual_id_number":1004996,"owner":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-22 12:36:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 11:48:01.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"state":"REVIEWER"},{"updated":"2026-09-10 11:48:01.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2026-09-10 12:05:44.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-10 12:39:53.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 12:40:00.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 12:40:18.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 12:44:43.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-10 13:00:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 14:12:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0258d32558f628246b8747fe86b9b2ee9d693d08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2026-09-10 11:46:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2644edc3b6d3f36f688cc3193b743a7e39495121","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 12:05:44.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/6185534ec88b471c8b6529611c1a3cfc/","accounts_in_message":[],"_revision_number":1},{"id":"ac0470e47c6041a171a4c5df65a27e1e8a7b823a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 12:37:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6185534ec88b471c8b6529611c1a3cfc\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5f44b54493bf40deb8120bc2d58ad9ba : SUCCESS in 47m 38s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b00c7ec919b44160acb256cd6105f04e : SUCCESS in 35m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/86ec76a8513c486d829c790e32b24435 : SUCCESS in 50m 48s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6b9478203634a62abfc1ac8035463c4 : SUCCESS in 32m 57s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/87d77fb74b4144b9ae4f4ea692924eac : FAILURE in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fe06467d2b7051a99ec252df68ee769e022d740","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-10 12:39:53.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1d482bf01ffd1489da9adb1c00de528f02707a86 .","accounts_in_message":[],"_revision_number":1},{"id":"03fb8d162200f84485563bf48524fbf539a0582e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-10 12:40:00.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 10s\n- FC       : SUCCESS in 0h 49m 1s\n- NVMe/TCP : SUCCESS in 0h 48m 2s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004996","accounts_in_message":[],"_revision_number":1},{"id":"8c99aee16adc4266b6f29f2715fb5802a674be1e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-10 12:40:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/f82a42da4b4c4553a609268d59ded37d : SUCCESS in 47m 56s\n- cinder-storpool-tempest-iscsi-multipath build/b47c66bee43f4c3a86d6924e6ad0668b : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"842df2c0c64d25a8cd8bde1c5a4306c1551465a9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-10 12:44:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5319/.","accounts_in_message":[],"_revision_number":1},{"id":"9f0aa69be12b06b3efec3362698137236e30e149","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-10 12:45:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5496/.","accounts_in_message":[],"_revision_number":1},{"id":"bc3b72157831d807aadedd9f416b8a6e50c0f98b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:00:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25d67b02755d49fea9de194c3b98b6dc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/703074180e8942268710d26cddd7e2f7 : FAILURE in 15m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/462f6ccaa16841a0b7ce511b3e7d8634 : FAILURE in 14m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0bbb2786bf4749b4ba584b0559288428 : FAILURE in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38c68aeddc977c98703bae4a0c65fd25781c7191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:12:45.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f5b782004d1449eb64e0aa4dda45cf1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd5af97b29dc4315bf7bc3c75aec40e7 : FAILURE in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54371ca44e83453785a394f402662d22 : FAILURE in 7m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eca84152b47b4521927bef9ad3145f3b : FAILURE in 10m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/25698f130746411db71b49d4aec57a6e : FAILURE in 16m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467d8a4645a54e8c8090ed0a4756930b : FAILURE in 5m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/acdd6c78025d4fbb8bbb45fbb4e4b98c : SUCCESS in 59m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/53095534b8c34efcbdbc1ef005d1f2c5 : SUCCESS in 1h 08m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ef769ae19504b9aa0e377f39ab8a6d7 : SUCCESS in 1h 22m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8a18cd89e84441b85a1448772823c74 : SUCCESS in 58m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed832fb5801347a4badc54354d066636 : SUCCESS in 10m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ba273aa87eca4691bd29b670677e197c : TIMED_OUT in 42m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9b48cee5c6c040cc8c1f5f99b437e905 : SUCCESS in 6m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/31e59157576448b0bfd066a845c3fcc9 : SUCCESS in 7m 25s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/52db623aaa0c46698efa5d537185260c : FAILURE in 8m 57s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0882c2be8cc845188ecb10ae7ceb276f : FAILURE in 9m 36s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b5d1177251984c9787bf48e2375c5ef4 : SUCCESS in 8m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/55ec9de306f341e48d47a4beca248d89 : SUCCESS in 6m 26s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/af19d0f231c24d68a65a6ebd85f756b7 : FAILURE in 8m 39s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c70b436abdab460a909335e558c3ce52 : SUCCESS in 8m 49s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cf45c52666204dff9dd9cf00d750d2af : SUCCESS in 44m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b8fea677d2394f66aba656c7c5c66158 : FAILURE in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/273c37f3598b4ddd8ca8b2d95dcb463d : SUCCESS in 54m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d3af1cf2895840269fb4be2cc1c91f95 : FAILURE in 2h 05m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ea9318254ad42459b6d890d619d898d : SUCCESS in 2h 18m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/48b842c3e8c249af9b054ec8960d36bd : FAILURE in 2h 11m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b861c2a755a44a218e05564696c65727 : FAILURE in 1h 57m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/66cce620e153433a9d0b89c2164abd7a : SUCCESS in 40m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6d2fac88a2674c11aacb9ddc382e3e00 : SUCCESS in 1h 15m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2cd7e2733e7847919aae241a714769cd : SUCCESS in 1h 28m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8730558ce88d404e866342ebd6695e61 : SUCCESS in 48m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/111362791c204cdcac3e8c195013e53f : SUCCESS in 1h 55m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfb60ae57df242b9942b074a6d3ade12 : SUCCESS in 56m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"3fe7082e4d8f086e8d77680ac3735f637054f7ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2026-09-11 07:35:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b8ea119347adcaa9016fbb87ffd5034262c447eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 07:57:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec624b2797d44ab48ad8ab459ae3288d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/09ce6ce6bdef45d4bc6564914ea8c93e : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/49dd89bfa9ca4bb68bb26504f41f0791 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d980f343fa3043fdbc344f6aca86f8d5 : SUCCESS in 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"75e70a0a9665def7cf20df96ec340806cf059338","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-11 08:39:44.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b787dbf2793a4f63a354cf724968d71b/","accounts_in_message":[],"_revision_number":2},{"id":"48bd7051111035a088d80d00581348637f25609b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-11 08:49:09.000000000","message":"Patch Set 2:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 56m 57s\n- FC       : FAILURE in 0h 55m 19s\n- NVMe/TCP : FAILURE in 0h 53m 35s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004996","accounts_in_message":[],"_revision_number":2},{"id":"b9c82a34d49a4e0c7291ea2bad75b81868aa3176","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-11 09:00:00.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1d482bf01ffd1489da9adb1c00de528f02707a86 .","accounts_in_message":[],"_revision_number":2},{"id":"7f6c19e8acaa8917f0419c422c5328364ec6fbbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 09:40:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3d530beab4642fa812630da41358c86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e4cda2634348e5be1931e7245670e3 : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05bc6cc1ce314992b12aff7b5f521bd9 : SUCCESS in 11m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5efc774a2f754707ba0c6698de6c7a62 : SUCCESS in 9m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/74eb13689ba64ba49ddaeb15d051b53a : SUCCESS in 15m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa0fc5ed9ab4b35bf265114c6cb6385 : FAILURE in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/335c8e4ad5fa44b389a0637a3f3040f7 : SUCCESS in 1h 08m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/428c29275b114a15891c9d41d9624f2c : SUCCESS in 35m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b2126060a7564b09a09686d45bf88254 : SUCCESS in 1h 21m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/531bb59abd6848289fa1afa8b1921349 : SUCCESS in 1h 01m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a0b2f4fb57042f0b0de362d01805424 : SUCCESS in 8m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6e5a57ea828543ee8da172e4da37f48d : SUCCESS in 17m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fd8220207b8c4abf82fa63b0d787e307 : SUCCESS in 3m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9e75a8e9ced84efd9347282dc4e777e3 : SUCCESS in 5m 57s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/4c8d9ef73e7f423f8b5ed24a12e64a40 : FAILURE in 19m 34s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6ea26ee2e3814a8a89478944089d6c07 : SUCCESS in 12m 56s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/86b511a0545a4370bb189bfd691c6779 : SUCCESS in 6m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c6a6836a14824cec872aa0fb6faba832 : SUCCESS in 5m 58s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3388f47ed95c40afbf391ae146c65408 : FAILURE in 7m 32s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f52147dbaabb4ee9b62fc8532f298221 : SUCCESS in 9m 20s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/801fc7bd6fe34785a88effed4c7def12 : SUCCESS in 25m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3fa2b69fe0464a7d98f798cc14fbb835 : FAILURE in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/77cb0c1b61c348c1bd2eae3666f4c8f6 : SUCCESS in 1h 30m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7051a698515a42a0b257891d3359762c : FAILURE in 1h 57m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/909f5cf78f4d4126b3ca15cf9c1a4a34 : SUCCESS in 1h 18m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/37c381d1ba334958a62c7ae7ad894e59 : FAILURE in 1h 58m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ec427ebc93ed44b9a4a0141a50a92e56 : SUCCESS in 1h 33m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1516ca3cef0d42f6b65be5110dc81c1f : SUCCESS in 37m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/07455ee30f49422db09af118ec58a48e : SUCCESS in 1h 16m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/919c8e389c504958a3e62e9249f331db : SUCCESS in 1h 36m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/902eea8fa46f469abb8d21d1963e6d85 : SUCCESS in 37m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40faf361220f42c9a8bc977417877b64 : SUCCESS in 1h 31m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/decb4d4457114a0b9b718456a058b5f9 : SUCCESS in 55m 42s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"6031f90a38760b9576ab4d5d76170090b6a7506f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-11 10:44:02.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b787dbf2793a4f63a354cf724968d71b\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/38b832c302c949dbb229bd871287a4f2 : SUCCESS in 1h 28m 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3ae27b4139a044e8bd56cb71d14636ea : FAILURE in 2h 06m 09s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93942c8e1d0c4e7d89f598f4ecd42554 : SUCCESS in 1h 06m 20s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a038bb79220f419d8859d80d66e352fa : SUCCESS in 1h 03m 24s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/761e1a42f8f84c52904ece6983ec869a : FAILURE in 42m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c76f72ed0880d557a0fa5268dbd2b5685a4d0f6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-11 11:13:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/1a6178e3e9024ef88820548fde0f5a0f : SUCCESS in 2h 00m 20s\n- cinder-storpool-tempest-iscsi-multipath build/7c0620a0faf64660b4717bf30884121e : SUCCESS in 1h 11m 49s","accounts_in_message":[],"_revision_number":2},{"id":"53dc859aa59591fc41720fbb7862de53855fde0a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-11 11:46:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5514/.","accounts_in_message":[],"_revision_number":2},{"id":"12095f1255ff4a546e8586e2a24995f41935d65d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-11 13:13:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5337/.","accounts_in_message":[],"_revision_number":2},{"id":"8825202f5edc2e711666ddb6dc3a48f0b6155711","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2026-09-22 10:16:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"2687a542e66fd954b70fdff2faab7d4cb36ea3a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2026-09-22 10:20:31.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"090ec9af1290d0bbc5193fab3d3205253741dd2a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-22 10:38:35.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d67412c3b3ed4953a4467993c76a3001/","accounts_in_message":[],"_revision_number":4},{"id":"f5f293e8fdb3c3fd294f45915994a140198a9d1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 10:46:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f022c21aaa34e4783c399d66ae3c4c2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c07cc02016564a4b849bc84952cfa4c3 : FAILURE in 17m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/282efd57da3d42558eec587af89e65fd : FAILURE in 16m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b84f66163bfa4b2b94a91a45ea7b1c6a : FAILURE in 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d4978c9eef2bbb5d793095c5518ea56cb4940ae4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-22 11:12:04.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1d482bf01ffd1489da9adb1c00de528f02707a86 .","accounts_in_message":[],"_revision_number":3},{"id":"efc8d8a2a4ac0665d4f632c4b699d9c07102e8e4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-22 11:13:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/9763ea0004224dfab47fab3dc371b84d : SUCCESS in 47m 30s\n- cinder-storpool-tempest-iscsi-multipath build/f57c20bc89c64ffc8027e9bed70101fb : SUCCESS in 51m 57s","accounts_in_message":[],"_revision_number":4},{"id":"c8094e6d3f1fde1f11593b78b124d6427fcd3164","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-22 11:14:44.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1d482bf01ffd1489da9adb1c00de528f02707a86 .","accounts_in_message":[],"_revision_number":4},{"id":"04f35ae5e4c2694ead3245e97ddfbbdb604fc419","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-22 11:16:36.000000000","message":"Patch Set 4:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 49m 2s\n- FC       : SUCCESS in 0h 48m 54s\n- NVMe/TCP : SUCCESS in 0h 47m 5s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004996","accounts_in_message":[],"_revision_number":4},{"id":"83814fe282beb03d6f034620c2aa666ea1037e32","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-22 11:24:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5626/.","accounts_in_message":[],"_revision_number":3},{"id":"1a767e6b0b24a7ef0e077326d039ff1e80472801","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-22 11:30:28.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5449/.","accounts_in_message":[],"_revision_number":3},{"id":"167033ae951589279d8d7091ac1638099576492e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-22 12:22:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5627/.","accounts_in_message":[],"_revision_number":4},{"id":"2d00bd571a57d3b06fabfa1cbf2326e7ad62b852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 12:36:53.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed3c238025cf48f7b6deac42187cc110\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59bd720fe727477a982fcc57336bcc31 : FAILURE in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e5628d116a4472bbc8b665a1fff2067 : FAILURE in 11m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77bb53f4fc684af2a46bf7306925f71b : FAILURE in 10m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d7833ec618ec430f972f2c0aaffcdbe2 : FAILURE in 13m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4f5bea538b464b99ba1df16fc5c67a : FAILURE in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/6d16f85968aa41b1b9cdf0455bdddb8c : SUCCESS in 1h 05m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db445f01f1ee4e33831dd3066ef57e99 : SUCCESS in 1h 04m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3c2023a21de24b15aaf64313da8793f6 : SUCCESS in 1h 21m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/996bf36714674b78a78e11e30e37d93c : SUCCESS in 56m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9863828df7b46dda985af23935ea985 : SUCCESS in 5m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a9b16fdfd75e4ad090075c1dad3f3496 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/de11948282d6433d8847b4c71190aceb : SUCCESS in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/94e1ef337f534ec9abf5f794bce4dc6e : SUCCESS in 3m 24s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/b798ace815814ffa96c9db6e5d9f8f72 : FAILURE in 8m 32s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/88bc2e0a16474544aa4f29860a3114d9 : FAILURE in 10m 54s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ee2516b285f84144ab8adae8dfe92e59 : SUCCESS in 6m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d9450d97fa744f8780ff068e2e4a7c2c : SUCCESS in 5m 54s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/be416f73f7404b149e74f0d07e344743 : FAILURE in 4m 30s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/368560d97e6b4425bab7789a9e73ce34 : SUCCESS in 7m 08s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5197e3efc5c2404eb5e40dc8e48de2b2 : SUCCESS in 36m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6dd8a98748a04c8fbba1bc9a29b49a0f : FAILURE in 3m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/765021b261f14b0b93e6b6d0e79f3afb : SUCCESS in 46m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7079a4e6cbb1491b9e073706805db5b6 : FAILURE in 2h 09m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/24dad9fbc1544cf2b033b29f0d7b00d5 : SUCCESS in 1h 26m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/fbb5c3fdbda74cff8460534e8299c495 : FAILURE in 2h 13m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9a8fef6fce394abe8ae5e87ce3ee382a : SUCCESS in 1h 50m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9c7f3f89fb784adeb9f06b7e571cfb2a : SUCCESS in 33m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5260008b3b02490f9125c14a2caca30d : SUCCESS in 57m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2b81f958566b462f914189f366ff5282 : SUCCESS in 1h 09m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/817a1cdb667243669c881968bc9104f8 : SUCCESS in 25m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/145fa9af46c24245b0567a24445e9e1b : SUCCESS in 1h 14m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2135693f45648db9e2b5487cb7589a1 : SUCCESS in 26m 36s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"332e8bd56fdcc2bc97a39bbde5f4cc29ccfdcd95","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-22 12:38:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5450/.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"603bca00320e886cc9c933f324bd694a48358867","revisions":{"75dc81299ebe117c5c1135dbf93b9027f481af56":{"kind":"REWORK","_number":1,"created":"2026-09-10 11:46:08.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/96/1004996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/1004996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/1004996/1"}}},"commit":{"parents":[{"commit":"31ec30d44f086bcd2e60fad0c4cbc461a1044518","subject":"Merge \"[SVf]: Enable support to extend a volume with snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31ec30d44f086bcd2e60fad0c4cbc461a1044518"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2026-09-10 11:32:57.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2026-09-10 11:45:41.000000000","tz":330},"subject":"[IBM SVf] Add NVMe/TCP host attach support for Storwize/SVC driver","message":"[IBM SVf] Add NVMe/TCP host attach support for Storwize/SVC driver\n\nImplements NVMe over TCP protocol support for IBM Storwize family\nand SVC storage systems running firmware 8.4.2.0 or later.\n\nA new driver class StorwizeSVCNVMeTCPDriver is introduced alongside\nan abstract base class StorwizeSVCNVMeBaseDriver, both building on\nthe existing StorwizeSVCCommonDriver infrastructure. The common\nlayer is extended with portset-driven NVMe/TCP IP discovery, NQN-\nbased host caching, and a new SSH wrapper for NVMe namespace mapping.\n\nSupported features are VGS (Volume Group Snapshot), PBR (Policy-Based\nReplication), multipath via NVMe-oF ANA, and multi-attach.\nHyperSwap, Metro Mirror, Global Mirror, and GMCV are explicitly\nnot supported and raise InvalidInput if requested.\n\nImplements: blueprint storwize-nvme-tcp-driver\nChange-Id: I1d482bf01ffd1489da9adb1c00de528f02707a86\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75dc81299ebe117c5c1135dbf93b9027f481af56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75dc81299ebe117c5c1135dbf93b9027f481af56"}]},"branch":"refs/heads/master"},"7f2df9f22d3b9cdf808039f57d5d85d6daf6484e":{"kind":"REWORK","_number":2,"created":"2026-09-11 07:35:52.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/96/1004996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/1004996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/1004996/2"}}},"commit":{"parents":[{"commit":"31ec30d44f086bcd2e60fad0c4cbc461a1044518","subject":"Merge \"[SVf]: Enable support to extend a volume with snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31ec30d44f086bcd2e60fad0c4cbc461a1044518"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2026-09-10 11:32:57.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2026-09-11 07:35:31.000000000","tz":330},"subject":"[IBM SVf] Add NVMe/TCP host attach support for Storwize/SVC driver","message":"[IBM SVf] Add NVMe/TCP host attach support for Storwize/SVC driver\n\nImplements NVMe over TCP protocol support for IBM Storwize family\nand SVC storage systems running firmware 8.4.2.0 or later.\n\nA new driver class StorwizeSVCNVMeTCPDriver is introduced alongside\nan abstract base class StorwizeSVCNVMeBaseDriver, both building on\nthe existing StorwizeSVCCommonDriver infrastructure. The common\nlayer is extended with portset-driven NVMe/TCP IP discovery, NQN-\nbased host caching, and a new SSH wrapper for NVMe namespace mapping.\n\nSupported features are VGS (Volume Group Snapshot), PBR (Policy-Based\nReplication), multipath via NVMe-oF ANA, and multi-attach.\nHyperSwap, Metro Mirror, Global Mirror, and GMCV are explicitly\nnot supported and raise InvalidInput if requested.\n\nImplements: blueprint storwize-nvme-tcp-driver\nChange-Id: I1d482bf01ffd1489da9adb1c00de528f02707a86\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f2df9f22d3b9cdf808039f57d5d85d6daf6484e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f2df9f22d3b9cdf808039f57d5d85d6daf6484e"}]},"branch":"refs/heads/master"},"b31e65474859cbeebbf43e62c7dd5e0b8a32b140":{"kind":"REWORK","_number":3,"created":"2026-09-22 10:16:46.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/96/1004996/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/1004996/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/1004996/3"}}},"commit":{"parents":[{"commit":"e5c1549f4d550d0193bc5f01f61d4b6aba0a3ad4","subject":"Merge \"api-ref: Explain why consistency groups are deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5c1549f4d550d0193bc5f01f61d4b6aba0a3ad4"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2026-09-10 11:32:57.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2026-09-22 10:16:28.000000000","tz":330},"subject":"[IBM SVf] Add NVMe/TCP host attach support for Storwize/SVC driver","message":"[IBM SVf] Add NVMe/TCP host attach support for Storwize/SVC driver\n\nImplements NVMe over TCP protocol support for IBM Storwize family\nand SVC storage systems running firmware 8.4.2.0 or later.\n\nA new driver class StorwizeSVCNVMeTCPDriver is introduced alongside\nan abstract base class StorwizeSVCNVMeBaseDriver, both building on\nthe existing StorwizeSVCCommonDriver infrastructure. The common\nlayer is extended with portset-driven NVMe/TCP IP discovery, NQN-\nbased host caching, and a new SSH wrapper for NVMe namespace mapping.\n\nSupported features are VGS (Volume Group Snapshot), PBR (Policy-Based\nReplication), multipath via NVMe-oF ANA, and multi-attach.\nHyperSwap, Metro Mirror, Global Mirror, and GMCV are explicitly\nnot supported and raise InvalidInput if requested.\n\nImplements: blueprint storwize-nvme-tcp-driver\nChange-Id: I1d482bf01ffd1489da9adb1c00de528f02707a86\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b31e65474859cbeebbf43e62c7dd5e0b8a32b140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b31e65474859cbeebbf43e62c7dd5e0b8a32b140"}]},"branch":"refs/heads/master"},"603bca00320e886cc9c933f324bd694a48358867":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-09-22 10:20:31.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/96/1004996/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/1004996/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1004996/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/1004996/4"}}},"commit":{"parents":[{"commit":"e5c1549f4d550d0193bc5f01f61d4b6aba0a3ad4","subject":"Merge \"api-ref: Explain why consistency groups are deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5c1549f4d550d0193bc5f01f61d4b6aba0a3ad4"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2026-09-10 11:32:57.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2026-09-22 10:20:12.000000000","tz":330},"subject":"[IBM SVf] Add NVMe/TCP host attach support for Storwize/SVC driver","message":"[IBM SVf] Add NVMe/TCP host attach support for Storwize/SVC driver\n\nImplements NVMe over TCP protocol support for IBM Storwize family\nand SVC storage systems running firmware 8.4.2.0 or later.\n\nA new driver class StorwizeSVCNVMeTCPDriver is introduced alongside\nan abstract base class StorwizeSVCNVMeBaseDriver, both building on\nthe existing StorwizeSVCCommonDriver infrastructure. The common\nlayer is extended with portset-driven NVMe/TCP IP discovery, NQN-\nbased host caching, and a new SSH wrapper for NVMe namespace mapping.\n\nSupported features are VGS (Volume Group Snapshot), PBR (Policy-Based\nReplication), multipath via NVMe-oF ANA, and multi-attach.\nHyperSwap, Metro Mirror, Global Mirror, and GMCV are explicitly\nnot supported and raise InvalidInput if requested.\n\nChange-Id: I1d482bf01ffd1489da9adb1c00de528f02707a86\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/603bca00320e886cc9c933f324bd694a48358867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/603bca00320e886cc9c933f324bd694a48358867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
