)]}'
{"id":"openstack%2Fcinder~1004999","triplet_id":"openstack%2Fcinder~master~I0fd46412be7a8f7520d5a57692cf06005d46fbdf","project":"openstack/cinder","branch":"master","attention_set":{"37888":{"account":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"last_update":"2026-09-10 12:17:32.000000000","reason":"\u003cGERRIT_ACCOUNT_38319\u003e replied on the change","reason_account":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"}},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2026-09-10 12:17:32.000000000","reason":"\u003cGERRIT_ACCOUNT_38319\u003e replied on the change","reason_account":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"}},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-09-10 12:17:32.000000000","reason":"\u003cGERRIT_ACCOUNT_38319\u003e replied on the change","reason_account":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"}},"38319":{"account":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"last_update":"2026-09-10 14:56:23.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0fd46412be7a8f7520d5a57692cf06005d46fbdf","subject":"Fixed IBM Storwize/FlashSystem GMCV volume resize failure by waiting for replication relationship synchronization and added configurable sync timeout and polling options.","status":"NEW","created":"2026-09-10 12:10:00.000000000","updated":"2026-09-10 14:56:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e204d77d020287dcc973cffd13b520dffe7afdcf","_number":1004999,"virtual_id_number":1004999,"owner":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-10 14:56:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 12:17:32.000000000","updated_by":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"REVIEWER"},{"updated":"2026-09-10 12:17:32.000000000","updated_by":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"reviewer":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"state":"REVIEWER"},{"updated":"2026-09-10 12:17:32.000000000","updated_by":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2026-09-10 12:45:05.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-10 12:55:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 13:03:21.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 13:32:58.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 13:33:12.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 13:44:54.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-10 14:56:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6ce05a5d658edb3b464d01e4090179b499db9c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"date":"2026-09-10 12:10:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3087b87c909fcc6840cb2c808515176f30d7cda3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 12:45:05.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/14b8b91969fa439daeebc4d85c59d62d/","accounts_in_message":[],"_revision_number":1},{"id":"6710710f4d5d6dc09de2f239894249608314498f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 12:55:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905abb474d644930bd59342fe304c18c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d3614c62a77c4b5d880d3fd64cdf1bd3 : FAILURE in 14m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94916eea43e74cdd941f6c683a31d783 : FAILURE in 16m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8523a6dd65764c0688829f36e7bcdbc4 : FAILURE in 18m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8f825c8e1f3c7b05e4043106f25c58da3d4bb29","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-10 13:03:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I0fd46412be7a8f7520d5a57692cf06005d46fbdf .","accounts_in_message":[],"_revision_number":1},{"id":"243a233dbf1eabbd67bdcfd3bcd120a45d674d67","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-10 13:32:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/4f48666bb89c4864bfc589ecca058559 : SUCCESS in 41m 22s\n- cinder-storpool-tempest-iscsi-multipath build/5420dc2480334a2ab49fe83a280f18a2 : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":1},{"id":"53ac396b9761c081fcf5c13bb2862661de7597e6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-10 13:33:12.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 30s\n- FC       : SUCCESS in 0h 48m 45s\n- NVMe/TCP : SUCCESS in 0h 47m 43s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1004999","accounts_in_message":[],"_revision_number":1},{"id":"b9d5f225fe54afb74664dc08113a7e44eb1fe7e1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-10 13:44:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5320/.","accounts_in_message":[],"_revision_number":1},{"id":"c797c900a1234ee2b6267edceb3e262d97ebcdca","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-10 13:53:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5497/.","accounts_in_message":[],"_revision_number":1},{"id":"2a8d6d43414ba5250ffb2b53eb64fbd56a3a17c9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 14:08:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/14b8b91969fa439daeebc4d85c59d62d\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/48e970fe046044b19021bb8ab2924105 : SUCCESS in 42m 18s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ef13423570d43bdbb257ea38a5e4cfe : SUCCESS in 39m 50s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/68fdfbd8c2f24e789a33634fd399d0cc : SUCCESS in 45m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b2b7694ebd14ee8abe61ab76500682c : SUCCESS in 34m 37s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f9228605fd7a412b8b672ab40f76496d : FAILURE in 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e204d77d020287dcc973cffd13b520dffe7afdcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:56:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36d1e4182bdc467cbacdf658accc2dae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f8cef690424ebba23267727c7fdbf1 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1719a93089ef41efb8a4fded46a75f5e : FAILURE in 11m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40d383673b1446dd9fd3b759722cb952 : FAILURE in 10m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/35811060a65c44359f4aa65b64c5d25f : TIMED_OUT in 40m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c589ad512acf4f509665f281b58e2575 : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/b8f0f79c61b24ef19c5d428d36148fe1 : SUCCESS in 1h 06m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd8f7d8ca94d43aa9e70dbf76c510984 : SUCCESS in 1h 02m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d6c0a9ad8436491fa29299e8e6f66025 : SUCCESS in 58m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c749b0002654b249ca6628498c68298 : SUCCESS in 58m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d1934f4813d4772bfee06e6b12eccf0 : SUCCESS in 9m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c2c4358a937d4aa480a3f2767f148192 : FAILURE in 20m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8635eec1f7434b77b2f892eae70b963c : SUCCESS in 3m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a676a05fcac0469580f096cab3850c45 : SUCCESS in 4m 01s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/d568d525a262406dae680a692469be1f : FAILURE in 12m 25s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/cae017679dac4d8bb60bb59fc2e5a47e : FAILURE in 11m 08s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/76588f8018f64784a4e4d74227e0255b : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/14264d19a43e4d7ebfbdb70efde855bf : SUCCESS in 5m 37s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/68cc53c0ca1b46e2a2a45add21b66553 : FAILURE in 5m 57s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f663086219f14b69897e068e8dedc2c7 : SUCCESS in 9m 14s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/765c84bf6c0d4fcc8519670ba0e31f12 : SUCCESS in 41m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48181fe69c29423f92a6da3fd7269691 : FAILURE in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ba71315baa9548b9bc21524999317882 : SUCCESS in 2h 08m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6201e3ded2e04e4b9294299cb28f61f1 : FAILURE in 2h 02m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb39a53105bf4b2ca801dfdfa2fde7f8 : SUCCESS in 1h 36m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/923152afce204694904de844f15f8ced : FAILURE in 2h 37m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7caf99fe8bc14857a2ceecbff094e643 : SUCCESS in 1h 49m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4e1104793b5e4016a72d010233bebf42 : SUCCESS in 34m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3bac7a2e6075478a9f835d35600f9965 : SUCCESS in 1h 10m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/86a4855bcca74d969726ef1f3a597f2e : SUCCESS in 1h 16m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5a7e24b1306940be8717e95cfb48fc31 : SUCCESS in 55m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2c562b711b24115ba6a4e02ad188acc : FAILURE in 53m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6798f6c063c4f2bb3fcd2f202984a1f : SUCCESS in 57m 43s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba72c7873751d140c533a139a6852ae2910a3569","revisions":{"ba72c7873751d140c533a139a6852ae2910a3569":{"kind":"REWORK","_number":1,"created":"2026-09-10 12:10:00.000000000","uploader":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"ref":"refs/changes/99/1004999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/99/1004999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/1004999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/1004999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/1004999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/99/1004999/1"}}},"commit":{"parents":[{"commit":"31ec30d44f086bcd2e60fad0c4cbc461a1044518","subject":"Merge \"[SVf]: Enable support to extend a volume with snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31ec30d44f086bcd2e60fad0c4cbc461a1044518"}]}],"author":{"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","date":"2026-09-10 11:57:47.000000000","tz":330},"committer":{"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","date":"2026-09-10 12:09:27.000000000","tz":330},"subject":"Fixed IBM Storwize/FlashSystem GMCV volume resize failure by waiting for replication relationship synchronization and added configurable sync timeout and polling options.","message":"Fixed IBM Storwize/FlashSystem GMCV volume resize failure by waiting for replication relationship synchronization and added configurable sync timeout and polling options.\n\nChange-Id: I0fd46412be7a8f7520d5a57692cf06005d46fbdf\nSigned-off-by: Rutuja Pawar \u003crutuja.pawar@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba72c7873751d140c533a139a6852ae2910a3569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba72c7873751d140c533a139a6852ae2910a3569"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
