)]}'
{"id":"openstack%2Fcinder~1005096","triplet_id":"openstack%2Fcinder~master~I8d736755e5743eb1f102ae8a5ed9c16a2f0c7e56","project":"openstack/cinder","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-17 12:48:06.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-17 12:25:59.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8d736755e5743eb1f102ae8a5ed9c16a2f0c7e56","subject":"Backup chunkeddriver: Refactor restore() to improve readability","status":"NEW","created":"2026-09-10 17:30:04.000000000","updated":"2026-09-17 16:22:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd3b689fc56486233b9cf93ade0843c544e609b2","_number":1005096,"virtual_id_number":1005096,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-17 14:40:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 17:53:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 18:02:18.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-10 18:25:26.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 19:00:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 19:33:09.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 19:52:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 19:54:27.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-10 21:12:15.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-11 04:02:27.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-17 03:26:16.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"3bfac2d3d1d7a9857e95d946c4411ed26c4f3627","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-10 17:30:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b1464e8d693c5872d8ba17f451506bd2e5c74b5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-10 17:30:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8164a40bf20f08560d5dcd4f46683374ed70799c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 17:53:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3ea765997b944939ede4ff16862d9d2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d8c77d631add4b46863e94d46e537d23 : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94d4a95987384a748c4f66defc89d8b9 : SUCCESS in 16m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/090eaedb31e24b0c989c1df62b90a18d : SUCCESS in 20m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b8e15dca78c2cb1f36c326c1291d5c7b56ddd2c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 18:02:18.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/1ce8d5b00ba44a788865117ddd1013d3/","accounts_in_message":[],"_revision_number":1},{"id":"aa2ef27f1f61959a4002fc77031cdf087b0282c5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-10 18:25:26.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8d736755e5743eb1f102ae8a5ed9c16a2f0c7e56 .","accounts_in_message":[],"_revision_number":1},{"id":"2a8a78bb0fa080cf7587889a5346f7e621c669e2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-10 19:00:44.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 1m 24s\n- FC       : FAILURE in 0h 1m 10s\n- NVMe/TCP : FAILURE in 0h 1m 44s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1005096","accounts_in_message":[],"_revision_number":1},{"id":"0eacfb6339a971f2af8efa2e7d72c756c15d5b7e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-10 19:33:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/622985cf4e4545ada9c711d115710efb : SUCCESS in 49m 30s\n- cinder-storpool-tempest-iscsi-multipath build/fa7b9aa2e4944cc1b0d002c1a89c1d13 : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6ab07042c52f59f592f38393903b4bd59de74781","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 19:39:05.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1ce8d5b00ba44a788865117ddd1013d3\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fe86c8bd37b147fa9bbb8cb903cb9df6 : SUCCESS in 41m 40s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bce2c5b89c1a4c03a5491cf32fc89b3a : FAILURE in 1h 50m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5c41466ac44f4fef929e428759180538 : SUCCESS in 52m 39s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ece55ba3e7f04c8e9627713d70f72601 : SUCCESS in 31m 28s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d64912730e1a4418baceb05b655d4af3 : FAILURE in 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"334ea45acde68dedaad9b1bf9aba36d375d4b466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 19:52:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/244f5234c5ea4e77bff0354a6cd7d29f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd56a2d07e54204b50c2b564a22273c : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53531777822f49ad92acb0cb6df634b1 : SUCCESS in 9m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3942034d22a3416a89514879204016ab : SUCCESS in 9m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d6b08f6548d4ef584a3c088a6dc2cc4 : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656d31a35e544fc2b8d9e1a9436dc5b2 : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/81306146380f423fb7f87cc38198c995 : SUCCESS in 39m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f628ec10863475eaf87daf413bc67ac : SUCCESS in 44m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f2da6705e1564a0da9023df7fdc013cd : SUCCESS in 1h 08m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/460d34d784a64ea6bb7fe74c45c9c8a6 : SUCCESS in 1h 02m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b049fabb7bfc45bd8471f650e2c5819c : SUCCESS in 8m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/80d9041c98fc4d9f9f3d8fc09f36a9b1 : SUCCESS in 17m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d3556581879d4348b8befe1b794abe05 : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c76018239dee45bebf88a073b3bba026 : SUCCESS in 3m 31s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/1837fc17bbaa4c91919c2fed910ffb09 : FAILURE in 9m 11s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6481b369a8fa40b19f92858253d9e3b7 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e4e81fef826a4b22bd61a344758ee5cf : SUCCESS in 6m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/91a2804b8ecc40008c312d803a38e36f : SUCCESS in 6m 29s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/42458f4591f24889bf62e0c1a9328261 : FAILURE in 7m 47s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b2d56a0c15ab468d9dbf14701725b0d1 : SUCCESS in 10m 48s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b20843e4c5dc41d7a5c976d65bbbc3e1 : SUCCESS in 29m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c231a00eec634ded86540a29055962ab : FAILURE in 5m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd9a2cf5cc244e88a4234f842174940b : SUCCESS in 52m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0018eb67547a4653a42aee4639ebee61 : SUCCESS in 1h 12m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bb83edebc6bf48e68ddcb61beb19d588 : SUCCESS in 1h 39m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/cc4765b4c6e1478fbd2a7c5721835f0a : FAILURE in 2h 15m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ca19ae207a234e979e5da045acdd1163 : SUCCESS in 1h 19m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/085a46ce861a4beab2ad6d1251b3c7b9 : SUCCESS in 23m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1ecbd806f5ba4d80bcd8b1d755c333b8 : SUCCESS in 1h 18m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/34f3e768254f46daa5c9af4b791286d7 : SUCCESS in 1h 22m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/40feeea7952e40d9a94c47fcf9d6358b : SUCCESS in 44m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd7718cc9ecc4875883fa29edbdb7466 : SUCCESS in 42m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/657f9596275f451c964a2b6e97b74984 : SUCCESS in 56m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"0fbadad9ab586dcff96b9df67b7ff163d8ab429a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-10 20:03:24.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_09_2026_19_57_52\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"c58752d78091a7b6fa56a2d3f93156282952b8b2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-10 21:12:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5327/.","accounts_in_message":[],"_revision_number":1},{"id":"e209379a2ea16c88b65e36e515dec449305b7090","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-10 22:01:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5504/.","accounts_in_message":[],"_revision_number":1},{"id":"eb233a422f3af6897f92bcdc3f66dba2d6a97d2a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-11 04:02:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-96-1005096-1-fc : FAILURE in 12m 18s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-96-1005096-1-iscsi : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9ac1d771f315ec37780a2dd5865e7ec03b7c6f61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-11 19:04:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f3f3a1a239eb19345b0d879e9f9ba9fa0dbc37a8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-11 19:21:37.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/ee5442568d264b5fb6bea9de9bc0c11a/","accounts_in_message":[],"_revision_number":2},{"id":"1492b59b0e7b2e2baeb4c742a5f1066be7e0a18b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 19:30:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ee5fc9476fe41aeb7d22b8b6d1ae2ef\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b55a8fa72bbe4187aa336973574647b7 : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cc87a6eab0154a958461bf5cb5a31036 : SUCCESS in 18m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/63932d6ef3c7444dabadffccb2f79686 : SUCCESS in 25m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"70b596f55fee1da011097812a342c5cabed72ac4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-11 19:50:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ee5442568d264b5fb6bea9de9bc0c11a\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8beb0cecc63448099540587481c2beeb : SUCCESS in 38m 25s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9181aef8b40b4a5b976d688c5f003d03 : SUCCESS in 37m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/59d6a24981484d0689d6af6b54382749 : SUCCESS in 46m 07s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c96a000655144a279d11075a7045dae4 : SUCCESS in 33m 33s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/510a92751332457da755aa95d12ae5f5 : FAILURE in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea6c656b481478fea598516816f9438c9e41530b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-11 19:57:42.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8d736755e5743eb1f102ae8a5ed9c16a2f0c7e56 .","accounts_in_message":[],"_revision_number":2},{"id":"66f9820ed18fd5929146822b28c4fc83fb085958","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-11 20:05:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5517/.","accounts_in_message":[],"_revision_number":2},{"id":"90f3a8149220855a1ff1ed652df4319d8fc149a2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-11 20:07:45.000000000","message":"Patch Set 2:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 58m 44s\n- FC       : FAILURE in 0h 55m 0s\n- NVMe/TCP : FAILURE in 0h 54m 52s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1005096","accounts_in_message":[],"_revision_number":2},{"id":"b8a10e076dca053fc8345f4229a979e6f3f4229f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-11 20:10:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5340/.","accounts_in_message":[],"_revision_number":2},{"id":"255adfe17068140d5ba4383d9f3520e99effe6a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 21:33:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c380e4cd10a849899c5ffb61a93d93fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efb52e1c247e4b0db7ba681a0f774989 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01715c8a42074fdeac55b83cba3cf549 : SUCCESS in 11m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1a61db4a5b040cd8aedabfad3167757 : SUCCESS in 10m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c28f414bd1b24b51bca1935be5e3435c : SUCCESS in 14m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d152213ddc4992afc788d3263b7b3f : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/58cab94e2daa446da373a87078878ee7 : SUCCESS in 1h 02m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cab232e8887448aeb7a2512fe00b4cc1 : SUCCESS in 53m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1774525a603a40cb902882fbfed5fe7b : SUCCESS in 1h 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af9e89adba7d498c96fa6d5221d70df1 : SUCCESS in 35m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3dca6b5c48841d58b348711ffacc6ab : SUCCESS in 9m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fd576070efc54f558ea6a5de5521de2d : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c254a1a1198d438ea35649bcb7f4b9aa : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1fe41333ba824aaea20302fe4d72dc4e : SUCCESS in 7m 27s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/d5ab6a747a63492098c3de07d8f5621d : FAILURE in 11m 18s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/dd2fef7f2fe645a9baf144d36d3ff0c5 : FAILURE in 11m 59s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f6d626e608eb4fd285921fe7566dee63 : SUCCESS in 6m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6911019ab2b34c86b0579b6f59d360ab : SUCCESS in 6m 19s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7f005ddf24c640b98f9b0a90c2936a8c : FAILURE in 3m 56s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a637681f4dfd4738a7290369905f3de7 : SUCCESS in 8m 28s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9efd164108d247c7ac73a98d0a492b73 : SUCCESS in 37m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/368eeaab799c4fbea5c3212e7d96bf1f : FAILURE in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ceae10f7a5114d2dbb56078afb69c947 : SUCCESS in 1h 21m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/00176ec1d3974504b473fb0a79cfcd79 : SUCCESS in 2h 05m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/14a4e3a095d349988ee4159ed59daaa4 : SUCCESS in 1h 50m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/c5d2897d03f442e8b229a8c363820b4b : FAILURE in 2h 24m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f2e62b8179bf446283541da17b1e80ff : SUCCESS in 1h 17m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ef42b094df7440e3a58aa414aa411861 : SUCCESS in 36m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a5ad37a097d94d95849b815416869407 : SUCCESS in 38m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/026a85fbe037452e8bdf822fbdd38754 : SUCCESS in 1h 39m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0bb6522c5eba4adfabfed0d26d38c29c : SUCCESS in 45m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/81302c668c804d0c8565271eabaaf8f1 : SUCCESS in 41m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d173a5b9a918492d8182292d34b689ac : SUCCESS in 27m 03s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"a680a20657456abca984998de6dbfa2a4a7180b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-11 21:43:43.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/11_09_2026_21_37_41\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"d6cc23488b95cc34ff15969f80fc9c1afcf20e78","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-11 22:09:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/7d02a7da5a0b40ce84da66697b7b7c0d : SUCCESS in 45m 00s\n- cinder-storpool-tempest-iscsi-multipath build/82ea040032d94f92a0f70827c07e5d23 : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":2},{"id":"70045476da9694be67641b49616dba58bd8dc7e4","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-11 23:24:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-96-1005096-2-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-96-1005096-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e21dbaeadccd4b21a0050fc04415a9adaa193f82","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-17 03:26:16.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3df8b9000cd5bd381239bc25388e9ec5a1d6586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-17 12:25:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"415defdb7eb9c267089d230cc511ae0ea31f9792","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-17 12:25:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1991064f8c207912769ce13efcd9311abf659473","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 12:48:06.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/239d8274ce9547c49050026d48d80501/","accounts_in_message":[],"_revision_number":3},{"id":"6776c289f4a740ade473f3f955cab1ca3ffce2b8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-17 13:16:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/39c9825577914d618e90d43cbf33d7a0 : SUCCESS in 44m 16s\n- cinder-storpool-tempest-iscsi-multipath build/ac49ac5455ac495d974b7a00c7ea6cc2 : SUCCESS in 49m 04s","accounts_in_message":[],"_revision_number":3},{"id":"736681bc2b3e4dbb528daa144286a61e150526d5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 13:18:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/239d8274ce9547c49050026d48d80501\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/882aaa36948748e89737709636f48b0e : SUCCESS in 39m 02s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/126eb6b863974acc87b357c87b701384 : SUCCESS in 37m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3439ae0b9f1f432596dbb633801ffbbc : SUCCESS in 52m 08s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3be952d45d07495dbc90e858f0c84463 : SUCCESS in 35m 33s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/11ccbd0f13504ef09fb0144bcf7ddf7d : FAILURE in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e9fc4fc4a0d809df1a7db5ccba534bc7c0b711bc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 13:19:16.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8d736755e5743eb1f102ae8a5ed9c16a2f0c7e56 .","accounts_in_message":[],"_revision_number":3},{"id":"ade8801cd38ad076f76abc573b1afd19a1a1dd65","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 13:59:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5581/.","accounts_in_message":[],"_revision_number":3},{"id":"d77a6808e2d728303eef6a7a0febc41eeea5d2b8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 14:15:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5404/.","accounts_in_message":[],"_revision_number":3},{"id":"6207dcb5c7202fa84a6cf4244086fb45b0bd2410","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 14:34:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f68543bf8c594c9ca11b878758b97fa7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d2690b75c5c44169b9bb46902206f66d : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f5c9bb9d83f7491f94e241c947bcf6ca : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/08ccfd06a0c34626bf48841a7b548f77 : TIMED_OUT in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d0b254f1dfbc15bde1485f7e41921c4a46647b2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 14:40:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5a414cc15774f9ca9af03c080e39dc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/902eebe4fec7454089e43efcbcd9302d : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa659a5582824618b1ea7a1581c12aae : SUCCESS in 12m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/983947c10a144c19a52a02bfaff3aafa : SUCCESS in 10m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e916b71b639547789a710c942ca66722 : TIMED_OUT in 41m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6591f33012a4d1481249ec299bd5336 : SUCCESS in 16m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/a68bb3a27681461eb9d1a953de44a40c : SUCCESS in 1h 00m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/930d4b5d2b91417d933a02c01308c5b9 : SUCCESS in 1h 07m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/891895cf28a7459592c85ec15a3e6228 : SUCCESS in 1h 36m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4fa469fa5264cf3b256941fb6424c3e : SUCCESS in 53m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1003a18bb7ce4a78a24b26f1da843951 : SUCCESS in 8m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5f4c0f84f93d47ff89bf2335ef427e64 : FAILURE in 26m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/395e17f153324411a47658dd235e2853 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/121a391ba2584dbaadbb89ead3b244c9 : SUCCESS in 6m 32s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/ed895f9fa3f24383aea63c6468271e15 : SUCCESS in 12m 12s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7c9860afd3f7489dbd178ed78fe0b656 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/27ab01ccc63049f492715e9574b6056a : SUCCESS in 7m 05s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a7c2353b8f8d4bdda2865b48859b675f : SUCCESS in 6m 38s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/bc83481a54b74c51b2ea5edf25f92193 : FAILURE in 5m 54s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8b5ad28ff93049669cddb9c86c2d4390 : SUCCESS in 10m 18s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/62ac37aacad04bf98901f10a53df9e67 : SUCCESS in 38m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f011c74b42a467fae391aec2df77269 : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/76d182dc384d4e2893fa05463e853f8d : SUCCESS in 2h 01m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8dd1992eb3df47268be98dbb31bd7a4e : FAILURE in 2h 08m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6f713caf56cd43c9849d13dd6159fcc0 : SUCCESS in 1h 47m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/6b382a9c992548b2a396a328f2780256 : FAILURE in 1h 51m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e91c4292a990489185d731e0c908dfea : SUCCESS in 1h 41m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bcaf0ab5f71a4f71b5f35d89f8887c66 : SUCCESS in 37m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e22c1cf5522428eba0a1a81def714c4 : SUCCESS in 1h 30m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aacf53f921564448afcfc168e4819cbf : SUCCESS in 1h 38m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9354af23695c48a2b28d025425e031d5 : SUCCESS in 34m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5366855028aa48b4b04f5fe9069fdf67 : SUCCESS in 1h 40m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d8d1c3dd4dc45d2921c07eeb9702aa1 : SUCCESS in 55m 56s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"8b2aedb34f2bc3c3700c41f24899683df7333db2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-17 14:50:57.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_09_2026_14_45_06\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"0a33215cb1aa0ce163393db34e0a08abab59a947","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-17 14:51:12.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-96-1005096-3-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-96-1005096-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"bd3b689fc56486233b9cf93ade0843c544e609b2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-17 16:22:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-96-1005096-3-fc : SUCCESS in 58m 4s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-96-1005096-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ba4e0f807746d52067eb0eccaa1dceb00422ff5f","revisions":{"8d8976094a0a05726e20f5a6ceba93e88a8423e7":{"kind":"REWORK","_number":1,"created":"2026-09-10 17:30:04.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/96/1005096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/1005096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1005096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1005096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1005096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/1005096/1"}}},"commit":{"parents":[{"commit":"884d8ff7f7c2179bd4554e6d57bb4b69ed4a5b8a","subject":"Fix data-to-zero corruption restoring an incremental backup chain","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/884d8ff7f7c2179bd4554e6d57bb4b69ed4a5b8a"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-10 17:07:08.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-10 17:28:42.000000000","tz":-240},"subject":"Backup chunkeddriver: Refactor restore() to improve readability","message":"Backup chunkeddriver: Refactor restore() to improve readability\n\nDo things in a more Pythonic style here, which makes it\neasier to understand the logic flow.\n\nChange-Id: I8d736755e5743eb1f102ae8a5ed9c16a2f0c7e56\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d8976094a0a05726e20f5a6ceba93e88a8423e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d8976094a0a05726e20f5a6ceba93e88a8423e7"}]},"branch":"refs/heads/master"},"e38a8c044114afb6feb9f50177f90e0eb2d25003":{"kind":"REWORK","_number":2,"created":"2026-09-11 19:04:12.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/96/1005096/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/1005096/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1005096/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1005096/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1005096/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/1005096/2"}}},"commit":{"parents":[{"commit":"952943db76743c17a9894046eee0051f2753bfd2","subject":"Merge \"Apply non-default qemu process limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/952943db76743c17a9894046eee0051f2753bfd2"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-10 17:07:08.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-11 19:03:44.000000000","tz":-240},"subject":"Backup chunkeddriver: Refactor restore() to improve readability","message":"Backup chunkeddriver: Refactor restore() to improve readability\n\nDo things in a more Pythonic style here, which makes it\neasier to understand the logic flow.\n\nChange-Id: I8d736755e5743eb1f102ae8a5ed9c16a2f0c7e56\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e38a8c044114afb6feb9f50177f90e0eb2d25003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e38a8c044114afb6feb9f50177f90e0eb2d25003"}]},"branch":"refs/heads/master"},"ba4e0f807746d52067eb0eccaa1dceb00422ff5f":{"kind":"REWORK","_number":3,"created":"2026-09-17 12:25:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/96/1005096/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/1005096/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1005096/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1005096/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/1005096/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/1005096/3"}}},"commit":{"parents":[{"commit":"952943db76743c17a9894046eee0051f2753bfd2","subject":"Merge \"Apply non-default qemu process limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/952943db76743c17a9894046eee0051f2753bfd2"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-10 17:07:08.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-17 12:25:44.000000000","tz":-240},"subject":"Backup chunkeddriver: Refactor restore() to improve readability","message":"Backup chunkeddriver: Refactor restore() to improve readability\n\nDo things in a more Pythonic style here, which makes it\neasier to understand the logic flow.\n\nChange-Id: I8d736755e5743eb1f102ae8a5ed9c16a2f0c7e56\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba4e0f807746d52067eb0eccaa1dceb00422ff5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba4e0f807746d52067eb0eccaa1dceb00422ff5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
