)]}'
{"id":"openstack%2Fcinder~1005097","triplet_id":"openstack%2Fcinder~master~Ic115180b3001f76e0f7f535c23ca00ec605f6aff","project":"openstack/cinder","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-10 20:40:45.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic115180b3001f76e0f7f535c23ca00ec605f6aff","subject":"WIP: Add mypy coverage to backup/chunkeddriver","status":"NEW","created":"2026-09-10 17:30:04.000000000","updated":"2026-09-10 23:02:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dac12776cad0a11cfa531f5ddfcc1906582e90b5","_number":1005097,"virtual_id_number":1005097,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-10 19:47:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 17:48:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 18:19:21.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-10 18:30:02.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 18:57:32.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 19:47:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 20:24:35.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 22:10:57.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"e848d9fc8fa2ede28fbe5c5c08130890dfd228d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-10 17:30:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cafa69cbe3e2ff0f6af93a29d558580d0e360d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 17:48:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/812ec41e4397407e8cebc7d2727b142f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73f8c6f4c4f7426db7815f480ff1ae1c : SUCCESS in 15m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4ba33f23aaa546d09a479b829e4fd8f5 : SUCCESS in 14m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0502811c0d8240db88d4d03457d35fc8 : SUCCESS in 16m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edecbfb4c04111b3d290c57dc513419d98756296","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 18:19:21.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/de28db579c4140108970a7b9c959d621/","accounts_in_message":[],"_revision_number":1},{"id":"3f651c1b3d67cfb6970ca3d2c8b62a30eaa6c41b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-10 18:30:02.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic115180b3001f76e0f7f535c23ca00ec605f6aff .","accounts_in_message":[],"_revision_number":1},{"id":"c2be22e2ccdc9270843a44f2424d56b14c93b52d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-10 18:57:32.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 34s\n- FC       : SUCCESS in 0h 46m 53s\n- NVMe/TCP : SUCCESS in 0h 47m 19s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1005097","accounts_in_message":[],"_revision_number":1},{"id":"a832df070c1cca7778951bf92ec19c864aa905bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 19:47:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94e62667b3ec406b84093a565fbb347a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f41f73394d8429a9d21b71af0443710 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20d3f671b5f04b0ab343674096a1c499 : SUCCESS in 8m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b565634e5df64ed98e4a6b80108d5836 : SUCCESS in 11m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5618cc8518b742bc9448c7d03ef6b6bf : SUCCESS in 10m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81a2fd9350a449e5a65713de976c4d80 : SUCCESS in 15m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d951cc805f14e34808c34a9534bd9d5 : SUCCESS in 14m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4e5b63ef2f046e59ea5c7c44fef61d3 : SUCCESS in 3m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/a0bffaeed8a8413595594aea978daad4 : SUCCESS in 1h 08m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd79cfe2cd90470db583f963abd79efb : SUCCESS in 1h 06m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dfa9cb0272b249479514126c89866e06 : SUCCESS in 1h 49m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7795abb0f3ef4da68296fa083272c02e : SUCCESS in 1h 03m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a855020d53344b3193bb0971f5fe42cc : SUCCESS in 10m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/915870494add40d98ed61155ad46c67c : TIMED_OUT in 41m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8451fd18a19641e2a12c8e7a4ca246c0 : FAILURE in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a264509314248fba0a789b7031998d9 : SUCCESS in 8m 13s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8b25a647fd12482db75206bdb1d44a26 : SUCCESS in 12m 25s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2d14c4739de74e30b0261fe810f893b7 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2ff24ca1ffb745588a3056ce386b7506 : SUCCESS in 8m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3acba30674ed405caee9da7f18e4215e : SUCCESS in 8m 14s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/cb15f183386e497e89bd9a7a0874b0de : FAILURE in 7m 09s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f9dc93e305cc4d989c80a5e4e9757069 : SUCCESS in 12m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/90210a02a8a945f28fc8b7968d9c464d : SUCCESS in 27m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50599826b0244daca065242f7673290c : FAILURE in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/10d38002db484f5882c44acadb6c229f : SUCCESS in 1h 33m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dba46267acb2461fa2549e65240d3d09 : FAILURE in 1h 55m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/16b5b87e4a044905adf313194cd63a97 : SUCCESS in 1h 25m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/0308973573604cec950cea7bbf0a391e : FAILURE in 2h 11m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6cf1304a724648f8b535d06e2e6ddbd9 : SUCCESS in 1h 52m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3232dffa147b4da98ce290ed43f27d60 : SUCCESS in 26m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3ab758d333f24e818921589cd0351870 : SUCCESS in 1h 14m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d0f9551a8b694125919e2ba440927808 : SUCCESS in 1h 23m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4d4a38eef8ad47c4a07813d0c9d88c68 : SUCCESS in 47m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b11c4ad037849a79e7ec7ebb14039ce : SUCCESS in 50m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6c2d66840d241a38c883ba0faffbf2b : SUCCESS in 59m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"24f2ab51806ac0980c34a35652f9d05ae2572c06","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-10 20:24:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/48712261c8eb4b9c9aa95851eafb0ea6 : SUCCESS in 47m 37s\n- cinder-storpool-tempest-iscsi-multipath build/040f668f2d7f4d38bd43956816c837cc : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d0bf1d90706537e9dc86fe27a539c54f91d2e4f7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 20:40:45.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/de28db579c4140108970a7b9c959d621\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6f5748306ba64bf7932e1b80546e871e : SUCCESS in 42m 01s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/595d35720a5a469ebd64358dcb08f165 : FAILURE in 2h 17m 06s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80584d3776f64390811b2852f123b282 : SUCCESS in 45m 59s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/950814a9b6ca455ebc2b694335e4b70b : SUCCESS in 33m 13s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/daa96c7403ba4469aca22baf1ef633d9 : FAILURE in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b32896a9eb6bad9760ae0edf5b41c96fc027356a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-10 22:10:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5328/.","accounts_in_message":[],"_revision_number":1},{"id":"dac12776cad0a11cfa531f5ddfcc1906582e90b5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-10 23:02:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5505/.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3abe4a1de56185258b55da744f8ab41ece6dadc6","revisions":{"3abe4a1de56185258b55da744f8ab41ece6dadc6":{"kind":"REWORK","_number":1,"created":"2026-09-10 17:30:04.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/97/1005097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/97/1005097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/97/1005097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/97/1005097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/97/1005097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/97/1005097/1"}}},"commit":{"parents":[{"commit":"8d8976094a0a05726e20f5a6ceba93e88a8423e7","subject":"Backup chunkeddriver: Refactor restore() to improve readability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d8976094a0a05726e20f5a6ceba93e88a8423e7"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-10 17:25:37.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-10 17:28:59.000000000","tz":-240},"subject":"WIP: Add mypy coverage to backup/chunkeddriver","message":"WIP: Add mypy coverage to backup/chunkeddriver\n\nA few items in here still need analysis/work.\n\nChange-Id: Ic115180b3001f76e0f7f535c23ca00ec605f6aff\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3abe4a1de56185258b55da744f8ab41ece6dadc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3abe4a1de56185258b55da744f8ab41ece6dadc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
