)]}'
{"id":"openstack%2Fcinder~1005377","triplet_id":"openstack%2Fcinder~stable%2F2025.1~I1b44a1395044fe4544050f9c34d80d1932707a20","project":"openstack/cinder","branch":"stable/2025.1","attention_set":{"38860":{"account":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"last_update":"2026-09-12 10:58:48.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1b44a1395044fe4544050f9c34d80d1932707a20","subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","status":"NEW","created":"2026-09-12 07:58:49.000000000","updated":"2026-09-12 10:58:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c38ef825d1ba5760615f81d5d120a9e6da18e52","_number":1005377,"virtual_id_number":1005377,"owner":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-12 10:58:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-12 08:17:34.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-12 09:18:30.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-12 10:58:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6198eb3e116f0e67cc4f21659a149f151411da8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"date":"2026-09-12 07:58:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba91dbd24f83a6d1afca1175807df89f60ce5b8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"date":"2026-09-12 08:12:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6c015c0c1187bda91660cb5fcf40e7aa3db1d60f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-12 08:17:34.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5522/.","accounts_in_message":[],"_revision_number":1},{"id":"58ce0cf2efd41c6a87552e68b8ed60f4af14b5f6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-12 08:17:53.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5345/.","accounts_in_message":[],"_revision_number":1},{"id":"8a70dbd3e91ecdd4b8477d991555014ea17e8639","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-12 09:18:30.000000000","message":"Patch Set 2:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 57m 52s\n- FC       : FAILURE in 0h 53m 19s\n- NVMe/TCP : FAILURE in 0h 53m 55s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1005377","accounts_in_message":[],"_revision_number":2},{"id":"9a6d45858f205352daa3a5933fdc061470669c33","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-12 09:28:08.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5523/.","accounts_in_message":[],"_revision_number":2},{"id":"0468eabe95731fe4ccf339dcb33a1634478e37a3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-12 09:30:41.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5346/.","accounts_in_message":[],"_revision_number":2},{"id":"2c38ef825d1ba5760615f81d5d120a9e6da18e52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 10:58:48.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00d3872bee5549fe821cef399cd70725\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1664299b16834668832ef5dbe6e46549 : FAILURE in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59415d5e6275429fa46b7ca86f122a4c : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b79a38603564282bcf418df994623df : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a31b0c73fe44d5bef0daf4a9f31362 : SUCCESS in 16m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/90e7b266470e4b759a3a0913a921819d : FAILURE in 45m 27s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36e5abdfe61645e8961d911198e768a3 : SUCCESS in 1h 22m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d94c9458c7a4996b0a77fa0017f0374 : SUCCESS in 1h 05m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49a6dd551027437e8fcba70d2a7550d0 : SUCCESS in 7m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a923adbd66fc42b78005cb2259b996f8 : SUCCESS in 23m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4336fd69fc044d959dc50edb986d5499 : SUCCESS in 8m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e4fc4fe74204b2685c9dc3bfb188ff0 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f5e6cf396c74ff6aa85f137cd201749 : SUCCESS in 6m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4fb3f4d8227f439d9788cb030d6bd584 : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9e990b1b6f3453e9ac43225b17f7edc : FAILURE in 14m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bf5e7a6035c54454a4326a6073fd04a9 : SUCCESS in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f89509bc37047b4be77b8693581405d : SUCCESS in 1h 07m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1877aa5467e144cfb7a7128c110e251f : SUCCESS in 2h 30m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c154900df9f6461b9526eb69be3cd13c : SUCCESS in 2h 14m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f84bde0f83764d19b5fe1eb36a90c5bd : FAILURE in 23m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/164649cfc4b548b8bf82a7d8886a6baf : SUCCESS in 36m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6e98ed5d3db446faa60b2b1b7187e6ef : SUCCESS in 1h 38m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e21005ba38554012ba3af4bae37bef44 : SUCCESS in 32m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/729e070db7304ed8a1c652865414e0db : SUCCESS in 1h 55m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a5dd90a6593b463a915c200b02fadb33 : FAILURE in 30m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f4b5a9caa14246bb954f7b4f70b69e52 : SUCCESS in 1h 10m 15s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66fd8d4fc1c345ddab4eca96c29b04d0 : SUCCESS in 1h 37m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57e8aea46ec14fa6a93f5a76afe1d663 : SUCCESS in 59m 04s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job devstack-plugin-nfs-tempest-full-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job devstack-plugin-nfs-tempest-full-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c2ccbbf9fd4424f0525a463a3e18ade032e7680b","revisions":{"3781d3a10efb9c1f5c5bd8231dc6184b779ebbf3":{"kind":"REWORK","_number":1,"created":"2026-09-12 07:58:49.000000000","uploader":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"ref":"refs/changes/77/1005377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/1005377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/1005377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/1005377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/1005377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/1005377/1"}}},"commit":{"parents":[{"commit":"ef0d50cb18d0ceb5d70d500c9a032b0eb66e749b","subject":"Merge \"Fix properties\u003d{} being sent to Glance\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ef0d50cb18d0ceb5d70d500c9a032b0eb66e749b"}]}],"author":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-02-24 08:27:44.000000000","tz":210},"committer":{"name":"soheil linred","email":"soheillinred@hotmail.com","date":"2026-09-12 07:58:39.000000000","tz":210},"subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","message":"Fix: avoid automatic checksum headers breaking S3-compatible backups\n\nS3BackupDriver currently sets request_checksum_calculation\u003d\u0027when_required\u0027\nin the botocore Config. On some S3-compatible backends, this causes\nput_object requests to include checksum headers that are not supported,\nleading to backup failures.\n\nCloses-Bug: #2142548\nChange-Id: I1b44a1395044fe4544050f9c34d80d1932707a20\nSigned-off-by: soheil \u003csoheillinred@hotmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3781d3a10efb9c1f5c5bd8231dc6184b779ebbf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3781d3a10efb9c1f5c5bd8231dc6184b779ebbf3"}]},"branch":"refs/heads/stable/2025.1"},"c2ccbbf9fd4424f0525a463a3e18ade032e7680b":{"kind":"REWORK","_number":2,"created":"2026-09-12 08:12:49.000000000","uploader":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"ref":"refs/changes/77/1005377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/1005377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/1005377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/1005377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/1005377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/1005377/2"}}},"commit":{"parents":[{"commit":"ef0d50cb18d0ceb5d70d500c9a032b0eb66e749b","subject":"Merge \"Fix properties\u003d{} being sent to Glance\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ef0d50cb18d0ceb5d70d500c9a032b0eb66e749b"}]}],"author":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-02-24 08:27:44.000000000","tz":210},"committer":{"name":"soheil linred","email":"soheillinred@hotmail.com","date":"2026-09-12 08:11:22.000000000","tz":210},"subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","message":"Fix: avoid automatic checksum headers breaking S3-compatible backups\n\nS3BackupDriver currently sets request_checksum_calculation\u003d\u0027when_required\u0027\nin the botocore Config. On some S3-compatible backends, this causes\nput_object requests to include checksum headers that are not supported,\nleading to backup failures.\n\nCloses-Bug: #2142548\nChange-Id: I1b44a1395044fe4544050f9c34d80d1932707a20\nSigned-off-by: soheil \u003csoheillinred@hotmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2ccbbf9fd4424f0525a463a3e18ade032e7680b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2ccbbf9fd4424f0525a463a3e18ade032e7680b"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
