)]}'
{"id":"openstack%2Fcinder~1005579","triplet_id":"openstack%2Fcinder~master~I1089ba70f84d587a4b69c34ba0f7808f81219867","project":"openstack/cinder","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-18 18:37:25.000000000","reason":"\u003cGERRIT_ACCOUNT_38797\u003e replied on the change","reason_account":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"}},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-09-18 17:30:39.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"38797":{"account":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"last_update":"2026-09-18 19:02:19.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1089ba70f84d587a4b69c34ba0f7808f81219867","subject":"Cleanup unused only_quotas path in quota_destroy_all_by_project","status":"NEW","created":"2026-09-14 14:13:24.000000000","updated":"2026-09-18 21:09:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d839c9087c522ef31d40467cdb4bc7514f5c3643","_number":1005579,"virtual_id_number":1005579,"owner":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-18 21:00:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 14:35:13.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-14 14:50:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-14 14:57:34.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-14 15:14:19.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-14 15:33:56.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-09-14 15:54:43.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-14 16:10:31.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-14 16:10:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 17:58:06.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-16 18:08:34.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-18 16:01:17.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"}],"messages":[{"id":"1101f51b0696eeaf475b4a5b4c1cd8e0716b3f79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-14 14:13:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d7aa4084476e4cf786628f7725c7d5b98dea465","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-14 14:18:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8a2cbcf8b06cbf04dc8c47fb5be022360479c22a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-14 14:35:13.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/35661462e1284994bb06e8de25ac5651/","accounts_in_message":[],"_revision_number":2},{"id":"a50c1526762818c49a6de0a0eaeb119a5bebbcce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 14:50:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ecd1d6f212e4060a102dbc815dcc3cb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/607f81f1f80a430fbb53e1564148fe5e : FAILURE in 20m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/21e0b6e918a24504a13880f354139301 : FAILURE in 13m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4a40d38eaa66413dae37d87854d8e791 : FAILURE in 27m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c9e022a6ada44c65ca683fc8fad47bec9742b5fc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-14 14:57:34.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1089ba70f84d587a4b69c34ba0f7808f81219867 .","accounts_in_message":[],"_revision_number":1},{"id":"f0f24abe0934505b1897a32e41f4832758f45686","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-14 15:02:45.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1089ba70f84d587a4b69c34ba0f7808f81219867 .","accounts_in_message":[],"_revision_number":2},{"id":"6c8340fda3833dafb051e07cf5a8e049ebfddb37","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-14 15:14:19.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5353/.","accounts_in_message":[],"_revision_number":1},{"id":"2787183790bbd81156ab37bd122b81408837c476","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-14 15:14:19.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5530/.","accounts_in_message":[],"_revision_number":1},{"id":"5b6dd76279fbdf0f8a508325852fde9f4ff7352e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 15:33:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"458dbce39b52ab281b09d660f313d38bbb9fe4d0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-14 15:53:33.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/35661462e1284994bb06e8de25ac5651\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93ebc605b20149eaa3945bb84ab89cb9 : FAILURE in 8m 25s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/92bfc23d8410468aa49f8717871c57cd : FAILURE in 9m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1125d83f64fe40bfbd60eef59fc66414 : NODE_FAILURE Node request 200-0001271717 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd9569f7265c4f62aa84269f6ca69050 : FAILURE in 36m 00s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b64c0a4b626f4c02b4399d350910dbfb : FAILURE in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58c659b8540d23423d89ce04ab202fe5889b506c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-14 15:54:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/751d20c5f620443e9ea828995f478af3 : FAILURE in 47m 28s\n- cinder-storpool-tempest-iscsi-multipath build/4fb4665c878b464096f98a8a5f94d52e : FAILURE in 48m 48s","accounts_in_message":[],"_revision_number":2},{"id":"36d31eb384ae4c2d139b2c8293282646d83d8837","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-14 16:10:31.000000000","message":"Patch Set 2:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 56m 44s\n- FC       : FAILURE in 0h 53m 54s\n- NVMe/TCP : FAILURE in 0h 52m 56s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1005579","accounts_in_message":[],"_revision_number":2},{"id":"ec043e2dba24bb39847507d6ab49cbc62dd91c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:10:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b1f52dfa11240f1a6ad217fecf80c8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fbd9a4140544694ae90e4591c4cff80 : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57d0e01a26fd41dfb200a2b9e8a6950d : FAILURE in 10m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65ffe0a065084dd892d4f44a9c359269 : TIMED_OUT in 40m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a2c47cfc5a9e4dd3837a8688697e61fe : FAILURE in 15m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09fdc0d914b94046bdfb1b08b3be7a7d : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/5d7688ad17ea4da9bb7db2d553040333 : SUCCESS in 31m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f195da030f17499b8b321d9f8f048c8c : SUCCESS in 1h 05m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eda2cc71a5ca4e4d8d6e888ca5e1020d : FAILURE in 1h 07m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d07b7965c9ae4e8a854d31a710f934d5 : FAILURE in 56m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56c94830c8114f3bae526d33cc9120bf : SUCCESS in 4m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/754244007673478f8f8c9106484a3a90 : FAILURE in 26m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0675979f2a134d0c8988f2282a52f4b1 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79bb4d2bcf364fbe8aa4652e0b34e049 : SUCCESS in 6m 15s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/38ab01e382d94b56ad9ca257ecc4ac47 : FAILURE in 10m 23s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/00c24cbbcecf4852a00b9f775cc9cc40 : FAILURE in 7m 45s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/afdf4fd3419f4289869bbb701b6d64f2 : SUCCESS in 6m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dd045fe27b9544548115aeca4f44bac9 : SUCCESS in 6m 10s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d02bcc934f714cd09b92bc1cba5a978c : FAILURE in 5m 45s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/13dd3c34f13a4807bf6a0b91e6cf1d19 : SUCCESS in 7m 56s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/861062aa05da41e8828260de2c286367 : FAILURE in 28m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6be77bd869fb47849b220861367e7280 : FAILURE in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3d8cddafb974438aa81460d8859ad1db : FAILURE in 1h 09m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d80871cc714b472f919ae401ba760d29 : FAILURE in 1h 22m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4b16db5b3d2c42a58f4c4bf864aa2cc6 : FAILURE in 1h 42m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/a76e84710ee442c0b592a370ff290616 : FAILURE in 15m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/423102c93a3a426eadc08c7316e1860b : FAILURE in 1h 17m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/63e3ea4403894adf871d9d9b903132f7 : SUCCESS in 17m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/71247daff5c24b0582b4dea4d9e88a6c : SUCCESS in 1h 20m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/79a936fd960d43529ffbb74ae4cc903d : FAILURE in 1h 09m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f8d8e1e206c247e98382b96b95799e99 : SUCCESS in 43m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c70dfa9ead44d3189bd438186769fc7 : SUCCESS in 1h 36m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a4f475c9dcb4facaaf039dfeb2e9845 : SUCCESS in 37m 07s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"d906d0088fd90effc14a4ed66acceb93ad52eed0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-14 16:16:42.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5354/.","accounts_in_message":[],"_revision_number":2},{"id":"bfd5cb2dc9c38e43d0934221714d78b884bcc2f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-14 16:23:39.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5531/.","accounts_in_message":[],"_revision_number":2},{"id":"06329c893d2d2a5a5f6963bec775bf8defc5d7c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-14 17:27:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c6f0a57bb67e824412721466aef20e7bace888e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:48:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feee2671d2a545c4b9d93b5b9870d45c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb6390813db7454c97c47f7cf9e23bff : FAILURE in 15m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/13e95217523e411e9c8c0dee8022bfad : FAILURE in 14m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7eff5454a4614941a43ff962afc91f8d : FAILURE in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"23b5e9ac7ecc08ae962e209e789e2eb9be881bf4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 18:01:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"32125cec599c43a4e75f40cab7f99299eb20faed","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-14 18:10:56.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1089ba70f84d587a4b69c34ba0f7808f81219867 .","accounts_in_message":[],"_revision_number":3},{"id":"d8f0ec06a5a138606419533760a3b75a69b82e8b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-14 18:27:44.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d832767580534f70aeae74a40ce103fc/","accounts_in_message":[],"_revision_number":3},{"id":"44fd054dbc88b765c5536bb9ce564c44b6c8c364","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-14 18:57:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d832767580534f70aeae74a40ce103fc\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4d6bba53ddb5432dbbd569bc922117ea : SUCCESS in 39m 33s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/070306686d594da5aa7500e456c898a0 : SUCCESS in 40m 50s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93909530abe347cbb1cbf1dee339530d : SUCCESS in 43m 13s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/db04f60d3af24eb2a795f862c208edac : SUCCESS in 33m 39s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0fa25129921f40f2aa77e02bfb9c822c : FAILURE in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e1ab281a080b779c3bb721be0c1c2535d493282e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-14 18:59:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"58cb4990fd825b8ee7b669aa657e621aa4c61b0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 19:20:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0b5beb74fe14264b28227ae4498f968\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f63e6942071f41c9ae334873dd85ad33 : FAILURE in 14m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/60e3ee95bb6d44ababf77b99fcc74ad0 : FAILURE in 11m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a72975c0adc845ea8450dc0f9581b4a8 : FAILURE in 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b7de02bff190477be604bd454f6b024832eabae5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-14 19:39:42.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/48e9bd8e60d143379b98f48b91cbddc4/","accounts_in_message":[],"_revision_number":4},{"id":"79a6f88f3264ef64626835f32ad6a61f0b708064","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-14 19:45:27.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1089ba70f84d587a4b69c34ba0f7808f81219867 .","accounts_in_message":[],"_revision_number":4},{"id":"05c444fae4073ef28915cfd99df047131524125e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-14 20:13:37.000000000","message":"Patch Set 4:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 57m 5s\n- FC       : FAILURE in 0h 53m 3s\n- NVMe/TCP : FAILURE in 0h 0m 12s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1005579","accounts_in_message":[],"_revision_number":4},{"id":"16ff6afb56e064add41c74e0406e2da7a9b215f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 20:50:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/274114d9853440ed9a68a081fc72793f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35ebd14724540cc8e81c5da3c813a09 : SUCCESS in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8591544254c24d8098f718a854efae54 : FAILURE in 11m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/612a2c306d844c94bcddfe30eda62040 : FAILURE in 10m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b94a036d785b408086746be887041276 : FAILURE in 13m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ba180a74b54a4f96fc7b9c98dd22f1 : SUCCESS in 15m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/73a859c6bfba477db56a4951edb17a27 : SUCCESS in 1h 09m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/224456ddaee4451bb0bde1969a3e9ea5 : SUCCESS in 1h 04m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec2312003f4742ee9227ddf59015fe39 : SUCCESS in 1h 36m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/405f83aa98b8453baa23f8cd673d6c9c : SUCCESS in 1h 00m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f83945cfc364aa1affb53c2e02a4a6b : SUCCESS in 9m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f1c67ac50bf940bbbb2762bb22abf542 : FAILURE in 20m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/674a02b700ae4981ac72a4a57df21821 : SUCCESS in 4m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d4381ddeba9046909673a65f5e7885f2 : SUCCESS in 7m 10s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/305d7ff75a954b70946d0f74775570b7 : FAILURE in 11m 13s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d10383fe57de44dcbf0e6e4ba7b28585 : FAILURE in 9m 42s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ece0efc2c3bf43e5bba3a8f1cf5fb806 : SUCCESS in 8m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/abb33a95d4a2427b995ef70b393ae3eb : SUCCESS in 6m 26s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/38f5a76d3c364735b374b3c4c07dd486 : FAILURE in 5m 31s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/562cd57e0e6247368a48bc05620caf04 : SUCCESS in 9m 12s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ab801ebff4b943d29a7d3cd8e94c0daa : SUCCESS in 21m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98643a30730543e289054ea88fc0a606 : FAILURE in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7f2c2406b8934a988c2fe4e0dbf5e6f4 : SUCCESS in 46m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1893b33ba0a946f1a40db172965fe26a : FAILURE in 1h 29m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6fe00f40f8aa4cfdabaa1eb1487afd86 : SUCCESS in 1h 26m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b231f61888a8471f84aa46adb360ac64 : FAILURE in 1h 49m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8cf93b575a4a4c5f8701bb0ed8fc922a : SUCCESS in 1h 32m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f4c8d68654c145a4a1ffa540329cda28 : SUCCESS in 32m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/00c2f2799e354f64ad40a6cdbb3ed6c7 : SUCCESS in 1h 19m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cd5a34d61db54b82b512c1bda566776b : SUCCESS in 1h 39m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/219ac3b741844a879556db0032e87e7b : SUCCESS in 46m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/77646451b6b34cab937c960b7a3a688f : FAILURE in 1h 40m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/126b43973ff04e3b9e70658d60498371 : SUCCESS in 58m 42s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"de0dfbd72637266051188f390fd541def335447f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-14 21:18:53.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/18c004c3083d4b15b92d7ca87432199d : SUCCESS in 44m 48s\n- cinder-storpool-tempest-iscsi-multipath build/6d3953d195344477ae4434d345544d46 : SUCCESS in 54m 23s","accounts_in_message":[],"_revision_number":4},{"id":"33d3b3817a8f0e1aabed7cd03b626659fc02f8af","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-14 21:32:43.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5359/.","accounts_in_message":[],"_revision_number":3},{"id":"91f7c9b95745c9f92b6468da986bcba3523784cf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-14 21:43:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5536/.","accounts_in_message":[],"_revision_number":3},{"id":"2ea3510a6d1b37fcff3172d0a2f48a52e4608435","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-14 22:07:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/48e9bd8e60d143379b98f48b91cbddc4\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e36d4dd2d094fc6a1e71194190ca09f : SUCCESS in 40m 22s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d59d08c80a2f454091070a774ca9f0fe : SUCCESS in 38m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4cd989a1a90f4c2f9d6e39a2688f4ecd : SUCCESS in 46m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d863aa3a60984a73996a5d10c90ccaf4 : SUCCESS in 35m 11s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/abed3f3fc3be47a0be81999072e31cf2 : FAILURE in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3c4ac0f0a24d65a679b57e250916f10303360696","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-14 23:31:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5361/.","accounts_in_message":[],"_revision_number":4},{"id":"e140c12356308757e9501b49ff35c445a5d20a1d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-15 00:01:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5538/.","accounts_in_message":[],"_revision_number":4},{"id":"8e5a46ea3d1c65cac57adfe770405e4b4cdaad30","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-15 19:02:43.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"373b1e0c6346fa639100c2a703ab4b532fb3dee5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-16 13:50:48.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b1a70f1375955e0295640ed0a30584a2375a1d5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 14:34:52.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"6f8428267cf88db102d8bb4161156f002dc4df00","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 14:37:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"916d673eff0f761691137bcb88fd6eabb5bedcf1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-16 14:55:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1d5c4caca8dcf779fca35d5ae64f644bc2ee68a7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 15:25:14.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/c688192e1af048988a350b5b02f5aeaf/","accounts_in_message":[],"_revision_number":5},{"id":"10c298a41dc9b74e8a88dab94a6c19b8b6eafe4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 15:28:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"80dced2e717cb4b40af327c8cc321424d87b844e","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 15:28:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a425398c80b5a29f174b75c5286561ed96010dbd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 15:30:51.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1089ba70f84d587a4b69c34ba0f7808f81219867 .","accounts_in_message":[],"_revision_number":5},{"id":"fa1e3e4bc4a5de81156f097cd07267205d21911b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 16:23:00.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1089ba70f84d587a4b69c34ba0f7808f81219867 .","accounts_in_message":[],"_revision_number":6},{"id":"26df1d63d40169443f2102be44f7e7b2f119353b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 16:23:30.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/418fc541710e4838ac17b190148d8e33/","accounts_in_message":[],"_revision_number":6},{"id":"96297881d377e2dbfa41796268f8066abf21d4c0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 17:09:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5390/.","accounts_in_message":[],"_revision_number":5},{"id":"a83eb89bda7c8795a54d4a5dd1ca3f7b1a0c1b86","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 17:16:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5567/.","accounts_in_message":[],"_revision_number":5},{"id":"e5dd45c50f50211012770cb6a7b1adb5812ecfaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 17:57:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2836d121104147da9d0ca8a93d4dea70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3097bb7f264aca801f2755a4b08bef : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0317cdb6f56f4538b8a9b08b1df12ade : SUCCESS in 12m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/407ea7c4c0e14187aed76c603cb9fcaf : SUCCESS in 11m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/de773c1854a14625bf69ac7c4c02336c : SUCCESS in 16m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559b222cd1384d3c848653022e840ffa : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/0dc61e2dcf774deca4936c52b5f3fa43 : SUCCESS in 1h 05m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c2f0a33be6d74443ba0bbbdf5f2b2eb9 : SUCCESS in 1h 07m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bcb26404ca0e434485cf367b33954402 : SUCCESS in 1h 55m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bd1ee4129ae44d3b49ebe60684dbdc8 : SUCCESS in 31m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e64ac94192f4831bc83118e2b48297d : SUCCESS in 9m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8a983ced794a432d8def79b2e3187ab2 : FAILURE in 24m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ea92c3e59b21461b81d260bff4d2f4b5 : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7336428bbb3f468b8765e9655bfa2630 : SUCCESS in 6m 58s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/b27dba651c104697bba682ebeb9d77b6 : SUCCESS in 12m 14s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/bbb42913b4ad4d9fa8f4f2bdabf9f336 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cc9ac82b6e9445c1889b14b1c998b508 : SUCCESS in 6m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cafad90ed78c4651a98ac9fb6531e50f : SUCCESS in 5m 56s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4eb280b828184413b8f042e5c25fe047 : FAILURE in 5m 32s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/73a1952cfd114825ad2d5ed5ef2d7d7f : SUCCESS in 8m 11s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e4807f1b6f644a5fbb0619c4019c93a8 : SUCCESS in 24m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae8454bc6dc44834b1416b5e7b6f5c89 : FAILURE in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/059d0acf282146afae4f7e5b1462d982 : SUCCESS in 1h 44m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b3e1bb81d3674d0d906ef393e3b926f1 : SUCCESS in 1h 19m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/09aac9b7952a4dfb882063c353fbf61f : SUCCESS in 1h 42m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/28a2903a68514aa3875f93e520cded3f : FAILURE in 2h 20m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/65955578d11541c0a11bb9a424e7076f : FAILURE in 1h 15m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c3cec0c31ee7440f90cd2c4bac9c874a : SUCCESS in 36m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e91d757dec36448a9dbfc291baa373ff : SUCCESS in 1h 12m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5f931b30b9c3479babc43e2741689e12 : SUCCESS in 1h 30m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e5ee9bd4c01f47ddb34301aed126f213 : SUCCESS in 38m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c2036dfa3fb641188c8a367f6b2643af : SUCCESS in 1h 39m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/551e4d7197664af38b989e22b833832e : SUCCESS in 55m 55s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"ef492bcb9cc54f45641ac5c66f8df7e9345a5d75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-16 18:07:47.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_09_2026_18_01_58\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"875508883ab3f4354b5e579f220ccd7fd158c1cd","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-16 18:08:34.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-1005579-6-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-1005579-6-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"98477ad0613a0ebd2b3a6db64c84df85da1f478a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-16 18:15:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/21838c9fe0504857bf3f2a0d7f0c4b0c : SUCCESS in 47m 26s\n- cinder-storpool-tempest-iscsi-multipath build/eeeaeed5748e49a3bf508316b9c0fcae : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":6},{"id":"23a9c602e00126abd081b576768901d539293217","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 19:01:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/418fc541710e4838ac17b190148d8e33\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/34bc7466fead46bcb329b9c6eb8db7c8 : SUCCESS in 38m 14s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5fbc9bb8eb914427a6d9efb76f1869ff : SUCCESS in 33m 38s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7850780984dc4ea983a0f6681e3ffdda : SUCCESS in 48m 21s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/19d6df970ed5493692f9b3aa81810a69 : SUCCESS in 39m 51s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/37925daa90a249779a401c7ba612f976 : FAILURE in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ed17539a771b2e127e9c9abf28d1bf5763220f2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 19:55:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bed91e32107c413f9f50a0eefcb8094a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b94f694f4a3f4132948cd5f744bc4dd0 : SUCCESS in 18m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5adc53feb20a4ade828aa3270a7f8d0d : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/979b0e59c32a4b95a97cd31a500cfc17 : TIMED_OUT in 1h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"64ef2c035e67bba82e420170ef2af94234bf1b8b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 20:10:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5393/.","accounts_in_message":[],"_revision_number":6},{"id":"4eba72cd1d5d42b9f4fa535ad2e78a0e6b26a39e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 20:15:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5570/.","accounts_in_message":[],"_revision_number":6},{"id":"76669cd6eb1b317adad5ccdff63e89e596892f7c","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-17 07:34:05.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-79-1005579-6-fc : SUCCESS in 50m 20s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-79-1005579-6-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"f3c81652e6b467ddfafcc4ea2e7425caa414daa3","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-18 13:38:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fa966f92f2af7ec12c88a9c4b62913ef21d0976d","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-09-18 16:01:17.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"cc4c104fd498ebca2976c49f4e6294846b7b46b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-18 17:30:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"2b9cbb25b82e569c5724c32f93e99bf52fd0bd3a","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-18 17:31:33.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f5937edd6aeee27187066aaec8180b7ba0f223ea","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-18 17:48:59.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6fbe4438fcec7d3a431f7d30565b096bdfccd2c7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 18:08:54.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/c8bb10e2f6744b76a8644fbccaa6b2bc/","accounts_in_message":[],"_revision_number":7},{"id":"21c77f0f2edcea153ac05b704177697b330b5ef8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 18:23:54.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1089ba70f84d587a4b69c34ba0f7808f81219867 .","accounts_in_message":[],"_revision_number":7},{"id":"944595eb40989a2563037aa42e0f46949dbcacc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-18 18:36:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a5f9186ad4a952f8bcd95e506cd4ce1768754449","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-18 18:37:25.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d40dc80073d767665d16044b07065394415a6208","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:00:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d402d1fff9460fa34d69ff2b08110b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e3b15cae7a64f16822467d47b8f4cd4 : SUCCESS in 16m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bde66160d0934334963295107dd14bf3 : SUCCESS in 16m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fb66819b4d0e489ca3e15aa6b70e5e5a : SUCCESS in 22m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"02e747a02674af45f300ac5f6b6645463c23583a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 19:02:19.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/de8db3dfc035482a8da30e62e4a52b70/","accounts_in_message":[],"_revision_number":8},{"id":"4f15e050b3e60eaa8311709bb07c6273cfa33f8f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 19:27:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5423/.","accounts_in_message":[],"_revision_number":7},{"id":"1bf829aa8e40082dbc9e51cecac37f91ed7612a9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 19:30:05.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/de8db3dfc035482a8da30e62e4a52b70\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71916018223044e883325afeb374ac39 : SUCCESS in 44m 44s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/539cf8284adc44a1985fc75e229e9e8f : SUCCESS in 35m 34s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/814332509f4a4202bdd73ec458e8caaa : SUCCESS in 53m 11s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/acabcc81e0f145309ae17ffbfc27d85b : SUCCESS in 32m 56s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8142f88057c04365a449d9e8f27930e2 : FAILURE in 12m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"79832fbd7fa60102c3401f3693e5bca90a734599","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-18 19:30:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/12ee4920ad354981a585788aa7476e01 : SUCCESS in 47m 12s\n- cinder-storpool-tempest-iscsi-multipath build/edaff107a1404afe99eae5586434d512 : SUCCESS in 52m 11s","accounts_in_message":[],"_revision_number":8},{"id":"5be0083f820f49ef78276253a9a6bcabecf99cfd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 19:31:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5600/.","accounts_in_message":[],"_revision_number":7},{"id":"42f75129f33c701192017b79c2e296dc11424bf8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-18 19:32:33.000000000","message":"Patch Set 8:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 7s\n- FC       : SUCCESS in 0h 47m 34s\n- NVMe/TCP : SUCCESS in 0h 48m 2s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1005579","accounts_in_message":[],"_revision_number":8},{"id":"65fc33451ea2c1d93c542c7de95c9b55beb3ffef","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 19:32:53.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1089ba70f84d587a4b69c34ba0f7808f81219867 .","accounts_in_message":[],"_revision_number":8},{"id":"f827a33fafe4030347d819ea5a6c08c55efaa728","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 20:42:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5601/.","accounts_in_message":[],"_revision_number":8},{"id":"5d1cb4ad3be62119b8a378931f2475094aae0e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 21:00:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b88fb3999e3449fa00d543a8d5539ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aafcc92baea54e9fa98eb58fcaedaed5 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf0d12162bbc4b23ac6c51efb398974e : SUCCESS in 9m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3af86d43a8584eb7a5fc7ddcb0434c72 : SUCCESS in 7m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d421b356fab4200a10bdc8da725fb85 : SUCCESS in 13m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff73808ffc74d88863964b49a41ac3d : SUCCESS in 16m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/40720b3f69c549f5a90b93cf56ce16a6 : SUCCESS in 1h 08m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/808b9ccf55664bda954df97870b09023 : SUCCESS in 31m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0599bdc84ca2459db15a5ccc6a357610 : SUCCESS in 1h 02m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1280b108f3c4fc68909534fa46206d8 : SUCCESS in 36m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a49a3e2a5aa496d9cab5ace262562c8 : SUCCESS in 7m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/702c9c64709e445c9c31c258dbb18b4f : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/02fdcc8ea0234ec68aa5faa03509ffb9 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86df9ef8ff534ca9bc982147896b94e7 : SUCCESS in 6m 27s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/42a06fdc0dfb4ccc8bf436aa1a14877c : FAILURE in 7m 53s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/00e08743329b48a5b8521fc4b563ac99 : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ef89453d301b4941a6297165ace0b818 : SUCCESS in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f0144c07f47a486e82b925c3d0148cbc : SUCCESS in 5m 46s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/680d8fea55334af79121c6371b38c2f8 : FAILURE in 5m 24s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/68dbd794990440009548f46ac60906ba : SUCCESS in 8m 56s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c002dce63e7d4a7198c95f4edae35486 : SUCCESS in 36m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ce125f560acd4cc9b894404c45fe3500 : FAILURE in 7m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3fb3ef91550d424aab689e2925421f61 : SUCCESS in 1h 51m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eedfc98653c2456f913eae745fa14323 : FAILURE in 2h 11m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0b32f4c657bc477f925eede7e950c5b3 : SUCCESS in 1h 59m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/210fd1181e4f48a4a348ab8ef540d5c8 : FAILURE in 2h 22m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/20bf3144ceaa4a62bcdd244c582740f2 : SUCCESS in 1h 23m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/175398c830ba4d3ba1fb74b59d41ab59 : SUCCESS in 18m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c8537a0a8b2e405ba299e92d569c1f7f : SUCCESS in 40m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7f93e6f4b0ca4b5bae32f0ccb469ec53 : SUCCESS in 1h 19m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4b0f92845eb340708469d800589a74a5 : SUCCESS in 34m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/711e2f3b4ac94805a766d68cb5058167 : SUCCESS in 1h 43m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf10a28283984589ba83d0946f7f0922 : SUCCESS in 36m 32s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"d839c9087c522ef31d40467cdb4bc7514f5c3643","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-18 21:09:58.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_09_2026_21_04_19\nLeave a comment with","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d4a2cd97a7c0481cac616047f53e1fd8fcbc047c","revisions":{"77b4f220d751bea678f37cdf6d6524c21344e521":{"kind":"REWORK","_number":1,"created":"2026-09-14 14:13:24.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/79/1005579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/1005579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/1005579/1"}}},"commit":{"parents":[{"commit":"bf495b13005a48a564337d356900408098415df2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf495b13005a48a564337d356900408098415df2"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 13:58:36.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 14:13:21.000000000","tz":-180},"subject":"Cleanup unused only_quotas path in quota_destroy_all_by_project As requested in the TODO above, the only_quotas\u003dFalse path was removed from the quota_destroy_all_by_project function since it was unused code.","message":"Cleanup unused only_quotas path in quota_destroy_all_by_project\nAs requested in the TODO above, the only_quotas\u003dFalse path was removed from the quota_destroy_all_by_project function since it was unused code.\n\nChange-Id: I1089ba70f84d587a4b69c34ba0f7808f81219867\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77b4f220d751bea678f37cdf6d6524c21344e521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77b4f220d751bea678f37cdf6d6524c21344e521"}]},"branch":"refs/heads/master"},"dd5c4e0c7bf3b32e071838a49272d0299c896305":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-14 14:18:49.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/79/1005579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/1005579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/1005579/2"}}},"commit":{"parents":[{"commit":"bf495b13005a48a564337d356900408098415df2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf495b13005a48a564337d356900408098415df2"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 13:58:36.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 14:15:18.000000000","tz":-180},"subject":"Cleanup unused only_quotas path in quota_destroy_all_by_project","message":"Cleanup unused only_quotas path in quota_destroy_all_by_project\n\nAs requested in the TODO above, the only_quotas\u003dFalse path was removed from the quota_destroy_all_by_project function since it was unused code.\n\nChange-Id: I1089ba70f84d587a4b69c34ba0f7808f81219867\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd5c4e0c7bf3b32e071838a49272d0299c896305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd5c4e0c7bf3b32e071838a49272d0299c896305"}]},"branch":"refs/heads/master"},"9d58e8a12c0122d13384bcc5300865edcff0d811":{"kind":"REWORK","_number":3,"created":"2026-09-14 17:27:19.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/79/1005579/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/1005579/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/1005579/3"}}},"commit":{"parents":[{"commit":"bf495b13005a48a564337d356900408098415df2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf495b13005a48a564337d356900408098415df2"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 13:58:36.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 17:26:42.000000000","tz":-180},"subject":"Cleanup unused only_quotas path in quota_destroy_all_by_project","message":"Cleanup unused only_quotas path in quota_destroy_all_by_project\n\nAs requested in the TODO above, the only_quotas\u003dFalse path was removed from the quota_destroy_all_by_project function since it was unused code.\n\nChange-Id: I1089ba70f84d587a4b69c34ba0f7808f81219867\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d58e8a12c0122d13384bcc5300865edcff0d811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d58e8a12c0122d13384bcc5300865edcff0d811"}]},"branch":"refs/heads/master"},"98b3a57924eee310562647a57d7c223df1f274f7":{"kind":"REWORK","_number":4,"created":"2026-09-14 18:59:02.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/79/1005579/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/1005579/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/1005579/4"}}},"commit":{"parents":[{"commit":"bf495b13005a48a564337d356900408098415df2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf495b13005a48a564337d356900408098415df2"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 13:58:36.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 18:58:50.000000000","tz":-180},"subject":"Cleanup unused only_quotas path in quota_destroy_all_by_project","message":"Cleanup unused only_quotas path in quota_destroy_all_by_project\n\nAs requested in the TODO above, the only_quotas\u003dFalse path was removed from the quota_destroy_all_by_project function since it was unused code.\n\nChange-Id: I1089ba70f84d587a4b69c34ba0f7808f81219867\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98b3a57924eee310562647a57d7c223df1f274f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98b3a57924eee310562647a57d7c223df1f274f7"}]},"branch":"refs/heads/master"},"ed65dbc13728232a5f4efd9baf294d58cf1cc97f":{"kind":"REWORK","_number":5,"created":"2026-09-16 14:34:52.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/79/1005579/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/1005579/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/1005579/5"}}},"commit":{"parents":[{"commit":"bf495b13005a48a564337d356900408098415df2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf495b13005a48a564337d356900408098415df2"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 13:58:36.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 14:34:34.000000000","tz":-180},"subject":"Cleanup unused only_quotas path in quota_destroy_all_by_project","message":"Cleanup unused only_quotas path in quota_destroy_all_by_project\n\nAs requested in the TODO above, the only_quotas\u003dFalse path was removed from the quota_destroy_all_by_project function since it was unused code.\n\nChange-Id: I1089ba70f84d587a4b69c34ba0f7808f81219867\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed65dbc13728232a5f4efd9baf294d58cf1cc97f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed65dbc13728232a5f4efd9baf294d58cf1cc97f"}]},"branch":"refs/heads/master"},"28f0021b705d5e167bd35bc654156a26162445f9":{"kind":"REWORK","_number":6,"created":"2026-09-16 15:28:36.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/79/1005579/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/1005579/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/1005579/6"}}},"commit":{"parents":[{"commit":"bf495b13005a48a564337d356900408098415df2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf495b13005a48a564337d356900408098415df2"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 13:58:36.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 15:28:21.000000000","tz":-180},"subject":"Cleanup unused only_quotas path in quota_destroy_all_by_project","message":"Cleanup unused only_quotas path in quota_destroy_all_by_project\n\nAs requested in the TODO above, the only_quotas\u003dFalse path was removed from the quota_destroy_all_by_project function since it was unused code.\n\nChange-Id: I1089ba70f84d587a4b69c34ba0f7808f81219867\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28f0021b705d5e167bd35bc654156a26162445f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28f0021b705d5e167bd35bc654156a26162445f9"}]},"branch":"refs/heads/master"},"bac97ca880cdcc2078e833553e6073e8df9189b5":{"kind":"REWORK","_number":7,"created":"2026-09-18 17:30:39.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/79/1005579/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/1005579/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/1005579/7"}}},"commit":{"parents":[{"commit":"bf495b13005a48a564337d356900408098415df2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf495b13005a48a564337d356900408098415df2"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 13:58:36.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-18 17:29:57.000000000","tz":-180},"subject":"Cleanup unused only_quotas path in quota_destroy_all_by_project","message":"Cleanup unused only_quotas path in quota_destroy_all_by_project\n\nAs requested in the TODO above, the only_quotas\u003dFalse path was\nremoved from the quota_destroy_all_by_project function since it was\nunused code.\n\nChange-Id: I1089ba70f84d587a4b69c34ba0f7808f81219867\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bac97ca880cdcc2078e833553e6073e8df9189b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bac97ca880cdcc2078e833553e6073e8df9189b5"}]},"branch":"refs/heads/master"},"d4a2cd97a7c0481cac616047f53e1fd8fcbc047c":{"kind":"REWORK","_number":8,"created":"2026-09-18 18:36:22.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/79/1005579/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/1005579/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/1005579/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/1005579/8"}}},"commit":{"parents":[{"commit":"bf495b13005a48a564337d356900408098415df2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf495b13005a48a564337d356900408098415df2"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-14 13:58:36.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-18 18:36:08.000000000","tz":-180},"subject":"Cleanup unused only_quotas path in quota_destroy_all_by_project","message":"Cleanup unused only_quotas path in quota_destroy_all_by_project\n\nAs requested in the TODO above, the only_quotas\u003dFalse path was\nremoved from the quota_destroy_all_by_project function since it was\nunused code.\n\nChange-Id: I1089ba70f84d587a4b69c34ba0f7808f81219867\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4a2cd97a7c0481cac616047f53e1fd8fcbc047c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4a2cd97a7c0481cac616047f53e1fd8fcbc047c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
