)]}'
{"id":"openstack%2Fcinder~1005749","triplet_id":"openstack%2Fcinder~master~I1fad7ab9b7af57ed6c1db3750a368a55940cee33","project":"openstack/cinder","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-16 14:43:27.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-16 13:52:22.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1fad7ab9b7af57ed6c1db3750a368a55940cee33","subject":"mypy: Add volume manager type coverage","status":"NEW","created":"2026-09-15 15:23:25.000000000","updated":"2026-09-16 20:11:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5804a22f274abd2096306c824a7f4e1abb4ff31a","_number":1005749,"virtual_id_number":1005749,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 16:25:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 15:42:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-15 15:58:38.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-15 16:18:23.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-15 16:18:24.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-15 16:22:16.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-15 17:10:14.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-09-15 17:50:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-15 20:00:37.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-16 16:34:42.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-16 16:36:35.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"cc5ff3cb157fb0cc0539b39b8b1a1c7ca096041e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-15 15:23:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"745cfb0308c834b8196b3c03c4f11f014e6da13c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 15:42:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81a7576a181e4753ad06731feea82344\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9a4953b14434cc9937ef7535924d9c6 : SUCCESS in 14m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a887eb265393497da8b026c8903bf6bd : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5205e8c7dd35464b99e27685d47e06c1 : SUCCESS in 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0713af47c23d8066d1c93192238459bc22cef8f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-15 15:58:38.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b2c6067e010f4f97a2600f738ca48a4b/","accounts_in_message":[],"_revision_number":1},{"id":"04927948e68e129fbd80c4ca8e3be5c1d63057a9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-15 16:12:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b2c6067e010f4f97a2600f738ca48a4b\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6ca481114a54f0eada77eb4f4975bbf : SUCCESS in 45m 17s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b9090f3315641fb96e6c1481867135c : SUCCESS in 36m 11s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1a7388454b9a472c84393981cd1ddee2 : SUCCESS in 47m 55s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c848578bf6c4ee09a158fb07bb0c2df : SUCCESS in 34m 44s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4d3436eb2edb43afad7889607eb92143 : FAILURE in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23eb8adbb2a26873de22859e6220c9b481737380","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-15 16:18:23.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 50m 2s\n- FC       : SUCCESS in 0h 50m 0s\n- NVMe/TCP : SUCCESS in 0h 49m 35s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1005749","accounts_in_message":[],"_revision_number":1},{"id":"27a3d7b7ab827d54c1764b9cfcc9460363141fd4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-15 16:18:24.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1fad7ab9b7af57ed6c1db3750a368a55940cee33 .","accounts_in_message":[],"_revision_number":1},{"id":"1234d31838cf67d517fd1998e2ffd222078d2ddc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-15 16:22:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/8c55be09895841478fa7b48aed6d508e : SUCCESS in 48m 40s\n- cinder-storpool-tempest-iscsi-multipath build/ab923bbaeca84bc0808dacc16dc0bf29 : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bec0f665053f5e8e1d5c07300cdc4c812c791771","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-15 17:10:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6de44130a8746e32c8dd41ecbc17aaa2e415172d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 17:50:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47f17d8b7f36427ea15e9e1028c53b44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f9c5d47d9b64c2fafb3b824bab2dbbb : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38cb51cab4bf40e38f5990c0964184df : SUCCESS in 8m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59d03e4f434c43f39468a6f68e59b89d : SUCCESS in 7m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ce42d66b853f4bce9523ac3bf7a70de4 : SUCCESS in 14m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2e76cbba6064dad81050ea18a784d9e : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/fd023509c1c64bd48046deaf09e4be56 : SUCCESS in 1h 11m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb8ca1d07e7d4905aa0c9973d06e73fd : SUCCESS in 30m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/15376b71c5764ae989a5b61ceacd3fdc : SUCCESS in 1h 05m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e5664683d734b349616cd36d5fda3ea : SUCCESS in 47m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d01aa46310a44470b53652b6da7d95fd : SUCCESS in 8m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/113d0b9731e74b1ba91ffce01b0d45d4 : SUCCESS in 22m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a175482afb8640a3975c367f04c5c6da : FAILURE in 3m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/097a39ba0e934f159f83fa1e115f9821 : SUCCESS in 4m 21s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/d956a068f2804026a7f276bd35fd5777 : SUCCESS in 12m 00s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/bc5f0f37796a472d88c2619f2cfd0ed7 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fb08c47b6ab940d191bf21b045f27128 : SUCCESS in 6m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7060b5c647c9456fb17e3ca05c9cf338 : SUCCESS in 6m 03s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e09fda00e665499492b77e80e85161da : FAILURE in 5m 56s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/97a8cf80d4c14c94938558b953eec556 : SUCCESS in 7m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b398cbd0b91f4d47994c86983b720988 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e75c2712d17c4777bfcba4e727f388e1 : FAILURE in 7m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3773f514f0c14a23853d09bb121e96e5 : SUCCESS in 1h 33m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ff653f177f1943f5b6082b9ac8a09447 : SUCCESS in 1h 35m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/42156914c8ef44f593eaecf85883bda4 : SUCCESS in 1h 19m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/13c513747755477cb1fcab01fd03d403 : FAILURE in 2h 12m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2270db1d25c54d0ebc9e44b5d06c3d2b : SUCCESS in 1h 16m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e0cc856f61eb4824980de0b107a333e1 : SUCCESS in 37m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0378cce1dd0f470295ceeb340137c7f0 : SUCCESS in 1h 21m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/37961664470e418597654e131d17bdb8 : SUCCESS in 1h 37m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da6a2e13dd29491d8b5685fb4332a889 : SUCCESS in 31m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dc8ed3b16f8d41a7acff4723cf26243b : SUCCESS in 1h 48m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c2e96ac52d840a88da7ca27b23b0410 : SUCCESS in 1h 13m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c7008c4402a4579bea8638c5977c655 : SUCCESS in 1h 00m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"08c650453c3fdc14b7c7f5c2b3df88df620957d5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-15 20:00:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5555/.","accounts_in_message":[],"_revision_number":1},{"id":"d45fef6475a1d7204fbca27f2f93f5056aecc170","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-15 20:17:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67218432b2dce4033c60eb2b17e2379d7b5719fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-15 20:17:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a821c4bd0935a2f99803f5692b73880434f3f808","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-15 20:36:14.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/2c44dffbc17741c0b7944e4a8fafbca4/","accounts_in_message":[],"_revision_number":2},{"id":"04b9732619bc5f75c04bc24a39d77c9b41d004a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 20:43:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60b42518a3f346ad99c8014a2a64f9b8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3fc42fa1bd0c4346a0346a57c0339a91 : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/82e486d1f0254f5391e7d57d43bac2f6 : SUCCESS in 16m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/53056c227140465cad1a834a0bdbc61b : SUCCESS in 24m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"af0c6f11a2b3e7a176b96eb3f811e87080f2c7df","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-15 21:04:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2c44dffbc17741c0b7944e4a8fafbca4\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1981e48c81154299a1924a2b2db3d51b : SUCCESS in 40m 37s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/026780a2cbb54a54b32fc3876f2a2a4b : SUCCESS in 43m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ca65c17221e4f22b023af17ace90117 : SUCCESS in 46m 57s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ffe79404153646bc8aeffe7608535a60 : SUCCESS in 38m 12s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/95253407b6174af998fee9ed5d2c0e5d : FAILURE in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86f17b41f3a8b55ac035f3c92367da373522c3df","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-15 21:11:12.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1fad7ab9b7af57ed6c1db3750a368a55940cee33 .","accounts_in_message":[],"_revision_number":2},{"id":"cd320764793237e35baa897e4441f2edb80d9adc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-15 21:11:45.000000000","message":"Patch Set 2:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 49m 2s\n- FC       : SUCCESS in 0h 49m 42s\n- NVMe/TCP : SUCCESS in 0h 48m 44s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1005749","accounts_in_message":[],"_revision_number":2},{"id":"08db1af949b5677a6e5248e1f69a11ec2d264099","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-15 21:18:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/afeb578c999b4efd8ef548dcba2860fb : SUCCESS in 45m 31s\n- cinder-storpool-tempest-iscsi-multipath build/01a3bc472cae41c48c1bcdcccfd67a79 : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":2},{"id":"6b908da6a2a0e74d88a17647fc7cd8db151ed2d8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-15 21:27:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5378/.","accounts_in_message":[],"_revision_number":1},{"id":"cce4d3cd429e3a4a413e63f96c2f659b21e6982e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-15 22:03:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5557/.","accounts_in_message":[],"_revision_number":2},{"id":"ca074c70a752db6db841483d8ea497c4f6571021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:31:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa0816d04a4747e59d0cce7992f20174\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d339bbb2418b4143a47120dde569c9f4 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bcd5423b8346407e8cbd22d1c28ddc3c : SUCCESS in 11m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b1c5c07b8f4424fb9f0f68e1f565e79 : SUCCESS in 10m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/97ffa3f698584ee980f4adee1abcd8c6 : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92a23656cc14b4baabba599d162cbe0 : SUCCESS in 15m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/cd64961a4a374de787784a6f0b497097 : SUCCESS in 51m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/96b839ce0e64408ab96e218e1b02ffae : SUCCESS in 30m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c638f3747ad44b1960d20abb3eadd73 : SUCCESS in 1h 24m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/654501ff45f3445aac486a28b4d7298b : SUCCESS in 32m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f88c4d65601b4e4e9c6ef0c7eaab7c66 : SUCCESS in 8m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3eac17696b584a91a295b8316ad2a6fd : SUCCESS in 20m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/adb503e0c64343b082edeef9970e277d : SUCCESS in 6m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8372fae9dd1a4193807dfa6bd4e1c23b : SUCCESS in 6m 43s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/054dbed92cb24415ac20841f9f4bf633 : SUCCESS in 8m 26s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/37c8aae6d415458ca3805493dc3ef7c0 : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/add5298e8ae6469e9869822dfc631b04 : SUCCESS in 6m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fd8499bbf2674d64b4661b0eccd4cb86 : SUCCESS in 6m 43s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f455c132bfdd4d4e9f4ff34e65dfd8c3 : FAILURE in 4m 56s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/634757a385de43aa9ade3e34e9090bf2 : SUCCESS in 6m 12s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b404f19ecd9a4810bd8bd11bfc677140 : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5774fc682c9b4a3cacc6e074f04e3622 : FAILURE in 3m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1286786207b1445b830afe28306bc444 : SUCCESS in 57m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b842374953d0479bac3de15490407cb7 : FAILURE in 1h 05m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7e600911d71b46db9a9b6de1fb30a734 : SUCCESS in 1h 59m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/d94448ed5a7444b4a14b15c2a0d39716 : FAILURE in 2h 06m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5491e7df3843412abf5ed1ab6790d60f : SUCCESS in 1h 41m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ae52f14079b645d582c0f4c69e5a3c6f : SUCCESS in 37m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/facca2a69a1841b19d16472d34b174ba : SUCCESS in 1h 15m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/113f0550d9274a2abb81dd9fa3e12c14 : SUCCESS in 1h 37m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/72c7333b772a4867913ca7948edf51bf : SUCCESS in 46m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/71e078e5d7fa418f8946f7a7dcfa87a1 : SUCCESS in 1h 47m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cbbe8cbfe371472bb7b05c6f7bf68977 : FAILURE in 1h 51m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d188cfaca194869b003f0cb24f4e844 : SUCCESS in 57m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"593e0ed070cbcd8ad638f06297d94fe1eabf8728","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-15 23:17:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5380/.","accounts_in_message":[],"_revision_number":2},{"id":"fdb43d32de9a79ab6779f4cd5fde2cd00a2b36c4","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-16 05:17:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"42d65da59451ce6d47d9298ad88a3d74196d43a5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-16 13:37:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2009ddd5663bf600c11a56898d352416f9598fe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-16 13:52:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"895959d1c9f7a749f70a199abbe2103dc4b2a14c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-16 13:52:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"620c58220909553145be40eef41803eed76c7866","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 14:43:27.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b0ca80891f324cf794b3556a445a5d7e/","accounts_in_message":[],"_revision_number":3},{"id":"2fea9ee94c9865004010c904ca286dbeb7469a86","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-16 14:44:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/af48afde1d054c0aaa04be6b16d48002 : SUCCESS in 49m 49s\n- cinder-storpool-tempest-iscsi-multipath build/8b8ebb33f8764c0b81240c6f2b08aaed : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":3},{"id":"b192322a38e18a8dcb64e94cfbedb87605741b37","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 14:49:24.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1fad7ab9b7af57ed6c1db3750a368a55940cee33 .","accounts_in_message":[],"_revision_number":3},{"id":"b3cab92f056f8e1a1dceeff9c80c60fc20a17f53","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 14:55:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5563/.","accounts_in_message":[],"_revision_number":3},{"id":"981230f05edbbda0efa519e0312e449029128191","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 14:55:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5386/.","accounts_in_message":[],"_revision_number":3},{"id":"ab8c4ec45df133686bd51851b05d1f41c6d525ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:25:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d07106209a704900afe9f886c58dccd6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad35e43e25f4151bd8e9fa2db8106b2 : SUCCESS in 9m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54bbd97246984a4495c17ca2dcc88e5b : SUCCESS in 11m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84d17fe3eac74157a4ebe3c1963d12dc : SUCCESS in 14m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a9a93ea34ae4c7e8c9164e45f76b25f : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a13bc2fd40114b96a3dda8a385d02290 : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/93379ff876ca4d2d8aa96ac1711c8fad : SUCCESS in 37m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1cfbd8509d3249bf9703412fd7a6f545 : SUCCESS in 36m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35568e73bbec4eb091064e6388d9911d : SUCCESS in 1h 34m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c385c0e2f81483b845ecbb39ebe158a : SUCCESS in 43m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dbd1476298740159e02b50c9b0cfaed : SUCCESS in 11m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e9e2ab6d8fe94fdda0ddeb05a6711b7d : SUCCESS in 29m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fa7ab5b47a014653b4d0c66967deb103 : SUCCESS in 3m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7fe62a8378f0405b8aa166d97be806fa : SUCCESS in 7m 16s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/93685921652b4c5ab32815c9393104d3 : SUCCESS in 11m 36s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a024ef35cda64143bd3e9180fd5c65ef : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/80eaa01242524dfbb5bd98983d0d33fe : SUCCESS in 5m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a8e970a2b1a54799a985cf65deb1a0e6 : SUCCESS in 5m 07s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/222064f425274752b6acb02e056bc579 : FAILURE in 5m 40s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c1c807a1a0ce49009b5595b9583fc951 : SUCCESS in 11m 23s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9ee4ccb7f45840458263d525e1a796a5 : SUCCESS in 28m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a06dacd8d1714643ac205d4bc1ba262e : FAILURE in 8m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/93314a06dd9a4687818463f3a7fba910 : SUCCESS in 1h 43m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/91cb6c45af554e589744b98c10826c18 : FAILURE in 1h 05m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc84ef9d46a448ce975b51393fa55fc2 : SUCCESS in 1h 36m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/c27eac67d8574b56a607436d62a8d4bd : FAILURE in 2h 15m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9cdf6931c4124c33bc7c9ab0551909ed : SUCCESS in 1h 21m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1465e52b9b084db593895ab7771ffb59 : SUCCESS in 39m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1f3e657f397f4014869d74d55a7c6bbe : SUCCESS in 1h 15m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9066cbf8c29f4d2284370fd84bd6962e : SUCCESS in 1h 25m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/885611e8702041aa89ed618b51ea61d1 : SUCCESS in 33m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3c97f85bc2e04c9986874291ebf9be7d : SUCCESS in 1h 42m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e90b6c79486f4ec5a1669517cedd764e : SUCCESS in 1h 42m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/034fcffad5e34e9da71ac708b18324d0 : SUCCESS in 30m 17s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"035bec4726bac52c5a180dcd32d0f861a3a58af6","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-16 16:36:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-49-1005749-3-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-49-1005749-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"a04c0d1ac5c58720b478c47556f9dea30b29a50d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-16 16:44:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_09_2026_16_38_22\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"0d20f29c19354ac1c216243ffadbd9d3693595f5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 18:43:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b0ca80891f324cf794b3556a445a5d7e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/47d37d67625f4bd993fcf15e7247f304 : SUCCESS in 46m 58s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ea96cfede1b0473b8db52938be2f6b63 : SUCCESS in 34m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/76d8eba00d30494b9706e67a8fb81de4 : SUCCESS in 48m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c25ca7ff3ac94b2ab0496f9fdd647de8 : SUCCESS in 36m 47s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/13768dd6ba2347af98d12a4110959cb0 : RETRY_LIMIT in 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e1fe9debe312c4b12299ed49d30a135854407872","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 19:42:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a30b0bec94114404862a4c91a250b8e4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4827600c16e540c8b54be01b8560a7b9 : SUCCESS in 17m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7107195565d34b5a8e2c928edb29d7c3 : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a5b986514045467e95a8a4112adfd86c : TIMED_OUT in 1h 01m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5804a22f274abd2096306c824a7f4e1abb4ff31a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-16 20:11:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-49-1005749-3-fc : SUCCESS in 48m 16s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-49-1005749-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6ec8f17514c973ee26fb09194caed07002518c0e","revisions":{"1b54f44a03282cd20556a79287172f7d032e930b":{"kind":"REWORK","_number":1,"created":"2026-09-15 15:23:25.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/49/1005749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/1005749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/1005749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/1005749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/1005749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/1005749/1"}}},"commit":{"parents":[{"commit":"5bde6d7b62391cdcc1f78b05ec10173b121abe69","subject":"Merge \"Replace remaining deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bde6d7b62391cdcc1f78b05ec10173b121abe69"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-15 14:16:06.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-15 15:22:41.000000000","tz":-240},"subject":"mypy: Add volume manager type coverage","message":"mypy: Add volume manager type coverage\n\nAdd more type coverage for methods in the volume\nmanager.\n\nChange-Id: I1fad7ab9b7af57ed6c1db3750a368a55940cee33\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b54f44a03282cd20556a79287172f7d032e930b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b54f44a03282cd20556a79287172f7d032e930b"}]},"branch":"refs/heads/master"},"67c1916dd6438c32831af4548089e7d67c5ec620":{"kind":"REWORK","_number":2,"created":"2026-09-15 20:17:16.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/49/1005749/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/1005749/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/1005749/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/1005749/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/1005749/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/1005749/2"}}},"commit":{"parents":[{"commit":"3a22da1f1bd102ee6b0d86b58a33097b9357f6cf","subject":"Merge \"Pure Storage: check SDK response on volume/snapshot operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a22da1f1bd102ee6b0d86b58a33097b9357f6cf"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-15 14:16:06.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-15 20:14:06.000000000","tz":-240},"subject":"mypy: Add volume manager type coverage","message":"mypy: Add volume manager type coverage\n\nAdd more type coverage for methods in the volume\nmanager.\n\nChange-Id: I1fad7ab9b7af57ed6c1db3750a368a55940cee33\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67c1916dd6438c32831af4548089e7d67c5ec620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67c1916dd6438c32831af4548089e7d67c5ec620"}]},"branch":"refs/heads/master"},"6ec8f17514c973ee26fb09194caed07002518c0e":{"kind":"REWORK","_number":3,"created":"2026-09-16 13:52:22.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/49/1005749/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/1005749/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/1005749/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/1005749/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/1005749/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/1005749/3"}}},"commit":{"parents":[{"commit":"a2e7131bdcf424e17f9e6498d1e62cfb39002c63","subject":"volume_utils: fail create_encryption_key on unencrypted types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2e7131bdcf424e17f9e6498d1e62cfb39002c63"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-15 14:16:06.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-16 13:49:04.000000000","tz":-240},"subject":"mypy: Add volume manager type coverage","message":"mypy: Add volume manager type coverage\n\nAdd more type coverage for methods in the volume\nmanager.\n\nChange-Id: I1fad7ab9b7af57ed6c1db3750a368a55940cee33\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ec8f17514c973ee26fb09194caed07002518c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ec8f17514c973ee26fb09194caed07002518c0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
