)]}'
{"id":"openstack%2Fcinder~1005884","triplet_id":"openstack%2Fcinder~master~I8280ee4f9a1f38db49265f31d29926e3e72b0bb1","project":"openstack/cinder","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-16 14:49:20.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8280ee4f9a1f38db49265f31d29926e3e72b0bb1","subject":"volume_utils: fail create_encryption_key on unencrypted types","status":"NEW","created":"2026-09-16 13:52:22.000000000","updated":"2026-09-17 05:59:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e6e35291118c34ee68b9ae2566df07a0c3c5ee39","_number":1005884,"virtual_id_number":1005884,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 16:27:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 14:49:20.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-16 14:49:21.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-16 15:36:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-16 15:51:53.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-16 16:27:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 16:44:09.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-16 16:47:14.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"3d15d20f211ac007a8bebb4b9225a8765e15c2ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-16 13:52:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db1a151922646c2f6b390e434fd85cb75a5162c1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 14:49:20.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/e69a28ea194e4da9a1cb014a0f451b60/","accounts_in_message":[],"_revision_number":1},{"id":"e71a3cd6789b8233378329611638fca17d446c6b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 14:49:21.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8280ee4f9a1f38db49265f31d29926e3e72b0bb1 .","accounts_in_message":[],"_revision_number":1},{"id":"44e7fd1b7cefcd3bb0a63fba690c8c8400a26406","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 15:33:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e69a28ea194e4da9a1cb014a0f451b60\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ad13e74cf51645008c4f3e6c4fdb9704 : SUCCESS in 46m 43s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f811c44a74f54635b45a90eeaf4590db : SUCCESS in 36m 55s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4187ee8a5f0146a8956ed8a19ae006c5 : SUCCESS in 46m 07s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33255f9261154c0cac2a97ecb45d5bea : SUCCESS in 33m 56s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bd9da6610ca644e9b00d7a3a5648715d : FAILURE in 13m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5fe504b80e4b68eb99ce6ccdd0151d5fd151c91","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-16 15:36:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ca3529274e4648918884f77be2f04db8 : SUCCESS in 45m 29s\n- cinder-storpool-tempest-iscsi-multipath build/6c89eaed82da4a4183c73b33cdf7b288 : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":1},{"id":"86b1129caa9f8e06469aa5908e33a48cdd27dbbe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 15:51:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5387/.","accounts_in_message":[],"_revision_number":1},{"id":"2afd665f4df3950588d841ff393aacf5948dccdc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 16:00:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5564/.","accounts_in_message":[],"_revision_number":1},{"id":"4d7f75730f566b9943f17c0a2a45b378baf44a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:27:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf98c489e4a140a183e97f6c2f6abad9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8786cd23284925b556a45ae9f695f4 : SUCCESS in 7m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ce343b503ab4f7a9b95369958d9d90a : SUCCESS in 12m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c55a01a235ef410facbdc6b0bd1dd3a4 : SUCCESS in 10m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/08e51159f04e41328a1ecac0091ce5bc : TIMED_OUT in 41m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5106a9a312394224b29cbe9a568923e0 : SUCCESS in 15m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/48b8f1e3906046329117950a57a01aeb : SUCCESS in 37m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8d5e68c6844542678e9070bd98b68c7a : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2eef302349c24ad3bb52de035622c8ec : SUCCESS in 1h 38m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d649f0e775d045aeba3ca47994937f50 : SUCCESS in 1h 00m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff1f2b920a9b41739f5293d20b10df0d : SUCCESS in 7m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/34c853672e3e40428a785cb166458067 : SUCCESS in 20m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ccc402248f8d414b84dccf70f535143b : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e837227d7d014282b9dc037d691fb6c3 : SUCCESS in 6m 54s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/b03345518e0e449ebba4e23edfe9891a : SUCCESS in 12m 58s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/24ae524ebeda467a813d3fd94aebf334 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/47eb53b12eb04f5faf610ebe0ea4af83 : SUCCESS in 7m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1767848462fc412282d9c32f3b4f8529 : SUCCESS in 6m 44s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a7a6682765ff47a0981491c3e1035941 : FAILURE in 6m 50s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/35e327347cc04a629a2f68748cafa2e4 : SUCCESS in 9m 35s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a55d67251994c2ab9e09587ccf0a545 : SUCCESS in 44m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/30fdbd93a8934ce3b51e9e1765c6c238 : FAILURE in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/88edf74a8fa84d8ea07351ddf5e5b985 : SUCCESS in 1h 47m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cf3d093da07e460082907a18bea180bd : SUCCESS in 1h 02m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/13552c17a8ea4da1a5b7aa818b0296cd : SUCCESS in 2h 09m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/adbb351f1d914cd2bab301998788e052 : FAILURE in 2h 18m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/249def212137466fb02568f7a3bd5522 : SUCCESS in 1h 55m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0995fac3e49648dca195e32e9b44fc97 : SUCCESS in 20m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a8e8acf7657247bfbeab84bd465abe6c : SUCCESS in 1h 11m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3aca2e1e296a40d9a347f9a8d50da25d : SUCCESS in 1h 36m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1e8eb4765ba1494089e23aed3ceb8046 : SUCCESS in 49m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9d3e47afef7146ea9f300f613883259c : SUCCESS in 1h 36m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23dce0371c9446f88f93e2d3cee5429a : SUCCESS in 1h 21m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cb4b577d37642a985d9a38585992f4b : SUCCESS in 59m 34s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"cc226e78342a36c81e3ad7f02101c9f7a3b279e3","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-16 16:47:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-84-1005884-1-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-84-1005884-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6c566bd1238db0905af52d55f473c5f09261bf12","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-16 16:53:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_09_2026_16_47_45\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"7c0615b3ab48947d9d096afcd8ac7c2e23a73dee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 19:08:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9867ed203fdc40ac8a2bad3f504e88e4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/50c37de56b0044e7aa40a898bae04386 : SUCCESS in 18m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a70862283c154bdb8c7ebc5e9a465eef : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0d1092a0c355470c8eb3db81fa1dd76e : SUCCESS in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6e35291118c34ee68b9ae2566df07a0c3c5ee39","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-17 05:59:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-84-1005884-1-fc : SUCCESS in 51m 50s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-84-1005884-1-iscsi : FAILURE in 6m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2e7131bdcf424e17f9e6498d1e62cfb39002c63","revisions":{"a2e7131bdcf424e17f9e6498d1e62cfb39002c63":{"kind":"REWORK","_number":1,"created":"2026-09-16 13:52:22.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/84/1005884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/84/1005884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/84/1005884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/84/1005884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/84/1005884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/84/1005884/1"}}},"commit":{"parents":[{"commit":"732a7d9bfe9908cda26798b254f0c1be2716ef33","subject":"Merge \"Fix exclusivity enforcement between replicated and non-replicated CGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/732a7d9bfe9908cda26798b254f0c1be2716ef33"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-16 13:37:54.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-16 13:49:03.000000000","tz":-240},"subject":"volume_utils: fail create_encryption_key on unencrypted types","message":"volume_utils: fail create_encryption_key on unencrypted types\n\nThis method currently returns None for an unencrypted type,\nwhich doesn\u0027t really serve a purpose, because it is never\ncalled for an unencrypted type. Change it to raise an exception\ninstead of returning None.\n\nChange-Id: I8280ee4f9a1f38db49265f31d29926e3e72b0bb1\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2e7131bdcf424e17f9e6498d1e62cfb39002c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2e7131bdcf424e17f9e6498d1e62cfb39002c63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
