)]}'
{"id":"openstack%2Fcinder~1005916","triplet_id":"openstack%2Fcinder~master~I0dda964cd7500127784f2f529c956b71c56bf95b","project":"openstack/cinder","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-16 17:03:23.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0dda964cd7500127784f2f529c956b71c56bf95b","subject":"WIP: mypy: Add coverage to cinder/quota.py","status":"NEW","created":"2026-09-16 16:23:22.000000000","updated":"2026-09-16 23:22:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"6ffe20e4db72ab86f0c0ca223aefc1b888d29f48","_number":1005916,"virtual_id_number":1005916,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 18:23:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":-1,"date":"2026-09-16 16:25:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 17:03:23.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-16 17:17:13.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-16 18:23:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 18:23:51.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-16 18:34:24.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-16 19:05:39.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-16 21:12:04.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"eff829cbf38a0e25f9f69072906dbe21b2ebdc1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-16 16:23:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebbca27cab7d52cdbc41e8f90e2f5a31ae244d85","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-16 16:25:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f3ddf598c1dd836fdcec6958f11e8427062fd6b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 17:03:23.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/599da4fe23114258bcdd3fb5c84ec64f/","accounts_in_message":[],"_revision_number":1},{"id":"4e801d1d2299e7bfa47d8b6b12bcc85037f06e41","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 17:17:13.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0dda964cd7500127784f2f529c956b71c56bf95b .","accounts_in_message":[],"_revision_number":1},{"id":"2c941f30aded551252802f12e17460285ee31d26","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 18:08:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/599da4fe23114258bcdd3fb5c84ec64f\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4fcccf9bf9c1472ebd1d5f165bd1d90a : SUCCESS in 40m 27s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/808c2785ccc846668439f52c44cffe6e : SUCCESS in 36m 14s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2d1cdf9ff1554a3784ef2061c26e56ea : SUCCESS in 56m 01s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1456e4f097ee4b9aae5b93fc6cd97c73 : SUCCESS in 39m 23s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d43f506c18f34e25b1d9c291706e5069 : FAILURE in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d27b2cb3ad63926ecf0b33a3394d62ef2d99db1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 18:23:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/524d94777fec474ba84ca3cf83a7c335\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6672c409c2344c8496c53fa7557c42ce : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c905f2d52444acd875bdd8b4da24daa : SUCCESS in 8m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1b1f8e4577c46caa425316e492b164e : SUCCESS in 12m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59077a98ac724547ac2fb22d5a3ec785 : SUCCESS in 12m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3780b0619d1e4657b9300a5bdb71540b : SUCCESS in 15m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/244aebabc86c43069d7afc2509401e06 : SUCCESS in 14m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb7d0444110148c2a21c2429e272e503 : SUCCESS in 2m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/3a57edadbd5d4569b20cba785c029f15 : SUCCESS in 1h 09m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5da05896a0ce436195939025dee09557 : SUCCESS in 1h 09m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f22fdaff3bf437a822c31ee5c098bf0 : SUCCESS in 1h 33m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0c5d2517dd146df8e6d9a4b4dedf1d0 : SUCCESS in 30m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa3c0ce0526b4e26bcad507dd9a45e27 : SUCCESS in 8m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/727f694c0951457fb3256c790e5053ae : TIMED_OUT in 41m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/acd5765ab0554acabdbb0c7986ae05ed : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b316ec90c5094677b8aca27a7b66175a : SUCCESS in 8m 05s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/53d10147cdc1449c9fc55d732dc0e96a : SUCCESS in 9m 38s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/5c6ef20c09d24516b10a7ac1faba5fc1 : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/65d5e56a3df34b92812928b8a8c76041 : SUCCESS in 7m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/72f5df7b8aee425e92a82d1222260319 : SUCCESS in 6m 39s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b3a10498b8ee4553b8c16c2b7b439d7b : FAILURE in 4m 22s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9e02532ab81546b49bce8c5266f46ca9 : SUCCESS in 8m 54s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b39c6cfb2dfb4ff38faa5307717b1c23 : SUCCESS in 38m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a58974d8e3bb43b3b33bed89bca5abe6 : FAILURE in 7m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/05e960c1fdd5423698d9206921436f4f : SUCCESS in 53m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f1a3201b3bb14368abf9a35e7ff7755d : FAILURE in 1h 02m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6a27e955cf8b4fe6866ab288899255c2 : SUCCESS in 1h 50m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/eb6c6543f82a402f972c4250d9ea2f35 : FAILURE in 1h 56m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e88e7709d66841f584e268812d5114ed : SUCCESS in 1h 56m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e323893bb31649e097e9151c98e697df : SUCCESS in 33m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1c7932a97c4245f5b3b4ce759b380329 : SUCCESS in 1h 03m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4dad15a59e5747e09eccfe63aad2bf0c : SUCCESS in 1h 29m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/32ab3f5fabdc44fc85d5362368248fbf : SUCCESS in 45m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fd29da7658047889750d9c035b7b8c3 : SUCCESS in 1h 32m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/231779c44b5b4af8aab887b1fef65e72 : SUCCESS in 54m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"29e8480fd4bfe159703bb44d95f06032d355c3b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-16 18:33:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_09_2026_18_27_27\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"9c978420eea165da296d4419e90ee454c59c316a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-16 18:34:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-16-1005916-1-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-16-1005916-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"34fef47e329fb05b100dba02956a17d01d2bff55","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-16 19:05:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/af38ff9c3cf44119bb8d9a1d3801197b : SUCCESS in 41m 03s\n- cinder-storpool-tempest-iscsi-multipath build/8a6628baa97a4a34bd3cbd229d3be9bb : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f9f33c0b56707eefa54a8504e0fb569c9811b192","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 21:05:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3b961d7c807408ab0ab0e816346a106\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/033ce141bc564b4a8dd984ada45665b6 : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e50b52f24a0b4335bd08baffe53cf91c : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/36fbfda535a64f48ae73fca3aead5689 : SUCCESS in 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13ab83abadd011eff4e51d528b0e42bb7c1d0126","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 21:12:04.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5394/.","accounts_in_message":[],"_revision_number":1},{"id":"9629f131f36eb8c9b24d27add7128ca46b9a1fff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 21:13:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5571/.","accounts_in_message":[],"_revision_number":1},{"id":"6ffe20e4db72ab86f0c0ca223aefc1b888d29f48","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-16 23:22:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-16-1005916-1-fc : SUCCESS in 51m 40s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-16-1005916-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7c0504442d9e17991fb6ed607c848c00d111bcc","revisions":{"f7c0504442d9e17991fb6ed607c848c00d111bcc":{"kind":"REWORK","_number":1,"created":"2026-09-16 16:23:22.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/16/1005916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/1005916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/1005916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/1005916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/1005916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/1005916/1"}}},"commit":{"parents":[{"commit":"6ec8f17514c973ee26fb09194caed07002518c0e","subject":"mypy: Add volume manager type coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ec8f17514c973ee26fb09194caed07002518c0e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-16 16:21:20.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-16 16:22:09.000000000","tz":-240},"subject":"WIP: mypy: Add coverage to cinder/quota.py","message":"WIP: mypy: Add coverage to cinder/quota.py\n\nGenerated-By: Claude Opus 4.8\nChange-Id: I0dda964cd7500127784f2f529c956b71c56bf95b\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f7c0504442d9e17991fb6ed607c848c00d111bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f7c0504442d9e17991fb6ed607c848c00d111bcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
