)]}'
{"id":"openstack%2Fcinder~1005946","triplet_id":"openstack%2Fcinder~master~I0c260524cecb340f5cf999cc768c9ac13a520601","project":"openstack/cinder","branch":"master","attention_set":{"38271":{"account":{"_account_id":38271,"name":"Carlos Henrique Felipe","display_name":"Carlos Henrique Felipe","email":"carlos.figueiredofelipe@windriver.com","username":"cfigueir"},"last_update":"2026-09-16 19:58:24.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0c260524cecb340f5cf999cc768c9ac13a520601","subject":"Fix NetApp ssc task stopping on first exception","status":"NEW","created":"2026-09-16 19:19:22.000000000","updated":"2026-09-17 02:46:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c831944358a8002b47db1eeba837d46054b1f9ca","_number":1005946,"virtual_id_number":1005946,"owner":{"_account_id":38271,"name":"Carlos Henrique Felipe","display_name":"Carlos Henrique Felipe","email":"carlos.figueiredofelipe@windriver.com","username":"cfigueir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 21:45:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 19:58:24.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-16 20:13:03.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-16 21:38:44.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-16 21:45:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 21:45:42.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-16 21:56:15.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-17 00:06:55.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"b13c6f766896fa5466d78db3898623e41c151c27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38271,"name":"Carlos Henrique Felipe","display_name":"Carlos Henrique Felipe","email":"carlos.figueiredofelipe@windriver.com","username":"cfigueir"},"date":"2026-09-16 19:19:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbccd2bfefb1574ea7489e89f70b5550a90edf49","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 19:58:24.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/02a893781201432196fb14c492001d73/","accounts_in_message":[],"_revision_number":1},{"id":"a04783201682ff835f45dc3e9b1faad5ebc7db86","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 20:13:03.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0c260524cecb340f5cf999cc768c9ac13a520601 .","accounts_in_message":[],"_revision_number":1},{"id":"15c272db7d48c67cb3c37271500c783f1d8198d1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 20:43:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/02a893781201432196fb14c492001d73\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/adbd915ecee240ff93329a17be2a8f7b : SUCCESS in 39m 10s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/74fb5631e0d64bf893297961eaf767f7 : SUCCESS in 35m 49s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d77c22abde264110a4a900318e17e9cf : SUCCESS in 50m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a579d74d79ff4afabe41125c3cb0b97b : SUCCESS in 38m 35s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3fb322728bb2490088ea0d7b1533b338 : FAILURE in 14m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b740cb5e85d9acc079fa5d9d2b44271786145672","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-16 21:38:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e59547df3ff34c87990ccc880cae8fb9 : SUCCESS in 50m 23s\n- cinder-storpool-tempest-iscsi-multipath build/55a693ae619b4bca8876e35099cd8619 : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cbe95bd598444aa48f4fe0a5611e86e9d57aa44c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 21:45:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53abe632cd18463b9348da7dda5c983b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7cfd3e3f54c4d16bb8f1101792f176b : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/939a513e1d6f4f4e83299a82e87ae95f : SUCCESS in 9m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b60fc710ee824e788b918c7f9e214f5b : SUCCESS in 9m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b28abf726c5446b8aeee49883c129551 : SUCCESS in 14m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bd02661898247ebbbef05e39944a156 : SUCCESS in 13m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/fb542f94456346dd88d96415d7ca3ad2 : SUCCESS in 59m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e163b7a112e442e80b7915ed28d4a9e : SUCCESS in 59m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/58e3e0b7b04a4a1095b470c18fec497c : SUCCESS in 1h 10m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4a87cabc59040d18c54fe381df6aa41 : SUCCESS in 40m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaf9e62ed3564481b210cf4d36df7830 : SUCCESS in 9m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ea12e763d0294823813dca0b9b156cb4 : SUCCESS in 24m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/49567d353b5d49cd907c828cae568427 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9a57edcb345d497a9c61d8bcc4ad17db : SUCCESS in 5m 04s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/3135e1517ef744e28ee3cf958ac52921 : SUCCESS in 10m 35s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4835f7b1bf8948c5aca98d12a2341a71 : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/06f2ecf16097448285ef73dbb9049e87 : SUCCESS in 8m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0029042bdf604349afc17ef417532ade : SUCCESS in 5m 57s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c92a0f67475f47dbb56eee3ec40e4b8d : FAILURE in 5m 18s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/601adfaf4faf40228f0e9f295bae9dd3 : SUCCESS in 7m 53s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b23b22ceec914e95835dfc8d4135a213 : SUCCESS in 18m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f18f72c53b7541178c22ab63b58a6cd6 : FAILURE in 3m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/75bf56bff13b4530996fd96875514a53 : SUCCESS in 1h 36m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/be5bb0da52114bf5ae1c991fcd804e18 : FAILURE in 2h 10m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/832c19f47e8440e6a7cf425aefaa3e20 : SUCCESS in 1h 44m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/f23e47316a3a49faa4d2c21e66758155 : FAILURE in 2h 20m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bb90e0fa2bd841028c2126c27e397f5b : SUCCESS in 1h 49m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1e3a0497b31149ef842bbcba2cea3ac2 : SUCCESS in 26m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bc7cbe959102459c8af303d6df00f0ca : SUCCESS in 1h 17m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6d95c06e2c0643cb9b746132a30ba1e4 : SUCCESS in 1h 52m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/19568ea4197c4c7c83bab5212e08c407 : SUCCESS in 27m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c287348131584eb4a71b5281fa134ad1 : SUCCESS in 59m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f56e04a11f6546019488b4fb0fce746b : SUCCESS in 53m 40s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"19a7e4b4e0c871d22a60bad8811fee4f47e3aea3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-16 21:55:22.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_09_2026_21_49_30\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"c107207416fa70400a15abbae75327cdffd2d521","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-16 21:56:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1005946-1-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-1005946-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5407b84bb2465800ebb9e5ddb04caa5e132295ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 22:16:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9d9326a0ba04b38a0c65e9bde1b230b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c77525667a654574828693fc1c5ee7ce : SUCCESS in 16m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a1b3098de3b44a03943c1547d4cb55bf : SUCCESS in 21m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f1b3878f87c84d068024418046eb89df : SUCCESS in 25m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c93e48a79eca6ee3e78b90dfefa2c4c774193826","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 00:06:55.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5397/.","accounts_in_message":[],"_revision_number":1},{"id":"c937784cfad34f0d46d469d3a028a9a63ed80370","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 00:34:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5574/.","accounts_in_message":[],"_revision_number":1},{"id":"c831944358a8002b47db1eeba837d46054b1f9ca","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-17 02:46:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-46-1005946-1-fc : SUCCESS in 1h 3m 32s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-46-1005946-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2ad86ad3823cda2892faf4c894dd81cefac92b2","revisions":{"a2ad86ad3823cda2892faf4c894dd81cefac92b2":{"kind":"REWORK","_number":1,"created":"2026-09-16 19:19:22.000000000","uploader":{"_account_id":38271,"name":"Carlos Henrique Felipe","display_name":"Carlos Henrique Felipe","email":"carlos.figueiredofelipe@windriver.com","username":"cfigueir"},"ref":"refs/changes/46/1005946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/1005946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1005946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1005946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/1005946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/1005946/1"}}},"commit":{"parents":[{"commit":"732a7d9bfe9908cda26798b254f0c1be2716ef33","subject":"Merge \"Fix exclusivity enforcement between replicated and non-replicated CGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/732a7d9bfe9908cda26798b254f0c1be2716ef33"}]}],"author":{"name":"cfigueir","email":"carlos.figueiredofelipe@windriver.com","date":"2026-09-15 19:35:26.000000000","tz":-180},"committer":{"name":"cfigueir","email":"carlos.figueiredofelipe@windriver.com","date":"2026-09-16 19:19:19.000000000","tz":-180},"subject":"Fix NetApp ssc task stopping on first exception","message":"Fix NetApp ssc task stopping on first exception\n\nWhen scheduling the `_update_ssc` looping task the NetApp ONTAP\nimplementation would not allow the task to run again when an exception\nwas raised during the task execution.\n\nIn a scenario where the `_update_ssc` routine might fail due to\ntemporary external failures (e.g. networking, DNS), storage catalog\ninformation would no longer be updated requiring a service restart.\n\nThis change exposes the looping call mechanism\u0027s parameter\n`stop_on_exception` and changes the registering of `_update_ssc` task\nin order to allow it to recover from an exception and not be stopped\nafter a failure.\n\nTEST PLAN:\n[PASS] Updated Cinder service unit tests (LoopingCallsTestCase,\n\tNetAppBlockStorageCmodeLibraryTestCase)\n[PASS] Test in live scenario with NetApp configured as backend storage\n\tprovisioned for iSCSI, NFS. Verified that the catalog update\n\ttask continuously runs after and exception has been raised.\n\nCloses-Bug: #2167491\nChange-Id: I0c260524cecb340f5cf999cc768c9ac13a520601\nSigned-off-by: cfigueir \u003ccarlos.figueiredofelipe@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2ad86ad3823cda2892faf4c894dd81cefac92b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2ad86ad3823cda2892faf4c894dd81cefac92b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
