)]}'
{"id":"openstack%2Fcinder~1006007","triplet_id":"openstack%2Fcinder~master~I4e39c58c87b3f0671f631c633bb2134a27ed228a","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-17 19:57:03.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-09-17 11:38:50.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-09-18 17:44:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e39c58c87b3f0671f631c633bb2134a27ed228a","subject":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured","status":"MERGED","created":"2026-09-17 09:49:16.000000000","updated":"2026-09-18 17:47:05.000000000","submitted":"2026-09-18 17:44:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006007","meta_rev_id":"8d382ef1f355ac509a1fc434b34fb0a16e94e095","_number":1006007,"virtual_id_number":1006007,"owner":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 17:44:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"all":[{"value":2,"date":"2026-09-18 14:46:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":2,"date":"2026-09-17 19:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"all":[{"value":1,"date":"2026-09-18 14:46:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-09-17 19:09:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 09:52:50.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-09-17 09:52:50.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-09-17 10:44:23.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 10:49:13.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-17 10:50:23.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-17 11:09:12.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 11:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 12:40:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 16:57:29.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-09-17 21:45:40.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-18 03:11:05.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-18 14:46:41.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"}],"messages":[{"id":"f28aa6a45e280c87daadcf84208e6cf88a5634bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-09-17 09:49:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aefe6a0e157324309d6076c897638ca4b3d2ba9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-09-17 10:13:04.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bc6ddb77e699d42f015f8647022b67a093b114fa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 10:44:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4e39c58c87b3f0671f631c633bb2134a27ed228a .","accounts_in_message":[],"_revision_number":1},{"id":"fcbeff67543950b92eebb3c3f95cfeead1bac670","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 10:49:13.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b3150dc4bf0b4eb985a101b815c01f4e/","accounts_in_message":[],"_revision_number":2},{"id":"23d4d9db6d55f08a1153991ab3069ec153c14915","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 10:50:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5578/.","accounts_in_message":[],"_revision_number":1},{"id":"f9e53c9b6dd738b8552cff6fa938f2d0a73ca794","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 10:54:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5401/.","accounts_in_message":[],"_revision_number":1},{"id":"57fbdce15184afab8b27e05e279e8034c90ebd37","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 11:06:32.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4e39c58c87b3f0671f631c633bb2134a27ed228a .","accounts_in_message":[],"_revision_number":2},{"id":"096989c3d54fa3e4d16c291b0dd1824abb0626b3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-17 11:09:12.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ac08b365307b44bca5b2ac29e58ea2a2 : SUCCESS in 46m 04s\n- cinder-storpool-tempest-iscsi-multipath build/217b3a1dd467462a908e255cf5b83a2d : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":2},{"id":"4cb88dd5b7f1b6ce78c76a3a9a99c8c2ad3b2a9a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 11:24:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b3150dc4bf0b4eb985a101b815c01f4e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/992903c28f3647c2b4d5927a5e76639d : SUCCESS in 40m 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/af216ec3c7404de78669a05ca1496f37 : SUCCESS in 35m 41s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/46130adae3a0447fb75f6b6ceff996b7 : SUCCESS in 49m 50s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8984fe3cd58744c4ae55438b843861ff : SUCCESS in 38m 16s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/470e5981a0b34d3e88fea8f197b5d334 : FAILURE in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"90758de8303bee9aaff423df34d06b1ef0080ce7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 11:36:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c78ac40983d4117a0073c7bcf794064\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb01d357cc9e4a99aceba3653f17c41a : FAILURE in 19m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b477193435314c5b9cdbb8b12d801284 : FAILURE in 11m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/47d41a7b7ca941149d6ca74cf00eebf7 : FAILURE in 29m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eb76ea22cbdc5be3ad30b33afd6196fccea81fc5","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-09-17 11:38:50.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"a7dc896eed8f072a79ac3490260ad54083a4e24d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 11:57:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5579/.","accounts_in_message":[],"_revision_number":2},{"id":"71e659c74c9e5ac5ff8d1fef8e508ecce255ebdc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 12:16:52.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5402/.","accounts_in_message":[],"_revision_number":2},{"id":"fdb45f62f5d04f4f491c183707edf16298e15c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 12:40:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9468c586db28411f9be3bb509bc4dd64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e35b7fd0fbb14f6ba0ee0488bf0f6998 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f89ca93e284347deb9d3de68952777eb : FAILURE in 11m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26725245f38c448f9966c70a92e710c4 : FAILURE in 12m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38a397aaf82b4c1fb580017a08cde1c6 : FAILURE in 15m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd306325b064a4f830f45b468b39b8b : SUCCESS in 14m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/b4a91de886b440c89765c957bc358dd0 : SUCCESS in 1h 04m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3f0e5518f884e87b48289a68df94bfb : SUCCESS in 59m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de7e77a32c7e4d3a8bc992c3b6ed7aec : SUCCESS in 1h 30m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c45efed7aaef418eb5edffe8b3d5c8ef : SUCCESS in 1h 03m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f92e4ca4c7d4ef7a93d211808a9f0a6 : SUCCESS in 9m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7ad72f9963704d5e915beb36accc6896 : TIMED_OUT in 41m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ec737e577e624d8eb5d092fc40ad694e : SUCCESS in 4m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6852f5182ebb4b6bbce72804a9790513 : SUCCESS in 8m 11s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/e543621087c34381984814a045f8f51e : FAILURE in 12m 13s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c6bd7324196249258915224f8d4615fc : FAILURE in 11m 08s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/50dfbf6b2ddb4ec6b514138af83114c9 : SUCCESS in 10m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0b1da199fe574f7fb47d6db7b7f015a5 : SUCCESS in 7m 28s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3a8f8e9c72134241a3226cf61a98543d : FAILURE in 7m 34s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e1fabec43ee84e429da1a41fdf525d55 : SUCCESS in 11m 53s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb2b251cb9bc4d5b9dd9826bb6ec3f89 : SUCCESS in 36m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/41248478188a42d38a874508a53b9bdc : FAILURE in 7m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2d89363412e441e9b1ad31d9d31e1c34 : SUCCESS in 1h 43m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e11862bc89844cf09f2ff36a29a546c2 : FAILURE in 2h 16m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea3359ea53524ffeaa0db7dc2f572dcf : SUCCESS in 1h 34m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b299eb648d4f4defa8e976cee4d6c64b : FAILURE in 1h 54m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e7def3961c2c4d3fb8077af22576c030 : SUCCESS in 1h 48m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c428021e75de4512aa20f8744ec34e0a : SUCCESS in 40m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1a596c8de8bf4eb7b40546f1c7cb2d02 : SUCCESS in 1h 17m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/10f666ba16f54ac185a5fd1d7441709a : SUCCESS in 1h 23m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/56c4fdc9a4224c189580a9ff61ed5582 : SUCCESS in 42m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9375bbecdf38479cbf9621bb0723f3ba : SUCCESS in 1h 34m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfb0eb6f624a47a997b7398b7b62f5fc : SUCCESS in 57m 42s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"3dded3da94a7e606221d688fd1afb0cfdbb2d13a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-17 13:29:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d0959275f6dd90feb7a45b4a36d04d351312377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-09-17 16:20:18.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ce11c00ce2be1477980d55c01eab2959a50b21e4","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-09-17 16:21:51.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c10e287965d251b8552b29e86f34f5f6ee929061","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:37:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe81a2110704b45b9f8787d860754c2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/989218e467da48dfa791043e587cc589 : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/36fc5678dd4f4d06ae8da86f2ae89362 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1653d859b4d34160b32691af2abac704 : SUCCESS in 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"78b0badbef5f037ed053173289418c3921d72f5b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 16:38:34.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/191dedae7c6846378e57206e1d641da9/","accounts_in_message":[],"_revision_number":3},{"id":"3cf59c31c178d9e765438dda72194db66b1ff0d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-09-17 16:57:29.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/972/","accounts_in_message":[],"_revision_number":3},{"id":"816c71a592e84029d773ca79db5a7fa0b9955563","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 17:07:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/191dedae7c6846378e57206e1d641da9\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4cc11cb578c74effb73635e744cce862 : SUCCESS in 36m 40s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/72a49648e4bf42b0a95caae7bee14b89 : SUCCESS in 37m 11s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d0e3ec57e8e9496b89c3323fbf742816 : SUCCESS in 46m 48s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/18870e61a60a46aaa91492add7ce2c29 : SUCCESS in 32m 59s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1055fd8844eb45b9af2c07de09bc272e : FAILURE in 14m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7fc965bb9c03a7b70960880ca80149da913374cc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 17:15:53.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4e39c58c87b3f0671f631c633bb2134a27ed228a .","accounts_in_message":[],"_revision_number":3},{"id":"61984b07308445fbf9c9ac72e8bde9cc6ad0f1ce","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-17 17:17:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/337c6ff4af8a4bb984f645210fdaa1dd : SUCCESS in 55m 13s\n- cinder-storpool-tempest-iscsi-multipath build/2b49b992b142402eb950ae8d871c9e31 : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":3},{"id":"a15ab4119c8ac2f1ecf9ba431d61ad73cff7197d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-09-17 17:23:49.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/1006007/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/1006007/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"77577686470114358e68bf0450a15cc8570a96be","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 17:23:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5407/.","accounts_in_message":[],"_revision_number":3},{"id":"7fd306656888df785331f8e539ee11784af92de9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 17:28:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5584/.","accounts_in_message":[],"_revision_number":3},{"id":"b7e360ae9f00ad5ee3ca134b5b165dca02ff009b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 18:39:40.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a05509710514e99a7a8e17439d93fd0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89cbf848778f4b42a40f4c2b05a43287 : FAILURE in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2947eff58a2f44339644b0b157e3f596 : SUCCESS in 12m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c748bdd2385841b8a507c7ded597b192 : SUCCESS in 11m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/52ab5cc877884f2587f5348e9a6f2206 : TIMED_OUT in 40m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb47cc0d70c4c2bae8ba17703b5d20c : SUCCESS in 15m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/288cfca2d0e049bf99f05489b47c10c4 : SUCCESS in 33m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c3d76a3ef1464921b4d1adfedfab8280 : SUCCESS in 1h 04m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f1fb85049891451abfb672e2a0db8c88 : SUCCESS in 1h 34m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dea6cafa77e24e1e807d9c499574e3e3 : SUCCESS in 1h 00m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22dcc75c078a4eb880142548ba378336 : SUCCESS in 8m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e1a69179ca614f2f9e49d92eab71f9ef : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a7549167624d48bdb6af41c48ec13356 : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3538d217d64c48168b5f3cd65c9340e5 : SUCCESS in 3m 27s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/0394a2fadbac4702854b1f25294d78ce : SUCCESS in 11m 03s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d39719a750804240b76254f61db1fdb4 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0022e7a84a4247cc9a8da3534e1e195b : SUCCESS in 5m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dd0a528744104a1d9329c0fa3ca7648a : SUCCESS in 6m 19s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b6142c8a396546f8b26b92f1d20736a5 : FAILURE in 6m 18s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/86a0b3b1d6fd4fd99865e047cfcda991 : SUCCESS in 8m 41s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b70b1d9fbd664d53a6181be803a45d0c : SUCCESS in 32m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/986f48b2a347429dad1302edf28f2448 : FAILURE in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d27f0800d96f4b0d925bf17a72ce71ae : SUCCESS in 1h 31m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a2aac7402eab412c807d45a2944f458b : FAILURE in 2h 02m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1b8ed47d74a84dbcaa37130eda64de0a : SUCCESS in 1h 57m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/8de13dc5f4f44a9dbae38792763a81b6 : FAILURE in 2h 16m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2d143ca1fd4a4f6cbd229fd2fb736665 : SUCCESS in 1h 20m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/38d90442a624436f84356535ace19bdc : SUCCESS in 24m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/096f9cbbe86a443e9353d8c316a546d2 : SUCCESS in 1h 06m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/75590beb064e410b8a650ad8e023158d : SUCCESS in 1h 17m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/69c62f61c6ff4791b2a9867befff168d : SUCCESS in 44m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a922e532b0f34d0b83ad716066896a4d : SUCCESS in 1h 33m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d67f10504e94c62882f83866988f467 : SUCCESS in 25m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"acc6957571d6a00f84c04219dc1be3105d8475ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-09-17 19:09:44.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"e698e3170d97fea68466f3199584d9db15591317","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 19:27:05.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b324bf38912c404782a7f1a0d64d0acd/","accounts_in_message":[],"_revision_number":4},{"id":"2d623972ff0f7ceb78227169faa2b7ae40577430","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 19:37:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daa5d1698e124011b9ca47bd181fa085\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c2df3928bda4d8bb6a3b1e2a90c8685 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/59a6946d08494cdf9d8cf1f5109cf7e9 : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f79403f5e9ba44f09e1674ebb259f775 : SUCCESS in 24m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b1b6923b2df1d28ddc0250d646197b9d251e73c5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-17 19:57:03.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2af2e71b57a8550854f8edc6058873072aec2996","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 19:58:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b324bf38912c404782a7f1a0d64d0acd\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/27bcd83b49ba44699c21423d2031a653 : SUCCESS in 42m 46s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/57e59cb3f1ec43aa9906a6fe2f63e65c : SUCCESS in 35m 21s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/932cb0986ed34ecc836a9ddc46bb6adc : SUCCESS in 48m 18s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/79d15288a05d480d97fe2ee328ea419d : SUCCESS in 34m 32s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/64c60cb4325a4109998020080da9e176 : FAILURE in 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8af0d07fdab8a195f51b3ea17acc8d8cadd70d74","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 20:04:02.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4e39c58c87b3f0671f631c633bb2134a27ed228a .","accounts_in_message":[],"_revision_number":4},{"id":"c942bb80a2c3e86486b31756453ca15f1149b850","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-17 20:28:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/aabf5c7f88484ff1923d3dd63af7ef86 : SUCCESS in 44m 31s\n- cinder-storpool-tempest-iscsi-multipath build/d57b49f3f96244f8ad4f53f2a2602395 : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":4},{"id":"30e452ae8a8ec71dd555b4f38304bb470104060d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 20:51:54.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5410/.","accounts_in_message":[],"_revision_number":4},{"id":"85dc03612253a0e9b6af6787b3db85a4f03551a0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 20:57:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5587/.","accounts_in_message":[],"_revision_number":4},{"id":"1aa460adcef352b21f6d7738727adbb09a87fb22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 21:45:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90ad40a9d7ec434c8ad1e2986cb4e2ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c312ef0f4eb44f67b37de8ae4aed1e33 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39d91f4cb6944885a1372cd8d0837f02 : SUCCESS in 11m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76145b4a0a0948a5aef295423a7b8e82 : SUCCESS in 9m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6cf4f687aa3f441796260821278078ce : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062f309a059247acad1ba5bafa96a074 : SUCCESS in 12m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/cddabce87efd4a2b96c303e4e377be30 : SUCCESS in 1h 00m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c560c9336e614617b3d599062ffed8ea : SUCCESS in 1h 02m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c94ebbf58c7494d9eba69e25f356629 : SUCCESS in 1h 29m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4003085728604c63b08d8810bd58ffc2 : SUCCESS in 53m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfba8bd26cd9466f9d00afc4cf562928 : SUCCESS in 11m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ad5f45f6a9354d91998b3c09e1fec6a4 : SUCCESS in 21m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/577085980c8e43ffa2ef826df7344fb3 : SUCCESS in 6m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8cb5e27f2f4f4f5ba5379b1a523da001 : SUCCESS in 7m 04s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/f90060b58f91487b861baf6cd7fb81af : SUCCESS in 11m 51s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9d449e4d84f2466aa15f913bd9705faa : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9013a2ba929542528d275153d5d0832d : SUCCESS in 7m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a837fca4590b4094bbb40c6c3b5781da : SUCCESS in 5m 28s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8e8d2d36b8634e29a6012cb34d1dca56 : FAILURE in 5m 38s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9858599e68f44e48876692309a8174a9 : SUCCESS in 8m 54s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/73b45e6578fd45eaab87f5769df40aa2 : SUCCESS in 31m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f7a9a4c960e4a509e24798b467fbd02 : FAILURE in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a057f08338fc42048c002372b87d6103 : SUCCESS in 1h 32m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/20fe4d23c612482b92b585c97730f861 : FAILURE in 2h 13m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d11b888e3574d399b84315ad22d6485 : SUCCESS in 1h 42m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/d80027d52c2f41b89d61fd2c5562d95e : FAILURE in 2h 30m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/70767700c9ea419687193ad1877ecde9 : SUCCESS in 1h 31m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0c0905b4d7ab4bf4b4885b8385cbadb5 : SUCCESS in 43m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/921ee7898e5a420d8f6433555f2fd98a : SUCCESS in 1h 12m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bd0bc9e5d1a141d4bd5b5dda900e0ec5 : SUCCESS in 1h 26m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ce3747b6f11840968f1439902c8c9894 : SUCCESS in 31m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7674461a4634e6cbfc2bcee904b1f26 : SUCCESS in 1h 48m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38d023d22bd04f0187c8035aba55f450 : SUCCESS in 1h 02m 06s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"6853a4cf4672cf401ef85d64b6fafa9487b5c93d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-17 21:55:15.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_09_2026_21_49_10\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"ae23caca9bc3ca323b30c4cd067af766dd2f1c65","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-18 03:11:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-07-1006007-4-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-07-1006007-4-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"c6b8f89f23253b8aabee0412118d5a92d03065dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-09-18 14:28:20.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/973/","accounts_in_message":[],"_revision_number":4},{"id":"91f70b35e81d2cc3e971920c12e51eb5bffe47da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-09-18 14:44:51.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/1006007/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/1006007/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"d6fa45dabe7b1d2684042af508fd2f871c26342e","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-18 14:46:41.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd7a5ec5b15192664fe895af47a0d7e07e5839fc","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-18 14:46:48.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5f74d3a645c3f2d43e68242e1409ff10f764ece0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 14:47:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2640084d5a8d97030c0db22ad0db95c410442bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:44:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10f5ff5822384780b22f453dfa001e2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8938730db01e49ea8f47cf31c85a2c88 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e98cc96bcf0f40978dba433db0646088 : SUCCESS in 11m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b400b0deb954dee873c2d086447b1e6 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d30d85e449840bdb10db7462e469840 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/c06a13a139e44a9fb4cb60932a958e02 : SUCCESS in 50m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ec5a19bab7a42b9ae3b93bad92f11e7 : SUCCESS in 1h 15m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d35c647506574751b3f91379c22974d5 : SUCCESS in 1h 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d3618578ea64337b9a6652f43bb67b8 : SUCCESS in 41m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab26dfd3694f434892b91fc9686134d6 : SUCCESS in 7m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/658ecfe4b67d49339a6b8433f3214088 : SUCCESS in 37m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6132bfa37b794f88b4714f1f930984d5 : SUCCESS in 49m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c7b6232a4844e849d1679499611a213 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":4},{"id":"9074c3b0dd22a08567e2a6f0e5d93a7488c69b3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:44:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8d382ef1f355ac509a1fc434b34fb0a16e94e095","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:47:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d7c69bc9f2e4b59b3b48bcd24b66c8a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e410a37743454ab2141c1cabbb4e12 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9017f8762a264cc882b5643aeb43b3a5 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"602213d59ac73ca67bf0ea4db866ccc6ddf81f63","revisions":{"7418d7f951cafa8e5d16c607b36b4887e1727d33":{"kind":"REWORK","_number":1,"created":"2026-09-17 09:49:16.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/07/1006007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/1006007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/1006007/1"}}},"commit":{"parents":[{"commit":"9b31bd509c5eed4bdc65184b6619c15445a20c02","subject":"Merge \"Replace eventlet spawn calls in volume drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b31bd509c5eed4bdc65184b6619c15445a20c02"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-09-17 09:44:49.000000000","tz":0},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-09-17 09:44:49.000000000","tz":0},"subject":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured","message":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured\n\nDuring driver initialization, _update_storwize_state() unconditionally\ncalled helper.add_iscsi_ip_addrs() to query iSCSI node IPs. On systems\nwith code level \u003e\u003d 8.4.2.0, this executes \u0027lsip\u0027 for the configured\nportset (default portset0).\n\nWhen running the FC driver (StorwizeSVCFCDriver) on a backend where no\niSCSI IP addresses or portsets are configured, add_iscsi_ip_addrs()\nraises an InvalidInput exception (\"No nodes found in the given portset\"),\ncausing driver initialization to fail before add_fc_wwpns() can run.\n\nThis change avoids running add_iscsi_ip_addrs() when the FC protocol is\nconfigured (and conversely avoids add_fc_wwpns() when iSCSI is configured),\nallowing FC-only environments without node IP configuration to initialize\nsuccessfully.\n\nRefer the below change that caused this issue-\nhttps://review.opendev.org/c/openstack/cinder/+/874633\n\nCloses-Bug: #2167561\nChange-Id: I4e39c58c87b3f0671f631c633bb2134a27ed228a\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7418d7f951cafa8e5d16c607b36b4887e1727d33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7418d7f951cafa8e5d16c607b36b4887e1727d33"}]},"branch":"refs/heads/master"},"8c1834fc80098d5312b7b711075253a8b4f16a36":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-17 10:13:04.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/07/1006007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/1006007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/1006007/2"}}},"commit":{"parents":[{"commit":"9b31bd509c5eed4bdc65184b6619c15445a20c02","subject":"Merge \"Replace eventlet spawn calls in volume drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b31bd509c5eed4bdc65184b6619c15445a20c02"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-09-17 09:44:49.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2026-09-17 10:13:04.000000000","tz":0},"subject":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured","message":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured\n\nDuring driver initialization, _update_storwize_state()\nunconditionally called helper.add_iscsi_ip_addrs() to\nquery iSCSI node IPs. On systems with code level \u003e\u003d 8.4.2.0,\nthis executes \u0027lsip\u0027 for the configured portset\n(default portset0).\n\nWhen running the FC driver (StorwizeSVCFCDriver) on a\nbackend where no iSCSI IP addresses or portsets are\nconfigured, add_iscsi_ip_addrs() raises an InvalidInput\nexception (\"No nodes found in the given portset\"),\ncausing driver initialization to fail before\nadd_fc_wwpns() can run.\n\nThis change avoids running add_iscsi_ip_addrs() when\nthe FC protocol is configured (and conversely avoids\nadd_fc_wwpns() when iSCSI is configured), allowing\nFC-only environments without node IP configuration\\\nto initialize successfully.\n\nRefer the below change that caused this issue-\nhttps://review.opendev.org/c/openstack/cinder/+/874633\n\nCloses-Bug: #2167561\nChange-Id: I4e39c58c87b3f0671f631c633bb2134a27ed228a\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c1834fc80098d5312b7b711075253a8b4f16a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c1834fc80098d5312b7b711075253a8b4f16a36"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9ea75f32bf4586893a31026da0f554d7f433c7df":{"kind":"REWORK","_number":3,"created":"2026-09-17 16:20:18.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/07/1006007/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/1006007/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/1006007/3"}}},"commit":{"parents":[{"commit":"9b31bd509c5eed4bdc65184b6619c15445a20c02","subject":"Merge \"Replace eventlet spawn calls in volume drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b31bd509c5eed4bdc65184b6619c15445a20c02"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-09-17 09:44:49.000000000","tz":0},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-09-17 16:19:52.000000000","tz":0},"subject":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured","message":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured\n\nDuring driver initialization, _update_storwize_state()\nunconditionally called helper.add_iscsi_ip_addrs() to\nquery iSCSI node IPs. On systems with code level \u003e\u003d 8.4.2.0,\nthis executes \u0027lsip\u0027 for the configured portset\n(default portset0).\n\nWhen running the FC driver (StorwizeSVCFCDriver) on a\nbackend where no iSCSI IP addresses or portsets are\nconfigured, add_iscsi_ip_addrs() raises an InvalidInput\nexception (\"No nodes found in the given portset\"),\ncausing driver initialization to fail before\nadd_fc_wwpns() can run.\n\nThis change avoids running add_iscsi_ip_addrs() when\nthe FC protocol is configured (and conversely avoids\nadd_fc_wwpns() when iSCSI is configured), allowing\nFC-only environments without node IP configuration\nto initialize successfully.\n\nRefer the below change that caused this issue-\nhttps://review.opendev.org/c/openstack/cinder/+/874633\n\nCloses-Bug: #2167561\nChange-Id: I4e39c58c87b3f0671f631c633bb2134a27ed228a\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ea75f32bf4586893a31026da0f554d7f433c7df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ea75f32bf4586893a31026da0f554d7f433c7df"}]},"branch":"refs/heads/master"},"602213d59ac73ca67bf0ea4db866ccc6ddf81f63":{"kind":"REWORK","_number":4,"created":"2026-09-17 19:09:44.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/07/1006007/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/1006007/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/1006007/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/1006007/4"}}},"commit":{"parents":[{"commit":"9b31bd509c5eed4bdc65184b6619c15445a20c02","subject":"Merge \"Replace eventlet spawn calls in volume drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b31bd509c5eed4bdc65184b6619c15445a20c02"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-09-17 09:44:49.000000000","tz":0},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-09-17 19:09:05.000000000","tz":0},"subject":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured","message":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured\n\nDuring driver initialization, _update_storwize_state()\nunconditionally called helper.add_iscsi_ip_addrs() to\nquery iSCSI node IPs. On systems with code level \u003e\u003d 8.4.2.0,\nthis executes \u0027lsip\u0027 for the configured portset\n(default portset0).\n\nWhen running the FC driver (StorwizeSVCFCDriver) on a\nbackend where no iSCSI IP addresses or portsets are\nconfigured, add_iscsi_ip_addrs() raises an InvalidInput\nexception (\"No nodes found in the given portset\"),\ncausing driver initialization to fail before\nadd_fc_wwpns() can run.\n\nThis change avoids running add_iscsi_ip_addrs() when\nthe FC protocol is configured (and conversely avoids\nadd_fc_wwpns() when iSCSI is configured), allowing\nFC-only environments without node IP configuration\nto initialize successfully.\n\nRefer the below change that caused this issue-\nhttps://review.opendev.org/c/openstack/cinder/+/874633\n\nCloses-Bug: #2167561\nChange-Id: I4e39c58c87b3f0671f631c633bb2134a27ed228a\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/602213d59ac73ca67bf0ea4db866ccc6ddf81f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/602213d59ac73ca67bf0ea4db866ccc6ddf81f63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
