)]}'
{"id":"openstack%2Fcinder~1006042","triplet_id":"openstack%2Fcinder~stable%2F2026.2~Iac62ffb5f208bd6916ffd0253065d7c4bda55167","project":"openstack/cinder","branch":"stable/2026.2","topic":"change-1003601","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-17 15:53:52.000000000","reason":"Change was abandoned"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-17 15:53:52.000000000","reason":"Change was abandoned"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-09-17 15:53:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iac62ffb5f208bd6916ffd0253065d7c4bda55167","subject":"NetApp: disable clone_across_pools","status":"ABANDONED","created":"2026-09-17 14:31:05.000000000","updated":"2026-09-17 17:55:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a3b6861bfa549661b122dc5c9a82d6e087e3d79","_number":1006042,"virtual_id_number":1006042,"owner":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-17 14:57:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 14:31:42.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-09-17 14:31:42.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-09-17 14:57:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 14:57:49.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-17 15:06:35.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-17 15:08:19.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-17 15:24:10.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"303ccf0299a6446e8bf0cda0a45b8214f26cf0da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-17 14:31:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26dc43d0e48a6d8621652bb3092a34e13ff1d8b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 14:57:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b68740de523348f9b10ccfd0dd632b36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21de4751820c4ba69a8f9c100c23de8f : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0fe14d7a6b14c31aa1f1eda55854af5 : SUCCESS in 12m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7152f4331214c6d868165beab233930 : SUCCESS in 11m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa8972cddc674514a603ad6f86a793ba : SUCCESS in 14m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c769f0f50c87452584fb49c0afdff292 : SUCCESS in 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01d62b483e6c464e92f300a64af53b51 : SUCCESS in 9m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5f8982459fad46749837885389a4bfae : SUCCESS in 20m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b1eb6f9c6c6e49c4a5ea1d44a4c63e6e : SUCCESS in 6m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b13501e3a15d4968a6321ebf067f3462 : SUCCESS in 5m 38s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/e5009e78ce94406d9ef0402ac0d8230c : FAILURE in 12m 07s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/bcf8a82f49854d38acb95725c6fc96fb : FAILURE in 11m 50s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2b3b5892273e4af8a7770bf5cbf2b799 : SUCCESS in 6m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7645a90d06e849bb9534fc5d31c7c6ce : SUCCESS in 6m 49s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/61b64abc479042e9b619456a35b6af28 : FAILURE in 6m 52s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/37f40360187e40969c7dfdb5a38f5c63 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3da81203d7384bd996d128d5e261e530 : FAILURE in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f187c01b2da8c4029f11dcd9ab2172dbae666d3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 15:06:35.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5583/.","accounts_in_message":[],"_revision_number":1},{"id":"466bbb122a6ff74e6f6cd3b4cf63994d2c376618","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-17 15:07:12.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_09_2026_15_01_31\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"e3a3f9fcafa24f362750627bd041efaeb17bcb11","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-17 15:08:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-42-1006042-1-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-42-1006042-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"48b9716f0bf16775df6f6cda1db11688858fa2ac","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 15:24:10.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iac62ffb5f208bd6916ffd0253065d7c4bda55167 .","accounts_in_message":[],"_revision_number":1},{"id":"79244847bac1a5a0951d29ab51ee5076458441b0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 15:29:03.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5406/.","accounts_in_message":[],"_revision_number":1},{"id":"2831a84193373ce8d0475641125e4e5097b80766","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-17 15:53:52.000000000","message":"Abandoned\n\nChange merged to master before stable/2026.2 was cut","accounts_in_message":[],"_revision_number":1},{"id":"6a3b6861bfa549661b122dc5c9a82d6e087e3d79","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-17 17:55:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-42-1006042-1-fc : SUCCESS in 49m 13s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-42-1006042-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf16a7f41f0b6a04bd8516bc7a1ca4eaa9bfa73d","revisions":{"bf16a7f41f0b6a04bd8516bc7a1ca4eaa9bfa73d":{"kind":"REWORK","_number":1,"created":"2026-09-17 14:31:05.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/42/1006042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/1006042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/1006042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/1006042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/1006042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/1006042/1"}}},"commit":{"parents":[{"commit":"8e68009860b632fdaf9aaa3a3b7d2181c10d273b","subject":"Merge \"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e68009860b632fdaf9aaa3a3b7d2181c10d273b"}]}],"author":{"name":"Jayaanand Borra","email":"jayaanand.borra@netapp.com","date":"2026-09-08 11:41:34.000000000","tz":-240},"committer":{"name":"Jayaanand Borra","email":"jayaanand.borra@netapp.com","date":"2026-09-17 14:30:23.000000000","tz":-240},"subject":"NetApp: disable clone_across_pools","message":"NetApp: disable clone_across_pools\n\nThe clone across pools capability was introduced by Change\nI3358109b1c6997961c1cb73e114dd938fb9ff1e2 during Hibiscus\ndevelopment. The implementation works, but subsequent performance\ntesting has revealed bugs and inefficiency that make it unsuitable\nto ship in this release. Those issues will be addressed in patches\nduring the Indri development cycle.\n\nDisable clone_across_pools for Hibiscus so image-cache clones use\nGlance download plus same-pool FlexClone instead of lun-copy.\nSame-pool clones are unchanged.\n\nChange-Id: Iac62ffb5f208bd6916ffd0253065d7c4bda55167\nSigned-off-by: Jayaanand Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf16a7f41f0b6a04bd8516bc7a1ca4eaa9bfa73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf16a7f41f0b6a04bd8516bc7a1ca4eaa9bfa73d"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
