)]}'
{"id":"openstack%2Fcinder~1006104","triplet_id":"openstack%2Fcinder~master~I0c6773ab4a0eb79d7f5cb1807d07c2881143049a","project":"openstack/cinder","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-17 20:07:05.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0c6773ab4a0eb79d7f5cb1807d07c2881143049a","subject":"Combine OSError and IOError in rtstool","status":"NEW","created":"2026-09-17 19:21:02.000000000","updated":"2026-09-23 11:21:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f3588565bf544ff565f9d52fd77c79626b336c15","_number":1006104,"virtual_id_number":1006104,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-17 21:22:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2026-09-18 14:45:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2026-09-23 11:21:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":1,"date":"2026-09-21 04:48:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 20:07:05.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-17 20:14:26.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 20:39:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 21:22:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 21:22:30.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-17 22:16:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 22:52:27.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-18 00:05:45.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-18 14:45:02.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2026-09-21 04:48:23.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-09-23 11:21:58.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"}],"messages":[{"id":"72ab9073f84b65f1347ee9a99e78a1b54c67c54c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-17 19:21:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de18a13a0c71855bdb71972d869cb6fd791ebac7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 20:07:05.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/477ef05d1bcc4869953302d04138bc45/","accounts_in_message":[],"_revision_number":1},{"id":"586f77a4869939e7a74df3a7806c8dc4a0404432","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 20:14:26.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0c6773ab4a0eb79d7f5cb1807d07c2881143049a .","accounts_in_message":[],"_revision_number":1},{"id":"b99cb1dacea735887ef49a3d9b8ca58bc099f16e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 20:39:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3b435ebc8ff4826822938c4870365dc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/43e0519f72ad41b2880b907b6c5197e6 : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2eb6cc0378324ce09a64e3b184fa2b49 : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8a6661337e224d06ad996e1fe0a9e4a4 : TIMED_OUT in 1h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e6b7091e20fe32416b566638288167b6c5291e9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 20:51:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/477ef05d1bcc4869953302d04138bc45\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7c4447ef4f4a4d8d9848cbfda55c253f : SUCCESS in 44m 38s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/19bf0501214947a7b8cb6279b308335b : SUCCESS in 38m 52s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8e2f3a19c4f74ebaab948dda66c2a7ba : SUCCESS in 48m 54s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/740984f7d7a74f0e9998e4d6277dea58 : SUCCESS in 35m 09s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b2f43a7d0ea43f9b785060aa620f875 : FAILURE in 19m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f65fda063a4154e041243582233d25d02c819113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 21:22:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efb5a5207251449cadb0f9ead420ad61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/addf7d511a4a4752af12c9947947734c : SUCCESS in 6m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da086ed5416c4d88881360888be25c60 : SUCCESS in 11m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71fe4dea26af4357b7310224a5891502 : SUCCESS in 11m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/634e77900daf4e42a2a2336b5e48125a : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c624b6ca1245e1987bd5c51eef0df9 : SUCCESS in 16m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/9cf05257d87f44faba4ad55db9a27c15 : SUCCESS in 1h 00m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/77d0300084eb4bbe96af2b37e662b4de : SUCCESS in 1h 00m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/743f1a4da007455083977f55e9c1b04b : SUCCESS in 1h 33m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc0057d86edf4faf8992a1c8e87e0125 : SUCCESS in 56m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df0a47e2f4bb42a29fe81e279e357aed : SUCCESS in 9m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/044e785be07145faa9663225020f6e6c : FAILURE in 24m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ce8789e11fc54604983ede5fe22c9489 : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc5523869b7649999a236c1207c26d97 : SUCCESS in 6m 22s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/db3e9dbb3ee34d299de9d15218658534 : SUCCESS in 11m 54s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a0789611050e43fc862442ba4b87aa3b : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d8683ea334c44f80a5e864f74c633794 : SUCCESS in 6m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f0cea6bf683c4a5aa8907fc14fb1edd2 : SUCCESS in 7m 50s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1bb95ae87403490d829c5a08e1ffec58 : FAILURE in 7m 23s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a013aa7140a440a586ab5248f35ea4c9 : SUCCESS in 8m 30s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/931df8093f964617b845c255d631b0e7 : SUCCESS in 41m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa2deb858c3b401e9b028a5aec976fd6 : FAILURE in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cb0e36c2233c473888892caa16433cf5 : SUCCESS in 1h 46m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/522b2b35a0fd43d1b7653927b6417f78 : FAILURE in 1h 54m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e974f37be4904af7a109956547f5877f : SUCCESS in 1h 43m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/04407940902d46e8ab89eff7ef5c6b49 : FAILURE in 1h 32m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9e97c0efc11e47e68c052386ccbced6f : SUCCESS in 1h 54m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/94b993d542994fd290ee56d2258da6d9 : SUCCESS in 37m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0936e2ce01c24fa4be207c6571f90d70 : SUCCESS in 1h 18m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/edcf24209c9d458e94e97e35ebbe1b3e : SUCCESS in 1h 33m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/39ea37593cdb4ae199f8c923d88fa222 : SUCCESS in 47m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b5b23082beb446adb90d63edf6248d4d : SUCCESS in 1h 18m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38d06474867643cd8c612c5da379806b : SUCCESS in 1h 03m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"0588168f6a3fbd6cd0e126b381830494b1282d10","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-17 21:32:19.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_09_2026_21_26_07\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"1170ea82a2e5cc3cc44f2c050984abd9ddd4f3c5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-17 22:16:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/3d6a00c2d3b44a33baef73d317542a0d : SUCCESS in 45m 33s\n- cinder-storpool-tempest-iscsi-multipath build/7fc7b3ae77074b66b905e3be4f72c131 : FAILURE in 48m 58s","accounts_in_message":[],"_revision_number":1},{"id":"65c965774e63cfbf87b961711454ebb18e18eb4f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 22:52:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5589/.","accounts_in_message":[],"_revision_number":1},{"id":"0c46c997f155ea5c5837d064f16f06312a9e1027","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 23:38:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5412/.","accounts_in_message":[],"_revision_number":1},{"id":"1f47f40f5b212e082ce80d77b7f3a6b94caec93b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-18 00:05:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-04-1006104-1-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-04-1006104-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ffbb39975bdee9a521796c057d7ff1cb494fed4b","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-18 14:45:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb0a0f60c007df6ebb5c658a75671cb7fcef7200","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-21 04:48:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f3588565bf544ff565f9d52fd77c79626b336c15","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-23 11:21:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"565b7763229d43128a0a5fc15477d25eb476b21e","revisions":{"565b7763229d43128a0a5fc15477d25eb476b21e":{"kind":"REWORK","_number":1,"created":"2026-09-17 19:21:02.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/04/1006104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/1006104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/1006104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/1006104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/1006104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/1006104/1"}}},"commit":{"parents":[{"commit":"9b31bd509c5eed4bdc65184b6619c15445a20c02","subject":"Merge \"Replace eventlet spawn calls in volume drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b31bd509c5eed4bdc65184b6619c15445a20c02"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-17 19:12:41.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-17 19:20:52.000000000","tz":-240},"subject":"Combine OSError and IOError in rtstool","message":"Combine OSError and IOError in rtstool\n\nIn Python 3.3+, IOError is equivalent to OSError, so\nwe don\u0027t need to catch IOError here.\n\nGenerated-By: Claude Code Sonnet 4.5\nChange-Id: I0c6773ab4a0eb79d7f5cb1807d07c2881143049a\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/565b7763229d43128a0a5fc15477d25eb476b21e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/565b7763229d43128a0a5fc15477d25eb476b21e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
