)]}'
{"id":"openstack%2Fcinder~1006118","triplet_id":"openstack%2Fcinder~master~I5c40545f82ddf63c6be82e4c3bcca5f3cd32d1ca","project":"openstack/cinder","branch":"master","topic":"indri_bump","attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2026-09-19 20:21:11.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5c40545f82ddf63c6be82e4c3bcca5f3cd32d1ca","subject":"[Everpure] Bump version numbers for 2027.1 (Indri)","status":"NEW","created":"2026-09-17 21:02:50.000000000","updated":"2026-09-21 04:15:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c4272efcef0c23526af0f33aee60a540f50cd193","_number":1006118,"virtual_id_number":1006118,"owner":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-19 22:18:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"all":[{"value":1,"date":"2026-09-21 04:15:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 21:36:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 21:42:22.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-17 21:57:37.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 23:06:33.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-17 23:19:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 23:48:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-18 01:06:16.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-09-19 22:18:48.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-21 04:15:44.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"4d3c8242518dbc890ed86883282a1cd1fa9db746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-17 21:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12fdcc0588f50501922a7a9dba005065fcfa8ce0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-17 21:08:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64e0b84a4a7a836736537e42f6496517b0864944","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 21:36:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e00884d6b6114ee1b543d764bf185c9f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a32ddf464b947768d16e1b2be356e52 : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d43c8e2d9bbd419c90124e9389a5d338 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c7d7dd623da848fc892d4ae8743eb789 : SUCCESS in 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8fbe88ca78a7898be02ac25f23d7392813261ff","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 21:42:22.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d26e0f8889964fa28d0e8cfc121c790b/","accounts_in_message":[],"_revision_number":2},{"id":"1b721daa833c9ba0738b7bf74d0ce816ea959de6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 21:57:37.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I5c40545f82ddf63c6be82e4c3bcca5f3cd32d1ca .","accounts_in_message":[],"_revision_number":1},{"id":"07a8fa73520cfd99c34c9155d348b2b8c9c3a0fd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 22:00:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d26e0f8889964fa28d0e8cfc121c790b\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0120607d588d4e998b7f2d71ff7e444c : SUCCESS in 46m 54s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/09035f9fbd564325becf37b9ef497a23 : SUCCESS in 37m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ec5d1d0af13b4095b1fb2fa6e603800f : SUCCESS in 51m 31s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5be78e766a8e4e9dad4d268678674b17 : SUCCESS in 33m 13s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f9b40bd4de384b65b72d0a0c7c5c56c3 : RETRY_LIMIT in 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3e81a7dcbebdf5485e32c20e40b2eb609376cd7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 22:01:13.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I5c40545f82ddf63c6be82e4c3bcca5f3cd32d1ca .","accounts_in_message":[],"_revision_number":2},{"id":"ea9c9abf87f36ea576a93151bbc449382b8034f0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-17 23:06:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/22cb6257411048aab790721c6b732293 : SUCCESS in 50m 29s\n- cinder-storpool-tempest-iscsi-multipath build/f03e5177acac4ea1a6ce795c172512e6 : SUCCESS in 48m 46s","accounts_in_message":[],"_revision_number":2},{"id":"de3389dcacfe292b56c5649f5d6971ae1137a9e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 23:19:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9c3349909ff4fc0a590812180f57358\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a360c6a9d2784d1a96e81c90c03af783 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01154f4a3ae441adb89d94fdf1c7b60b : SUCCESS in 11m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02f2695880b14e2a9e732790a5c65bfe : SUCCESS in 11m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/528d0217e22a478cbe75a6e1227d67da : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e7d363072c46daac86233b8dd49e09 : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/94a9a800be554568acfc88a6f3fe075c : SUCCESS in 1h 06m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14ad22b08c784b8386f662928ad6d534 : SUCCESS in 1h 07m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4a86e4e16a246849b44401bd303eb5d : SUCCESS in 1h 22m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2b42adb8c30419db0cad8635ee096cf : SUCCESS in 56m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/209475b72499436a90d2d53a35ffc121 : SUCCESS in 9m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a4f8fecf69f646fea205074eac7d631c : TIMED_OUT in 41m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2d8b4433bba547cb94ad82a058952dd9 : SUCCESS in 3m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/31c5d22435174fd6a2efe0ba0ffd225b : SUCCESS in 5m 55s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/48a376356e4940ff8c30934ceded71ca : SUCCESS in 8m 46s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e4ebc03cb3e441de9939f8487fbe67b0 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6f6c63415892480fa2f9f32238d1ea48 : SUCCESS in 6m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/92d296f6d2264e2abd60b72ad1402497 : SUCCESS in 6m 00s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/12f91ae2f239492786324918abe0d352 : FAILURE in 4m 45s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/138125e8a67b40caab45019cf96e10b3 : SUCCESS in 8m 54s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c3daef5ff0545ce8b54eab2500d3d9e : SUCCESS in 32m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/994566bda2e848498259d5450917a7f5 : FAILURE in 3m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/368c0e9a7c6f4303bbf67c8ca5e90519 : SUCCESS in 1h 36m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ae2e984a039f42e5a71ec21c7ebecf88 : FAILURE in 1h 36m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/45edf69b8cfd4207b437c4411c882ab0 : SUCCESS in 1h 17m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b10fdd80698141609ee10bff6df22562 : FAILURE in 2h 07m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5fa8005a11094f61b14b987fa015618f : SUCCESS in 1h 22m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/574201c436ba402e9c165bdc55e5ecdd : SUCCESS in 26m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4deb5fb53e4740e584350a76d203fc0f : SUCCESS in 1h 16m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3214dcf875a74a99b707b851e0c1cd34 : SUCCESS in 1h 41m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/23c08e9a5eb645a1a43c17a83f314b68 : SUCCESS in 49m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/203735abb0fe48a38dd24b5f9587b155 : FAILURE in 1h 32m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99282b64121144c7a957470194cbad2c : SUCCESS in 1h 01m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"6956e070c2e8e3bb7cf9631f766b73fb64f93d94","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 23:48:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5590/.","accounts_in_message":[],"_revision_number":1},{"id":"cc56450d518e70c937a35daf73535ff0b3e1a79b","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-17 23:54:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c8a528d4fa9389210e28369349099505da5a60b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 00:19:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5187d3ed36f4e02afe2048cc9d0086d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/789ac74732cf4db5999a62205aa14f97 : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4b95385e681a444fa55cb12458f043dc : SUCCESS in 16m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9aa6c2fc4dfe482a85e6fcdd9dc466e5 : SUCCESS in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19fdaf9253174defb4c32c89170ba7fbd0eb6ffc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 00:37:04.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/f7b93ef48f624330b821b0b293de3ff5/","accounts_in_message":[],"_revision_number":2},{"id":"25872e6d0c21b1e7a989a62d0f86ac3c792c5065","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 00:40:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5413/.","accounts_in_message":[],"_revision_number":1},{"id":"faa167f62c4fdb5964f4ce812716ed09a636634d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 00:49:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f7b93ef48f624330b821b0b293de3ff5\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81380854d3a74baf8ea0237d2b9356ab : SUCCESS in 44m 10s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/14ebb922928c42a9a6d15d72f4dc126b : SUCCESS in 31m 59s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f1a349c461854f5fb8f80066c02d30bd : SUCCESS in 53m 53s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6dc95a2863bd4cc0ab01fb739cca4e46 : SUCCESS in 36m 22s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/18fc000735a24f3595a1383265249185 : FAILURE in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c44b41f3fb1f071a4222d238c99a4e18c5867535","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 00:52:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5591/.","accounts_in_message":[],"_revision_number":2},{"id":"75f0b2d15fcbb0846a00cf8549c3bc847c2eac38","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-18 01:06:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-18-1006118-2-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-18-1006118-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0d210b50e62213eccffd049dee1aa3f3d0e01c88","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 01:39:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5414/.","accounts_in_message":[],"_revision_number":2},{"id":"1262b32e80696d7a2305441d854b6329b626fa46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 02:08:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/412b5d51f50e4d8a9d18201c9441aaad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34970ac482b342ca81290e35ec7bf4cd : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2432e2abdd144c3380c388f9e7af0575 : FAILURE in 12m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c3628134a3b452781b89fd400f57b79 : SUCCESS in 10m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/189e66be54534c408f53c0d5cba6c7c0 : TIMED_OUT in 40m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a1642d746f14785953f13828815744f : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/a86e50ff168145a281e6f66877ac93d5 : SUCCESS in 56m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc19a3a64ee04048843adaf94786677e : SUCCESS in 1h 02m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b3b9121f72f416d8e261284a063b683 : SUCCESS in 1h 25m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8050d474afa3466991267b750e0d7839 : SUCCESS in 57m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a5e6f1a6ebd4fdc8b4f3d43cdbb2cd0 : SUCCESS in 9m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a5b3486a9d1b4f11889858d36eb24b20 : SUCCESS in 22m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/86fa14254df843f390e8782938600f9c : SUCCESS in 3m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d91be0ea251348248c7a086c50873cfb : SUCCESS in 5m 51s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/889a86e693a642aab94f5efdd1372a54 : SUCCESS in 9m 40s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/038a79bd4e944842a935bdc8b9697219 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/42d2a58bd49045d489da5cb7f9a27494 : SUCCESS in 6m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/79d5e0bc970d4a97b4bcc4f33c34ac4a : SUCCESS in 6m 12s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8824aae8162e4e1f82b7c574e352c6fc : FAILURE in 4m 17s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c79515d981e140df937d1ee1d91914ac : SUCCESS in 7m 46s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d019a2a1c3224fd8b82dca110ebc57dc : SUCCESS in 36m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7d5c655d1ae941d6be59f0df86ee9765 : FAILURE in 3m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/665b23bb525b4ed0bc62f81670b79d9b : SUCCESS in 1h 39m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8e28305e8aab488f8e73633bd0582e29 : FAILURE in 1h 07m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f8b86289ef6947619a7525d4e25c14b6 : SUCCESS in 1h 23m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/152bf1a9e51e424598e55a2bd874be9e : FAILURE in 2h 11m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0a027dc8d85842c890869c67f0797c36 : SUCCESS in 1h 20m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/704ae45e5a9f42bca979a9ab064de44f : SUCCESS in 21m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4ba860406af4d9aab8700edbf0e24cc : SUCCESS in 1h 16m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a452a4b1cc19457fb935fbf5a3747c06 : SUCCESS in 1h 26m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3b66b7c0985446fb8941fa9374a007aa : SUCCESS in 28m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d1d2e4a9502846c99519ec88c180f5d5 : SUCCESS in 50m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e4465a53b7d40be9e1d208dd7ca6df7 : SUCCESS in 28m 35s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"cb1acc322b82811b70d45550e1bf3a5cbf1cdcde","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-18 02:24:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/76f74dc74b9a48f38682eef1420aa6e1 : SUCCESS in 42m 19s\n- cinder-storpool-tempest-iscsi-multipath build/04c3ec5d9827470d966914ad1d90b3c9 : SUCCESS in 49m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a5d87852dc854545dedced11924b8100117caf74","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-19 20:02:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc3a31226333b8b3e692d782435637aeb7912981","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-19 20:21:11.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/c38f41d6418f45bfb04f5284597fba65/","accounts_in_message":[],"_revision_number":2},{"id":"18e7ae7a9ef12583f5ec00bbd605b36035ed165e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 20:31:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a4bca9eb0cd4bc3afcc173e9b0adf57\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/417063698c64405daeeff464eea3fa28 : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9837932ae89745e082f7f2de1ab6bd9c : SUCCESS in 18m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a9dbdb8b5cba49e5a6393e178598735c : SUCCESS in 26m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f9ab6f8451fa78a2c6fc663e85247350e6988c91","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-19 20:50:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c38f41d6418f45bfb04f5284597fba65\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/57dd229f69f14a11be2463ea9c341572 : SUCCESS in 40m 00s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d12e2a770e6048bda101cdcccd152fc1 : SUCCESS in 32m 19s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/31130dd563914cf296aeb8fba10a5a23 : SUCCESS in 46m 36s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca157f6b60084acdaa746588d2443c79 : SUCCESS in 38m 32s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c962a01f2e804f05b12c02e33ca51bcc : FAILURE in 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c1d29bd4a9a2ad1d4167a997e519e1048c1b92bc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-19 20:56:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ab43aa4891f24f98827d01f8bd21283e : SUCCESS in 51m 11s\n- cinder-storpool-tempest-iscsi-multipath build/7a4b223f58d24020af02a1cd17ce2630 : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":2},{"id":"30f276ee36473c366b7d2aedb24ba2f785789bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 22:18:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9905c2a50e6412887d1b7ab16d09938\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7716dfda40994276a4ebf2bb9dac65cf : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2806ada8efa435a99c9eaed5fae1cce : SUCCESS in 10m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/216fb3ca068e4479a0e9120cf9912902 : SUCCESS in 7m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/752508700e59404bb0eae376773f8ad6 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f243438033c14c2683c0e52f9ba20e0c : SUCCESS in 15m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/55f5cbede4184cd589ffb0af19cf2e67 : SUCCESS in 58m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/77ba928d40af42ceb317a9c4a640e4ee : SUCCESS in 59m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/932fc5c6d0284458954216eba3b78776 : SUCCESS in 1h 29m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79a70342591046c3a5e2c41a4d4e2186 : SUCCESS in 40m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29b3c2706b2146be9fee8b5a96e8a46f : SUCCESS in 9m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/78e97ab6ef644644b330344c1f55cb14 : SUCCESS in 16m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/718fcf778f3344cf9faa73e136e5757a : SUCCESS in 6m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1705407891ca4f64816949698aad44f4 : SUCCESS in 6m 14s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/442d1bd83cc341659aa4db093d916876 : FAILURE in 10m 55s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/1e11ff62c74e4c06957084acbff08325 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b3c5e2c9ab9649869523cca65fe6466f : SUCCESS in 6m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b23439d8dc4347efae49a87040d2af0b : SUCCESS in 4m 39s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/088c941e87934f22b7b3df2292ac957b : FAILURE in 3m 57s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/29f742ab271340cab4dc545de94da237 : SUCCESS in 8m 39s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8891d47ce4f94e06a663c3eb722fb685 : SUCCESS in 21m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/53fd7f9bef2c44a9af420d0f861b7cea : FAILURE in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e0cd58462a9f455eb3c4ea8c72a862ea : SUCCESS in 51m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e28ee4e9ea024b4db6b2b15ede4ebe0a : SUCCESS in 1h 06m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3d049b4b848349b592d03577ee64b262 : SUCCESS in 1h 53m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/31a23f54b2a44b60b870bb84f21ef803 : FAILURE in 2h 08m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7455443275964eb28a7ecf249f0932b0 : SUCCESS in 1h 49m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e46ca92991cd445bb04e273a451148ef : SUCCESS in 33m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c55b7df179ad4f9c9b8257a8400d18e9 : SUCCESS in 52m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/891f8b31de4e4f1ba444a75ebfda42b6 : SUCCESS in 1h 26m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/747f041fdbb540549ec5230635693504 : SUCCESS in 43m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b035b4b2a0014f05a3640ac763478972 : SUCCESS in 1h 04m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1360c0cb676475a8d156cf210ae1929 : SUCCESS in 26m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"b9a20aa157ff8a4e173a60feb9e8eba255c10d1b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-19 22:28:07.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_09_2026_22_22_19\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"c4272efcef0c23526af0f33aee60a540f50cd193","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-21 04:15:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"56f5ab79996bf0e83371a0b212662439a4c58e58","revisions":{"2463b6b888cd04892801e7e8a4944152a05a19ad":{"kind":"REWORK","_number":1,"created":"2026-09-17 21:02:50.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/18/1006118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/1006118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/1006118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/1006118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/1006118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/1006118/1"}}},"commit":{"parents":[{"commit":"8313581b64a1b5b9422714f769a04d1a87f110ca","subject":"Merge \"Optimize getting parent backup for new incremental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8313581b64a1b5b9422714f769a04d1a87f110ca"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-09-17 21:00:23.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-09-17 21:00:23.000000000","tz":-240},"subject":"[Everpure] Bump version numbers for 2027.1 (Indri)","message":"[Everpure] Bump version numbers for 2027.1 (Indri)\n\nEverpure uses the version number of the driver to identify which\ncustomers are using which versions of OpenStack.\n\nThis helps granular analysis of which driver is being\nused in each OpenStack version\n\nChange-Id: I5c40545f82ddf63c6be82e4c3bcca5f3cd32d1ca\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2463b6b888cd04892801e7e8a4944152a05a19ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2463b6b888cd04892801e7e8a4944152a05a19ad"}]},"branch":"refs/heads/master"},"56f5ab79996bf0e83371a0b212662439a4c58e58":{"kind":"REWORK","_number":2,"created":"2026-09-17 21:08:40.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/18/1006118/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/1006118/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/1006118/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/1006118/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/1006118/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/1006118/2"}}},"commit":{"parents":[{"commit":"8313581b64a1b5b9422714f769a04d1a87f110ca","subject":"Merge \"Optimize getting parent backup for new incremental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8313581b64a1b5b9422714f769a04d1a87f110ca"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-09-17 21:00:23.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-09-17 21:08:31.000000000","tz":-240},"subject":"[Everpure] Bump version numbers for 2027.1 (Indri)","message":"[Everpure] Bump version numbers for 2027.1 (Indri)\n\nEverpure uses the version number of the driver to identify which\ncustomers are using which versions of OpenStack.\n\nThis helps granular analysis of which driver is being\nused in each OpenStack version\n\nChange-Id: I5c40545f82ddf63c6be82e4c3bcca5f3cd32d1ca\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56f5ab79996bf0e83371a0b212662439a4c58e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56f5ab79996bf0e83371a0b212662439a4c58e58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
