)]}'
{"id":"openstack%2Fcinder~1006343","triplet_id":"openstack%2Fcinder~stable%2F2026.2~I4e39c58c87b3f0671f631c633bb2134a27ed228a","project":"openstack/cinder","branch":"stable/2026.2","topic":"c-hibiscus-rc-2","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-24 20:49:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e39c58c87b3f0671f631c633bb2134a27ed228a","subject":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured","status":"MERGED","created":"2026-09-18 21:13:21.000000000","updated":"2026-09-24 20:49:34.000000000","submitted":"2026-09-24 20:49:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006343-c-hibiscus-rc-2","cherry_pick_of_change":1006007,"cherry_pick_of_patch_set":4,"meta_rev_id":"dba626ac604e42aa77d7f7e2ae557ff2f8a19f42","_number":1006343,"virtual_id_number":1006343,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-24 20:49:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-09-24 15:56:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-09-24 15:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-09-19 06:35:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":1,"date":"2026-09-21 06:08:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-09-24 15:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 21:13:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"CC"},{"updated":"2026-09-18 21:20:35.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-18 22:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 22:01:39.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-18 22:06:32.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-19 01:39:13.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-19 06:35:23.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2026-09-21 06:08:35.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-09-21 16:38:04.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-09-24 15:57:03.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"1b757c2ce7f721a62d96ad978dfcbabe76e39521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-18 21:13:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b1bf4070b934f1368b06ab548bb8b063158fb99c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-18 21:14:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f6ae389f75766e502d341574605d20844620bc8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-18 21:20:35.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 1m 50s\n- FC       : FAILURE in 0h 1m 34s\n- NVMe/TCP : FAILURE in 0h 1m 46s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1006343","accounts_in_message":[],"_revision_number":1},{"id":"1046a3883f7e87de9b945af39a301a440da7c828","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:01:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db9a27eefe2347c3b9489b5a1fc71a85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1806770fc234492daf3006e95a358f22 : SUCCESS in 10m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94b4991a3e6e4e6ba00efe0415fcaf4b : SUCCESS in 8m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d40704949224d559385eb3634ad8aab : SUCCESS in 10m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/002716928a8146afa7395f5f71c51482 : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2297e550af7649669e700fe3f505bec4 : SUCCESS in 15m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/424a9355afe0451f87c92fed0c9e5639 : SUCCESS in 12m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b6a08ed901f44d51a4fdede33737a6ca : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d70fa93be15e43b6a12fd75227541e4b : SUCCESS in 4m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34d90f25da4d4b44bf1047d54aa8f3af : SUCCESS in 3m 53s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/ae3f6fb874994e0b97a8effac0a1a6ad : FAILURE in 13m 58s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9982dca24678463295188141c49cedce : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5ba1a4b42af5451eb79558b5a558fe2c : SUCCESS in 5m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a32b5acef9c14f93bdc2f2176ab76102 : SUCCESS in 4m 50s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4895ea5d00d9401cb8432de05afb049e : FAILURE in 5m 56s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9aaf47a659d444a9ab4dcaeac7ac4a57 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89b7dcf3274e4aa2aba02659d7cdb5ec : FAILURE in 7m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"251911524d74a72ed9534498ef76060b7c223dd1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 22:06:32.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4e39c58c87b3f0671f631c633bb2134a27ed228a .","accounts_in_message":[],"_revision_number":1},{"id":"6a31132c73f1de893d22358b6358ed84ffb74c00","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-18 22:10:57.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_09_2026_22_05_06\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"ea016c3d03a074d9c748d91b3473e436ed97b648","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-19 01:39:13.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5606/.","accounts_in_message":[],"_revision_number":1},{"id":"f9e510d7dd4a5cf8291215ecc77fb543e86ec79f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-19 02:00:50.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5429/.","accounts_in_message":[],"_revision_number":1},{"id":"248825f2b37079bc4c60fed715960eabec53e72a","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-09-19 06:35:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f8f9f69ae2d50835b73ab09320ba2372a131cfe","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-21 06:08:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e19eb84c4011fc39234a8a4d215c81e442015c70","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-09-21 16:38:04.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/977/","accounts_in_message":[],"_revision_number":1},{"id":"43198f64eb7dcf4874e3ec63fa7b19dfbba9180f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-09-21 17:12:07.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/1006343/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/1006343/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"7cbda6fce4b8314a245393daefc14071c6991af0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-24 15:56:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06ead48e64f29ddc898de088d4fcfdd1ccfe0c4a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-09-24 15:57:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9f6226283d1ac9bf9bfe1e3194244cf71fbc3580","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-09-24 15:57:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"482904a47cb161ba43dd9121fc5e35997bf88012","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 15:58:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e8524a1b1101877c7241d895991f026f5222d30","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-24 16:12:47.000000000","message":"Topic set to c-hibiscus-rc-2","accounts_in_message":[],"_revision_number":1},{"id":"8dc7c079b28dd9b9155e7550a869e6097a2ce7ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 20:49:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71d89a29cd364137b75c990204417b30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f77615cec2746dc89af39de484f445e : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2694adfa73ad4b2e8126133d27915181 : SUCCESS in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9df19669368e4bb2a9e0c85fa06bd41e : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5bf7ec89654b8bbad39a6172c879d5 : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/99cd44b9d7404a4cb37a972968677ac9 : SUCCESS in 1h 00m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7099629d5b6b45cd9f6bf446dea28aab : SUCCESS in 1h 05m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efd8b97864334cb8b204a0e4bf21742a : SUCCESS in 57m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/541d74413abc4318a6d922a3eed2f833 : SUCCESS in 9m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6118e842fcd44e0099cfa61c676fb92c : SUCCESS in 1h 19m 31s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a9662cb350834b3cbd88c2ce9a825dce : SUCCESS in 1h 39m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cd5b691eef4414ab8154c8274558bca : SUCCESS in 32m 05s","accounts_in_message":[],"_revision_number":1},{"id":"dba626ac604e42aa77d7f7e2ae557ff2f8a19f42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 20:49:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9463c065d86dc9781a375ad87a622acb7f5bea0e","revisions":{"9463c065d86dc9781a375ad87a622acb7f5bea0e":{"kind":"REWORK","_number":1,"created":"2026-09-18 21:13:21.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/43/1006343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/43/1006343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/43/1006343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/43/1006343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/43/1006343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/43/1006343/1"}}},"commit":{"parents":[{"commit":"5014451bce8de8a888b03a842d1f5c1e0445d525","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5014451bce8de8a888b03a842d1f5c1e0445d525"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-09-17 09:44:49.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-09-18 21:13:21.000000000","tz":0},"subject":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured","message":"[IBM SVf] Skip iSCSI IP discovery when FC protocol is configured\n\nDuring driver initialization, _update_storwize_state()\nunconditionally called helper.add_iscsi_ip_addrs() to\nquery iSCSI node IPs. On systems with code level \u003e\u003d 8.4.2.0,\nthis executes \u0027lsip\u0027 for the configured portset\n(default portset0).\n\nWhen running the FC driver (StorwizeSVCFCDriver) on a\nbackend where no iSCSI IP addresses or portsets are\nconfigured, add_iscsi_ip_addrs() raises an InvalidInput\nexception (\"No nodes found in the given portset\"),\ncausing driver initialization to fail before\nadd_fc_wwpns() can run.\n\nThis change avoids running add_iscsi_ip_addrs() when\nthe FC protocol is configured (and conversely avoids\nadd_fc_wwpns() when iSCSI is configured), allowing\nFC-only environments without node IP configuration\nto initialize successfully.\n\nRefer the below change that caused this issue-\nhttps://review.opendev.org/c/openstack/cinder/+/874633\n\nCloses-Bug: #2167561\nChange-Id: I4e39c58c87b3f0671f631c633bb2134a27ed228a\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n(cherry picked from commit 602213d59ac73ca67bf0ea4db866ccc6ddf81f63)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9463c065d86dc9781a375ad87a622acb7f5bea0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9463c065d86dc9781a375ad87a622acb7f5bea0e"}]},"branch":"refs/heads/stable/2026.2","conflicts":{"base":"9b31bd509c5eed4bdc65184b6619c15445a20c02","ours":"5014451bce8de8a888b03a842d1f5c1e0445d525","theirs":"602213d59ac73ca67bf0ea4db866ccc6ddf81f63","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
