)]}'
{"id":"openstack%2Fcinder~1006358","triplet_id":"openstack%2Fcinder~stable%2F2025.2~I2abff84798f54f207ce53ab5afedf41e26348648","project":"openstack/cinder","branch":"stable/2025.2","topic":"bug/2139068-stable/2026.1-stable/2025.2","attention_set":{"36686":{"account":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"last_update":"2026-09-21 21:52:00.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I2abff84798f54f207ce53ab5afedf41e26348648","subject":"Fix missing volume delete notifications on group delete","status":"NEW","created":"2026-09-19 07:00:30.000000000","updated":"2026-09-21 22:01:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":997681,"cherry_pick_of_patch_set":1,"meta_rev_id":"599105e9dcb8d46c85b9c5377cc6f6ee84df0214","_number":1006358,"virtual_id_number":1006358,"owner":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-21 21:51:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-19 07:04:55.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-19 07:41:00.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-19 07:45:18.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-19 08:15:13.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-19 08:24:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 20:04:09.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2026-09-21 21:52:00.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"923da7c52ac4067f4f9296be0b3b1f867ecc3845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-09-19 07:00:30.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.1.","accounts_in_message":[],"_revision_number":1},{"id":"f8a15301d61d6fceb517d1c171b5202534a531e7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-19 07:04:55.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 1m 43s\n- FC       : FAILURE in 0h 1m 46s\n- NVMe/TCP : FAILURE in 0h 1m 52s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1006358","accounts_in_message":[],"_revision_number":1},{"id":"83feb1b38b37eadc5d3cefc6a78be6bb12a69e50","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-19 07:41:00.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/46fd464a7f504a7da0a126172d094328/","accounts_in_message":[],"_revision_number":1},{"id":"a18330f5bea22e6332567c38f1aa9ceed0f2c7d6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-19 07:45:18.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2abff84798f54f207ce53ab5afedf41e26348648 .","accounts_in_message":[],"_revision_number":1},{"id":"9548dfa76bae2fa68c65fc67c8dc842c775a1fa6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-19 08:15:13.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5430/.","accounts_in_message":[],"_revision_number":1},{"id":"157394202eaa386f5e3caa645d5839be6e90f6e7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-19 08:17:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5607/.","accounts_in_message":[],"_revision_number":1},{"id":"8c69a590e9ff2a62320ab12887ba34aac72c6c59","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-19 08:20:51.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/46fd464a7f504a7da0a126172d094328\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8cc2e1ceec42400eaec68c30122fc53a : FAILURE in 43m 50s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/67f735655ec64e11a7485381b97bc005 : SUCCESS in 39m 44s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd002b2a8d534eb5b0a70cc30556d4e5 : SUCCESS in 58m 54s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fe9d21747e4b48cfaacb0509df9c288a : SUCCESS in 40m 03s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fedb4bba7ca84965b5e7eba5953172eb : FAILURE in 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7d3ac83d69f9bc0431838bac0ed1be6bda4876e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 08:24:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cffea010e2274f95b2b87862bcc1c98a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce9e43d0e964921a98c80f9672cc13d : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b980ca9b21e54e6cad9fe55d048a66af : RETRY_LIMIT\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f50b641575b94ef1ae62ff64550f667a : SUCCESS in 11m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab9a0086cbdf4f79944b04fa9310fe04 : RETRY_LIMIT (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c55cdcb11e10478fb03b3dbade9a8798 : RETRY_LIMIT\n- grenade https://zuul.opendev.org/t/openstack/build/df850020188343ec9755c244b23c213e : RETRY_LIMIT\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5dd822111454500b1095563efae4a11 : SUCCESS in 1h 22m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db556564d6e748c1be7a09110851cea4 : SUCCESS in 1h 03m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10d64159d0624889ad4a85b988c74428 : SUCCESS in 5m 36s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6093b61aef124f5db5a1f2fca2f164bd : SUCCESS in 3m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73dd8fc86f464067b62e053b99a5207e : RETRY_LIMIT\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/17aa0a5cee55466091e4501a6ed0381b : SUCCESS in 4m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f404684816864d388254d30b62b52cb7 : RETRY_LIMIT\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb02f90a059a43c7b233becc6e06562e : RETRY_LIMIT\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e358b198bbdb45e8b9fa0cadc6bcb992 : FAILURE in 9m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9d3fcadbc4574e69a4c3f99845dce31a : RETRY_LIMIT\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8519557e24f14b9d9981846b348c410a : RETRY_LIMIT\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a3073d53f0934751934b1aaf53fb64cd : RETRY_LIMIT\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/adfa61e5b3e646b38401eebc7cbe290c : RETRY_LIMIT (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a097ec11b50f4754b1c7074bb1a1436f : RETRY_LIMIT\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ffd8e8e0bb5439ab85177bc08ca87a6 : RETRY_LIMIT\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"14bf3f7232bc6ead983a0b83a7043695e4c5ff3f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-21 20:04:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd348daae5fcf5630dfdbc6ff64ff22441892888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 21:51:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35ae7c73f81a4810a5a7bc30fc4bc686\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ed6a65c1c54fd6b7f6f3421b7e2944 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/704915d6ba0e455c9a01ecf49eceb96c : SUCCESS in 12m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/147edbf3ea16400c8d78c675fb333fb6 : SUCCESS in 9m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a6f9a822b919465bacb0d815fc5fb9b4 : SUCCESS in 15m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea4cd1b828d41d8bcfaf96627559736 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/f2df6077d4d643369b44d22a983c7d93 : SUCCESS in 1h 17m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c230bf0e6b341fd8f30f98f3e2ff580 : SUCCESS in 1h 34m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad7c19ecd8bf4c6c80dd9f462e7b5a2a : SUCCESS in 1h 04m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93918bf1ac5943aea91cac10cc317cc7 : SUCCESS in 9m 05s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/227d1c1e48e04e2aba5ae9caae406c0d : SUCCESS in 3m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/00ce34680bf94a10887abd7eddffba06 : SUCCESS in 5m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4c5e6a83ad664f6c9af64ee637c3e016 : SUCCESS in 5m 54s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1f66de9541db44529c26a7ad89d7cf39 : SUCCESS in 1h 06m 31s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fa153c6a7d934e9b9fabb5bcc427ff25 : SUCCESS in 1h 35m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f482b637d4df4a07875d1f172e4b4c00 : FAILURE in 10m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5db23e451f9a4a418e0d5a413e25409e : SUCCESS in 38m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/704d71bd718541f68795ee937a12b9db : SUCCESS in 1h 17m 58s\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/81b5b6d7b3db4ec9b4ad770c3483c84f : SUCCESS in 1h 10m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/517753bbf24145e6bd8cbec9426ddf2a : FAILURE in 10m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c29abfce8f1b4d75975cb9e36d2a945f : SUCCESS in 1h 39m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cc9f212b4e945b2992f401a70258690 : SUCCESS in 44m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job devstack-plugin-nfs-tempest-full-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job devstack-plugin-nfs-tempest-full-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"599105e9dcb8d46c85b9c5377cc6f6ee84df0214","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-21 22:01:26.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_09_2026_21_55_34\nLeave a comment with","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25c1d75321fa712516014700db9839f8fdaa9b6e","revisions":{"25c1d75321fa712516014700db9839f8fdaa9b6e":{"kind":"REWORK","_number":1,"created":"2026-09-19 07:00:30.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/58/1006358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/1006358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/1006358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/1006358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/1006358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/1006358/1"}}},"commit":{"parents":[{"commit":"5d59ca2e9818ec6cbc7fffa4ae09d604f62912d3","subject":"Merge \"Fix _is_encrypted comparison in volume api\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d59ca2e9818ec6cbc7fffa4ae09d604f62912d3"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-01-26 11:30:20.000000000","tz":540},"committer":{"name":"Inyong Hong","email":"inyong.hong@samsung.com","date":"2026-09-19 07:00:30.000000000","tz":0},"subject":"Fix missing volume delete notifications on group delete","message":"Fix missing volume delete notifications on group delete\n\nWhen deleting a group with the delete-volumes flag set to True,\nthe volumes associated with the group are deleted, but the\n\u0027volume.delete.end\u0027 notifications are not emitted. This occurs\nbecause the group delete flow in the manager does not trigger\nthe standard notification logic used in individual volume deletes.\n\nThis patch ensures that \u0027volume.delete.start\u0027 and\n\u0027volume.delete.end\u0027 notifications are correctly sent for each\nvolume processed during a group deletion.\n\nCloses-Bug: #2139068\nChange-Id: I2abff84798f54f207ce53ab5afedf41e26348648\nSigned-off-by: hongp \u003cinyong.hong@samsung.com\u003e\n(cherry picked from commit cb63d6f58c6d01fc249e5f77b7f71f9098b315a8)\n(cherry picked from commit 429a763449c46e693fe60b95f73022273f1f77d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25c1d75321fa712516014700db9839f8fdaa9b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25c1d75321fa712516014700db9839f8fdaa9b6e"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"1fdd593358bf8ff0a8c108dfa1b6f9026ce011b5","ours":"5d59ca2e9818ec6cbc7fffa4ae09d604f62912d3","theirs":"429a763449c46e693fe60b95f73022273f1f77d0","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
