)]}'
{"id":"openstack%2Fcinder~1007706","triplet_id":"openstack%2Fcinder~master~Ie742d2585cb4e9f7d6775d012ac4fa6f8883b6f2","project":"openstack/cinder","branch":"master","attention_set":{"36686":{"account":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"last_update":"2026-09-29 04:04:14.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie742d2585cb4e9f7d6775d012ac4fa6f8883b6f2","subject":"Preserve volume like-filter search strings","status":"NEW","created":"2026-09-28 12:22:22.000000000","updated":"2026-09-29 06:22:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"663b345c21c16f45020586113f29620ec23c9f74","_number":1007706,"virtual_id_number":1007706,"owner":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-29 05:59:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-28 12:39:03.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-28 12:41:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-28 13:17:33.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-28 13:26:40.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-28 15:26:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-28 16:38:56.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-28 16:50:57.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-29 06:00:01.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"985c09ac09dbbbbddfab0f8847aeca6028f8319d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-09-28 12:22:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03c7ac53346311dbc2368e0aada134f2aff77783","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-28 12:39:03.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/0dffb6e40ce04fe2b6f41b3d4f2e21e2/","accounts_in_message":[],"_revision_number":1},{"id":"f3b6a7a24c0cf9351ef5db6a762338cf31defc1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 12:41:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9323284b98334a4a955c8be5db92f151\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e85d864ed2243188f169dd6dc78a02c : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d86b50ec15b64a5cb6e1bc352f2f875c : SUCCESS in 16m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e53a416b0f16fe4d90871752f4a0b563354a128c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-28 13:17:33.000000000","message":"Patch Set 1:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 29s\n- FC       : SUCCESS in 0h 50m 18s\n- NVMe/TCP : SUCCESS in 0h 48m 9s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/1007706","accounts_in_message":[],"_revision_number":1},{"id":"043bf0ea11c7fc860e7c5b67be1de4ebae68dd07","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-28 13:17:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0dffb6e40ce04fe2b6f41b3d4f2e21e2\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cabebffc0dad48b19793471b150fa03d : SUCCESS in 35m 46s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/63a734f5dee34199baf96738a218d7ea : SUCCESS in 31m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5eb5ab02d4d4942be79f6d7e38a026c : SUCCESS in 40m 54s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0cac68e41a8a494283ee3bcde5266433 : SUCCESS in 30m 32s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e4e7d57a8864f4fae23bc58df54b2ca : FAILURE in 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9a9effa4c9fadaffe68c28708a9e5a3a9884a15","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-28 13:26:40.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ie742d2585cb4e9f7d6775d012ac4fa6f8883b6f2 .","accounts_in_message":[],"_revision_number":1},{"id":"27f7e4eed20d95befc37e23c71feef55e0e00b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 15:26:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ed7e35ff71848138166c97a32dc19be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183c789686264bfe86e56e33ace7bd2e : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c4ca96c396e4607a28a07612be22c12 : SUCCESS in 10m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7c1adeaa9b9c48adbc34cc276d7c8787 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed6653196d54bb4ba7e624879700160 : SUCCESS in 17m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/6adde4021ec94b8fa62b2591a7ae3f34 : SUCCESS in 53m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b59850fdf44437796ab4d090befc98f : SUCCESS in 55m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ba28521ef1f48f7bfef2fb5c950c1fd : SUCCESS in 1h 35m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db2811dbf76747f6bda6b1576b228d73 : SUCCESS in 1h 06m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/270d15ec73c9405fa833d935dc371136 : SUCCESS in 9m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c013f0733cab4a13ab21f3faef8d711c : TIMED_OUT in 41m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/84d2c58b614c41dda0663d637d010922 : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5fbbe64c41d643818e643ded1a0dcbc6 : SUCCESS in 3m 47s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/4afc0ae4566a4b9abc75905585e254b7 : SUCCESS in 13m 16s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c1e2249423834233aee09f67d8670f7e : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c5c2bc9e973b4f59b45c63ff0cf3a3b9 : SUCCESS in 5m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d6d4025911e143b38c3c00f12071a6e1 : SUCCESS in 6m 09s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/2b61ebcd606441428e69d9c04763e43c : FAILURE in 4m 30s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/578301828a1140308688314955e6cf6c : SUCCESS in 6m 17s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/22a3c49530854be4bc8752d7d510b57b : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fb28e7cdbbcf4b39ab74260ee1281ef2 : FAILURE in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5d6932dad82e4353991e0db087a12a92 : SUCCESS in 1h 50m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f36ca376a432473aa3bf0f31c15c35b9 : FAILURE in 1h 11m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/334a3b1032c044cb889a388e4a5f2a00 : SUCCESS in 1h 52m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/63520cf06c7c41e2826666b3f8382e9d : FAILURE in 3h 02m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/52e9fd263bd84856bfb60202358f5b64 : FAILURE in 41m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7d138ea7af014c188db491d077b87a18 : SUCCESS in 34m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/26bc1c0d52fe4b38b56b121b1b8d8267 : SUCCESS in 1h 14m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/27614afeab714cb5be0629711b8ad8db : SUCCESS in 1h 25m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3b236258773b4286aaf67b925f8cebb7 : SUCCESS in 40m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c723fc58e293459983646fb08ce79409 : FAILURE in 2h 25m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8456c69edd94e5ea08a3b99a249168b : SUCCESS in 58m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"fa3d8f5f5285fee5ba60f669a2646ee505f4e572","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-28 16:38:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/c640dd89f82b41b6b7ac32356fbff26e : SUCCESS in 44m 28s\n- cinder-storpool-tempest-iscsi-multipath build/50b7dbc9c2be4026a2c7e0691c30a2de : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a94300d00ba772407bd9f16ad31f95474780c19f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-28 16:50:57.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5553/.","accounts_in_message":[],"_revision_number":1},{"id":"8b6e70234f18a9267b5e871d6c7174cc578c112a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-28 17:12:11.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5730/.","accounts_in_message":[],"_revision_number":1},{"id":"c8b39b9da42edf1aa694b7f9326751d495c45f52","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-09-29 03:45:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d971216cd79bb1db2b5c3ef2de84dee3865708c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-29 04:04:14.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/fde81900588a4aa79630164163fc600c/","accounts_in_message":[],"_revision_number":1},{"id":"1bda9a30767e6ccee854ea33555bbae5d77729f7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-29 04:32:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/fde81900588a4aa79630164163fc600c\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b89b351c1b4b444e91e04a8a73f08219 : SUCCESS in 45m 42s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0119f0b21fee4697aa4f352cb6d37d50 : SUCCESS in 33m 23s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8f2893d9a0b94fe18abe4d7d2cfd3024 : SUCCESS in 41m 40s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0b641921365b4f66a84b7661e2efda78 : SUCCESS in 31m 39s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/59ccc4da8b3c45699dfdf2d8b4c80fa4 : FAILURE in 15m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d61766f5e9490783e410c48d7944ee47e8cd4aa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 04:49:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d98dce7087d49aaac51fab54ba6fabf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b406de064eab4b11980be5efc14f21af : SUCCESS in 19m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d9828fe28aeb4a4382a5aea43eaf09ee : TIMED_OUT in 1h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d182fe971371f6cf3f38ab77984622e9dd9b5576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 05:59:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fff027c169a42c7ab4d8694c1102fa0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a63ad42cc52a474db9853ccc2c370630 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85823af9228143d5afd8549c7bc30c19 : SUCCESS in 8m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea85fab9a944456f88d174aa3c9dbaee : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0aa4a6a2144d3aa8ff70b3fdc05192 : SUCCESS in 7m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/80b15705f34745729c77f1b9abcf906b : SUCCESS in 31m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9c4ebf66a314b12a8b9ed706858d26c : SUCCESS in 44m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e51816e517c4f99a790aeb6fd81a899 : SUCCESS in 58m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71473029e41e47acad1091d32a3bbbc2 : SUCCESS in 44m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21e519f2c89c40a5b845c76bc7ea76bc : SUCCESS in 5m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f2c2c2c9f4a644e6953588fe435a0051 : FAILURE in 25m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/286c7ef5861b4e87b8445c9db668ca3f : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/09f258d1ceb14898a5600b8507263929 : SUCCESS in 3m 54s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8104bd2e812b42e29a60da7cae1556b3 : SUCCESS in 9m 06s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0d4ebcbb41bc4bb0b9339ce4450c5e83 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1e815dfbc5ce4a5295be169ec9526293 : SUCCESS in 6m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d54ef42ce78044cfb0c3904b888b02ed : SUCCESS in 5m 20s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/155338dc3ee74267aa41b003f76fc077 : FAILURE in 5m 10s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4c12d886eac748298762a802b6ac0cc9 : SUCCESS in 6m 46s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d51f3866c9eb4b23b464c8e9e8869725 : SUCCESS in 20m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6fbd0975136d471c8e56056dee1b44a3 : FAILURE in 5m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e9a11072f41b4a6694d0f767ec3e0704 : SUCCESS in 1h 24m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0ee926f1dfef4953a47f0c87fd403028 : FAILURE in 1h 08m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9e54d97481bd4826ad85a3f9059d07bc : SUCCESS in 1h 43m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b72f899c31614fc38c8b45bdfa9c158a : FAILURE in 2h 07m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0e5d9d42ad9744399f4764a35453274a : SUCCESS in 1h 45m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2e0a8f933dba4eb5a8a8e1ab46c30164 : SUCCESS in 19m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/70db0091642d462c9b80ddf6e0296f54 : SUCCESS in 1h 09m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df21268d2f7d4c0ba3f40d9e5701e98a : SUCCESS in 1h 08m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/86dd248429a8452891ec4073a0fb4ca9 : SUCCESS in 29m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c209b45192204c62b2fa48d2837a633d : SUCCESS in 1h 08m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdd29ab441454758828d511e99d60b3f : SUCCESS in 27m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"1729b1bbf8d4c23a174e2058cf16933c5e359371","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-29 06:09:32.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_09_2026_06_03_37\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"663b345c21c16f45020586113f29620ec23c9f74","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-29 06:22:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/cb3da117752749b28cb6b87346857718 : SUCCESS in 46m 11s\n- cinder-storpool-tempest-iscsi-multipath build/de0c8ad607744623afc3af0bd5d49b69 : SUCCESS in 48m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"326c2444da19d57760ecd78091a04235e9be6c36","revisions":{"326c2444da19d57760ecd78091a04235e9be6c36":{"kind":"REWORK","_number":1,"created":"2026-09-28 12:22:22.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/06/1007706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/1007706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/1007706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/1007706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/1007706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/1007706/1"}}},"commit":{"parents":[{"commit":"e3b9582a60a58d6ea630e845bbe5d96c8780fd8c","subject":"Merge \"mypy: annotate lvm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3b9582a60a58d6ea630e845bbe5d96c8780fd8c"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-09-28 12:19:43.000000000","tz":540},"committer":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-09-28 12:22:18.000000000","tz":540},"subject":"Preserve volume like-filter search strings","message":"Preserve volume like-filter search strings\n\nSkip literal evaluation of like-filter values to prevent changed search\ntext or silently ignored filters.\n\nCloses-Bug: #2168746\nChange-Id: Ie742d2585cb4e9f7d6775d012ac4fa6f8883b6f2\nSigned-off-by: hongp \u003cinyong.hong@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/326c2444da19d57760ecd78091a04235e9be6c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/326c2444da19d57760ecd78091a04235e9be6c36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
