)]}'
{"id":"openstack%2Fcinder~28256","triplet_id":"openstack%2Fcinder~master~I4e94fc751658fbe1b6b9b74859d800445a1c2e5d","project":"openstack/cinder","branch":"master","topic":"fix_log_message_spacing","hashtags":[],"change_id":"I4e94fc751658fbe1b6b9b74859d800445a1c2e5d","subject":"Fix missing spaces in Huawei Logging","status":"MERGED","created":"2013-05-05 17:15:00.000000000","updated":"2013-05-10 17:38:15.000000000","submitted":"2013-05-10 17:38:14.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cdf5992304476ba53f8393bc3d7e582e3248b787","_number":28256,"virtual_id_number":28256,"owner":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-05-10 17:38:14.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},{"value":0,"_account_id":1207,"name":"Duncan Thomas","email":"duncan.thomas@gmail.com","username":"duncan-thomas"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":6043,"name":"Kurt Martin","email":"kurt.f.martin@hpe.com","username":"kurt-f-martin"},{"value":0,"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-05-10 14:58:16.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},{"value":2,"date":"2013-05-10 16:52:47.000000000","_account_id":1207,"name":"Duncan Thomas","email":"duncan.thomas@gmail.com","username":"duncan-thomas"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":6043,"name":"Kurt Martin","email":"kurt.f.martin@hpe.com","username":"kurt-f-martin"},{"value":0,"date":"2013-05-10 14:02:46.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},{"value":1,"date":"2013-05-10 15:46:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2013-05-10 16:24:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1207,"name":"Duncan Thomas","email":"duncan.thomas@gmail.com","username":"duncan-thomas"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-05-10 14:58:16.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},{"value":1,"date":"2013-05-10 16:52:47.000000000","_account_id":1207,"name":"Duncan Thomas","email":"duncan.thomas@gmail.com","username":"duncan-thomas"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":6043,"name":"Kurt Martin","email":"kurt.f.martin@hpe.com","username":"kurt-f-martin"},{"value":0,"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},{"value":0,"_account_id":1207,"name":"Duncan Thomas","email":"duncan.thomas@gmail.com","username":"duncan-thomas"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":6043,"name":"Kurt Martin","email":"kurt.f.martin@hpe.com","username":"kurt-f-martin"},{"value":0,"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"_account_id":1207,"name":"Duncan Thomas","email":"duncan.thomas@gmail.com","username":"duncan-thomas"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},{"_account_id":6043,"name":"Kurt Martin","email":"kurt.f.martin@hpe.com","username":"kurt-f-martin"},{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-07 01:02:14.000000000","updated_by":{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},"reviewer":{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},"state":"REVIEWER"},{"updated":"2013-05-09 14:54:09.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-05-09 16:34:14.000000000","updated_by":{"_account_id":6043,"name":"Kurt Martin","email":"kurt.f.martin@hpe.com","username":"kurt-f-martin"},"reviewer":{"_account_id":6043,"name":"Kurt Martin","email":"kurt.f.martin@hpe.com","username":"kurt-f-martin"},"state":"REVIEWER"},{"updated":"2013-05-10 10:15:51.000000000","updated_by":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"reviewer":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"state":"REVIEWER"},{"updated":"2013-05-10 14:02:46.000000000","updated_by":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"reviewer":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"state":"REVIEWER"},{"updated":"2013-05-10 15:46:44.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2013-05-10 16:24:44.000000000","updated_by":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"reviewer":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"state":"REVIEWER"},{"updated":"2013-05-10 16:52:47.000000000","updated_by":{"_account_id":1207,"name":"Duncan Thomas","email":"duncan.thomas@gmail.com","username":"duncan-thomas"},"reviewer":{"_account_id":1207,"name":"Duncan Thomas","email":"duncan.thomas@gmail.com","username":"duncan-thomas"},"state":"REVIEWER"},{"updated":"2013-05-10 17:38:14.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"9768ab48021194344b55e009672350e659423339","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-05-05 17:34:18.000000000","message":"Patch Set 1: Works for me\n\nSmokeStack Results (patch set 1):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/85579\n","accounts_in_message":[],"_revision_number":1},{"id":"5dab35c4df56b40426e11cb4b0e2047978e07144","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-05 17:59:59.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28256/1/check/gate-cinder-pep8/3277 : SUCCESS in 00:01:49\n- http://docs-draft.openstack.org/28256/1/check/gate-cinder-docs/3169/doc/build/html/ : SUCCESS in 00:04:47\n- http://logs.openstack.org/28256/1/check/gate-cinder-python26/3191 : SUCCESS in 00:03:48\n- http://logs.openstack.org/28256/1/check/gate-cinder-python26-centos6/27 : SUCCESS in 00:03:09 (non-voting)\n- http://logs.openstack.org/28256/1/check/gate-cinder-python27/3207 : SUCCESS in 00:02:42\n- http://logs.openstack.org/28256/1/check/gate-tempest-devstack-vm-full/14625 : SUCCESS in 00:42:01\n- http://logs.openstack.org/28256/1/check/gate-tempest-devstack-vm-postgres-full/14340 : FAILURE in 00:44:38 (non-voting)\n- http://logs.openstack.org/28256/1/check/gate-tempest-devstack-vm-quantum/20779 : SUCCESS in 00:20:54\n- http://logs.openstack.org/28256/1/check/gate-grenade-devstack-vm/2428 : SUCCESS in 00:15:11 (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"dd12a1beec3eeba55dc7a372f64969ae033b2869","author":{"_account_id":6043,"name":"Kurt Martin","email":"kurt.f.martin@hpe.com","username":"kurt-f-martin"},"date":"2013-05-05 19:33:38.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"ac98acb3d06a966cd81935e3a930a37c64d6b690","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2013-05-05 22:20:23.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\nGood fix.  Stuff like that drives me nuts!  :-)","accounts_in_message":[],"_revision_number":1},{"id":"9844d8e38a946adeb9000286de58c059d01d2c62","author":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"date":"2013-05-06 04:12:14.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(9 inline comments)\n\nA couple of missed spots on my two last comments. Others are probably unrelated, but your call.","accounts_in_message":[],"_revision_number":1},{"id":"89d4f4f558d3bd1e781aa9a0c8617e083d1d9b6f","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-06 16:48:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8a5465be8da0b00cfd37e24d83ce8ebadc101fa","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-06 16:49:07.000000000","message":"Patch Set 1: (7 inline comments)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"ba61243849d80546ab089c16ad3ac45d98ead7b2","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-06 17:03:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ac823270074a58cecc52e7b749780243cfdbdb8","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-05-06 17:27:07.000000000","message":"Patch Set 3: Works for me\n\nSmokeStack Results (patch set 3):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/85646\n","accounts_in_message":[],"_revision_number":3},{"id":"184d2a7dcd569317571791c1ddb3b1995fb594ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-06 17:49:32.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/28256/3/check/gate-cinder-pep8/3283 : FAILURE in 00:01:37\n- http://docs-draft.openstack.org/28256/3/check/gate-cinder-docs/3175/doc/build/html/ : SUCCESS in 00:04:22\n- http://logs.openstack.org/28256/3/check/gate-cinder-python26/3197 : SUCCESS in 00:03:26\n- http://logs.openstack.org/28256/3/check/gate-cinder-python26-centos6/33 : SUCCESS in 00:02:55 (non-voting)\n- http://logs.openstack.org/28256/3/check/gate-cinder-python27/3213 : SUCCESS in 00:02:31\n- http://logs.openstack.org/28256/3/check/gate-tempest-devstack-vm-full/14704 : SUCCESS in 00:44:52\n- http://logs.openstack.org/28256/3/check/gate-tempest-devstack-vm-postgres-full/14419 : FAILURE in 00:45:31 (non-voting)\n- http://logs.openstack.org/28256/3/check/gate-tempest-devstack-vm-quantum/20885 : SUCCESS in 00:21:32\n- http://logs.openstack.org/28256/3/check/gate-grenade-devstack-vm/2501 : SUCCESS in 00:14:45 (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"f305733476d270415f589c6444dbec561d456ff5","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-06 22:53:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a659805443eab81e90596422d9c2073b4e218b9e","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-05-06 23:17:00.000000000","message":"Patch Set 4: Works for me\n\nSmokeStack Results (patch set 4):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/85672\n","accounts_in_message":[],"_revision_number":4},{"id":"be14e396178b55e0fa1b1746c0f6d8ec79cc3958","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-06 23:39:40.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28256/4/check/gate-cinder-pep8/3286 : SUCCESS in 00:01:56\n- http://docs-draft.openstack.org/28256/4/check/gate-cinder-docs/3178/doc/build/html/ : SUCCESS in 00:04:58\n- http://logs.openstack.org/28256/4/check/gate-cinder-python26/3200 : SUCCESS in 00:03:32\n- http://logs.openstack.org/28256/4/check/gate-cinder-python26-centos6/36 : SUCCESS in 00:03:09 (non-voting)\n- http://logs.openstack.org/28256/4/check/gate-cinder-python27/3216 : SUCCESS in 00:02:48\n- http://logs.openstack.org/28256/4/check/gate-tempest-devstack-vm-full/14743 : SUCCESS in 00:44:21\n- http://logs.openstack.org/28256/4/check/gate-tempest-devstack-vm-postgres-full/14458 : FAILURE in 00:45:48 (non-voting)\n- http://logs.openstack.org/28256/4/check/gate-tempest-devstack-vm-quantum/20933 : SUCCESS in 00:21:04\n- http://logs.openstack.org/28256/4/check/gate-grenade-devstack-vm/2530 : SUCCESS in 00:14:40 (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"589a6eeb29da89acb508b99cdf5d27a3983a7864","author":{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},"date":"2013-05-07 01:02:14.000000000","message":"Patch Set 4: I would prefer that you didn\u0027t merge this\n\n(13 inline comments)\n\nsome more spaces are needed","accounts_in_message":[],"_revision_number":4},{"id":"ce50b1fbd25970750a4ad76456fa65b1c464275f","author":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"date":"2013-05-08 13:52:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f353d00694af2d9b3fce4aea478b689e5c3e1c58","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-08 14:37:49.000000000","message":"Patch Set 5: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28256/5/check/gate-cinder-pep8/3304 : SUCCESS in 00:01:37\n- http://docs-draft.openstack.org/28256/5/check/gate-cinder-docs/3196/doc/build/html/ : SUCCESS in 00:04:33\n- http://logs.openstack.org/28256/5/check/gate-cinder-python26/3218 : SUCCESS in 00:03:21\n- http://logs.openstack.org/28256/5/check/gate-cinder-python26-centos6/49 : SUCCESS in 00:02:58 (non-voting)\n- http://logs.openstack.org/28256/5/check/gate-cinder-python27/3234 : SUCCESS in 00:02:34\n- http://logs.openstack.org/28256/5/check/gate-tempest-devstack-vm-full/15018 : SUCCESS in 00:43:24\n- http://logs.openstack.org/28256/5/check/gate-tempest-devstack-vm-postgres-full/14733 : FAILURE in 00:45:17 (non-voting)\n- http://logs.openstack.org/28256/5/check/gate-tempest-devstack-vm-quantum/21254 : SUCCESS in 00:20:56\n- http://logs.openstack.org/28256/5/check/gate-grenade-devstack-vm/2772 : SUCCESS in 00:18:36 (non-voting)\n","accounts_in_message":[],"_revision_number":5},{"id":"21113f66011868e66dacd3e719eca211535cd32c","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-09 05:07:33.000000000","message":"Patch Set 5:\n\nzhangchao, not sure what happened with your upload here but I\u0027m no longer able to review?\n\nFeel free to abandon and resubmit if you like, unless you know how to get this activated again.","accounts_in_message":[],"_revision_number":5},{"id":"99fbdcea8f806bcde9d81bc9e3406be6e53d1378","author":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"date":"2013-05-09 14:28:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6dcfb0bd572db19147d424a9a6c176547ec2848f","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-05-09 14:54:09.000000000","message":"Patch Set 6: Works for me\n\nSmokeStack Results (patch set 6):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/85957\n","accounts_in_message":[],"_revision_number":6},{"id":"5605aa21ecb85e4e3b7dc62e65ed45bf1f88720a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-09 15:14:15.000000000","message":"Patch Set 6: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28256/6/check/gate-cinder-pep8/3318 : SUCCESS in 00:01:37\n- http://docs-draft.openstack.org/28256/6/check/gate-cinder-docs/3210/doc/build/html/ : SUCCESS in 00:04:13\n- http://logs.openstack.org/28256/6/check/gate-cinder-python26/3232 : SUCCESS in 00:03:26\n- http://logs.openstack.org/28256/6/check/gate-cinder-python26-centos6/60 : SUCCESS in 00:02:54 (non-voting)\n- http://logs.openstack.org/28256/6/check/gate-cinder-python27/3248 : SUCCESS in 00:02:25\n- http://logs.openstack.org/28256/6/check/gate-tempest-devstack-vm-full/15219 : SUCCESS in 00:43:33\n- http://logs.openstack.org/28256/6/check/gate-tempest-devstack-vm-postgres-full/14934 : FAILURE in 00:45:14 (non-voting)\n- http://logs.openstack.org/28256/6/check/gate-tempest-devstack-vm-quantum/21482 : SUCCESS in 00:19:56\n- http://logs.openstack.org/28256/6/check/gate-grenade-devstack-vm/2958 : FAILURE in 00:16:41 (non-voting)\n","accounts_in_message":[],"_revision_number":6},{"id":"90bef5a4017bb726edc45ea247f6b4d7d3caf21c","author":{"_account_id":6043,"name":"Kurt Martin","email":"kurt.f.martin@hpe.com","username":"kurt-f-martin"},"date":"2013-05-09 16:34:14.000000000","message":"Patch Set 6: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nStill missing one space character","accounts_in_message":[],"_revision_number":6},{"id":"9d4c4386f15ca6beb019749d27bc70b79d980027","author":{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},"date":"2013-05-10 07:16:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4860b96c1c2ffcd2e8861fac3acccc5a306b1536","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-10 08:05:59.000000000","message":"Patch Set 7: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28256/7/check/gate-cinder-pep8/3335 : SUCCESS in 00:01:31\n- http://docs-draft.openstack.org/28256/7/check/gate-cinder-docs/3227/doc/build/html/ : SUCCESS in 00:04:32\n- http://logs.openstack.org/28256/7/check/gate-cinder-python26/3249 : SUCCESS in 00:02:50\n- http://logs.openstack.org/28256/7/check/gate-cinder-python27/3265 : SUCCESS in 00:02:18\n- http://logs.openstack.org/28256/7/check/gate-tempest-devstack-vm-full/15368 : SUCCESS in 00:49:02\n- http://logs.openstack.org/28256/7/check/gate-tempest-devstack-vm-postgres-full/15084 : FAILURE in 00:45:15 (non-voting)\n- http://logs.openstack.org/28256/7/check/gate-tempest-devstack-vm-quantum/21653 : SUCCESS in 00:20:36\n- http://logs.openstack.org/28256/7/check/gate-grenade-devstack-vm/3099 : FAILURE in 00:09:49 (non-voting)\n","accounts_in_message":[],"_revision_number":7},{"id":"657901fd88dfedd4a23e81f99ff4db3414d953c8","author":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"date":"2013-05-10 10:15:51.000000000","message":"Patch Set 7: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":7},{"id":"bbae46578f830a8b74aad79a0fde8a38d97adb8c","author":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"date":"2013-05-10 12:30:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b7d02145565e270214d9febcabebb9be1402ad88","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-10 13:16:10.000000000","message":"Patch Set 8: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28256/8/check/gate-cinder-pep8/3341 : SUCCESS in 00:01:28\n- http://docs-draft.openstack.org/28256/8/check/gate-cinder-docs/3233/doc/build/html/ : SUCCESS in 00:04:24\n- http://logs.openstack.org/28256/8/check/gate-cinder-python26/3255 : SUCCESS in 00:02:48\n- http://logs.openstack.org/28256/8/check/gate-cinder-python27/3271 : SUCCESS in 00:02:18\n- http://logs.openstack.org/28256/8/check/gate-tempest-devstack-vm-full/15390 : SUCCESS in 00:44:47\n- http://logs.openstack.org/28256/8/check/gate-tempest-devstack-vm-postgres-full/15106 : FAILURE in 00:45:18 (non-voting)\n- http://logs.openstack.org/28256/8/check/gate-tempest-devstack-vm-quantum/21678 : SUCCESS in 00:22:50\n- http://logs.openstack.org/28256/8/check/gate-grenade-devstack-vm/3119 : FAILURE in 00:10:19 (non-voting)\n","accounts_in_message":[],"_revision_number":8},{"id":"ec04b2f78e49c283cb6a52c46a13a9a90cbf4c1a","author":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"date":"2013-05-10 13:59:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0f5b958b233bad4b1d30673e23cb51c34b6fc30f","author":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"date":"2013-05-10 14:02:46.000000000","message":"Patch Set 9:\n\nSorry,I made a wrong submit.I have no idea how to delete the draft so I have to resubmit the patch.","accounts_in_message":[],"_revision_number":9},{"id":"6a329650408690a27dac880b0c8e7f9b005c7c95","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-10 14:44:44.000000000","message":"Patch Set 9: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28256/9/check/gate-cinder-pep8/3343 : SUCCESS in 00:01:31\n- http://docs-draft.openstack.org/28256/9/check/gate-cinder-docs/3235/doc/build/html/ : SUCCESS in 00:04:17\n- http://logs.openstack.org/28256/9/check/gate-cinder-python26/3257 : SUCCESS in 00:02:51\n- http://logs.openstack.org/28256/9/check/gate-cinder-python27/3273 : SUCCESS in 00:02:21\n- http://logs.openstack.org/28256/9/check/gate-tempest-devstack-vm-full/15402 : SUCCESS in 00:44:26\n- http://logs.openstack.org/28256/9/check/gate-tempest-devstack-vm-postgres-full/15118 : FAILURE in 00:45:09 (non-voting)\n- http://logs.openstack.org/28256/9/check/gate-tempest-devstack-vm-quantum/21690 : SUCCESS in 00:20:38\n- http://logs.openstack.org/28256/9/check/gate-grenade-devstack-vm/3131 : FAILURE in 00:11:11 (non-voting)\n","accounts_in_message":[],"_revision_number":9},{"id":"76d768544e03022fede8ed8a15a40af2d586b00f","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-10 14:58:16.000000000","message":"Patch Set 9: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":9},{"id":"464ad61d38bd89947da6e5242c94ef90c1cd5b8e","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2013-05-10 15:46:44.000000000","message":"Patch Set 9: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":9},{"id":"3b4a8f7a4c26aa6e3facc20f4aa258d51fcdc3c6","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2013-05-10 16:24:44.000000000","message":"Patch Set 9: Looks good to me, but someone else must approve\n\nWould you guys mind to chime in here[0] wrt translating exceptions?\n\nhttp://lists.openstack.org/pipermail/openstack-dev/2013-May/008881.html","accounts_in_message":[],"_revision_number":9},{"id":"d716129a8cabe101631730134c087b3447b8c1b9","author":{"_account_id":1207,"name":"Duncan Thomas","email":"duncan.thomas@gmail.com","username":"duncan-thomas"},"date":"2013-05-10 16:52:47.000000000","message":"Patch Set 9: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":9},{"id":"d14a84524877a4835d017009c38884b11108ff90","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-10 16:52:49.000000000","message":"Patch Set 9: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":9},{"id":"3f0acec37b4080cf0b81c95b5fc65feb190e8fb1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-10 17:38:14.000000000","message":"Patch Set 9: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/28256/9/gate/gate-cinder-docs/3237/doc/build/html/ : SUCCESS in 00:04:58\n- http://logs.openstack.org/28256/9/gate/gate-cinder-pep8/3345 : SUCCESS in 00:01:39\n- http://logs.openstack.org/28256/9/gate/gate-cinder-python26/3259 : SUCCESS in 00:02:55\n- http://logs.openstack.org/28256/9/gate/gate-cinder-python27/3275 : SUCCESS in 00:02:26\n- http://logs.openstack.org/28256/9/gate/gate-tempest-devstack-vm-full/15431 : SUCCESS in 00:44:48\n- http://logs.openstack.org/28256/9/gate/gate-tempest-devstack-vm-postgres-full/15147 : FAILURE in 00:45:14 (non-voting)\n- http://logs.openstack.org/28256/9/gate/gate-tempest-devstack-vm-quantum/21721 : SUCCESS in 00:20:11\n- http://logs.openstack.org/28256/9/gate/gate-grenade-devstack-vm/3153 : FAILURE in 00:09:43 (non-voting)\n","accounts_in_message":[],"_revision_number":9},{"id":"2c51a08c12798fddf817015cf68f6f4f8e956fb8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-10 17:38:15.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"caeb4a215fa3a1c0b142e3dfbe5c73c20e537ac3","revisions":{"a266dd2434a13a47bf67c49ff20eb5512f5dd61f":{"kind":"REWORK","_number":1,"created":"2013-05-05 17:15:00.000000000","uploader":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"ref":"refs/changes/56/28256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/28256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/28256/1"}}},"commit":{"parents":[{"commit":"822cda06ca664e87ff01c8a184d5744a424f7315","subject":"Merge \"Fixes 3par driver methods that were double locking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/822cda06ca664e87ff01c8a184d5744a424f7315"}]}],"author":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-05 17:12:53.000000000","tz":-360},"committer":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-05 17:14:58.000000000","tz":-360},"subject":"Fix missing spaces in Huawei Logging","message":"Fix missing spaces in Huawei Logging\n\nAdd spaces between line concatenations in Log messages.\n\nChange-Id: I4e94fc751658fbe1b6b9b74859d800445a1c2e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a266dd2434a13a47bf67c49ff20eb5512f5dd61f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a266dd2434a13a47bf67c49ff20eb5512f5dd61f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"822cda06ca664e87ff01c8a184d5744a424f7315","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a9fa9e2925ebb1553488060bac828fd6468bd5f8":{"kind":"REWORK","_number":2,"created":"2013-05-06 16:48:09.000000000","uploader":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"ref":"refs/changes/56/28256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/28256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/28256/2"}}},"commit":{"parents":[{"commit":"822cda06ca664e87ff01c8a184d5744a424f7315","subject":"Merge \"Fixes 3par driver methods that were double locking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/822cda06ca664e87ff01c8a184d5744a424f7315"}]}],"author":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-05 17:12:53.000000000","tz":-360},"committer":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-06 16:48:08.000000000","tz":-360},"subject":"Fix missing spaces in Huawei Logging","message":"Fix missing spaces in Huawei Logging\n\nAdd spaces between line concatenations in Log messages.\n\nChange-Id: I4e94fc751658fbe1b6b9b74859d800445a1c2e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a9fa9e2925ebb1553488060bac828fd6468bd5f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a9fa9e2925ebb1553488060bac828fd6468bd5f8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"822cda06ca664e87ff01c8a184d5744a424f7315","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"23f0232e8c5cac26b8ad561444054f13026d65f5":{"kind":"REWORK","_number":3,"created":"2013-05-06 17:03:48.000000000","uploader":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"ref":"refs/changes/56/28256/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/28256/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/28256/3"}}},"commit":{"parents":[{"commit":"822cda06ca664e87ff01c8a184d5744a424f7315","subject":"Merge \"Fixes 3par driver methods that were double locking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/822cda06ca664e87ff01c8a184d5744a424f7315"}]}],"author":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-05 17:12:53.000000000","tz":-360},"committer":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-06 17:03:36.000000000","tz":-360},"subject":"Fix missing spaces in Huawei Logging","message":"Fix missing spaces in Huawei Logging\n\nAdd spaces between line concatenations in Log messages.\n\nChange-Id: I4e94fc751658fbe1b6b9b74859d800445a1c2e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23f0232e8c5cac26b8ad561444054f13026d65f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23f0232e8c5cac26b8ad561444054f13026d65f5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"822cda06ca664e87ff01c8a184d5744a424f7315","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"18be052eb48caaff274dd44d79414a07bf48daf6":{"kind":"REWORK","_number":4,"created":"2013-05-06 22:53:36.000000000","uploader":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"ref":"refs/changes/56/28256/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/28256/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/28256/4"}}},"commit":{"parents":[{"commit":"822cda06ca664e87ff01c8a184d5744a424f7315","subject":"Merge \"Fixes 3par driver methods that were double locking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/822cda06ca664e87ff01c8a184d5744a424f7315"}]}],"author":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-05 17:12:53.000000000","tz":-360},"committer":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-06 22:53:26.000000000","tz":-360},"subject":"Fix missing spaces in Huawei Logging","message":"Fix missing spaces in Huawei Logging\n\nAdd spaces between line concatenations in Log messages.\n\nChange-Id: I4e94fc751658fbe1b6b9b74859d800445a1c2e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18be052eb48caaff274dd44d79414a07bf48daf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18be052eb48caaff274dd44d79414a07bf48daf6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"822cda06ca664e87ff01c8a184d5744a424f7315","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"4a398e4be08bf47ee2d4283cbd088fe58ccb762b":{"kind":"REWORK","_number":5,"created":"2013-05-08 13:52:18.000000000","uploader":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"ref":"refs/changes/56/28256/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/28256/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/28256/5"}}},"commit":{"parents":[{"commit":"822cda06ca664e87ff01c8a184d5744a424f7315","subject":"Merge \"Fixes 3par driver methods that were double locking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/822cda06ca664e87ff01c8a184d5744a424f7315"}]}],"author":{"name":"zhangchao010","email":"zhangchao010@huawei.com","date":"2013-05-08 13:48:26.000000000","tz":480},"committer":{"name":"zhangchao010","email":"zhangchao010@huawei.com","date":"2013-05-08 13:48:26.000000000","tz":480},"subject":"Fix missing spaces in Huawei Logging","message":"Fix missing spaces in Huawei Logging\n\nAdd spaces between line concatenations in Log messages.\n\nChange-Id: I4e94fc751658fbe1b6b9b74859d800445a1c2e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a398e4be08bf47ee2d4283cbd088fe58ccb762b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a398e4be08bf47ee2d4283cbd088fe58ccb762b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"822cda06ca664e87ff01c8a184d5744a424f7315","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f8cf3029defe3bb067be87f6157192066e4ddfa9":{"kind":"REWORK","_number":6,"created":"2013-05-09 14:28:40.000000000","uploader":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"ref":"refs/changes/56/28256/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/28256/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/28256/6"}}},"commit":{"parents":[{"commit":"822cda06ca664e87ff01c8a184d5744a424f7315","subject":"Merge \"Fixes 3par driver methods that were double locking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/822cda06ca664e87ff01c8a184d5744a424f7315"}]}],"author":{"name":"zhangchao010","email":"zhangchao010@huawei.com","date":"2013-05-09 14:08:37.000000000","tz":480},"committer":{"name":"zhangchao010","email":"zhangchao010@huawei.com","date":"2013-05-09 14:24:03.000000000","tz":480},"subject":"Fix missing spaces in Huawei Logging","message":"Fix missing spaces in Huawei Logging\n\nAdd spaces between line concatenations in Log messages.\n\nChange-Id: I4e94fc751658fbe1b6b9b74859d800445a1c2e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8cf3029defe3bb067be87f6157192066e4ddfa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8cf3029defe3bb067be87f6157192066e4ddfa9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"822cda06ca664e87ff01c8a184d5744a424f7315","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"baf5940a9ac8e629948801b88b4fa4264f9f9ef6":{"kind":"REWORK","_number":7,"created":"2013-05-10 07:16:33.000000000","uploader":{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},"ref":"refs/changes/56/28256/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/28256/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/28256/7"}}},"commit":{"parents":[{"commit":"7dad062dd694c85af3ecd85e07a3246cef204a61","subject":"Remove outdated cinder test doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7dad062dd694c85af3ecd85e07a3246cef204a61"}]}],"author":{"name":"zhangchao010","email":"zhangchao010@huawei.com","date":"2013-05-09 14:08:37.000000000","tz":480},"committer":{"name":"Zhiteng Huang","email":"zhiteng.huang@intel.com","date":"2013-05-10 07:15:22.000000000","tz":480},"subject":"Fix missing spaces in Huawei Logging","message":"Fix missing spaces in Huawei Logging\n\nAdd spaces between line concatenations in Log messages.\n\nChange-Id: I4e94fc751658fbe1b6b9b74859d800445a1c2e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/baf5940a9ac8e629948801b88b4fa4264f9f9ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/baf5940a9ac8e629948801b88b4fa4264f9f9ef6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7dad062dd694c85af3ecd85e07a3246cef204a61","is_merged_in_target_branch":true,"change_id":"Id096ec97b37a2019baf7a0b55bfd5bce4e71908f","change_number":28726,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"351b3556251c74a20d349be7598958808ea2e805":{"kind":"REWORK","_number":8,"created":"2013-05-10 12:30:39.000000000","uploader":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"ref":"refs/changes/56/28256/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/28256/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/28256/8"}}},"commit":{"parents":[{"commit":"d8c9e68d43323032e8e0d007aa29e85ff5afedbf","subject":"Merge \"Fix ability to add custom volume_backend_name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8c9e68d43323032e8e0d007aa29e85ff5afedbf"}]}],"author":{"name":"zhangchao010","email":"zhangchao010@huawei.com","date":"2013-05-09 14:08:37.000000000","tz":480},"committer":{"name":"zhangchao010","email":"zhangchao010@huawei.com","date":"2013-05-10 12:23:28.000000000","tz":480},"subject":"Fix missing spaces in Huawei Logging","message":"Fix missing spaces in Huawei Logging\n\nAdd spaces between line concatenations in Log messages.\n\nChange-Id: I4e94fc751658fbe1b6b9b74859d800445a1c2e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/351b3556251c74a20d349be7598958808ea2e805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/351b3556251c74a20d349be7598958808ea2e805"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d8c9e68d43323032e8e0d007aa29e85ff5afedbf","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"caeb4a215fa3a1c0b142e3dfbe5c73c20e537ac3":{"kind":"REWORK","_number":9,"created":"2013-05-10 13:59:21.000000000","uploader":{"_account_id":6604,"name":"zhangchao","email":"zhangchao010@huawei.com","username":"zhangchao010"},"ref":"refs/changes/56/28256/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/28256/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/28256/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/28256/9"}}},"commit":{"parents":[{"commit":"d8c9e68d43323032e8e0d007aa29e85ff5afedbf","subject":"Merge \"Fix ability to add custom volume_backend_name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8c9e68d43323032e8e0d007aa29e85ff5afedbf"}]}],"author":{"name":"zhangchao010","email":"zhangchao010@huawei.com","date":"2013-05-10 13:41:23.000000000","tz":480},"committer":{"name":"zhangchao010","email":"zhangchao010@huawei.com","date":"2013-05-10 13:54:30.000000000","tz":480},"subject":"Fix missing spaces in Huawei Logging","message":"Fix missing spaces in Huawei Logging\n\nAdd spaces between line concatenations in Log messages.\n\nChange-Id: I4e94fc751658fbe1b6b9b74859d800445a1c2e5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/caeb4a215fa3a1c0b142e3dfbe5c73c20e537ac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/caeb4a215fa3a1c0b142e3dfbe5c73c20e537ac3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d8c9e68d43323032e8e0d007aa29e85ff5afedbf","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
