)]}'
{"id":"openstack%2Fcinder~28440","triplet_id":"openstack%2Fcinder~stable%2Fgrizzly~I4197f99bebd6ca8e4969795ae6edb537bc066c90","project":"openstack/cinder","branch":"stable/grizzly","topic":"bug/1175222","hashtags":[],"change_id":"I4197f99bebd6ca8e4969795ae6edb537bc066c90","subject":"Fix ability to add custom volume_backend_name","status":"MERGED","created":"2013-05-07 17:13:23.000000000","updated":"2013-05-08 17:13:24.000000000","submitted":"2013-05-08 17:13:23.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2910da07d73fcd11913672994e1c0fe6a456b6eb","_number":28440,"virtual_id_number":28440,"owner":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},{"value":2,"date":"2013-05-08 17:13:23.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":2,"date":"2013-05-07 19:31:20.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":2,"date":"2013-05-08 15:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":2,"date":"2013-05-08 16:23:26.000000000","_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-05-08 16:22:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"value":1,"date":"2013-05-07 17:20:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2013-05-07 19:53:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"all":[{"value":0,"date":"2013-05-07 19:31:20.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"date":"2013-05-08 15:49:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"date":"2013-05-08 16:23:26.000000000","_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-05-08 16:22:31.000000000","_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"value":0,"date":"2013-05-07 17:13:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-07 17:13:23.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2013-05-07 19:53:13.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2013-05-08 15:49:00.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2013-05-08 16:22:31.000000000","updated_by":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"reviewer":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"state":"REVIEWER"},{"updated":"2013-05-08 16:23:26.000000000","updated_by":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"reviewer":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"state":"REVIEWER"},{"updated":"2013-05-08 17:13:23.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"b182d191ec3abfc6b5733126f5c0846338e56a8c","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2013-05-07 17:20:47.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"1d8ddb28a2489b7f29136b9bb2557df256ecd588","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-07 17:51:43.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28440/1/check/gate-cinder-pep8/3296 : SUCCESS in 00:01:31\n- http://docs-draft.openstack.org/28440/1/check/gate-cinder-docs/3188/doc/build/html/ : SUCCESS in 00:04:18\n- http://logs.openstack.org/28440/1/check/gate-cinder-python26/3210 : SUCCESS in 00:03:26\n- http://logs.openstack.org/28440/1/check/gate-cinder-python26-centos6/42 : SUCCESS in 00:02:54 (non-voting)\n- http://logs.openstack.org/28440/1/check/gate-cinder-python27/3226 : SUCCESS in 00:02:29\n- http://logs.openstack.org/28440/1/check/gate-tempest-devstack-vm-full/14872 : SUCCESS in 00:38:05\n- http://logs.openstack.org/28440/1/check/gate-tempest-devstack-vm-postgres-full/14587 : SUCCESS in 00:35:54 (non-voting)\n- http://logs.openstack.org/28440/1/check/gate-tempest-devstack-vm-quantum/21090 : SUCCESS in 00:19:21\n- http://logs.openstack.org/28440/1/check/gate-grenade-devstack-vm/2643 : SUCCESS in 00:16:01 (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"07a18a9dc64cd945579ea205840b113e7c97c2e7","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2013-05-07 18:38:51.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(2 inline comments)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"380d49b8ad8fcedf366724a9f8312f659669ee9f","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2013-05-07 19:01:59.000000000","message":"Patch Set 1: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"cf92cb9ca32547a4464ec09ac62498030430ce53","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2013-05-07 19:21:02.000000000","message":"Patch Set 1: No score\n\n(1 inline comment)\n\nResponded ...","accounts_in_message":[],"_revision_number":1},{"id":"0d9c3cf112b49f40114d6b38f16239ed02f27192","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-07 19:31:20.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\n(1 inline comment)\n\nHi Jay, good feed-back and there are a number of issues that we still would like to address here.  However, this is just a minimal patch to get the expected functionality in stable.\n\nI plan to fix up the code a bit in master in a separate patch but those changes are do not warrant a backport to stable IMO.\n\nPlease see my comments in line and let me know if you disagree.","accounts_in_message":[],"_revision_number":1},{"id":"24ae28fc21d587b2b3e3f5f6259434a26ff6f441","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2013-05-07 19:53:13.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\nJohn,\n\nThanks for the additional comments.  I see your argument about this being a backport.  It would be best to keep this consistent with the original in that case.\n\nTechnically, there isn\u0027t a big difference between what is being done versus what I am suggesting.  It is just a readability issue.  I had to stare at those two locations for a bit trying to figure out why the variable was being preset and then possibly not used.  Felt since it wasn\u0027t clear I would comment.  \n\nCertainly not a show-stopper.  I recant my -1. :-)","accounts_in_message":[],"_revision_number":1},{"id":"e861852fdf205212a65d909c8df08948ae72204e","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2013-05-08 15:49:00.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\nProposed as an exception candidate for 2013.1.1, it enables documented functionality.\nChanges a lot of lines but all are safe, keeping existing default values for backend name.","accounts_in_message":[],"_revision_number":1},{"id":"0ca175c99d28f109e9cba164596c53bc84f91847","author":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"date":"2013-05-08 16:22:31.000000000","message":"Patch Set 1: Approved\n\nSimilar fix was already backported to solidfire driver in https://review.openstack.org/#/c/28207/.  Only concerns seem to be style concerns, which should have been voiced during review of the original patch to master.","accounts_in_message":[],"_revision_number":1},{"id":"f71a51bd4f5efb41481c8ac6f3d476bf14420cdb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-08 16:22:33.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"4800685d946d8df3b2ed23be41ab11c589cde554","author":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"date":"2013-05-08 16:23:26.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\nfwiw +2 from me as well","accounts_in_message":[],"_revision_number":1},{"id":"eb437d76e5f6d832d3449387943d1be04a7c5236","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-08 17:13:23.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/28440/1/gate/gate-cinder-docs/3200/doc/build/html/ : SUCCESS in 00:04:07\n- http://logs.openstack.org/28440/1/gate/gate-cinder-pep8/3308 : SUCCESS in 00:01:30\n- http://logs.openstack.org/28440/1/gate/gate-cinder-python26/3222 : SUCCESS in 00:03:17\n- http://logs.openstack.org/28440/1/gate/gate-cinder-python27/3238 : SUCCESS in 00:02:23\n- http://logs.openstack.org/28440/1/gate/gate-tempest-devstack-vm-full/15053 : SUCCESS in 00:50:40\n- http://logs.openstack.org/28440/1/gate/gate-tempest-devstack-vm-postgres-full/14768 : SUCCESS in 00:36:03 (non-voting)\n- http://logs.openstack.org/28440/1/gate/gate-tempest-devstack-vm-quantum/21294 : SUCCESS in 00:18:46\n- http://logs.openstack.org/28440/1/gate/gate-grenade-devstack-vm/2804 : SUCCESS in 00:18:28 (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"56b2efc2359bbbb1ff073ddf8717ec7af84cc8a4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-08 17:13:24.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"995a3100fcaa7eb445b2838c4b4799a482c817cc","revisions":{"995a3100fcaa7eb445b2838c4b4799a482c817cc":{"kind":"REWORK","_number":1,"created":"2013-05-07 17:13:23.000000000","uploader":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"ref":"refs/changes/40/28440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/28440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/28440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/28440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/28440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/28440/1"}}},"commit":{"parents":[{"commit":"79b9582367402ecdf2fd0ed66ae67f63a18da49c","subject":"Merge \"Fixes 3PAR FC driver synchronization\" into stable/grizzly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79b9582367402ecdf2fd0ed66ae67f63a18da49c"}]}],"author":{"name":"Walter A. Boring IV","email":"walter.boring@hp.com","date":"2013-05-03 21:36:12.000000000","tz":-420},"committer":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-07 17:13:19.000000000","tz":-360},"subject":"Fix ability to add custom volume_backend_name","message":"Fix ability to add custom volume_backend_name\n\nThis patch fixes the ability to set the volume_backend_name\nfor each driver.   The original bug was filed against LeftHand\nand 3PAR, but several drivers exhibited a fixed backend name.\n\nFixes: bug #1175222\n       bug #1172286\n\nChange-Id: I4197f99bebd6ca8e4969795ae6edb537bc066c90\n(cherry picked from commit 54a2ee48a4dc31cbf116b1e9d5add37efe66bc20)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/995a3100fcaa7eb445b2838c4b4799a482c817cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/995a3100fcaa7eb445b2838c4b4799a482c817cc"}]},"branch":"refs/heads/stable/grizzly"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
