)]}'
{"id":"openstack%2Fcinder~29591","triplet_id":"openstack%2Fcinder~master~I10136f8d1628128a93c4adfe2cb9447b99720558","project":"openstack/cinder","branch":"master","topic":"remotefs-image","hashtags":[],"change_id":"I10136f8d1628128a93c4adfe2cb9447b99720558","subject":"RemoteFsDriver: copy_image_to_volume and copy_volume_to_image","status":"MERGED","created":"2013-05-17 19:04:43.000000000","updated":"2013-05-18 18:04:16.000000000","submitted":"2013-05-18 18:04:16.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9e5e3387d8f36dcab1bfda09f28707c4c22179e5","_number":29591,"virtual_id_number":29591,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":2,"date":"2013-05-18 18:04:16.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-05-17 19:42:46.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":2,"date":"2013-05-17 20:16:02.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":2,"date":"2013-05-17 22:26:15.000000000","_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"date":"2013-05-18 16:12:24.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":1,"date":"2013-05-17 20:38:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"all":[{"value":0,"date":"2013-05-17 20:16:02.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":1,"date":"2013-05-17 22:26:15.000000000","_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"date":"2013-05-17 19:18:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-17 19:42:46.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-05-17 20:16:02.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2013-05-17 20:38:19.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2013-05-17 22:26:15.000000000","updated_by":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"reviewer":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"state":"REVIEWER"},{"updated":"2013-05-18 18:04:16.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"85762c532d3623ced05112c1ccf7047c0dd8e4dc","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-05-17 19:18:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c96a930cb91d7a32712d1ec1f53ab5aeb5f461da","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-05-17 19:42:46.000000000","message":"Patch Set 2: Works for me\n\nSmokeStack Results (patch set 2):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/86869\n","accounts_in_message":[],"_revision_number":2},{"id":"0643931c524fe2a3d2331d499eeadc56a1fad318","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-17 20:04:00.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/29591/2/check/gate-cinder-pep8/3456 : SUCCESS in 1m 40s\n- http://docs-draft.openstack.org/29591/2/check/gate-cinder-docs/3348/doc/build/html/ : SUCCESS in 3m 25s\n- http://logs.openstack.org/29591/2/check/gate-cinder-python26/3370 : SUCCESS in 2m 53s\n- http://logs.openstack.org/29591/2/check/gate-cinder-python27/3386 : SUCCESS in 2m 24s\n- http://logs.openstack.org/29591/2/check/gate-cinder-pylint/19 : SUCCESS in 3m 12s (non-voting)\n- http://logs.openstack.org/29591/2/check/gate-tempest-devstack-vm-full/16595 : SUCCESS in 44m 30s\n- http://logs.openstack.org/29591/2/check/gate-tempest-devstack-vm-postgres-full/16310 : FAILURE in 44m 47s (non-voting)\n- http://logs.openstack.org/29591/2/check/gate-tempest-devstack-vm-quantum/23151 : SUCCESS in 21m 26s\n- http://logs.openstack.org/29591/2/check/gate-grenade-devstack-vm/4173 : FAILURE in 9m 56s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"a23efe89c8ae2f941628fa2ca3f8239b0f8bdd56","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-17 20:16:02.000000000","message":"Patch Set 2: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"182807ebafed074ba2008b43337d244ce5dee69e","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2013-05-17 20:38:19.000000000","message":"Patch Set 2: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":2},{"id":"d5e7d2da80f2540bab94af741ada3957117a6387","author":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"date":"2013-05-17 22:26:15.000000000","message":"Patch Set 2: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":2},{"id":"49a94ba2a252605c08cd0c4e9cbd81645940f128","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-17 22:26:18.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"350b409cd6ea85cda7f36dbbec9d0193a0968db2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-17 23:16:18.000000000","message":"Patch Set 2: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://docs-draft.openstack.org/29591/2/gate/gate-cinder-docs/3357/doc/build/html/ : SUCCESS in 3m 25s\n- http://logs.openstack.org/29591/2/gate/gate-cinder-pep8/3465 : SUCCESS in 1m 34s\n- http://logs.openstack.org/29591/2/gate/gate-cinder-python26/3379 : SUCCESS in 2m 50s\n- http://logs.openstack.org/29591/2/gate/gate-cinder-python27/3395 : SUCCESS in 2m 21s\n- http://logs.openstack.org/29591/2/gate/gate-tempest-devstack-vm-full/16631 : SUCCESS in 49m 50s\n- http://logs.openstack.org/29591/2/gate/gate-tempest-devstack-vm-postgres-full/16346 : FAILURE in 44m 57s (non-voting)\n- http://logs.openstack.org/29591/2/gate/gate-tempest-devstack-vm-quantum/23189 : FAILURE in 28m 11s\n- http://logs.openstack.org/29591/2/gate/gate-grenade-devstack-vm/4207 : FAILURE in 10m 09s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"7dc30d1d858234fc52301f3963e9702754c8dfb1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-05-18 16:12:24.000000000","message":"Patch Set 2:\n\nreverify bug 1181567","accounts_in_message":[],"_revision_number":2},{"id":"6e53dda3746ca714e9b3fc18280bc51f4b732b83","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-18 16:49:40.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"6be72483eff5bed8cc2dba041d665c14e7aa4d72","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-18 18:04:16.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/29591/2/gate/gate-cinder-docs/3363/doc/build/html/ : SUCCESS in 3m 39s\n- http://logs.openstack.org/29591/2/gate/gate-cinder-pep8/3471 : SUCCESS in 1m 47s\n- http://logs.openstack.org/29591/2/gate/gate-cinder-python26/3385 : SUCCESS in 3m 03s\n- http://logs.openstack.org/29591/2/gate/gate-cinder-python27/3401 : SUCCESS in 2m 33s\n- http://logs.openstack.org/29591/2/gate/gate-tempest-devstack-vm-full/16714 : SUCCESS in 49m 42s\n- http://logs.openstack.org/29591/2/gate/gate-tempest-devstack-vm-postgres-full/16429 : FAILURE in 44m 59s (non-voting)\n- http://logs.openstack.org/29591/2/gate/gate-tempest-devstack-vm-quantum/23291 : SUCCESS in 19m 03s\n- http://logs.openstack.org/29591/2/gate/gate-grenade-devstack-vm/4290 : FAILURE in 10m 51s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"d9a3751b9ecb90f043ef32c9d157439042c7ae9e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-18 18:04:16.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"615b4508a5608301c4f6962f0d11bdca31a70c11","revisions":{"7245f0742ff6f9995a27276d699024d8bca28be3":{"kind":"REWORK","_number":1,"created":"2013-05-17 19:04:43.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/91/29591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/29591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/29591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/29591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/29591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/29591/1"}}},"commit":{"parents":[{"commit":"1d6de0f42cb36861338209a4c9ead1dc677d4f2f","subject":"Merge \"solidfire: Make sure src_uuid is passed correctly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d6de0f42cb36861338209a4c9ead1dc677d4f2f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-05-17 18:48:24.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-05-17 19:02:03.000000000","tz":-240},"subject":"RemoteFsDriver: copy_image_to_volume and copy_volume_to_image","message":"RemoteFsDriver: copy_image_to_volume and copy_volume_to_image\n\nThis functionality was recently added to the NFS driver --\nit should instead be added to the RemoteFsDriver class to cover\nboth NFS and GlusterFS drivers.\n\nChange-Id: I10136f8d1628128a93c4adfe2cb9447b99720558\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7245f0742ff6f9995a27276d699024d8bca28be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7245f0742ff6f9995a27276d699024d8bca28be3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d6de0f42cb36861338209a4c9ead1dc677d4f2f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"615b4508a5608301c4f6962f0d11bdca31a70c11":{"kind":"REWORK","_number":2,"created":"2013-05-17 19:18:24.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/91/29591/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/29591/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/29591/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/29591/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/29591/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/29591/2"}}},"commit":{"parents":[{"commit":"1d6de0f42cb36861338209a4c9ead1dc677d4f2f","subject":"Merge \"solidfire: Make sure src_uuid is passed correctly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d6de0f42cb36861338209a4c9ead1dc677d4f2f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-05-17 18:48:24.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-05-17 19:17:53.000000000","tz":-240},"subject":"RemoteFsDriver: copy_image_to_volume and copy_volume_to_image","message":"RemoteFsDriver: copy_image_to_volume and copy_volume_to_image\n\nThis functionality was recently added to the NFS driver --\nit should instead be added to the RemoteFsDriver class to cover\nboth NFS and GlusterFS drivers.\n\nChange-Id: I10136f8d1628128a93c4adfe2cb9447b99720558\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/615b4508a5608301c4f6962f0d11bdca31a70c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/615b4508a5608301c4f6962f0d11bdca31a70c11"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d6de0f42cb36861338209a4c9ead1dc677d4f2f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
