)]}'
{"id":"openstack%2Fcinder~29599","triplet_id":"openstack%2Fcinder~master~Ic551f512c5a0cf51bafa8b60f0b6b7debbe22692","project":"openstack/cinder","branch":"master","topic":"bug/1180976","hashtags":[],"change_id":"Ic551f512c5a0cf51bafa8b60f0b6b7debbe22692","subject":"Update/Publish volume service updates on delete.","status":"MERGED","created":"2013-05-17 20:10:15.000000000","updated":"2013-05-17 23:03:44.000000000","submitted":"2013-05-17 23:03:44.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a6ba9a59bab231207c6293ee3184af72419ffc2e","_number":29599,"virtual_id_number":29599,"owner":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":2,"date":"2013-05-17 23:03:44.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-05-17 21:00:10.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":2,"date":"2013-05-17 22:12:02.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":2,"date":"2013-05-17 22:06:07.000000000","_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":1,"date":"2013-05-17 20:50:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2013-05-17 20:56:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":1,"date":"2013-05-17 22:12:02.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"date":"2013-05-17 22:06:07.000000000","_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-17 20:31:44.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2013-05-17 20:33:33.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2013-05-17 20:50:17.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2013-05-17 20:56:37.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2013-05-17 21:00:10.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-05-17 22:06:07.000000000","updated_by":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"reviewer":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"state":"REVIEWER"},{"updated":"2013-05-17 23:03:44.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"a618799f2be4ddcdddd245cadec9bac760803832","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-17 20:13:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4663b161bb014d992f02ea16f0ad91bdc0575ce9","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-05-17 20:26:14.000000000","message":"Patch Set 2: Looks good to me, but someone else must approve\n\n(1 inline comment)\n\nthough I\u0027m surprised you left the lp bug number in the code","accounts_in_message":[],"_revision_number":2},{"id":"506bfb41153f59b75f21a1998fb85cadf0e07428","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2013-05-17 20:29:18.000000000","message":"Patch Set 2: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"721d7fd3f4bfab50843417589ce9c7558b4162a6","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2013-05-17 20:31:44.000000000","message":"Patch Set 2: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":2},{"id":"844e3da4a8f507a5bee350abe3a84c0dbc926eee","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-05-17 20:33:33.000000000","message":"Patch Set 2:\n\nI have the same comments as Sean and Walter, would prefer to change both.","accounts_in_message":[],"_revision_number":2},{"id":"1d3d3f495950d660e26b3c9882f1dd7c68ded32a","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2013-05-17 20:34:55.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\nShouldn\u0027t we use self.publish_service_capabilities() here instead of making the internal calls manually?","accounts_in_message":[],"_revision_number":2},{"id":"39c4642b35d74596eedbccdc748af4804aec8e31","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-17 20:38:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd7157dc5e172d934feb640201f50cfc0bcc6fbb","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-17 20:39:03.000000000","message":"Patch Set 2: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"25a8ce62812b23780598572aaca836f829797d14","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2013-05-17 20:50:17.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"0c9c89df9825581f5713f19624e9046bf5c7394b","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-05-17 20:56:37.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"a2d1a7e279f5c24f8a0aaf7bf55cb81e2b1df972","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-05-17 21:00:10.000000000","message":"Patch Set 3: Works for me\n\nSmokeStack Results (patch set 3):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/86879\n","accounts_in_message":[],"_revision_number":3},{"id":"7a3aeec2c541a48c53ed6ea9e4eccc97f8cf9828","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-17 21:22:41.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/29599/3/check/gate-cinder-pep8/3461 : SUCCESS in 1m 35s\n- http://docs-draft.openstack.org/29599/3/check/gate-cinder-docs/3353/doc/build/html/ : SUCCESS in 3m 36s\n- http://logs.openstack.org/29599/3/check/gate-cinder-python26/3375 : SUCCESS in 2m 51s\n- http://logs.openstack.org/29599/3/check/gate-cinder-python27/3391 : SUCCESS in 2m 25s\n- http://logs.openstack.org/29599/3/check/gate-cinder-pylint/22 : SUCCESS in 3m 10s (non-voting)\n- http://logs.openstack.org/29599/3/check/gate-tempest-devstack-vm-full/16612 : SUCCESS in 43m 38s\n- http://logs.openstack.org/29599/3/check/gate-tempest-devstack-vm-postgres-full/16327 : FAILURE in 40m 36s (non-voting)\n- http://logs.openstack.org/29599/3/check/gate-tempest-devstack-vm-quantum/23168 : SUCCESS in 21m 22s\n- http://logs.openstack.org/29599/3/check/gate-grenade-devstack-vm/4189 : FAILURE in 10m 07s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"ad55dcaf18ad7e9c92f6785fd960a7b3f07f253f","author":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"date":"2013-05-17 22:06:07.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"a79b450cb40f2197b76995d29654e2a124c2076b","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-05-17 22:12:02.000000000","message":"Patch Set 3: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":3},{"id":"6492f3f14815fd2f3d1e9668d7c238e9a16e123b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-17 22:12:30.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"d9fb83322939c50b8a385708c3aac437cb726325","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-17 23:03:44.000000000","message":"Patch Set 3: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/29599/3/gate/gate-cinder-docs/3356/doc/build/html/ : SUCCESS in 4m 22s\n- http://logs.openstack.org/29599/3/gate/gate-cinder-pep8/3464 : SUCCESS in 2m 07s\n- http://logs.openstack.org/29599/3/gate/gate-cinder-python26/3378 : SUCCESS in 3m 22s\n- http://logs.openstack.org/29599/3/gate/gate-cinder-python27/3394 : SUCCESS in 4m 14s\n- http://logs.openstack.org/29599/3/gate/gate-tempest-devstack-vm-full/16629 : SUCCESS in 49m 23s\n- http://logs.openstack.org/29599/3/gate/gate-tempest-devstack-vm-postgres-full/16344 : FAILURE in 51m 04s (non-voting)\n- http://logs.openstack.org/29599/3/gate/gate-tempest-devstack-vm-quantum/23187 : SUCCESS in 21m 49s\n- http://logs.openstack.org/29599/3/gate/gate-grenade-devstack-vm/4205 : FAILURE in 10m 17s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"079ce724cd46ccfa14b7e5c1ba907e3b917f66dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-17 23:03:44.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"642526013c855e88873ecb96e8680698e095d83a","revisions":{"28f96cfa25469c79911a0d48defa550258011bac":{"kind":"REWORK","_number":1,"created":"2013-05-17 20:10:15.000000000","uploader":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"ref":"refs/changes/99/29599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/99/29599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/29599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/29599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/29599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/99/29599/1"}}},"commit":{"parents":[{"commit":"1d6de0f42cb36861338209a4c9ead1dc677d4f2f","subject":"Merge \"solidfire: Make sure src_uuid is passed correctly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d6de0f42cb36861338209a4c9ead1dc677d4f2f"}]}],"author":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-17 20:03:38.000000000","tz":-360},"committer":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-17 20:07:38.000000000","tz":-360},"subject":"Update/Publish volume service updates on delete.","message":"Update/Publish volume service updates on delete.\n\nThe Filter scheduler weighing function was failing because\nwe were only updating the capacity info either on the periodic\n60 second interval or after succesful creates.\n\nThe result was that if you exceeded your capacity within the\nupdate interval you would no longer be able to create volumes even\nif you deleted existing volumes until the update was performed and\npublished.\n\nSince we\u0027re updating on on create, there\u0027s no reason why we shouldn\u0027t\ndo the same on delete for the applicable volume-service.\n\nThe periodic updates are still important for other purposes so they\u0027re left\nalone, but with this change we keep things up to date so the filter scheduler\nhas the correct picture of the world.\n\nFixes bug: 1180976\n\nChange-Id: Ic551f512c5a0cf51bafa8b60f0b6b7debbe22692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28f96cfa25469c79911a0d48defa550258011bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28f96cfa25469c79911a0d48defa550258011bac"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d6de0f42cb36861338209a4c9ead1dc677d4f2f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9cbe7c55e9b39c0727eb18be9f6dcd5e26f537d0":{"kind":"REWORK","_number":2,"created":"2013-05-17 20:13:38.000000000","uploader":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"ref":"refs/changes/99/29599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/99/29599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/29599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/29599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/29599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/99/29599/2"}}},"commit":{"parents":[{"commit":"1d6de0f42cb36861338209a4c9ead1dc677d4f2f","subject":"Merge \"solidfire: Make sure src_uuid is passed correctly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d6de0f42cb36861338209a4c9ead1dc677d4f2f"}]}],"author":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-17 20:03:38.000000000","tz":-360},"committer":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-17 20:10:53.000000000","tz":-360},"subject":"Update/Publish volume service updates on delete.","message":"Update/Publish volume service updates on delete.\n\nThe Filter scheduler weighing function was failing because\nwe were only updating the capacity info either on the periodic\n60 second interval or after succesful creates.\n\nThe result was that if you exceeded your capacity within the\nupdate interval you would no longer be able to create volumes even\nif you deleted existing volumes until the update was performed and\npublished.\n\nSince we\u0027re updating on on create, there\u0027s no reason why we shouldn\u0027t\ndo the same on delete for the applicable volume-service.\n\nThe periodic updates are still important for other purposes so they\u0027re left\nalone, but with this change we keep things up to date so the filter scheduler\nhas the correct picture of the world.\n\nFixes bug: 1180976\n\nChange-Id: Ic551f512c5a0cf51bafa8b60f0b6b7debbe22692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9cbe7c55e9b39c0727eb18be9f6dcd5e26f537d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9cbe7c55e9b39c0727eb18be9f6dcd5e26f537d0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d6de0f42cb36861338209a4c9ead1dc677d4f2f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"642526013c855e88873ecb96e8680698e095d83a":{"kind":"REWORK","_number":3,"created":"2013-05-17 20:38:52.000000000","uploader":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"ref":"refs/changes/99/29599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/99/29599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/29599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/29599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/29599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/99/29599/3"}}},"commit":{"parents":[{"commit":"1d6de0f42cb36861338209a4c9ead1dc677d4f2f","subject":"Merge \"solidfire: Make sure src_uuid is passed correctly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d6de0f42cb36861338209a4c9ead1dc677d4f2f"}]}],"author":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-17 20:03:38.000000000","tz":-360},"committer":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-05-17 20:35:48.000000000","tz":-360},"subject":"Update/Publish volume service updates on delete.","message":"Update/Publish volume service updates on delete.\n\nThe Filter scheduler weighing function was failing because\nwe were only updating the capacity info either on the periodic\n60 second interval or after succesful creates.\n\nThe result was that if you exceeded your capacity within the\nupdate interval you would no longer be able to create volumes even\nif you deleted existing volumes until the update was performed and\npublished.\n\nSince we\u0027re updating on on create, there\u0027s no reason why we shouldn\u0027t\ndo the same on delete for the applicable volume-service.\n\nThe periodic updates are still important for other purposes so they\u0027re left\nalone, but with this change we keep things up to date so the filter scheduler\nhas the correct picture of the world.\n\nFixes bug: 1180976\n\nChange-Id: Ic551f512c5a0cf51bafa8b60f0b6b7debbe22692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/642526013c855e88873ecb96e8680698e095d83a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/642526013c855e88873ecb96e8680698e095d83a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d6de0f42cb36861338209a4c9ead1dc677d4f2f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
