)]}'
{"id":"openstack%2Fcinder~318737","triplet_id":"openstack%2Fcinder~master~I2a93852bbd5fc528f6f73d99212ac6040eaa004a","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"I2a93852bbd5fc528f6f73d99212ac6040eaa004a","subject":"IBM Flashsystem: Add missing items to common","status":"MERGED","created":"2016-05-19 15:17:43.000000000","updated":"2016-08-12 14:27:07.000000000","submitted":"2016-08-09 23:14:37.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2519d37d4d32489e944cc5ec7cac30373aaca043","_number":318737,"virtual_id_number":318737,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2016-08-09 23:23:18.000000000","post_submit":true,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":2,"date":"2016-08-09 23:14:37.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"date":"2016-08-10 18:07:48.000000000","post_submit":true,"_account_id":12778,"name":"HDS HNAS CI","email":"OpenStackDevelopment@hds.com","username":"hds-hnas-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2016-08-10 21:33:14.000000000","post_submit":true,"_account_id":18444,"name":"FalconStor CI","email":"openstack-ci@falconstor.com","username":"falconstor-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11215,"name":"Vivek Dhayaal","email":"vivekdhayaal@gmail.com","username":"vivekdhayaal"},{"value":1,"date":"2016-05-20 10:07:34.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2016-08-12 14:27:07.000000000","post_submit":true,"_account_id":16880,"name":"XP Cinder CI","email":"xp.cinder.ci@groups.ext.hpe.com","username":"xp-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2016-08-11 11:17:09.000000000","post_submit":true,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21144,"name":"Xinli Guan","email":"xinli@us.ibm.com","username":"xinli"},{"value":0,"_account_id":14907,"name":"Alyson","email":"alyson.rodrigues.rosa@gmail.com","username":"Alyson"},{"value":0,"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2016-07-21 22:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-07-28 19:23:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2016-07-26 16:03:11.000000000","_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":12778,"name":"HDS HNAS CI","email":"OpenStackDevelopment@hds.com","username":"hds-hnas-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18444,"name":"FalconStor CI","email":"openstack-ci@falconstor.com","username":"falconstor-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-08-09 17:01:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11215,"name":"Vivek Dhayaal","email":"vivekdhayaal@gmail.com","username":"vivekdhayaal"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":16880,"name":"XP Cinder CI","email":"xp.cinder.ci@groups.ext.hpe.com","username":"xp-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-05-23 16:22:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21144,"name":"Xinli Guan","email":"xinli@us.ibm.com","username":"xinli"},{"value":1,"date":"2016-07-22 11:18:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14907,"name":"Alyson","email":"alyson.rodrigues.rosa@gmail.com","username":"Alyson"},{"value":2,"date":"2016-07-29 05:32:25.000000000","_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2016-08-08 17:20:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":12778,"name":"HDS HNAS CI","email":"OpenStackDevelopment@hds.com","username":"hds-hnas-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18444,"name":"FalconStor CI","email":"openstack-ci@falconstor.com","username":"falconstor-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11215,"name":"Vivek Dhayaal","email":"vivekdhayaal@gmail.com","username":"vivekdhayaal"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":16880,"name":"XP Cinder CI","email":"xp.cinder.ci@groups.ext.hpe.com","username":"xp-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21144,"name":"Xinli Guan","email":"xinli@us.ibm.com","username":"xinli"},{"value":0,"_account_id":14907,"name":"Alyson","email":"alyson.rodrigues.rosa@gmail.com","username":"Alyson"},{"value":0,"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":12778,"name":"HDS HNAS CI","email":"OpenStackDevelopment@hds.com","username":"hds-hnas-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18444,"name":"FalconStor CI","email":"openstack-ci@falconstor.com","username":"falconstor-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11215,"name":"Vivek Dhayaal","email":"vivekdhayaal@gmail.com","username":"vivekdhayaal"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":16880,"name":"XP Cinder CI","email":"xp.cinder.ci@groups.ext.hpe.com","username":"xp-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21144,"name":"Xinli Guan","email":"xinli@us.ibm.com","username":"xinli"},{"value":0,"_account_id":14907,"name":"Alyson","email":"alyson.rodrigues.rosa@gmail.com","username":"Alyson"},{"value":0,"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11215,"name":"Vivek Dhayaal","email":"vivekdhayaal@gmail.com","username":"vivekdhayaal"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12778,"name":"HDS HNAS CI","email":"OpenStackDevelopment@hds.com","username":"hds-hnas-ci","tags":["SERVICE_USER"]},{"_account_id":14907,"name":"Alyson","email":"alyson.rodrigues.rosa@gmail.com","username":"Alyson"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"_account_id":16880,"name":"XP Cinder CI","email":"xp.cinder.ci@groups.ext.hpe.com","username":"xp-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":18444,"name":"FalconStor CI","email":"openstack-ci@falconstor.com","username":"falconstor-ci","tags":["SERVICE_USER"]},{"_account_id":21144,"name":"Xinli Guan","email":"xinli@us.ibm.com","username":"xinli"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-05-20 10:07:34.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-23 16:22:46.000000000","updated_by":{"_account_id":21144,"name":"Xinli Guan","email":"xinli@us.ibm.com","username":"xinli"},"reviewer":{"_account_id":21144,"name":"Xinli Guan","email":"xinli@us.ibm.com","username":"xinli"},"state":"REVIEWER"},{"updated":"2016-07-22 11:18:39.000000000","updated_by":{"_account_id":14907,"name":"Alyson","email":"alyson.rodrigues.rosa@gmail.com","username":"Alyson"},"reviewer":{"_account_id":14907,"name":"Alyson","email":"alyson.rodrigues.rosa@gmail.com","username":"Alyson"},"state":"REVIEWER"},{"updated":"2016-07-26 16:03:11.000000000","updated_by":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"reviewer":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"state":"REVIEWER"},{"updated":"2016-07-28 19:23:00.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2016-07-29 05:32:25.000000000","updated_by":{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},"reviewer":{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},"state":"REVIEWER"},{"updated":"2016-08-08 17:20:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2016-08-09 17:01:52.000000000","updated_by":{"_account_id":11215,"name":"Vivek Dhayaal","email":"vivekdhayaal@gmail.com","username":"vivekdhayaal"},"reviewer":{"_account_id":11215,"name":"Vivek Dhayaal","email":"vivekdhayaal@gmail.com","username":"vivekdhayaal"},"state":"REVIEWER"},{"updated":"2016-08-09 23:14:37.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-08-09 23:23:18.000000000","updated_by":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-08-10 18:07:48.000000000","updated_by":{"_account_id":12778,"name":"HDS HNAS CI","email":"OpenStackDevelopment@hds.com","username":"hds-hnas-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12778,"name":"HDS HNAS CI","email":"OpenStackDevelopment@hds.com","username":"hds-hnas-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-08-10 21:33:14.000000000","updated_by":{"_account_id":18444,"name":"FalconStor CI","email":"openstack-ci@falconstor.com","username":"falconstor-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18444,"name":"FalconStor CI","email":"openstack-ci@falconstor.com","username":"falconstor-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-08-11 11:17:09.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-08-12 14:27:07.000000000","updated_by":{"_account_id":16880,"name":"XP Cinder CI","email":"xp.cinder.ci@groups.ext.hpe.com","username":"xp-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16880,"name":"XP Cinder CI","email":"xp.cinder.ci@groups.ext.hpe.com","username":"xp-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a71dce45d30ce2b1f70242bac088d64db478e3a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2016-05-19 15:17:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c1e38197b4b7bc8d35384599c99cc56c556223c","author":{"_account_id":16160,"name":"Blockbridge OpenStack CI","email":"openstack-ci@blockbridge.com","username":"blockbridge","tags":["SERVICE_USER"]},"date":"2016-05-19 15:57:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-tempest-dsvm-volume-blockbridge http://logs.openstack.blockbridge.com/37/318737/1/check/check-tempest-dsvm-volume-blockbridge/124ffb5 : SUCCESS in 39m 54s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Blockbridge_EPS_CI","accounts_in_message":[],"_revision_number":1},{"id":"054c7b9e1ef30552e8ffc11601b5534b25feb69a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-19 17:03:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-cinder-docs http://docs-draft.openstack.org/37/318737/1/check/gate-cinder-docs/8fc3cc8//doc/build/html/ : SUCCESS in 11m 06s\n- gate-cinder-pep8 http://logs.openstack.org/37/318737/1/check/gate-cinder-pep8/99f4f34/ : SUCCESS in 5m 36s\n- gate-cinder-python27-db http://logs.openstack.org/37/318737/1/check/gate-cinder-python27-db/fe06b7b/ : SUCCESS in 6m 14s\n- gate-cinder-python34-db http://logs.openstack.org/37/318737/1/check/gate-cinder-python34-db/f721d49/ : SUCCESS in 7m 32s\n- gate-tempest-dsvm-full http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full/b0727ce/ : SUCCESS in 50m 14s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-postgres-full/88e69c1/ : SUCCESS in 41m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-neutron-full/0b234cb/ : SUCCESS in 1h 04m 18s\n- gate-grenade-dsvm http://logs.openstack.org/37/318737/1/check/gate-grenade-dsvm/bd6cb4b/ : SUCCESS in 32m 34s\n- gate-cinder-pylint http://logs.openstack.org/37/318737/1/check/gate-cinder-pylint/e256bc1/ : SUCCESS in 11m 51s (non-voting)\n- gate-rally-dsvm-cinder http://logs.openstack.org/37/318737/1/check/gate-rally-dsvm-cinder/ab3befc/ : SUCCESS in 44m 25s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/2b2480b/ : SUCCESS in 46m 24s\n- gate-cinder-dsvm-apache http://logs.openstack.org/37/318737/1/check/gate-cinder-dsvm-apache/523c9ce/ : SUCCESS in 46m 03s (non-voting)\n- gate-cinder-tox-db-functional http://logs.openstack.org/37/318737/1/check/gate-cinder-tox-db-functional/98106ab/ : SUCCESS in 4m 03s (non-voting)\n- gate-cinder-tempest-dsvm-glusterfs-nv http://logs.openstack.org/37/318737/1/check/gate-cinder-tempest-dsvm-glusterfs-nv/cae30c0/ : FAILURE in 1h 06m 43s (non-voting)\n- gate-tempest-dsvm-full-sheepdog-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-sheepdog-nv/750b7b0/ : SUCCESS in 47m 31s (non-voting)\n- gate-tempest-dsvm-full-drbd-devstack-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-drbd-devstack-nv/0b9ff8b/ : SUCCESS in 37m 20s (non-voting)\n- gate-tempest-dsvm-full-lio http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-lio/dc3df9f/ : SUCCESS in 42m 03s (non-voting)\n- gate-tempest-dsvm-full-bdd-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-bdd-nv/58b07ae/ : FAILURE in 1h 10m 16s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/017722c/ : SUCCESS in 1h 00m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25a4aae30a1988222f325b10b9288bc45eadf6a2","author":{"_account_id":13394,"name":"Microsoft iSCSI CI","username":"microsoft-iscsi-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-05-19 17:16:26.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Cinder CI, add a review comment with check microsoft-iscsi-ci\n\n- microsoft-ci-iscsi http://64.119.130.115/cinder/318737/1/iscsi/ : SUCCESS in 1h 15m 33s\n- microsoft-ci-smb3-linux http://64.119.130.115/cinder/318737/1/smb3_linux/ : SUCCESS in 31m 03s\n- microsoft-ci-smb3-windows http://64.119.130.115/cinder/318737/1/smb3_windows : FAILURE in 1h 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a0261655fd016cee52e896a8b387d3e5837b8fc9","author":{"_account_id":21193,"name":"IBM GPFS CINDER CI","email":"ibmgpfci@in.ibm.com","username":"ibmgpfscinderci"},"date":"2016-05-19 17:22:06.000000000","message":"Patch Set 1:\n\nBuild failure. For rechecking only on IBM GPFS CI, add a review comment with run ibmgpfs-ci.\n\n- dsvm-tempest-full http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-GPFS-CI/37/318737/1/check/dsvm-tempest-full/ac9e381 : FAILURE in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"dc77da471bc603b972332945128a3a4970aed100","author":{"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]},"date":"2016-05-19 17:35:53.000000000","message":"Patch Set 1:\n\n* SolidFire-dsvm-full http://54.164.167.86/solidfire-ci-logs/refs-changes-37-318737-1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8a2101fba804c9525704bad79948331c0812447d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-05-19 17:42:38.000000000","message":"Patch Set 1:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-37-318737-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"488eccfccaa8622cf0f4359484aa45396832453c","author":{"_account_id":10621,"name":"NetApp CI","email":"xdl-openstack-jenkins@netapp.com","username":"netapp-ci","inactive":true},"date":"2016-05-19 18:19:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-7mode-iSCSI http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-7mode-iSCSI/b8cb738/ : SUCCESS in 38m 46s\n- cinder-7mode-FCP http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-7mode-FCP/6c74f01/ : SUCCESS in 1h 15m 36s\n- cinder-7mode-NFS http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-7mode-NFS/f846dcc/ : FAILURE in 37m 11s\n- cinder-cDOT-iSCSI http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-cDOT-iSCSI/88f393e/ : SUCCESS in 38m 27s\n- cinder-cDOT-FCP http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-cDOT-FCP/5e8e0ea/ : SUCCESS in 49m 28s\n- cinder-cDOT-NFS http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-cDOT-NFS/5ba2c7a/ : SUCCESS in 41m 52s\n- cinder-eseries-iSCSI http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-eseries-iSCSI/171ce45/ : SUCCESS in 52m 36s\n- cinder-eseries-FCP http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-eseries-FCP/3b25ed5/ : SUCCESS in 1h 06m 50s\n\nComment \u0027recheck-netapp-ci\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"599f5f6bc198f30bc6d08e262a68e0c6091e05b1","author":{"_account_id":18402,"name":"HPE Storage CI","email":"hp.cinder.ci@groups.ext.hpe.com","username":"hp-storage-ci","tags":["SERVICE_USER"]},"date":"2016-05-19 18:42:26.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-eos10-dsvm http://54.201.44.218/37/318737/1/check/3par-iscsi-driver-master-client-pip-eos10-dsvm/866a9bb : SUCCESS in 1h 39m 06s\n- 3par-iscsi-driver-master-client-pip-c8k02-dsvm http://54.201.44.218/37/318737/1/check/3par-iscsi-driver-master-client-pip-c8k02-dsvm/24e1e95 : FAILURE in 8m 34s\n- 3par-fc-driver-master-client-pip-c8k02-dsvm http://54.201.44.218/37/318737/1/check/3par-fc-driver-master-client-pip-c8k02-dsvm/2e91b96 : SUCCESS in 1h 33m 09s\n- 3par-fc-driver-master-client-pip-eos10-dsvm http://54.201.44.218/37/318737/1/check/3par-fc-driver-master-client-pip-eos10-dsvm/6013e25 : SUCCESS in 1h 37m 58s\n- lefthand-iscsi-driver-master-client-pip-vsa673-dsvm http://54.201.44.218/37/318737/1/check/lefthand-iscsi-driver-master-client-pip-vsa673-dsvm/a0c9b3c : SUCCESS in 1h 36m 07s\n- lefthand-iscsi-driver-master-client-pip-vsa725http-dsvm http://54.201.44.218/37/318737/1/check/lefthand-iscsi-driver-master-client-pip-vsa725http-dsvm/8959662 : SUCCESS in 1h 25m 08s","accounts_in_message":[],"_revision_number":1},{"id":"bfdd1f677c3080f5fdfe391a51a1c17c16e435a9","author":{"_account_id":16880,"name":"XP Cinder CI","email":"xp.cinder.ci@groups.ext.hpe.com","username":"xp-cinder-ci","tags":["SERVICE_USER"]},"date":"2016-05-19 19:12:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- xp-fc-master-xp7-dsvm http://54.201.44.218/37/318737/1/check/xp-fc-master-xp7-dsvm/172d2b2 : SUCCESS in 1h 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"92f23f28400e9c9266449b41b22178c0b7afc10a","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-19 20:15:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/37/318737/1/check/check-dsvm-tempest-vz7-exe-minimal/180bb08 : SUCCESS in 1h 00m 14s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"73cd0bbe11899b5d0154bc9b8a4fad7c5ccd2c20","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2016-05-19 22:19:23.000000000","message":"Patch Set 1:\n\n* datera-dsvm-full http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/refs-changes-37-318737-1/index.html : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8c028730492fa34e62fc4466b2019a2ad7c36df1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2016-05-20 00:05:24.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- EMC_ScaleIO http://publiclogs.emc.com/vipr_ostack/scaleio/EMC_ScaleIO/1018 : SUCCESS in 53m 19s\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":1},{"id":"90e877306601be8304950df65440ddc10c321281","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2016-05-20 02:42:12.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/xio_ostack/EMC_XtremIO_FC/740 : SUCCESS in 56m 13s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/xio_ostack/EMC_XtremIO_iSCSI/782 : SUCCESS in 42m 31s\n\nLeave a comment with \u0027run emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":1},{"id":"856c1860d71f877a2c7e426d7821510750c6a58b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2016-05-20 03:10:34.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/37/318737/1/check/fujitsu-eternus-dx-iscsi/6cc78d1 : SUCCESS in 1h 27m 29s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/37/318737/1/check/fujitsu-eternus-dx-fc/11456a1 : SUCCESS in 3h 38m 19s","accounts_in_message":[],"_revision_number":1},{"id":"cf7ad444c9affa1f5e849bb3f8bdfd161f845314","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2016-05-20 04:17:35.000000000","message":"Patch Set 1:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-37-318737-1/ns5_iscsi : SUCCESS in 41m 14s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-37-318737-1/ns_iscsi : SUCCESS in 43m 46s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-37-318737-1/ns5_nfs : SUCCESS in 44m 19s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-37-318737-1/ns_nfs : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b968c5e7b7dbd8328ddf29ee94e46db4d350f85e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2016-05-20 10:00:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline). Use run-zadara to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/37/318737/1/check/dsvm-tempest-zadara-driver/00561bd : FAILURE in 1h 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82479ca75a85cb66ae09695228cad67fbe1f72ed","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-05-20 10:07:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. \n\n* cinder-dsvm-tempest http://208.91.1.172/logs/cinder-dsvm-tempest/318737/1/12252 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d6df8bb610c84595e0218f3dc5ecfb24d62a0d74","author":{"_account_id":15249,"name":"Dell Storage CI","email":"openstack-cinder-sc-ci@dell.com","username":"dell-storage-ci","tags":["SERVICE_USER"]},"date":"2016-05-20 15:09:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* dell-eql-dsvm-volume http://oslogs.compellent.com/dell-eql-318737-1/ : SUCCESS \n* dell-sc-iscsi-dsvm-volume http://oslogs.compellent.com/dell-sc-iscsi-318737-1/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ee2c2698b7fee2ece2d2bb7277bb37a847276b5f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2016-05-20 16:02:21.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- EMC_ScaleIO http://publiclogs.emc.com/vipr_ostack/scaleio/EMC_ScaleIO/1053 : SUCCESS in 53m 55s\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":1},{"id":"7d54e30306885e3c23c8243b822b4b41040f34f1","author":{"_account_id":18752,"name":"Tegile Storage CI","email":"openstack.ci@tegile.com","username":"tegile","tags":["SERVICE_USER"]},"date":"2016-05-20 16:04:18.000000000","message":"Patch Set 1:\n\nBuild failed. For rechecking only on Tegile Storage CI, add a review comment with run tegile\n\n- TegileISCSIDriver-dsvm-tempest http://openstack-logs.tegile.com/37/318737/1/check/TegileISCSIDriver-dsvm-tempest/d9eaa74 : FAILURE in 58m 03s\n- TegileFCDriver-dsvm-tempest http://openstack-logs.tegile.com/37/318737/1/check/TegileFCDriver-dsvm-tempest/cf28f38 : SUCCESS in 43m 55s\n\nFor Tegile Storage CI info, see https://wiki.openstack.org/wiki/ThirdPartySystems/Tegile_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"7958229244801f960b246dfa6025c9193a608b8a","author":{"_account_id":17852,"name":"IBM XIV CI","email":"ibmxivci@il.ibm.com","username":"ibmxivci","tags":["SERVICE_USER"]},"date":"2016-05-20 18:12:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- ibm-xiv-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-XIV-CI/37/318737/1/check/ibm-xiv-iscsi/03bcae1 : SUCCESS in 53m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5fa026d766552732e497c6a9cd287d517e352b96","author":{"_account_id":19146,"name":"Coho Data Storage CI","email":"openstack-ci@cohodata.com","username":"coho-openstack-ci","tags":["SERVICE_USER"]},"date":"2016-05-20 19:04:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-coho-cinder-driver http://logs.cohodata.com/37/318737/1/check/dsvm-tempest-coho-cinder-driver/4488abe : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":1},{"id":"dfc78aebeb8b2362602fce4b51e635d888a8be48","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2016-05-20 23:49:17.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/xio_ostack/EMC_XtremIO_FC/765 : SUCCESS in 49m 59s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/xio_ostack/EMC_XtremIO_iSCSI/808 : SUCCESS in 41m 51s\n\nLeave a comment with \u0027run emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":1},{"id":"56ad37ecf63caf028a2ca613863964c1e084d750","author":{"_account_id":16941,"name":"Vedams-HPMSA FCISCSIDriver CI","email":"vedams_hp_ci@vedams.com","username":"vedams_hp_ci","tags":["SERVICE_USER"]},"date":"2016-05-21 03:44:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- vedams-hpmsa-dsvm-tempest-full-iscsi http://tristero.cloudapp.net/37/318737/1/check/vedams-hpmsa-dsvm-tempest-full-iscsi/1ccee3c : SUCCESS in 1h 55m 47s\n- vedams-hpmsa-dsvm-tempest-full-fc http://tristero.cloudapp.net/37/318737/1/check/vedams-hpmsa-dsvm-tempest-full-fc/7a30078 : SUCCESS in 1h 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ab7d5a3e9dd9ae1712ced0bbe273e9e48e708061","author":{"_account_id":12540,"name":"Eddie Lin","email":"edlin@us.ibm.com","username":"ed3ed3","tags":["SERVICE_USER"]},"date":"2016-05-21 06:29:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- ibm-ds8k-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-DS8000-CI/37/318737/1/check/ibm-ds8k-fc/1dc25ef : SUCCESS in 2h 02m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f53715a230db2895635917fcb8490b2148a940e2","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2016-05-21 07:50:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/37/318737/1/check/infortrend-cinder-volume-iscsi/57c9d28 : SUCCESS in 2h 26m 15s\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/37/318737/1/check/infortrend-cinder-volume-fc/a3d4474 : SUCCESS in 2h 27m 11s","accounts_in_message":[],"_revision_number":1},{"id":"971d5e1e71e3a358ba6de0aad4e807f183732a34","author":{"_account_id":14259,"name":"CloudByte CI","email":"openstack-dev@cloudbyte.com","username":"CloudbyteCI","tags":["SERVICE_USER"]},"date":"2016-05-21 08:32:55.000000000","message":"Patch Set 1:\n\nBuild Succeded\n\n- dsvm-tempest-full http://117.218.56.3:9090/logs/37/318737/1/check/dsvm-tempest-full/6210e03 : SUCCESS in 1h 28m 36s","accounts_in_message":[],"_revision_number":1},{"id":"392a21908420baaa8aa67c4eb71a40c0fcaf8b02","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2016-05-21 08:40:20.000000000","message":"Patch Set 1:\n\n* nexenta-dsvm-volume-ns_edge http://140.174.232.106/refs-changes-37-318737-1/ns_edge : SUCCESS in 41m 43s\n* nexenta-dsvm-volume-cinder_nedge11_nbd http://140.174.232.106/refs-changes-37-318737-1/cinder_nedge11_nbd : FAILURE in 8m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b7035ddb83016f549e5ba7ea8618048f64d5a507","author":{"_account_id":12778,"name":"HDS HNAS CI","email":"OpenStackDevelopment@hds.com","username":"hds-hnas-ci","tags":["SERVICE_USER"]},"date":"2016-05-21 09:35:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi-hnas-iscsi http://openstack.fit-tecnologia.org.br:10000/37/318737/1/check/hitachi-hnas-iscsi/a2cc40b : SUCCESS in 1h 26m 12s\n- hitachi-hnas-nfs http://openstack.fit-tecnologia.org.br:10000/37/318737/1/check/hitachi-hnas-nfs/7fbb23d : SUCCESS in 1h 25m 20s\n\nLeave a comment with \u0027run hnas-ci\u0027 to trigger a recheck. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_HNAS_CI","accounts_in_message":[],"_revision_number":1},{"id":"250fc80ea0bded88acdf8b44e2a842f2ff7bd0a3","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2016-05-21 14:52:16.000000000","message":"Patch Set 1:\n\n* oracle-zfssa-dsvm-nfs https://openstackci.oracle.com/zfssa_ci/refs-changes-37-318737-1-nfs : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"766d76757e6c80aeaa00ba063ea3f45cc4ab73ae","author":{"_account_id":16660,"name":"Hitachi HBSD2 CI","email":"OpenStackCI2@hds.com","username":"hitachi-hbsd2-ci","tags":["SERVICE_USER"]},"date":"2016-05-21 18:23:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi-hbsd2-fc http://openstack.fit-tecnologia.org.br:10000/37/318737/1/check/hitachi-hbsd2-fc/22394c5 : SUCCESS in 1h 27m 17s\n\nLeave a comment with \u0027run hbsd2-ci\u0027 to trigger a recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_HBSD2_CI","accounts_in_message":[],"_revision_number":1},{"id":"e55910c635f2a596fdafa60483443a6b9fb80f79","author":{"_account_id":16269,"name":"Hitachi HBSD CI","email":"OpenStackCI@hds.com","username":"hitachi-hbsd-ci","tags":["SERVICE_USER"]},"date":"2016-05-22 00:41:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi-hbsd-iscsi http://openstack.fit-tecnologia.org.br:10000/37/318737/1/check/hitachi-hbsd-iscsi/53c6677 : SUCCESS in 1h 40m 13s\n- hitachi-hbsd-fc http://openstack.fit-tecnologia.org.br:10000/37/318737/1/check/hitachi-hbsd-fc/5aa1db6 : SUCCESS in 1h 29m 35s\n\nLeave a comment with \u0027run hbsd-ci\u0027 to trigger a recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_HBSD_CI","accounts_in_message":[],"_revision_number":1},{"id":"5e5a32d344b5153c43a2c31e5315f97f23c760ac","author":{"_account_id":16862,"name":"IBM Storwize CI","email":"ibmswci@cn.ibm.com","username":"ibm-storewize-ci","tags":["SERVICE_USER"]},"date":"2016-05-23 07:42:28.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027myvendor: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- storwize-cinder-ci-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-Storwize-CI/37/318737/1/check/storwize-cinder-ci-fc/24a3190 : SUCCESS in 1h 16m 38s\n- storwize-cinder-ci-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-Storwize-CI/37/318737/1/check/storwize-cinder-ci-iscsi/d96082a : FAILURE in 41m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0cfd577419082d0e58f64aec4c3920d4aa3c05bb","author":{"_account_id":21144,"name":"Xinli Guan","email":"xinli@us.ibm.com","username":"xinli"},"date":"2016-05-23 16:22:46.000000000","message":"Patch Set 1: Code-Review+1\n\nThe code looks good to me.","accounts_in_message":[],"_revision_number":1},{"id":"c2f1d472366116191db681820fac6138af9e0d8d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2016-07-21 22:00:32.000000000","message":"Patch Set 1: Code-Review+2\n\nrun-IBM FlashSystem\n\nLooks good, just need the IBM CI to report.","accounts_in_message":[],"_revision_number":1},{"id":"7eb3ec661de155e8f1f64f4881b213ee01b73b4b","author":{"_account_id":14907,"name":"Alyson","email":"alyson.rodrigues.rosa@gmail.com","username":"Alyson"},"date":"2016-07-22 11:18:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9e3d04b9b45685ef61dda650299c8c72c05b48e7","author":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"date":"2016-07-26 16:03:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d08fce66d1c9cc4dc098d3162ab425ef0e363510","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2016-07-28 19:23:00.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good to me. Thanks for cleaning that up Eric.","accounts_in_message":[],"_revision_number":1},{"id":"70f6df468847eb649b2b783cdfa747c15e552b6c","author":{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},"date":"2016-07-29 05:32:25.000000000","message":"Patch Set 1: Code-Review+2\n\nWaiting for IBM FlashSystem CI.","accounts_in_message":[],"_revision_number":1},{"id":"6f3ef8f3fc06ddd3d97873d0cf6e2433190632a7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2016-08-08 14:00:38.000000000","message":"Patch Set 1:\n\nrecheck flashci","accounts_in_message":[],"_revision_number":1},{"id":"24bb8e50ddde9b3d9ff134cb0797ef580e7b6008","author":{"_account_id":13394,"name":"Microsoft iSCSI CI","username":"microsoft-iscsi-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-08 15:25:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- microsoft-ci-iscsi http://64.119.130.115/cinder/318737/1/iscsi/ : SUCCESS in 1h 07m 41s\n- microsoft-ci-smb3-linux http://64.119.130.115/cinder/318737/1/smb3_linux/ : SUCCESS in 46m 45s\n- microsoft-ci-smb3-windows http://64.119.130.115/cinder/318737/1/smb3_windows : SUCCESS in 1h 24m 28s","accounts_in_message":[],"_revision_number":1},{"id":"577e55c931274585bc59c284306a38a6423d210f","author":{"_account_id":15374,"name":"IBM FlashSystem CI","email":"flashci@cn.ibm.com","username":"flashci","tags":["SERVICE_USER"]},"date":"2016-08-08 16:27:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- ibm-flashsystem-cinder-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-FlashSystem-CI/37/318737/1/check/ibm-flashsystem-cinder-fc/97c9030 : SUCCESS\n- ibm-flashsystem-cinder-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-FlashSystem-CI/37/318737/1/check/ibm-flashsystem-cinder-iscsi/16a42d2 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ed2d02ad4a957a0aae3d9eab5a3b25977223903c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2016-08-08 17:20:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1edeb5bab54b5584c9180a2ca2d193f9a8bf26c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-08 17:20:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"23c5fecff052ed60eddfdc1bb92684d854897e92","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-08 18:18:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-cinder-docs-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-cinder-docs-ubuntu-xenial/c04d055/ : POST_FAILURE in 4m 31s\n- gate-cinder-pep8-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-cinder-pep8-ubuntu-xenial/d3164bd/ : SUCCESS in 3m 16s\n- gate-cinder-python27-db-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-cinder-python27-db-ubuntu-xenial/0678f57/ : SUCCESS in 15m 38s\n- gate-cinder-python34-db http://logs.openstack.org/37/318737/1/check/gate-cinder-python34-db/3d375ea/ : SUCCESS in 5m 49s\n- gate-cinder-python35-db-nv http://logs.openstack.org/37/318737/1/check/gate-cinder-python35-db-nv/9609037/ : SUCCESS in 9m 30s (non-voting)\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-ubuntu-xenial/7b1b1c9/ : SUCCESS in 53m 45s\n- gate-tempest-dsvm-postgres-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-postgres-full-ubuntu-xenial/c23714b/ : SUCCESS in 50m 44s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/065231e/ : SUCCESS in 1h 20m 43s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/37/318737/1/check/gate-grenade-dsvm-ubuntu-trusty/a3441ae/ : SUCCESS in 55m 46s\n- gate-cinder-pylint http://logs.openstack.org/37/318737/1/check/gate-cinder-pylint/a5dd12d/ : SUCCESS in 11m 12s (non-voting)\n- gate-rally-dsvm-cinder http://logs.openstack.org/37/318737/1/check/gate-rally-dsvm-cinder/c65dab9/ : FAILURE in 47m 33s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/9244a23/ : SUCCESS in 1h 04m 20s\n- gate-cinder-dsvm-apache http://logs.openstack.org/37/318737/1/check/gate-cinder-dsvm-apache/8812df2/ : SUCCESS in 47m 23s (non-voting)\n- gate-cinder-tox-db-functional http://logs.openstack.org/37/318737/1/check/gate-cinder-tox-db-functional/daea847/ : SUCCESS in 5m 53s (non-voting)\n- gate-cinder-tox-compliance-ubuntu-trusty http://logs.openstack.org/37/318737/1/check/gate-cinder-tox-compliance-ubuntu-trusty/8234b97/ : SUCCESS in 3m 46s\n- gate-cinder-tox-compliance-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-cinder-tox-compliance-ubuntu-xenial/d0ae5fc/ : SUCCESS in 2m 45s\n- gate-cinder-tempest-dsvm-glusterfs-nv http://logs.openstack.org/37/318737/1/check/gate-cinder-tempest-dsvm-glusterfs-nv/19622a3/ : POST_FAILURE in 1h 07m 54s (non-voting)\n- gate-tempest-dsvm-full-sheepdog-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-sheepdog-nv/56c111f/ : SUCCESS in 45m 43s (non-voting)\n- gate-tempest-dsvm-full-drbd-devstack-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-drbd-devstack-nv/dfae16d/ : FAILURE in 1h 05m 43s (non-voting)\n- gate-tempest-dsvm-full-lio http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-lio/31c4f42/ : SUCCESS in 59m 28s (non-voting)\n- gate-tempest-dsvm-full-bdd-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-bdd-nv/6504703/ : FAILURE in 1h 46m 23s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/b080eb8/ : FAILURE in 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ac8a608fc4b67b7f082d87c764099c654e2a5c6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2016-08-08 18:44:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9510a40a854939431f60269273f8481060cff1cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-08 19:46:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-cinder-docs-ubuntu-xenial http://docs-draft.openstack.org/37/318737/1/gate/gate-cinder-docs-ubuntu-xenial/8e5ba55//doc/build/html/ : SUCCESS in 6m 44s\n- gate-cinder-pep8-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-cinder-pep8-ubuntu-xenial/1c14754/ : SUCCESS in 8m 48s\n- gate-cinder-python27-db-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-cinder-python27-db-ubuntu-xenial/515a934/ : SUCCESS in 9m 22s\n- gate-cinder-python34-db http://logs.openstack.org/37/318737/1/gate/gate-cinder-python34-db/89d3130/ : SUCCESS in 9m 41s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-full-ubuntu-xenial/84c782a/ : SUCCESS in 48m 42s\n- gate-tempest-dsvm-postgres-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-postgres-full-ubuntu-xenial/93804ac/ : SUCCESS in 56m 43s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/987a697/ : FAILURE in 14m 23s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/37/318737/1/gate/gate-grenade-dsvm-ubuntu-trusty/dfc3682/ : SUCCESS in 1h 24m 37s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-full-devstack-plugin-ceph/5f43958/ : SUCCESS in 1h 02m 20s\n- gate-cinder-tox-compliance-ubuntu-trusty http://logs.openstack.org/37/318737/1/gate/gate-cinder-tox-compliance-ubuntu-trusty/7f5dd9a/ : SUCCESS in 6m 13s\n- gate-cinder-tox-compliance-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-cinder-tox-compliance-ubuntu-xenial/44296a0/ : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7623dcab72f335da3d9c3b4dee34ab84770e3f1b","author":{"_account_id":13394,"name":"Microsoft iSCSI CI","username":"microsoft-iscsi-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-08 21:53:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Cinder CI, add a review comment with check microsoft-iscsi-ci\n\n- microsoft-ci-iscsi http://64.119.130.115/cinder/318737/1/iscsi/ : SUCCESS in 3h 08m 57s\n- microsoft-ci-smb3-linux http://64.119.130.115/cinder/318737/1/smb3_linux/ : SUCCESS in 33m 31s\n- microsoft-ci-smb3-windows http://64.119.130.115/cinder/318737/1/smb3_windows : FAILURE in 1h 17m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6cff76fd70768e70faee9a03e3b728a1f78b956e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-08 23:10:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-cinder-docs-ubuntu-xenial http://docs-draft.openstack.org/37/318737/1/check/gate-cinder-docs-ubuntu-xenial/defff01//doc/build/html/ : SUCCESS in 9m 34s\n- gate-cinder-pep8-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-cinder-pep8-ubuntu-xenial/01e326c/ : SUCCESS in 5m 40s\n- gate-cinder-python27-db-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-cinder-python27-db-ubuntu-xenial/af3dfa1/ : SUCCESS in 7m 17s\n- gate-cinder-python34-db http://logs.openstack.org/37/318737/1/check/gate-cinder-python34-db/d1b201d/ : SUCCESS in 7m 53s\n- gate-cinder-python35-db-nv http://logs.openstack.org/37/318737/1/check/gate-cinder-python35-db-nv/7d40fca/ : SUCCESS in 9m 21s (non-voting)\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-ubuntu-xenial/a12728d/ : SUCCESS in 56m 53s\n- gate-tempest-dsvm-postgres-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-postgres-full-ubuntu-xenial/37ec64f/ : SUCCESS in 1h 12m 52s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6b2dfec/ : SUCCESS in 1h 23m 08s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/37/318737/1/check/gate-grenade-dsvm-ubuntu-trusty/cd37cf1/ : SUCCESS in 59m 15s\n- gate-cinder-pylint http://logs.openstack.org/37/318737/1/check/gate-cinder-pylint/7b7f329/ : SUCCESS in 10m 55s (non-voting)\n- gate-rally-dsvm-cinder http://logs.openstack.org/37/318737/1/check/gate-rally-dsvm-cinder/c02e4f9/ : SUCCESS in 45m 41s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/b9d125b/ : SUCCESS in 1h 08m 07s\n- gate-cinder-dsvm-apache http://logs.openstack.org/37/318737/1/check/gate-cinder-dsvm-apache/e13cf61/ : SUCCESS in 1h 00m 48s (non-voting)\n- gate-cinder-tox-db-functional http://logs.openstack.org/37/318737/1/check/gate-cinder-tox-db-functional/2fcb48a/ : SUCCESS in 3m 58s (non-voting)\n- gate-cinder-tox-compliance-ubuntu-trusty http://logs.openstack.org/37/318737/1/check/gate-cinder-tox-compliance-ubuntu-trusty/465485d/ : SUCCESS in 2m 35s\n- gate-cinder-tox-compliance-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-cinder-tox-compliance-ubuntu-xenial/1823a18/ : SUCCESS in 5m 28s\n- gate-cinder-tempest-dsvm-glusterfs-nv http://logs.openstack.org/37/318737/1/check/gate-cinder-tempest-dsvm-glusterfs-nv/b04f811/ : POST_FAILURE in 1h 11m 15s (non-voting)\n- gate-tempest-dsvm-full-sheepdog-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-sheepdog-nv/17538f9/ : SUCCESS in 52m 30s (non-voting)\n- gate-tempest-dsvm-full-drbd-devstack-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-drbd-devstack-nv/e4245f1/ : FAILURE in 1h 13m 09s (non-voting)\n- gate-tempest-dsvm-full-lio http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-lio/060740c/ : SUCCESS in 1h 00m 56s (non-voting)\n- gate-tempest-dsvm-full-bdd-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-bdd-nv/942f124/ : FAILURE in 1h 47m 35s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/6c205da/ : FAILURE in 32m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3a72a7290c4e4fda514a0e2c6db2c9c1175e78b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-08 23:11:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"f401d5a6c18b9728f7a1f93fed02a10d2ad9c98a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-08-08 23:11:19.000000000","message":"Patch Set 1:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-37-318737-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"f1d7867479d5a084df13e193c7cb934ff2785c9f","author":{"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]},"date":"2016-08-09 00:12:35.000000000","message":"Patch Set 1:\n\n* SolidFire-dsvm-full http://54.164.167.86/solidfire-ci-logs/refs-changes-37-318737-1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a51ebf984c7fd48db80a818c8e7226711477f8a3","author":{"_account_id":18402,"name":"HPE Storage CI","email":"hp.cinder.ci@groups.ext.hpe.com","username":"hp-storage-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 01:01:56.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-c8k02-dsvm http://54.201.44.218/37/318737/1/check/3par-iscsi-driver-master-client-pip-c8k02-dsvm/095de7a : SUCCESS in 1h 33m 33s\n- 3par-fc-driver-master-client-pip-c8k02-dsvm http://54.201.44.218/37/318737/1/check/3par-fc-driver-master-client-pip-c8k02-dsvm/b737865 : FAILURE in 1h 40m 54s\n- lefthand-iscsi-driver-master-client-pip-vsa673-dsvm http://54.201.44.218/37/318737/1/check/lefthand-iscsi-driver-master-client-pip-vsa673-dsvm/0c03d2f : SUCCESS in 1h 41m 18s\n- lefthand-iscsi-driver-master-client-pip-vsa725http-dsvm http://54.201.44.218/37/318737/1/check/lefthand-iscsi-driver-master-client-pip-vsa725http-dsvm/adad2f1 : SUCCESS in 1h 50m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1f024f9a6b1dace3e5ba5c042e32cde2995b07d1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-09 01:28:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-cinder-docs-ubuntu-xenial http://docs-draft.openstack.org/37/318737/1/gate/gate-cinder-docs-ubuntu-xenial/95741da//doc/build/html/ : SUCCESS in 4m 42s\n- gate-cinder-pep8-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-cinder-pep8-ubuntu-xenial/b520833/ : SUCCESS in 2m 56s\n- gate-cinder-python27-db-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-cinder-python27-db-ubuntu-xenial/edf6cba/ : FAILURE in 18m 02s\n- gate-cinder-python34-db http://logs.openstack.org/37/318737/1/gate/gate-cinder-python34-db/dbafcdd/ : SUCCESS in 8m 35s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-full-ubuntu-xenial/777f610/ : SUCCESS in 50m 13s\n- gate-tempest-dsvm-postgres-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-postgres-full-ubuntu-xenial/f8a2e64/ : SUCCESS in 52m 33s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/862657a/ : SUCCESS in 1h 21m 55s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/37/318737/1/gate/gate-grenade-dsvm-ubuntu-trusty/5f5252c/ : SUCCESS in 45m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-full-devstack-plugin-ceph/e96cef8/ : SUCCESS in 57m 54s\n- gate-cinder-tox-compliance-ubuntu-trusty http://logs.openstack.org/37/318737/1/gate/gate-cinder-tox-compliance-ubuntu-trusty/e359c37/ : SUCCESS in 3m 03s\n- gate-cinder-tox-compliance-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-cinder-tox-compliance-ubuntu-xenial/c376674/ : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6992ffcf3f6664a8042f09233ec24d4214986fbc","author":{"_account_id":10621,"name":"NetApp CI","email":"xdl-openstack-jenkins@netapp.com","username":"netapp-ci","inactive":true},"date":"2016-08-09 01:44:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-7mode-iSCSI http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-7mode-iSCSI/99b1189/ : SUCCESS in 54m 10s\n- cinder-7mode-FCP http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-7mode-FCP/3ebbfa6/ : SUCCESS in 1h 03m 16s\n- cinder-7mode-NFS http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-7mode-NFS/78c33f4/ : SUCCESS in 48m 25s\n- cinder-cDOT-iSCSI http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-cDOT-iSCSI/03e9d3a/ : SUCCESS in 1h 00m 22s\n- cinder-cDOT-FCP http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-cDOT-FCP/f752854/ : SUCCESS in 1h 17m 06s\n- cinder-cDOT-NFS http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-cDOT-NFS/a04c4b1/ : SUCCESS in 58m 11s\n- cinder-eseries-iSCSI http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-eseries-iSCSI/c983ecb/ : SUCCESS in 1h 11m 37s\n- cinder-eseries-FCP http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-eseries-FCP/071efb3/ : FAILURE in 1h 02m 03s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"14c70d0c3b3367a844ba471b8713fc981c1721b8","author":{"_account_id":16880,"name":"XP Cinder CI","email":"xp.cinder.ci@groups.ext.hpe.com","username":"xp-cinder-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 02:10:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- xp-fc-master-xp7-dsvm http://54.201.44.218/37/318737/1/check/xp-fc-master-xp7-dsvm/a400ebc : SUCCESS in 1h 55m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b31e1c5d45a54f9a9f85a32db325fcf82cc10b3a","author":{"_account_id":15249,"name":"Dell Storage CI","email":"openstack-cinder-sc-ci@dell.com","username":"dell-storage-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 03:57:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* dell-sc-fc-dsvm-volume http://oslogs.compellent.com/dell-sc-fc-318737-1/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ac48a6d646f2d07d81d8a3ad827721460311172a","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 04:13:50.000000000","message":"Patch Set 1:\n\n* datera-dsvm-full http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/refs-changes-37-318737-1/index.html : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9a78423488a7b042abd51bcb8413f8b30fa1f71d","author":{"_account_id":18444,"name":"FalconStor CI","email":"openstack-ci@falconstor.com","username":"falconstor-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 12:14:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-cinder-fss-fc-custom http://openstack.falconstor.com/37/318737/1/check/dsvm-cinder-fss-fc-custom/bf2b4d7 : SUCCESS in 1h 02m 43s\n- dsvm-cinder-fss-iscsi-custom http://openstack.falconstor.com/37/318737/1/check/dsvm-cinder-fss-iscsi-custom/3d059fe : SUCCESS in 1h 17m 19s\n\nLeave a comment with \u0027run-FalconStor\u0027 to trigger another test. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/FalconStor_CI","accounts_in_message":[],"_revision_number":1},{"id":"b205e69a7f8a5a3f4abc604338e7032ff292a73e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 12:42:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/xio_ostack/EMC_XtremIO_FC/2298 : SUCCESS in 44m 21s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/xio_ostack/EMC_XtremIO_iSCSI/2389 : FAILURE in 1h 17m 10s\n\nLeave a comment with \u0027run emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":1},{"id":"4dc2756ca3a2efc41b8a422b92921ab83820551b","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2016-08-09 12:54:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/37/318737/1/check/synology-dsm-cinder-iscsi/92c10ce : SUCCESS in 1h 09m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c6779c5bca190af73f147d5fde091a580362a619","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2016-08-09 15:04:04.000000000","message":"Patch Set 1:\n\nmessage:\"self._assertEqualsVolumeOrderResult(vols)\" AND tags:\"console.html\"","accounts_in_message":[],"_revision_number":1},{"id":"eca7fc63b82a74798532ec8d383346a2faff0b9b","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2016-08-09 15:04:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60ffb79258c40fb43d95949c6f3f72e38c8d64c6","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2016-08-09 15:42:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/37/318737/1/check/dsvm-tempest-zadara-driver/00d99cf : SUCCESS in 1h 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebece58e8bef3ef6b91c51159b9a4cc1c6591404","author":{"_account_id":13394,"name":"Microsoft iSCSI CI","username":"microsoft-iscsi-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-08-09 16:55:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- microsoft-ci-iscsi http://64.119.130.115/cinder/318737/1/iscsi/ : SUCCESS in 1h 50m 41s\n- microsoft-ci-smb3-linux http://64.119.130.115/cinder/318737/1/smb3_linux/ : SUCCESS in 40m 59s\n- microsoft-ci-smb3-windows http://64.119.130.115/cinder/318737/1/smb3_windows : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5e4414992c919bad6786164fff6537c09ce66c8f","author":{"_account_id":11215,"name":"Vivek Dhayaal","email":"vivekdhayaal@gmail.com","username":"vivekdhayaal"},"date":"2016-08-09 17:01:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8a98652bdd7375429ea04c423d5770252c5a9017","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-09 21:08:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-cinder-docs-ubuntu-xenial http://docs-draft.openstack.org/37/318737/1/check/gate-cinder-docs-ubuntu-xenial/112a17a//doc/build/html/ : SUCCESS in 8m 20s\n- gate-cinder-pep8-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-cinder-pep8-ubuntu-xenial/b31acd9/ : SUCCESS in 4m 37s\n- gate-cinder-python27-db-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-cinder-python27-db-ubuntu-xenial/c8373e8/ : SUCCESS in 8m 30s\n- gate-cinder-python34-db http://logs.openstack.org/37/318737/1/check/gate-cinder-python34-db/e294536/ : SUCCESS in 9m 20s\n- gate-cinder-python35-db-nv http://logs.openstack.org/37/318737/1/check/gate-cinder-python35-db-nv/bf0efd5/ : SUCCESS in 7m 22s (non-voting)\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-ubuntu-xenial/f5973c6/ : SUCCESS in 43m 46s\n- gate-tempest-dsvm-postgres-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-postgres-full-ubuntu-xenial/d6a708c/ : SUCCESS in 47m 23s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/bfb50c4/ : SUCCESS in 1h 10m 48s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/37/318737/1/check/gate-grenade-dsvm-ubuntu-trusty/87da039/ : SUCCESS in 51m 23s\n- gate-cinder-pylint http://logs.openstack.org/37/318737/1/check/gate-cinder-pylint/39220f0/ : SUCCESS in 12m 44s (non-voting)\n- gate-rally-dsvm-cinder http://logs.openstack.org/37/318737/1/check/gate-rally-dsvm-cinder/c7dcf79/ : SUCCESS in 41m 31s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/7ba5e59/ : SUCCESS in 54m 58s\n- gate-cinder-dsvm-apache http://logs.openstack.org/37/318737/1/check/gate-cinder-dsvm-apache/eeb4607/ : SUCCESS in 1h 00m 21s (non-voting)\n- gate-cinder-tox-db-functional http://logs.openstack.org/37/318737/1/check/gate-cinder-tox-db-functional/f7d218a/ : SUCCESS in 5m 16s (non-voting)\n- gate-cinder-tox-compliance-ubuntu-xenial http://logs.openstack.org/37/318737/1/check/gate-cinder-tox-compliance-ubuntu-xenial/00b7fa5/ : SUCCESS in 2m 52s\n- gate-cinder-tempest-dsvm-glusterfs-nv http://logs.openstack.org/37/318737/1/check/gate-cinder-tempest-dsvm-glusterfs-nv/d871f16/ : POST_FAILURE in 48m 14s (non-voting)\n- gate-tempest-dsvm-full-sheepdog-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-sheepdog-nv/a93d3ad/ : SUCCESS in 48m 33s (non-voting)\n- gate-tempest-dsvm-full-drbd-devstack-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-drbd-devstack-nv/abf2e37/ : SUCCESS in 55m 26s (non-voting)\n- gate-tempest-dsvm-full-lio http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-lio/8429c4e/ : SUCCESS in 53m 56s (non-voting)\n- gate-tempest-dsvm-full-bdd-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-full-bdd-nv/98db650/ : SUCCESS in 1h 05m 44s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/37/318737/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/9fe32a6/ : FAILURE in 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d47cf4bc917a0640cd2bc1874185e68ba4fda6c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-09 21:09:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"e681e1788f5403ef74211e15712e89fe71b4b14f","author":{"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]},"date":"2016-08-09 21:37:37.000000000","message":"Patch Set 1:\n\n* SolidFire-dsvm-full http://54.164.167.86/solidfire-ci-logs/refs-changes-37-318737-1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ba1400fabcfb22451cb71d08052a9d0dc2dc6137","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-08-09 22:07:12.000000000","message":"Patch Set 1:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-37-318737-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"dbf503125b65136cb85fb10aeda818531c491fc3","author":{"_account_id":15249,"name":"Dell Storage CI","email":"openstack-cinder-sc-ci@dell.com","username":"dell-storage-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 22:17:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n* dell-eql-dsvm-volume http://oslogs.compellent.com/dell-eql-318737-1/ : FAILURE \n* dell-sc-iscsi-dsvm-volume http://oslogs.compellent.com/dell-sc-iscsi-318737-1/ : SUCCESS \n* dell-sc-fc-dsvm-volume http://oslogs.compellent.com/dell-sc-fc-318737-1/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"3defc2e571d8858242429ea3ed2dd3eec63ae876","author":{"_account_id":16422,"name":"Tintri CI","email":"openstack-ci@tintri.com","username":"tintri-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 22:17:19.000000000","message":"Patch Set 1:\n\n* tintri-ci http://openstack-ci.tintri.com/tintri/refs-changes-37-318737-1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"69b62cc74cede55f9091e54cd6c18b402c19a9e1","author":{"_account_id":10621,"name":"NetApp CI","email":"xdl-openstack-jenkins@netapp.com","username":"netapp-ci","inactive":true},"date":"2016-08-09 22:18:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-7mode-iSCSI http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-7mode-iSCSI/d72ff42/ : SUCCESS in 49m 18s\n- cinder-7mode-FCP http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-7mode-FCP/227f8e8/ : SUCCESS in 46m 11s\n- cinder-7mode-NFS http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-7mode-NFS/24cf2c4/ : SUCCESS in 44m 31s\n- cinder-cDOT-iSCSI http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-cDOT-iSCSI/dd03abc/ : SUCCESS in 52m 46s\n- cinder-cDOT-FCP http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-cDOT-FCP/9f7e11d/ : SUCCESS in 38m 07s\n- cinder-cDOT-NFS http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-cDOT-NFS/b8210d7/ : FAILURE in 1h 09m 54s\n- cinder-eseries-iSCSI http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-eseries-iSCSI/a326633/ : SUCCESS in 1h 02m 39s\n- cinder-eseries-FCP http://dcf901611175aa43f968-c54047c910227e27e1d6f03bb1796fd7.r95.cf5.rackcdn.com/37/318737/1/check/cinder-eseries-FCP/2d0aa97/ : SUCCESS in 49m 57s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"68855730cdaa26f09531da7cf375408ecd10e4a6","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 22:31:44.000000000","message":"Patch Set 1:\n\n* datera-dsvm-full http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/refs-changes-37-318737-1/index.html : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e66389c6acdea1841aa3bab252ea77db6ae00f69","author":{"_account_id":18402,"name":"HPE Storage CI","email":"hp.cinder.ci@groups.ext.hpe.com","username":"hp-storage-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 22:53:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-c8k02-dsvm http://54.201.44.218/37/318737/1/check/3par-iscsi-driver-master-client-pip-c8k02-dsvm/2a48a51 : SUCCESS in 1h 31m 19s\n- 3par-fc-driver-master-client-pip-c8k02-dsvm http://54.201.44.218/37/318737/1/check/3par-fc-driver-master-client-pip-c8k02-dsvm/d07010d : SUCCESS in 1h 37m 26s\n- lefthand-iscsi-driver-master-client-pip-vsa673-dsvm http://54.201.44.218/37/318737/1/check/lefthand-iscsi-driver-master-client-pip-vsa673-dsvm/1e9e571 : SUCCESS in 1h 42m 36s\n- lefthand-iscsi-driver-master-client-pip-vsa725http-dsvm http://54.201.44.218/37/318737/1/check/lefthand-iscsi-driver-master-client-pip-vsa725http-dsvm/d360d99 : SUCCESS in 1h 45m 05s","accounts_in_message":[],"_revision_number":1},{"id":"67c9bc299baed7c6212817957f4d698ba7cc3b2e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 23:11:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO http://publiclogs.emc.com/vipr_ostack/scaleio/EMC_ScaleIO/2978 : FAILURE in 44m 27s\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":1},{"id":"0cc477104a1545ef4402577cb013f6abc9c0a195","date":"2016-08-09 23:14:37.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"6d7adc9cfcf26d5854bd52c53e37f930e951c06b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-09 23:14:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-cinder-docs-ubuntu-xenial http://docs-draft.openstack.org/37/318737/1/gate/gate-cinder-docs-ubuntu-xenial/8696cab//doc/build/html/ : SUCCESS in 4m 54s\n- gate-cinder-pep8-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-cinder-pep8-ubuntu-xenial/0f132af/ : SUCCESS in 5m 15s\n- gate-cinder-python27-db-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-cinder-python27-db-ubuntu-xenial/c9ad0c1/ : SUCCESS in 7m 05s\n- gate-cinder-python34-db http://logs.openstack.org/37/318737/1/gate/gate-cinder-python34-db/4c59498/ : SUCCESS in 8m 00s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-full-ubuntu-xenial/0d63e28/ : SUCCESS in 49m 35s\n- gate-tempest-dsvm-postgres-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-postgres-full-ubuntu-xenial/41bccc7/ : SUCCESS in 1h 07m 21s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/ae0ac4c/ : SUCCESS in 1h 02m 40s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/37/318737/1/gate/gate-grenade-dsvm-ubuntu-trusty/1d03fb3/ : SUCCESS in 50m 50s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/318737/1/gate/gate-tempest-dsvm-full-devstack-plugin-ceph/7da859d/ : SUCCESS in 59m 44s\n- gate-cinder-tox-compliance-ubuntu-xenial http://logs.openstack.org/37/318737/1/gate/gate-cinder-tox-compliance-ubuntu-xenial/5cd0935/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4d41156cf194cbd687a42ebf254216ae33557b20","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2016-08-09 23:23:18.000000000","message":"Patch Set 1:\n\n* nexenta-dsvm-volume-ns_edge http://140.174.232.106/refs-changes-37-318737-1/ns_edge : FAILURE in 84m 8s\n* nexenta-dsvm-volume-cinder_nedge11_nbd http://140.174.232.106/refs-changes-37-318737-1/cinder_nedge11_nbd : SUCCESS in 49m 1s","accounts_in_message":[],"_revision_number":1},{"id":"4e1fd97d737b728c79cc1d9c6fa78bf2cf2c7e06","author":{"_account_id":12778,"name":"HDS HNAS CI","email":"OpenStackDevelopment@hds.com","username":"hds-hnas-ci","tags":["SERVICE_USER"]},"date":"2016-08-10 18:07:48.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- hitachi-hnas-iscsi http://openstack.fit-tecnologia.org.br:10000/37/318737/1/check/hitachi-hnas-iscsi/72b8309 : FAILURE in 6m 44s\n- hitachi-hnas-nfs http://openstack.fit-tecnologia.org.br:10000/37/318737/1/check/hitachi-hnas-nfs/00b9bf2 : FAILURE in 6m 41s\n\nLeave a comment with \u0027run hnas-ci\u0027 to trigger a recheck. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_HNAS_CI","accounts_in_message":[],"_revision_number":1},{"id":"92a0f19797d1668818907962357fc8f74d447d6e","author":{"_account_id":18444,"name":"FalconStor CI","email":"openstack-ci@falconstor.com","username":"falconstor-ci","tags":["SERVICE_USER"]},"date":"2016-08-10 21:33:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-cinder-fss-fc-custom http://openstack.falconstor.com/37/318737/1/check/dsvm-cinder-fss-fc-custom/d1fc51d : SUCCESS in 1h 04m 07s\n- dsvm-cinder-fss-iscsi-custom http://openstack.falconstor.com/37/318737/1/check/dsvm-cinder-fss-iscsi-custom/6a25ff7 : SUCCESS in 1h 18m 36s\n\nLeave a comment with \u0027run-FalconStor\u0027 to trigger another test. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/FalconStor_CI","accounts_in_message":[],"_revision_number":1},{"id":"4d5601b6dc8d3dd813891f6ebcb9f8d8528792e6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2016-08-11 11:17:09.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- EMC_VNX_FC http://publiclogs.emc.com/vnx_ostack/EMC_VNX_FC/1433 : SUCCESS in 1h 21m 27s\n- EMC_VNX_iSCSI http://publiclogs.emc.com/vnx_ostack/EMC_VNX_iSCSI/1288 : SUCCESS in 1h 32m 08s\n\nLeave a comment with \u0027run emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":1},{"id":"2519d37d4d32489e944cc5ec7cac30373aaca043","author":{"_account_id":16880,"name":"XP Cinder CI","email":"xp.cinder.ci@groups.ext.hpe.com","username":"xp-cinder-ci","tags":["SERVICE_USER"]},"date":"2016-08-12 14:27:07.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-XP Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/XP_Storage_CI\n\n- xp-fc-master-xp7-dsvm http://54.201.44.218/37/318737/1/check/xp-fc-master-xp7-dsvm/b4892b1 : FAILURE in 1h 41m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0300df74e158824564083ad8243e163cd126a8e","revisions":{"a0300df74e158824564083ad8243e163cd126a8e":{"kind":"REWORK","_number":1,"created":"2016-05-19 15:17:43.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/37/318737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/318737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/318737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/318737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/318737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/318737/1"}}},"commit":{"parents":[{"commit":"8291c0bb433630ed6caf560782ccfe95aa62a92b","subject":"Merge \"3PAR Clean up VLUN deletion on detach\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8291c0bb433630ed6caf560782ccfe95aa62a92b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2016-05-19 14:35:07.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2016-05-19 15:17:30.000000000","tz":-240},"subject":"IBM Flashsystem: Add missing items to common","message":"IBM Flashsystem: Add missing items to common\n\nThere are a handful of methods that are\nintended to be inherited from this base class\nthat aren\u0027t defined, and some member variables\nwhich are not guaranteed to be defined before use.\n\nAdd missing items to the base common class\nto clean things up.\n\nThis removes seven issues from our pylint run.\n\nChange-Id: I2a93852bbd5fc528f6f73d99212ac6040eaa004a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0300df74e158824564083ad8243e163cd126a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0300df74e158824564083ad8243e163cd126a8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
