)]}'
{"id":"openstack%2Fcinder~33704","triplet_id":"openstack%2Fcinder~master~I5cca287e8204c4647a5f09df961d9a76d0b38c42","project":"openstack/cinder","branch":"master","topic":"thin_lvm_snap_create_fix","hashtags":[],"change_id":"I5cca287e8204c4647a5f09df961d9a76d0b38c42","subject":"Add _create_volume to ThinLVMVolumeDriver","status":"MERGED","created":"2013-06-19 20:43:40.000000000","updated":"2013-06-20 17:28:59.000000000","submitted":"2013-06-20 17:28:59.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d8a573036a04237c683998002a2ae10e250c31b3","_number":33704,"virtual_id_number":33704,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":2,"date":"2013-06-20 17:28:59.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-06-20 14:46:19.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":2,"date":"2013-06-20 15:36:45.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":1,"date":"2013-06-20 15:36:45.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-06-20 14:46:19.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-06-20 15:36:45.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2013-06-20 17:28:59.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"0ed94126ffa39446d7de5d0f92717940f7208cc7","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-19 21:41:12.000000000","message":"Patch Set 1: Works for me\n\nSmokeStack Results (patch set 1):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/92383\n","accounts_in_message":[],"_revision_number":1},{"id":"c4271c0ea36440b3b27be97c2d6424088f3c8284","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-19 21:46:33.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/33704/1/check/gate-cinder-pep8/4305 : SUCCESS in 1m 49s\n- http://docs-draft.openstack.org/33704/1/check/gate-cinder-docs/4198/doc/build/html/ : SUCCESS in 3m 43s\n- http://logs.openstack.org/33704/1/check/gate-cinder-python26/4216 : SUCCESS in 4m 59s\n- http://logs.openstack.org/33704/1/check/gate-cinder-python27/4232 : SUCCESS in 3m 06s\n- http://logs.openstack.org/33704/1/check/gate-cinder-pylint/568 : SUCCESS in 3m 31s (non-voting)\n- http://logs.openstack.org/33704/1/check/gate-tempest-devstack-vm-full/24765 : SUCCESS in 53m 30s\n- http://logs.openstack.org/33704/1/check/gate-tempest-devstack-vm-postgres-full/24471 : SUCCESS in 43m 41s\n- http://logs.openstack.org/33704/1/check/gate-tempest-devstack-vm-quantum/32257 : SUCCESS in 20m 15s\n- http://logs.openstack.org/33704/1/check/gate-grenade-devstack-vm/11569 : FAILURE in 42m 35s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"8a8cfe5ea5df7830338c35f47d16f52eb33bf9d8","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-06-20 05:37:43.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(2 inline comments)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"d0c6967dea677a82204d1b9b4b43cd7a447a15f6","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-06-20 05:39:56.000000000","message":"Patch Set 1: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"854d2e23b3019b582728b3744f742dd22263a919","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-06-20 14:15:40.000000000","message":"Patch Set 1: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"95c0b9d1ec3ddc188235e4ebb2391c84b9433b7c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-06-20 14:18:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8b5dd51232260f91ba0b12eab2d3c56c200a6af","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-20 14:46:19.000000000","message":"Patch Set 2: Works for me\n\nSmokeStack Results (patch set 2):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/92578\n","accounts_in_message":[],"_revision_number":2},{"id":"17d9b77da9e80950dccd34a18cc8f9495968192a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-20 15:07:18.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/33704/2/check/gate-cinder-pep8/4337 : SUCCESS in 1m 46s\n- http://docs-draft.openstack.org/33704/2/check/gate-cinder-docs/4230/doc/build/html/ : SUCCESS in 3m 45s\n- http://logs.openstack.org/33704/2/check/gate-cinder-python26/4248 : SUCCESS in 4m 33s\n- http://logs.openstack.org/33704/2/check/gate-cinder-python27/4264 : SUCCESS in 2m 56s\n- http://logs.openstack.org/33704/2/check/gate-cinder-pylint/595 : SUCCESS in 3m 29s (non-voting)\n- http://logs.openstack.org/33704/2/check/gate-tempest-devstack-vm-full/24984 : SUCCESS in 41m 04s\n- http://logs.openstack.org/33704/2/check/gate-tempest-devstack-vm-postgres-full/24690 : SUCCESS in 48m 16s\n- http://logs.openstack.org/33704/2/check/gate-tempest-devstack-vm-quantum/32495 : SUCCESS in 17m 19s\n- http://logs.openstack.org/33704/2/check/gate-grenade-devstack-vm/11733 : FAILURE in 9m 42s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"b2bb05a446ea20191bd4586dc31f461a54020b18","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-06-20 15:36:45.000000000","message":"Patch Set 2: Looks good to me (core reviewer); Approved\n\nThanks Eric, sorry about my confusion earlier","accounts_in_message":[],"_revision_number":2},{"id":"e0d6cc6a1a16e60ff86bef9a8f9e60468b894ffd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-20 15:36:47.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"384c68f723cdbae0f43bfe8388d7dd816395e982","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-20 17:28:59.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/33704/2/gate/gate-cinder-docs/4237/doc/build/html/ : SUCCESS in 3m 58s\n- http://logs.openstack.org/33704/2/gate/gate-cinder-pep8/4344 : SUCCESS in 2m 00s\n- http://logs.openstack.org/33704/2/gate/gate-cinder-python26/4255 : SUCCESS in 5m 19s\n- http://logs.openstack.org/33704/2/gate/gate-cinder-python27/4271 : SUCCESS in 3m 07s\n- http://logs.openstack.org/33704/2/gate/gate-tempest-devstack-vm-full/25040 : SUCCESS in 43m 05s\n- http://logs.openstack.org/33704/2/gate/gate-tempest-devstack-vm-postgres-full/24746 : SUCCESS in 43m 11s\n- http://logs.openstack.org/33704/2/gate/gate-tempest-devstack-vm-quantum/32557 : SUCCESS in 22m 49s\n- http://logs.openstack.org/33704/2/gate/gate-grenade-devstack-vm/11779 : FAILURE in 13m 12s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"4794b90a84b899b514f4b393450066f9041d2a3b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-20 17:28:59.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"79a402d07502ac34bc0ec485b21b0be50ba807a7","revisions":{"98d952c0c34c7a79c163949591d72c7a8c381e79":{"kind":"REWORK","_number":1,"created":"2013-06-19 20:43:40.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/04/33704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/33704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/33704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/33704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/33704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/33704/1"}}},"commit":{"parents":[{"commit":"1482c4045b94319aff0bd5999a52eb892152773b","subject":"Merge \"Use Python 3.x compatible except construct\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1482c4045b94319aff0bd5999a52eb892152773b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-06-19 20:31:07.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-06-19 20:43:12.000000000","tz":-240},"subject":"Add _create_volume to ThinLVMVolumeDriver","message":"Add _create_volume to ThinLVMVolumeDriver\n\nPreviously, _create_volume from LVMISCSIDriver would be called\nwhen the volume manager calls create_volume_from_snapshot.\n\nFixes bug: 1192601\n\nChange-Id: I5cca287e8204c4647a5f09df961d9a76d0b38c42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98d952c0c34c7a79c163949591d72c7a8c381e79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98d952c0c34c7a79c163949591d72c7a8c381e79"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1482c4045b94319aff0bd5999a52eb892152773b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"79a402d07502ac34bc0ec485b21b0be50ba807a7":{"kind":"REWORK","_number":2,"created":"2013-06-20 14:18:47.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/04/33704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/33704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/33704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/33704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/33704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/33704/2"}}},"commit":{"parents":[{"commit":"1482c4045b94319aff0bd5999a52eb892152773b","subject":"Merge \"Use Python 3.x compatible except construct\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1482c4045b94319aff0bd5999a52eb892152773b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-06-19 20:31:07.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-06-20 14:15:56.000000000","tz":-240},"subject":"Add _create_volume to ThinLVMVolumeDriver","message":"Add _create_volume to ThinLVMVolumeDriver\n\nPreviously, _create_volume from LVMISCSIDriver would be called\nwhen the volume manager calls create_volume_from_snapshot.\n\nFixes bug: 1192601\n\nChange-Id: I5cca287e8204c4647a5f09df961d9a76d0b38c42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79a402d07502ac34bc0ec485b21b0be50ba807a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79a402d07502ac34bc0ec485b21b0be50ba807a7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1482c4045b94319aff0bd5999a52eb892152773b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
