)]}'
{"id":"openstack%2Fcinder~34290","triplet_id":"openstack%2Fcinder~master~I0a532fe06798d07f21d79ee45444bb9c0a025430","project":"openstack/cinder","branch":"master","topic":"bp/gpfs-volume-driver","hashtags":[],"change_id":"I0a532fe06798d07f21d79ee45444bb9c0a025430","subject":"Add support for storing volumes on GPFS","status":"MERGED","created":"2013-06-25 01:29:14.000000000","updated":"2013-07-11 13:31:59.000000000","submitted":"2013-07-11 13:31:58.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":85,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"693f2e14608431fb97de8ba752fe4b6e426568c9","_number":34290,"virtual_id_number":34290,"owner":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},{"value":0,"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":2,"date":"2013-07-11 13:31:58.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-07-09 19:32:03.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":1773,"name":"oliver-leahy-l","email":"oliver.leahy@hpe.com","username":"oliver-leahy-l"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"all":[{"value":0,"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},{"value":2,"date":"2013-07-10 04:29:54.000000000","_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},{"value":0,"date":"2013-07-11 03:09:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":1,"date":"2013-07-10 09:39:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1773,"name":"oliver-leahy-l","email":"oliver.leahy@hpe.com","username":"oliver-leahy-l"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"all":[{"value":0,"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},{"value":1,"date":"2013-07-11 11:34:06.000000000","_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},{"value":0,"date":"2013-07-11 03:09:23.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":1773,"name":"oliver-leahy-l","email":"oliver.leahy@hpe.com","username":"oliver-leahy-l"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},{"value":0,"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":1773,"name":"oliver-leahy-l","email":"oliver.leahy@hpe.com","username":"oliver-leahy-l"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"_account_id":1773,"name":"oliver-leahy-l","email":"oliver.leahy@hpe.com","username":"oliver-leahy-l"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},{"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-06-25 10:27:38.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2013-06-25 23:42:19.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2013-07-01 23:44:14.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2013-07-02 03:43:50.000000000","updated_by":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"reviewer":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"state":"REVIEWER"},{"updated":"2013-07-02 16:27:43.000000000","updated_by":{"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},"reviewer":{"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},"state":"REVIEWER"},{"updated":"2013-07-09 19:32:03.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-07-10 09:39:10.000000000","updated_by":{"_account_id":1773,"name":"oliver-leahy-l","email":"oliver.leahy@hpe.com","username":"oliver-leahy-l"},"reviewer":{"_account_id":1773,"name":"oliver-leahy-l","email":"oliver.leahy@hpe.com","username":"oliver-leahy-l"},"state":"REVIEWER"},{"updated":"2013-07-11 03:09:23.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2013-07-11 11:34:06.000000000","updated_by":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"reviewer":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"state":"REVIEWER"},{"updated":"2013-07-11 13:31:58.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"009c9ad90ac6df7e6e0e60b6516f783c2574a77e","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-25 01:48:34.000000000","message":"Patch Set 1: Works for me\n\nSmokeStack Results (patch set 1):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/93484\n","accounts_in_message":[],"_revision_number":1},{"id":"5579ac4b4f75705abacbf676df14b76657f5e7b4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-25 02:13:07.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/34290/1/check/gate-cinder-pep8/4412 : FAILURE in 1m 55s\n- http://docs-draft.openstack.org/34290/1/check/gate-cinder-docs/4305/doc/build/html/ : SUCCESS in 3m 43s\n- http://logs.openstack.org/34290/1/check/gate-cinder-python26/4322 : FAILURE in 4m 46s\n- http://logs.openstack.org/34290/1/check/gate-cinder-python27/4339 : FAILURE in 2m 57s\n- http://logs.openstack.org/34290/1/check/gate-cinder-pylint/649 : SUCCESS in 3m 25s (non-voting)\n- http://logs.openstack.org/34290/1/check/gate-tempest-devstack-vm-full/25944 : SUCCESS in 39m 44s\n- http://logs.openstack.org/34290/1/check/gate-tempest-devstack-vm-postgres-full/25650 : SUCCESS in 43m 42s\n- http://logs.openstack.org/34290/1/check/gate-tempest-devstack-vm-quantum/33560 : SUCCESS in 16m 20s\n- http://logs.openstack.org/34290/1/check/gate-grenade-devstack-vm/12586 : FAILURE in 20m 41s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"929ec79cdd07bcaefedc425b3396614cf981040e","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-06-25 03:14:16.000000000","message":"Patch Set 1: Do not merge\n\nYour commit message doesn\u0027t reference a valid BP, and if you intended to use: ibm-gpfs-driver;\n\nThat particular BP is incomplete and needs significantly more detail.  Intro do GPFS, features implemented, user story etc.","accounts_in_message":[],"_revision_number":1},{"id":"18db92da631cd19183c806f6c8f14e29911d3cd2","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-06-25 10:27:38.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(4 inline comments)\n\nIn addition to John\u0027s comments, there are also some other fundamental issues with the patch as well. Comments inline.","accounts_in_message":[],"_revision_number":1},{"id":"218e3cce9f011e0cc893e49f69bb696bbe749cdb","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-06-25 20:22:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee5502b92d237d5f49a580bfdb7294ef4d275962","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-25 20:45:31.000000000","message":"Patch Set 2: Works for me\n\nSmokeStack Results (patch set 2):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/93743\n","accounts_in_message":[],"_revision_number":2},{"id":"7d6a1a66075dea6e26d2ce44c6207af652d800cf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-25 21:02:42.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/34290/2/check/gate-cinder-pep8/4458 : SUCCESS in 1m 54s\n- http://docs-draft.openstack.org/34290/2/check/gate-cinder-docs/4351/doc/build/html/ : SUCCESS in 4m 00s\n- http://logs.openstack.org/34290/2/check/gate-cinder-python26/4368 : FAILURE in 5m 20s\n- http://logs.openstack.org/34290/2/check/gate-cinder-python27/None : LOST\n- http://logs.openstack.org/34290/2/check/gate-cinder-pylint/None : LOST (non-voting)\n- http://logs.openstack.org/34290/2/check/gate-tempest-devstack-vm-full/26178 : SUCCESS in 39m 36s\n- http://logs.openstack.org/34290/2/check/gate-tempest-devstack-vm-postgres-full/25884 : SUCCESS in 40m 03s\n- http://logs.openstack.org/34290/2/check/gate-tempest-devstack-vm-quantum/33813 : SUCCESS in 20m 30s\n- http://logs.openstack.org/34290/2/check/gate-grenade-devstack-vm/12793 : FAILURE in 22m 33s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"c3f358f853734e6cbf09f5d8fe06a0cc8f563265","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-06-25 21:31:45.000000000","message":"Patch Set 2:\n\nJohn, this change implements the following BP: https://blueprints.launchpad.net/cinder/+spec/gpfs-volume-driver\n\nThe other BP (ibm-gpfs-driver) is an abandoned effort.","accounts_in_message":[],"_revision_number":2},{"id":"bd9e14a7524dcebf0c354541a9957e6ecd0b76e2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2013-06-25 23:42:19.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(8 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"0820f43d8319ab9e9d295aef2e7ec8ebd177dcf1","author":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"date":"2013-06-26 04:26:55.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(20 inline comments)\n\nI haven\u0027t completely gone through the logic, but this is just what I found at a glance.","accounts_in_message":[],"_revision_number":2},{"id":"3370264d4761de9c13acd0e395763c5a94d79d08","author":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"date":"2013-06-26 09:12:44.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nI think that this driver should inherit from RemoteFsDriver, so that you get functions like _create_sparsed_file(), _create_regular_file(), and _set_rw_permissions_for_all() without duplicating code.\n\nAlso, there is a problem that the blueprint doesn\u0027t show up in searches - does anyone have any insight into that?","accounts_in_message":[],"_revision_number":2},{"id":"7e1e65d1caab4d7f37816de98b5928d62befdda0","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-06-26 14:31:22.000000000","message":"Patch Set 2:\n\nThank you all for the comments so far.  Will upload a new patch set later this afternoon.","accounts_in_message":[],"_revision_number":2},{"id":"02ecf5920850c0b2f7cb071488a2b1856c6cf1e4","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-06-26 21:55:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a1bb9c36ef7e32c0932653a9bd2f9c22667869e6","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-06-26 22:06:51.000000000","message":"Patch Set 3:\n\nAvishay,  those private functions will invoke GPFS specific features such as per-volume replication level, placement etc. in a later revision.","accounts_in_message":[],"_revision_number":3},{"id":"a73a82876a92ab4986b94976bf1858bfe68b0446","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-26 22:16:26.000000000","message":"Patch Set 3: Works for me\n\nSmokeStack Results (patch set 3):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/94163\n","accounts_in_message":[],"_revision_number":3},{"id":"0adb1ae45e86dc59d6e2031b35eb7be4f0009f06","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-26 22:44:36.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/34290/3/check/gate-cinder-pep8/4494 : SUCCESS in 2m 01s\n- http://docs-draft.openstack.org/34290/3/check/gate-cinder-docs/4388/doc/build/html/ : SUCCESS in 4m 03s\n- http://logs.openstack.org/34290/3/check/gate-cinder-python26/4405 : FAILURE in 5m 04s\n- http://logs.openstack.org/34290/3/check/gate-cinder-python27/4420 : FAILURE in 2m 56s\n- http://logs.openstack.org/34290/3/check/gate-cinder-pylint/712 : SUCCESS in 3m 34s (non-voting)\n- http://logs.openstack.org/34290/3/check/gate-tempest-devstack-vm-full/26589 : SUCCESS in 48m 26s\n- http://logs.openstack.org/34290/3/check/gate-tempest-devstack-vm-postgres-full/26295 : SUCCESS in 43m 41s\n- http://logs.openstack.org/34290/3/check/gate-tempest-devstack-vm-quantum/34249 : SUCCESS in 19m 09s\n- http://logs.openstack.org/34290/3/check/gate-grenade-devstack-vm/13142 : FAILURE in 21m 09s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"78eb7af82a63df8ccc50c28a0c1a57aa8ad42dbd","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-06-27 19:47:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a7785bfaded5f6de4754daf581d3825929e2b4e0","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-27 20:10:20.000000000","message":"Patch Set 4: Works for me\n\nSmokeStack Results (patch set 4):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/94532\n","accounts_in_message":[],"_revision_number":4},{"id":"58170c04f08b3f7c5697f3dc025a49e577db750d","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-06-27 20:42:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1ebfd274eb30a55a0525ebe48093a143597f8b78","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-27 21:04:16.000000000","message":"Patch Set 5: Works for me\n\nSmokeStack Results (patch set 5):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/94549\n","accounts_in_message":[],"_revision_number":5},{"id":"c79f6daef7fd8231ed99d2f6b031950a0dc6c2ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-27 21:25:14.000000000","message":"Patch Set 5: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/34290/5/check/gate-cinder-pep8/4519 : SUCCESS in 2m 16s\n- http://docs-draft.openstack.org/34290/5/check/gate-cinder-docs/4413/doc/build/html/ : SUCCESS in 4m 26s\n- http://logs.openstack.org/34290/5/check/gate-cinder-python26/4430 : FAILURE in 5m 28s\n- http://logs.openstack.org/34290/5/check/gate-cinder-python27/4445 : SUCCESS in 4m 18s\n- http://logs.openstack.org/34290/5/check/gate-cinder-pylint/733 : SUCCESS in 3m 38s (non-voting)\n- http://logs.openstack.org/34290/5/check/gate-tempest-devstack-vm-full/26913 : SUCCESS in 41m 50s\n- http://logs.openstack.org/34290/5/check/gate-tempest-devstack-vm-postgres-full/26619 : SUCCESS in 42m 41s\n- http://logs.openstack.org/34290/5/check/gate-tempest-devstack-vm-quantum/34615 : SUCCESS in 17m 18s\n- http://logs.openstack.org/34290/5/check/gate-grenade-devstack-vm/13419 : FAILURE in 20m 21s (non-voting)\n","accounts_in_message":[],"_revision_number":5},{"id":"c1b9faccefd8568b44091bc16fd9086e44606176","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-06-28 16:32:55.000000000","message":"Patch Set 5: No score\n\n","accounts_in_message":[],"_revision_number":5},{"id":"4f502a00572412c907be4e19263c8da1023b1136","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-06-28 17:29:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"65ddeb1dd07e6951017bc5e7d1ac13918325f3bc","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-06-28 17:48:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0e4b5b32803ac4d785ae419048dbb42c0f00507d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-28 18:09:44.000000000","message":"Patch Set 7: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/34290/7/check/gate-cinder-pep8/4541 : SUCCESS in 2m 10s\n- http://docs-draft.openstack.org/34290/7/check/gate-cinder-docs/4435/doc/build/html/ : SUCCESS in 4m 08s\n- http://logs.openstack.org/34290/7/check/gate-cinder-python26/4452 : FAILURE in 5m 58s\n- http://logs.openstack.org/34290/7/check/gate-cinder-python27/4467 : SUCCESS in 3m 31s\n- http://logs.openstack.org/34290/7/check/gate-cinder-pylint/748 : SUCCESS in 4m 13s (non-voting)\n- http://logs.openstack.org/34290/7/check/gate-tempest-devstack-vm-full/27212 : FAILURE in 10m 42s\n- http://logs.openstack.org/34290/7/check/gate-tempest-devstack-vm-postgres-full/26918 : FAILURE in 12m 11s\n- http://logs.openstack.org/34290/7/check/gate-tempest-devstack-vm-quantum/34947 : FAILURE in 12m 22s\n- http://logs.openstack.org/34290/7/check/gate-grenade-devstack-vm/13683 : FAILURE in 20m 50s (non-voting)\n","accounts_in_message":[],"_revision_number":7},{"id":"f8e312c47eaafad98fbefbe9783661c7f2c2eb8f","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-28 18:11:31.000000000","message":"Patch Set 7: Works for me\n\nSmokeStack Results (patch set 7):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/94857\n","accounts_in_message":[],"_revision_number":7},{"id":"09521407ecfad75ff7497b9cee379fde92cf7f48","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-07-01 19:46:19.000000000","message":"Patch Set 7: I would prefer that you didn\u0027t merge this\n\n(4 inline comments)\n\nThe test class is named GPFSDriverTestCase but the driver is named GpfsDriver.  These should probably match.\n\nSince users have to type in the volume class name in cinder.conf, it may be better to go with GPFSDriver for the volume name, as I assume that is the more \"expected\" capitalization.  But either way.","accounts_in_message":[],"_revision_number":7},{"id":"62c32dd28a9b72ff7be84f3dab79fd45bbe9ede1","author":{"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},"date":"2013-07-01 23:44:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c05c49eb573ba6827548167d974ce3440b278fe9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-07-01 23:44:14.000000000","message":"Patch Set 8: I would prefer that you didn\u0027t merge this\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":8},{"id":"2fbdba5194e37f73dcf5be9123ac4a5b66187645","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-02 00:01:42.000000000","message":"Patch Set 8: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/34290/8/check/gate-cinder-pep8/4561 : SUCCESS in 2m 08s\n- http://docs-draft.openstack.org/34290/8/check/gate-cinder-docs/4455/doc/build/html/ : SUCCESS in 3m 58s\n- http://logs.openstack.org/34290/8/check/gate-cinder-python26/4472 : FAILURE in 4m 45s\n- http://logs.openstack.org/34290/8/check/gate-cinder-python27/4487 : SUCCESS in 3m 31s\n- http://logs.openstack.org/34290/8/check/gate-cinder-pylint/764 : SUCCESS in 3m 39s (non-voting)\n- http://logs.openstack.org/34290/8/check/gate-tempest-devstack-vm-full/27687 : FAILURE in 11m 10s\n- http://logs.openstack.org/34290/8/check/gate-tempest-devstack-vm-postgres-full/27393 : FAILURE in 10m 32s\n- http://logs.openstack.org/34290/8/check/gate-tempest-devstack-vm-quantum/35519 : FAILURE in 13m 05s\n- http://logs.openstack.org/34290/8/check/gate-grenade-devstack-vm/14116 : FAILURE in 17m 18s (non-voting)\n","accounts_in_message":[],"_revision_number":8},{"id":"0490c9c89f6c6c9c4972c4cdeeb97a993d960ad6","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-02 00:16:03.000000000","message":"Patch Set 8: Works for me\n\nSmokeStack Results (patch set 8):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/95467\n","accounts_in_message":[],"_revision_number":8},{"id":"2b393fe3d0ba8fa866b4088cd79634cdb9d432ec","author":{"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},"date":"2013-07-02 00:42:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"53b451bd7a622564a8842c27fcf1e1525692eeab","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-02 01:02:16.000000000","message":"Patch Set 9: Works for me\n\nSmokeStack Results (patch set 9):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/95479\n","accounts_in_message":[],"_revision_number":9},{"id":"c5c55e97de5e71931b9273ad66f34a408809b67d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-02 01:36:07.000000000","message":"Patch Set 9: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34290/9/check/gate-cinder-pep8/4563 : SUCCESS in 2m 01s\n- http://docs-draft.openstack.org/34290/9/check/gate-cinder-docs/4457/doc/build/html/ : SUCCESS in 4m 06s\n- http://logs.openstack.org/34290/9/check/gate-cinder-python26/4474 : SUCCESS in 5m 00s\n- http://logs.openstack.org/34290/9/check/gate-cinder-python27/4489 : SUCCESS in 3m 15s\n- http://logs.openstack.org/34290/9/check/gate-cinder-pylint/766 : SUCCESS in 3m 33s (non-voting)\n- http://logs.openstack.org/34290/9/check/gate-tempest-devstack-vm-full/27694 : SUCCESS in 53m 25s\n- http://logs.openstack.org/34290/9/check/gate-tempest-devstack-vm-postgres-full/27400 : SUCCESS in 43m 59s\n- http://logs.openstack.org/34290/9/check/gate-tempest-devstack-vm-quantum/35526 : SUCCESS in 18m 28s\n- http://logs.openstack.org/34290/9/check/gate-grenade-devstack-vm/14123 : FAILURE in 21m 35s (non-voting)\n","accounts_in_message":[],"_revision_number":9},{"id":"b72292acc4253a764b1ca5ceabeb19019293c56e","author":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"date":"2013-07-02 03:43:50.000000000","message":"Patch Set 9: Do not merge\n\nAccording to Bill Owen we\u0027re using this other one. https://review.openstack.org/#/c/35228/","accounts_in_message":[],"_revision_number":9},{"id":"0e6517a8bdcc24fc29d3ba42544c4ca54c5d9858","author":{"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},"date":"2013-07-02 16:27:43.000000000","message":"Patch Set 9:\n\nThis is the correct change to review.  Mike Perez, I incorrectly submitted the one you link to in your comment.  I have abandoned that change.  Sorry for the confusion!","accounts_in_message":[],"_revision_number":9},{"id":"052398344dde9d2fa68db455327234fe5f89d13d","author":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"date":"2013-07-02 19:52:49.000000000","message":"Patch Set 9: No score\n\n","accounts_in_message":[],"_revision_number":9},{"id":"2773566c9af1a2c817f35afd10552aa6b4dc5990","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-02 23:55:11.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1b652ad5e41bc6ef98c0a3855845dd5ca35448a6","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-07-03 00:05:40.000000000","message":"Patch Set 10:\n\nsorry, but I was wondering if you could clear things up a bit.  Your BP mentions what GPFS is in good detail, but I\u0027m trying to understand the use model in OpenStack/Cinder?\n\nI\u0027ve noticed you have the iscsi export methods all implemented but just doing a pass... how do you attach a GPFS volume from Cinder to an instance/Compute Node?","accounts_in_message":[],"_revision_number":10},{"id":"e223bf0aa2d7934cf1387df6fd3ef91a5449e7de","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-07-03 00:06:54.000000000","message":"Patch Set 10:\n\nIs this something that assumes GPFS is installed on the compute nodes and works more directly with Nova than actually with Cinder?","accounts_in_message":[],"_revision_number":10},{"id":"51b357a66c4fca89c55666f01cabd8dd4662f32f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-03 00:49:11.000000000","message":"Patch Set 10: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34290/10/check/gate-cinder-pep8/4591 : SUCCESS in 1m 59s\n- http://docs-draft.openstack.org/34290/10/check/gate-cinder-docs/4485/doc/build/html/ : SUCCESS in 4m 03s\n- http://logs.openstack.org/34290/10/check/gate-cinder-python26/4502 : SUCCESS in 6m 38s\n- http://logs.openstack.org/34290/10/check/gate-cinder-python27/4517 : SUCCESS in 3m 33s\n- http://logs.openstack.org/34290/10/check/gate-cinder-pylint/784 : SUCCESS in 3m 36s (non-voting)\n- http://logs.openstack.org/34290/10/check/gate-tempest-devstack-vm-full/28065 : SUCCESS in 43m 42s\n- http://logs.openstack.org/34290/10/check/gate-tempest-devstack-vm-postgres-full/27771 : SUCCESS in 53m 45s\n- http://logs.openstack.org/34290/10/check/gate-tempest-devstack-vm-quantum/35924 : SUCCESS in 21m 02s\n- http://logs.openstack.org/34290/10/check/gate-grenade-devstack-vm/14450 : FAILURE in 21m 31s (non-voting)\n","accounts_in_message":[],"_revision_number":10},{"id":"11957077029812e26a6181b764743b051d8ceecb","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-03 01:59:03.000000000","message":"Patch Set 10:\n\nThanks for bringing it up John.  Yes, the driver assumes that GPFS is installed and mounted across the nodes running cinder-volume and nova-compute services -- the physical storage backing the shared file system namespace could be DAS/SAN devices attached to any/some/all of these nodes.  Cinder driver maintains the volumes through its local GPFS interface and Nova connects them to respective instances again through its local interface to GPFS on the compute nodes.  No explicit change to Nova is required because Nova just views it as a \u0027local\u0027 store and uses base LibvirtVolumeDriver to do the attach.","accounts_in_message":[],"_revision_number":10},{"id":"42d1655558a0d2c4f3dcb087af262f97c9c66aac","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-03 02:06:10.000000000","message":"Patch Set 10:\n\nFor curiosity, here is a paper that describes the internals of GPFS and its usage in the Cloud:  http://dl.acm.org/citation.cfm?id\u003d2481709","accounts_in_message":[],"_revision_number":10},{"id":"9582ae54c560b0382cce28a244580c7e23defa33","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-03 23:44:42.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ddc25a4387d60483e80cf23fc93d05580c9b9c9d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-04 00:28:05.000000000","message":"Patch Set 11: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34290/11/check/gate-cinder-pep8/4608 : SUCCESS in 2m 04s\n- http://docs-draft.openstack.org/34290/11/check/gate-cinder-docs/4502/doc/build/html/ : SUCCESS in 3m 57s\n- http://logs.openstack.org/34290/11/check/gate-cinder-python26/4519 : SUCCESS in 6m 17s\n- http://logs.openstack.org/34290/11/check/gate-cinder-python27/4534 : SUCCESS in 3m 56s\n- http://logs.openstack.org/34290/11/check/gate-cinder-pylint/796 : SUCCESS in 3m 27s (non-voting)\n- http://logs.openstack.org/34290/11/check/gate-tempest-devstack-vm-full/28467 : SUCCESS in 42m 17s\n- http://logs.openstack.org/34290/11/check/gate-tempest-devstack-vm-postgres-full/28173 : SUCCESS in 43m 05s\n- http://logs.openstack.org/34290/11/check/gate-tempest-devstack-vm-quantum/36363 : SUCCESS in 18m 33s\n- http://logs.openstack.org/34290/11/check/gate-grenade-devstack-vm/14806 : FAILURE in 25m 21s (non-voting)\n","accounts_in_message":[],"_revision_number":11},{"id":"b56eff6e49d9b19e46fa855a011e80c711851834","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-04 00:57:05.000000000","message":"Patch Set 11: Works for me\n\nSmokeStack Results (patch set 11):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/96220\n","accounts_in_message":[],"_revision_number":11},{"id":"58470155f4471c407f8f6fc511e1d5082dfff518","author":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"date":"2013-07-04 15:00:15.000000000","message":"Patch Set 11: I would prefer that you didn\u0027t merge this\n\n(3 inline comments)\n\nOne nit and a couple questions.  Otherwise, looks good to me.","accounts_in_message":[],"_revision_number":11},{"id":"0f9b4d9da9ac11918b5d4559812431ee4febd2ab","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-04 16:37:28.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6a66007ee8ebb1b114cb5c2793dbe874d0edd135","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-04 17:31:53.000000000","message":"Patch Set 12: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34290/12/check/gate-cinder-pep8/4632 : SUCCESS in 2m 20s\n- http://docs-draft.openstack.org/34290/12/check/gate-cinder-docs/4526/doc/build/html/ : SUCCESS in 4m 41s\n- http://logs.openstack.org/34290/12/check/gate-cinder-python26/4543 : SUCCESS in 6m 41s\n- http://logs.openstack.org/34290/12/check/gate-cinder-python27/4558 : SUCCESS in 3m 37s\n- http://logs.openstack.org/34290/12/check/gate-cinder-pylint/816 : SUCCESS in 3m 50s (non-voting)\n- http://logs.openstack.org/34290/12/check/gate-tempest-devstack-vm-full/28637 : SUCCESS in 54m 13s\n- http://logs.openstack.org/34290/12/check/gate-tempest-devstack-vm-postgres-full/28343 : SUCCESS in 48m 12s\n- http://logs.openstack.org/34290/12/check/gate-tempest-devstack-vm-quantum/36558 : SUCCESS in 18m 24s\n- http://logs.openstack.org/34290/12/check/gate-grenade-devstack-vm/14958 : FAILURE in 19m 14s (non-voting)\n","accounts_in_message":[],"_revision_number":12},{"id":"a45443a2f91573440a6687cbc3d21dc18661c929","author":{"_account_id":1773,"name":"oliver-leahy-l","email":"oliver.leahy@hpe.com","username":"oliver-leahy-l"},"date":"2013-07-05 09:17:22.000000000","message":"Patch Set 12: I would prefer that you didn\u0027t merge this\n\n(4 inline comments)\n\nSome questions inline","accounts_in_message":[],"_revision_number":12},{"id":"4e4bea1d01e7f3734d1bd48ed6346bfab7b122b0","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-07 00:08:13.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7226310783b77a7bdf8dc5fd047d61e4a05eec09","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-07 00:30:50.000000000","message":"Patch Set 13: Works for me\n\nSmokeStack Results (patch set 13):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/97049\n","accounts_in_message":[],"_revision_number":13},{"id":"0fa7c53cf6e04b161dbc64c3b0ed0f4fc79711b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-07 01:22:25.000000000","message":"Patch Set 13: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34290/13/check/gate-cinder-pep8/4672 : SUCCESS in 2m 44s\n- http://docs-draft.openstack.org/34290/13/check/gate-cinder-docs/4566/doc/build/html/ : SUCCESS in 4m 47s\n- http://logs.openstack.org/34290/13/check/gate-cinder-python26/4583 : SUCCESS in 6m 58s\n- http://logs.openstack.org/34290/13/check/gate-cinder-python27/4598 : SUCCESS in 3m 27s\n- http://logs.openstack.org/34290/13/check/gate-cinder-pylint/846 : SUCCESS in 3m 42s (non-voting)\n- http://logs.openstack.org/34290/13/check/gate-tempest-devstack-vm-full/29020 : SUCCESS in 53m 12s\n- http://logs.openstack.org/34290/13/check/gate-tempest-devstack-vm-postgres-full/28726 : SUCCESS in 42m 56s\n- http://logs.openstack.org/34290/13/check/gate-tempest-devstack-vm-neutron/49 : SUCCESS in 21m 02s\n- http://logs.openstack.org/34290/13/check/gate-grenade-devstack-vm/15294 : FAILURE in 26m 01s (non-voting)\n","accounts_in_message":[],"_revision_number":13},{"id":"018684681552f5138b6b1e8afee09124568c0f71","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-08 15:54:44.000000000","message":"Patch Set 12: (3 inline comments)\n\nThanks for the comments, response inline.\n\n(I have been saving replies to inline comments but didn\u0027t realize that I had to explicitly review/publish them.  FWIW, I just published all old ones as well. )","accounts_in_message":[],"_revision_number":12},{"id":"fdfb40fb5ed191fd65c2ab5ef2d14f413812872f","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-08 15:54:59.000000000","message":"Patch Set 2: (28 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"723f048a76d8831c78ad2b67184b718cb070018d","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-08 15:55:38.000000000","message":"Patch Set 7: (4 inline comments)\n\n","accounts_in_message":[],"_revision_number":7},{"id":"03ec99b793a02e73ac8f198dedb0ccc9b3bab07b","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-08 15:55:56.000000000","message":"Patch Set 11: (3 inline comments)\n\n","accounts_in_message":[],"_revision_number":11},{"id":"5476471178a41cbeabdbe7ce60d75e268012017c","author":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"date":"2013-07-08 16:31:32.000000000","message":"Patch Set 13: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":13},{"id":"23580d7f1c920fc7d57590664e99bfce87b898b7","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-07-08 17:06:18.000000000","message":"Patch Set 13: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nNits on single line docstring inline, other than that I think it\u0027s about ready to go.","accounts_in_message":[],"_revision_number":13},{"id":"8dfdc864a598de094b4c3f21bbf50d33c2685899","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-08 18:46:41.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c3a6fdcf1ac28c0446b1c4a32c8f9c85bd5951f1","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-08 19:27:54.000000000","message":"Patch Set 14: Works for me\n\nSmokeStack Results (patch set 14):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/97445\n","accounts_in_message":[],"_revision_number":14},{"id":"65cb0aab4ce25a60372ca2af3eafa696a8aa8a25","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-08 19:40:49.000000000","message":"Patch Set 14: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34290/14/check/gate-cinder-pep8/4702 : SUCCESS in 2m 12s\n- http://docs-draft.openstack.org/34290/14/check/gate-cinder-docs/4596/doc/build/html/ : SUCCESS in 4m 22s\n- http://logs.openstack.org/34290/14/check/gate-cinder-python26/4613 : SUCCESS in 6m 33s\n- http://logs.openstack.org/34290/14/check/gate-cinder-python27/4628 : SUCCESS in 3m 22s\n- http://logs.openstack.org/34290/14/check/gate-cinder-pylint/874 : SUCCESS in 3m 47s (non-voting)\n- http://logs.openstack.org/34290/14/check/gate-tempest-devstack-vm-full/29367 : SUCCESS in 53m 05s\n- http://logs.openstack.org/34290/14/check/gate-tempest-devstack-vm-postgres-full/29073 : SUCCESS in 54m 00s\n- http://logs.openstack.org/34290/14/check/gate-tempest-devstack-vm-neutron/465 : SUCCESS in 17m 05s\n- http://logs.openstack.org/34290/14/check/gate-grenade-devstack-vm/15606 : FAILURE in 11m 24s (non-voting)\n","accounts_in_message":[],"_revision_number":14},{"id":"6f7545d26a2ba6a2127e97449ab7a239435a21fc","author":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"date":"2013-07-09 09:32:55.000000000","message":"Patch Set 14: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":14},{"id":"1a10b202544965b540a92bd7abb08c198f7839c8","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-09 16:13:36.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"e06113a268c83f8a8dcda5c42937739c8887f498","author":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"date":"2013-07-09 16:14:49.000000000","message":"Patch Set 14: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":14},{"id":"a7a4c1136a7e7ba5dcb1d065723209e68d61749f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-09 17:08:27.000000000","message":"Patch Set 15: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34290/15/check/gate-cinder-pep8/4741 : SUCCESS in 2m 11s\n- http://docs-draft.openstack.org/34290/15/check/gate-cinder-docs/4635/doc/build/html/ : SUCCESS in 4m 47s\n- http://logs.openstack.org/34290/15/check/gate-cinder-python26/4647 : SUCCESS in 7m 55s\n- http://logs.openstack.org/34290/15/check/gate-cinder-python27/4667 : SUCCESS in 3m 30s\n- http://logs.openstack.org/34290/15/check/gate-cinder-pylint/885 : SUCCESS in 4m 42s (non-voting)\n- http://logs.openstack.org/34290/15/check/gate-tempest-devstack-vm-full/29857 : SUCCESS in 44m 01s\n- http://logs.openstack.org/34290/15/check/gate-tempest-devstack-vm-postgres-full/29563 : SUCCESS in 54m 41s\n- http://logs.openstack.org/34290/15/check/gate-tempest-devstack-vm-neutron/1020 : SUCCESS in 19m 25s\n- http://logs.openstack.org/34290/15/check/gate-grenade-devstack-vm/16037 : FAILURE in 20m 59s (non-voting)\n","accounts_in_message":[],"_revision_number":15},{"id":"f80dd2154cda57e5d4aa158b84d9c8343b7b4154","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-09 19:32:03.000000000","message":"Patch Set 15: Works for me\n\nSmokeStack Results (patch set 15):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/97961\n","accounts_in_message":[],"_revision_number":15},{"id":"0576aa2145a137fc3f60a96793ef32b9a4462c3e","author":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"date":"2013-07-10 04:29:54.000000000","message":"Patch Set 15: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":15},{"id":"c6c42fd9c93902f687aa2428fd9ecae07902e6f2","author":{"_account_id":1773,"name":"oliver-leahy-l","email":"oliver.leahy@hpe.com","username":"oliver-leahy-l"},"date":"2013-07-10 09:39:10.000000000","message":"Patch Set 15: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":15},{"id":"d01ef82c969d654f6f98521a0f95dbdf85caea28","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-07-10 22:41:31.000000000","message":"Patch Set 15: Looks good to me (core reviewer); Approved\n\nCool stuff Dinesh, thanks!","accounts_in_message":[],"_revision_number":15},{"id":"b21f0cf0ab01ee4df3fa4766599f1e0be6224c86","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-10 22:41:34.000000000","message":"Patch Set 15: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":15},{"id":"d12708be8188989d1aa4eec1596c665c3b20424b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-11 02:46:19.000000000","message":"Patch Set 15: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://docs-draft.openstack.org/34290/15/gate/gate-cinder-docs/4722/doc/build/html/ : SUCCESS in 5m 09s\n- http://logs.openstack.org/34290/15/gate/gate-cinder-pep8/4828 : SUCCESS in 2m 51s\n- http://logs.openstack.org/34290/15/gate/gate-cinder-python26/4724 : SUCCESS in 6m 12s\n- http://logs.openstack.org/34290/15/gate/gate-cinder-python27/4754 : SUCCESS in 4m 24s\n- http://logs.openstack.org/34290/15/gate/gate-tempest-devstack-vm-full/30658 : FAILURE in 23m 50s\n- http://logs.openstack.org/34290/15/gate/gate-tempest-devstack-vm-postgres-full/30361 : FAILURE in 22m 53s\n- http://logs.openstack.org/34290/15/gate/gate-tempest-devstack-vm-neutron/1916 : FAILURE in 24m 35s\n- http://logs.openstack.org/34290/15/gate/gate-grenade-devstack-vm/16792 : FAILURE in 31m 52s (non-voting)\n","accounts_in_message":[],"_revision_number":15},{"id":"38aa7d6b64ccc61da343f03e7b61c7f5fcaa90e9","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-07-11 03:09:23.000000000","message":"Patch Set 15: No score; No score\n\nI think you\u0027re going to be hung up until this https://review.openstack.org/#/c/36559/\n\nlands and you do a rebase off of it...","accounts_in_message":[],"_revision_number":15},{"id":"91499c029fd38c37027987f945f6f6d03632f13f","author":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"date":"2013-07-11 11:34:06.000000000","message":"Patch Set 15: Approved\n\nreverify no bug","accounts_in_message":[],"_revision_number":15},{"id":"4dc94e8a4676346afff5c17d7850f7cd28f45a6f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-11 11:34:08.000000000","message":"Patch Set 15: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":15},{"id":"b3b50ad0dba8eabd411020a80d57218d320efaa2","author":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"date":"2013-07-11 11:43:52.000000000","message":"Patch Set 15:\n\nreverify no bug","accounts_in_message":[],"_revision_number":15},{"id":"81db4d6e38520c24672ed9fa752cfd17a27b396d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-11 13:31:58.000000000","message":"Patch Set 15: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/34290/15/gate/gate-cinder-docs/4782/doc/build/html/ : SUCCESS in 6m 00s\n- http://logs.openstack.org/34290/15/gate/gate-cinder-pep8/4888 : SUCCESS in 2m 41s\n- http://logs.openstack.org/34290/15/gate/gate-cinder-python26/4771 : SUCCESS in 5m 13s\n- http://logs.openstack.org/34290/15/gate/gate-cinder-python27/4814 : SUCCESS in 3m 57s\n- http://logs.openstack.org/34290/15/gate/gate-tempest-devstack-vm-full/31105 : SUCCESS in 45m 04s\n- http://logs.openstack.org/34290/15/gate/gate-tempest-devstack-vm-postgres-full/30798 : SUCCESS in 56m 17s\n- http://logs.openstack.org/34290/15/gate/gate-tempest-devstack-vm-neutron/2398 : SUCCESS in 20m 27s\n- http://logs.openstack.org/34290/15/gate/gate-grenade-devstack-vm/17180 : FAILURE in 25m 41s (non-voting)\n","accounts_in_message":[],"_revision_number":15},{"id":"3d9fb557c1c935ee9a17580058629a7db5536f6f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-11 13:31:59.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"b58911a95c4399bf54ece6ecc88d3e4417f08fe1","revisions":{"5d13e37b33f2e8ca833b62712cfef91baa5ec8c8":{"kind":"REWORK","_number":1,"created":"2013-06-25 01:29:14.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/1"}}},"commit":{"parents":[{"commit":"6f723a6b26090f3a04bc45d57df5997e20849914","subject":"Merge \"Remove usage of locals() for formatting from cinder.tests.*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f723a6b26090f3a04bc45d57df5997e20849914"}]}],"author":{"name":"Dinesh Subhraveti","email":"subhraveti@gmail.com","date":"2013-06-25 01:15:24.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"subhraveti@gmail.com","date":"2013-06-25 01:15:24.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism.  Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d13e37b33f2e8ca833b62712cfef91baa5ec8c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d13e37b33f2e8ca833b62712cfef91baa5ec8c8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6f723a6b26090f3a04bc45d57df5997e20849914","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"55bae7eb28689aff77c587ed4f89699d2ec11eb2":{"kind":"REWORK","_number":2,"created":"2013-06-25 20:22:22.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/2"}}},"commit":{"parents":[{"commit":"9d881790df49f8e602571a20b0b9a98fb2056ece","subject":"Merge \"Imported Translations from Transifex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d881790df49f8e602571a20b0b9a98fb2056ece"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-25 19:45:27.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-25 20:00:23.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism.  Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55bae7eb28689aff77c587ed4f89699d2ec11eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55bae7eb28689aff77c587ed4f89699d2ec11eb2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9d881790df49f8e602571a20b0b9a98fb2056ece","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c1f8c457e8c6d4801640115ca9640e59d5459f96":{"kind":"REWORK","_number":3,"created":"2013-06-26 21:55:56.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/3"}}},"commit":{"parents":[{"commit":"5e8ab1cd857eb5cedfd367251f3d6fa34fb6c861","subject":"Content-length missing in put_object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e8ab1cd857eb5cedfd367251f3d6fa34fb6c861"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-25 19:45:27.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-26 21:50:01.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism.  Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1f8c457e8c6d4801640115ca9640e59d5459f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1f8c457e8c6d4801640115ca9640e59d5459f96"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5e8ab1cd857eb5cedfd367251f3d6fa34fb6c861","is_merged_in_target_branch":true,"change_id":"I4c56823d70fcd1a801a75bfd14f474a2780dc1ac","change_number":33639,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"50d5c6d585c0f7363074bd3e4c2b774d17608a08":{"kind":"REWORK","_number":4,"created":"2013-06-27 19:47:38.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/4"}}},"commit":{"parents":[{"commit":"4c506246223b1ace10101a65521600233f45b430","subject":"Merge \"Fix check_for_setup_error for sheepdog driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c506246223b1ace10101a65521600233f45b430"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-25 19:45:27.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-27 19:41:34.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism.  Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50d5c6d585c0f7363074bd3e4c2b774d17608a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50d5c6d585c0f7363074bd3e4c2b774d17608a08"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4c506246223b1ace10101a65521600233f45b430","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e913eae0d6441b39eefe378f54ecaac9b12c59bc":{"kind":"REWORK","_number":5,"created":"2013-06-27 20:42:15.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/5"}}},"commit":{"parents":[{"commit":"4c506246223b1ace10101a65521600233f45b430","subject":"Merge \"Fix check_for_setup_error for sheepdog driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c506246223b1ace10101a65521600233f45b430"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-25 19:45:27.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-27 20:35:37.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism.  Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e913eae0d6441b39eefe378f54ecaac9b12c59bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e913eae0d6441b39eefe378f54ecaac9b12c59bc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4c506246223b1ace10101a65521600233f45b430","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a75d5efd914eb9a88b9504c54b9f2b8e22e0af2f":{"kind":"REWORK","_number":6,"created":"2013-06-28 17:29:07.000000000","uploader":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"ref":"refs/changes/90/34290/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/6"}}},"commit":{"parents":[{"commit":"066a6760f9c7c64411b40918fe7f5c218b391218","subject":"Merge \"Fix service alive information in os-services extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/066a6760f9c7c64411b40918fe7f5c218b391218"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-25 19:45:27.000000000","tz":-420},"committer":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-06-28 17:27:59.000000000","tz":-360},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism.  Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\n!!!!!!!!!!!!!!!!!!\n! Experimental mod\n!!!!!!!!!!!!!!!!!!\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a75d5efd914eb9a88b9504c54b9f2b8e22e0af2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a75d5efd914eb9a88b9504c54b9f2b8e22e0af2f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"066a6760f9c7c64411b40918fe7f5c218b391218","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"182481c101136ed66a34c63249600c4350514a23":{"kind":"REWORK","_number":7,"created":"2013-06-28 17:48:18.000000000","uploader":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"ref":"refs/changes/90/34290/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/7"}}},"commit":{"parents":[{"commit":"066a6760f9c7c64411b40918fe7f5c218b391218","subject":"Merge \"Fix service alive information in os-services extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/066a6760f9c7c64411b40918fe7f5c218b391218"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-25 19:45:27.000000000","tz":-420},"committer":{"name":"John Griffith","email":"john.griffith@solidfire.com","date":"2013-06-28 17:48:05.000000000","tz":-360},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism.  Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/182481c101136ed66a34c63249600c4350514a23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/182481c101136ed66a34c63249600c4350514a23"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"066a6760f9c7c64411b40918fe7f5c218b391218","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7269c1b8c759593513db5a73abec2c37339b0892":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2013-07-01 23:44:03.000000000","uploader":{"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},"ref":"refs/changes/90/34290/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/8"}}},"commit":{"parents":[{"commit":"e102c28b6407914ccb5d9d1837ecd43e494a0bd2","subject":"Merge \"Calculate count for customized dd blocksize\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e102c28b6407914ccb5d9d1837ecd43e494a0bd2"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-06-25 19:45:27.000000000","tz":-420},"committer":{"name":"Bill Owen","email":"billowen@us.ibm.com","date":"2013-07-01 23:43:52.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism.  Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7269c1b8c759593513db5a73abec2c37339b0892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7269c1b8c759593513db5a73abec2c37339b0892"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e102c28b6407914ccb5d9d1837ecd43e494a0bd2","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"be846cb82d6d1333baa5f0ee507275e8948f032f":{"kind":"REWORK","_number":9,"created":"2013-07-02 00:42:35.000000000","uploader":{"_account_id":7689,"name":"Bill Owen","email":"billowen@us.ibm.com","username":"billowen"},"ref":"refs/changes/90/34290/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/9"}}},"commit":{"parents":[{"commit":"572e34592c6f2afb4b48c8cc84f4514458281d2d","subject":"Merge \"Added Cinder volume backup to Ceph support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/572e34592c6f2afb4b48c8cc84f4514458281d2d"}]}],"author":{"name":"Bill Owen","email":"billowen@us.ibm.com","date":"2013-07-02 00:29:34.000000000","tz":-420},"committer":{"name":"Bill Owen","email":"billowen@us.ibm.com","date":"2013-07-02 00:42:04.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism. Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be846cb82d6d1333baa5f0ee507275e8948f032f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be846cb82d6d1333baa5f0ee507275e8948f032f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"572e34592c6f2afb4b48c8cc84f4514458281d2d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ba00533ebad72ad2383935017299f471f394a700":{"kind":"REWORK","_number":10,"created":"2013-07-02 23:55:11.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/10"}}},"commit":{"parents":[{"commit":"5e3b00e6d37b1944225aed7c3600e18abc713536","subject":"Merge \"Adding host attaching support to Cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e3b00e6d37b1944225aed7c3600e18abc713536"}]}],"author":{"name":"Bill Owen","email":"billowen@us.ibm.com","date":"2013-07-02 00:29:34.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-02 23:46:09.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism. Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba00533ebad72ad2383935017299f471f394a700"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba00533ebad72ad2383935017299f471f394a700"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5e3b00e6d37b1944225aed7c3600e18abc713536","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"33aaabe10c145bc266773f8d5f49d114ef3ea919":{"kind":"REWORK","_number":11,"created":"2013-07-03 23:44:42.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/11"}}},"commit":{"parents":[{"commit":"3b0921fbdc9e2fdf2b35801143fddbf09a9b84e2","subject":"Merge \"Skip brick_initiator test in virtual environments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3b0921fbdc9e2fdf2b35801143fddbf09a9b84e2"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-03 22:56:42.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-03 23:06:10.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism. Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33aaabe10c145bc266773f8d5f49d114ef3ea919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33aaabe10c145bc266773f8d5f49d114ef3ea919"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3b0921fbdc9e2fdf2b35801143fddbf09a9b84e2","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a030f5c208228805dbf287d81e22dc2807900d32":{"kind":"REWORK","_number":12,"created":"2013-07-04 16:37:28.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/12"}}},"commit":{"parents":[{"commit":"3b0921fbdc9e2fdf2b35801143fddbf09a9b84e2","subject":"Merge \"Skip brick_initiator test in virtual environments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3b0921fbdc9e2fdf2b35801143fddbf09a9b84e2"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-03 22:56:42.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-04 16:29:21.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism. Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a030f5c208228805dbf287d81e22dc2807900d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a030f5c208228805dbf287d81e22dc2807900d32"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3b0921fbdc9e2fdf2b35801143fddbf09a9b84e2","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"54a061014858abeadb9db2d69e0a935a34ec5b7d":{"kind":"REWORK","_number":13,"created":"2013-07-07 00:08:13.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/13"}}},"commit":{"parents":[{"commit":"faffe83d88276633fb701996b0e44f34b7a2daa0","subject":"Merge \"Add check for snapshot to Brick LVM.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/faffe83d88276633fb701996b0e44f34b7a2daa0"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-03 22:56:42.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-06 23:54:09.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism. Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54a061014858abeadb9db2d69e0a935a34ec5b7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54a061014858abeadb9db2d69e0a935a34ec5b7d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"faffe83d88276633fb701996b0e44f34b7a2daa0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e8e3dd38e10bd59363816d096a0e47093a193f9b":{"kind":"REWORK","_number":14,"created":"2013-07-08 18:46:41.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/14"}}},"commit":{"parents":[{"commit":"ca0e7295f44235c7d64954a225a172718a074dd8","subject":"Prevent wrongly privilege escalation of a context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca0e7295f44235c7d64954a225a172718a074dd8"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-03 22:56:42.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-08 18:38:31.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism. Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e8e3dd38e10bd59363816d096a0e47093a193f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e8e3dd38e10bd59363816d096a0e47093a193f9b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ca0e7295f44235c7d64954a225a172718a074dd8","is_merged_in_target_branch":true,"change_id":"I2822b1612ec741209b278fc65f18d0f8f3243e0a","change_number":35496,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b58911a95c4399bf54ece6ecc88d3e4417f08fe1":{"kind":"REWORK","_number":15,"created":"2013-07-09 16:13:36.000000000","uploader":{"_account_id":7201,"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","username":"subhraveti"},"ref":"refs/changes/90/34290/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/34290/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/34290/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/34290/15"}}},"commit":{"parents":[{"commit":"811d5145cdc4117eb072b355c5022fea4a735aa8","subject":"Merge \"Fix extent size issue when creating thin pool.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/811d5145cdc4117eb072b355c5022fea4a735aa8"}]}],"author":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-03 22:56:42.000000000","tz":-420},"committer":{"name":"Dinesh Subhraveti","email":"dineshs@us.ibm.com","date":"2013-07-09 16:05:17.000000000","tz":-420},"subject":"Add support for storing volumes on GPFS","message":"Add support for storing volumes on GPFS\n\nKey functions include efficient creation of snapshots and clones using\nfile system level copy-on-write mechanism. Copying of image blocks is\nalso avoided in case Glance backend repository happens to be on GPFS.\n\nChange-Id: I0a532fe06798d07f21d79ee45444bb9c0a025430\nImplements: blueprint gpfs-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b58911a95c4399bf54ece6ecc88d3e4417f08fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b58911a95c4399bf54ece6ecc88d3e4417f08fe1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"811d5145cdc4117eb072b355c5022fea4a735aa8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
