)]}'
{"id":"openstack%2Fcinder~36211","triplet_id":"openstack%2Fcinder~master~I760e8f419be529cda969e848f151a8cab1053dce","project":"openstack/cinder","branch":"master","topic":"master","hashtags":[],"change_id":"I760e8f419be529cda969e848f151a8cab1053dce","subject":"Add unit tests for cinder/api/versions","status":"MERGED","created":"2013-07-09 11:50:46.000000000","updated":"2013-07-11 11:34:35.000000000","submitted":"2013-07-11 11:34:34.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2b53561aecb331fae0e53fdd904f3b833c6f9ee3","_number":36211,"virtual_id_number":36211,"owner":{"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-07-11 11:34:34.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},{"value":1,"date":"2013-07-10 14:08:04.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":6507,"name":"Shturm Svetlana","email":"svetlana.v.shturm@gmail.com","username":"sshturm"},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":0,"_account_id":6509,"name":"Roman Bogorodskiy","email":"rbogorodskiy@mirantis.com","username":"novel"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-07-11 01:07:19.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"date":"2013-07-11 09:14:49.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":2,"date":"2013-07-11 03:00:31.000000000","_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":6507,"name":"Shturm Svetlana","email":"svetlana.v.shturm@gmail.com","username":"sshturm"},{"value":1,"date":"2013-07-11 09:46:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":0,"_account_id":6509,"name":"Roman Bogorodskiy","email":"rbogorodskiy@mirantis.com","username":"novel"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-07-11 01:07:19.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":1,"date":"2013-07-11 03:00:31.000000000","_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"date":"2013-07-10 14:45:21.000000000","_account_id":6507,"name":"Shturm Svetlana","email":"svetlana.v.shturm@gmail.com","username":"sshturm"},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":0,"date":"2013-07-10 14:44:39.000000000","_account_id":6509,"name":"Roman Bogorodskiy","email":"rbogorodskiy@mirantis.com","username":"novel"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"value":0,"_account_id":6507,"name":"Shturm Svetlana","email":"svetlana.v.shturm@gmail.com","username":"sshturm"},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":0,"_account_id":6509,"name":"Roman Bogorodskiy","email":"rbogorodskiy@mirantis.com","username":"novel"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"_account_id":6507,"name":"Shturm Svetlana","email":"svetlana.v.shturm@gmail.com","username":"sshturm"},{"_account_id":6509,"name":"Roman Bogorodskiy","email":"rbogorodskiy@mirantis.com","username":"novel"},{"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-07-10 14:08:04.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-07-10 14:44:39.000000000","updated_by":{"_account_id":6509,"name":"Roman Bogorodskiy","email":"rbogorodskiy@mirantis.com","username":"novel"},"reviewer":{"_account_id":6509,"name":"Roman Bogorodskiy","email":"rbogorodskiy@mirantis.com","username":"novel"},"state":"REVIEWER"},{"updated":"2013-07-10 14:45:21.000000000","updated_by":{"_account_id":6507,"name":"Shturm Svetlana","email":"svetlana.v.shturm@gmail.com","username":"sshturm"},"reviewer":{"_account_id":6507,"name":"Shturm Svetlana","email":"svetlana.v.shturm@gmail.com","username":"sshturm"},"state":"REVIEWER"},{"updated":"2013-07-11 01:07:19.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2013-07-11 03:00:31.000000000","updated_by":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"reviewer":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"state":"REVIEWER"},{"updated":"2013-07-11 09:46:14.000000000","updated_by":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"reviewer":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"state":"REVIEWER"},{"updated":"2013-07-11 11:34:34.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"2f3f1f983a4c2f819193268a53d12bb3d7fb46e0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-09 12:34:36.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/36211/1/check/gate-cinder-pep8/4729 : FAILURE in 2m 03s\n- http://docs-draft.openstack.org/36211/1/check/gate-cinder-docs/4623/doc/build/html/ : SUCCESS in 3m 58s\n- http://logs.openstack.org/36211/1/check/gate-cinder-python26/4635 : SUCCESS in 6m 26s\n- http://logs.openstack.org/36211/1/check/gate-cinder-python27/4655 : SUCCESS in 3m 44s\n- http://logs.openstack.org/36211/1/check/gate-cinder-pylint/879 : SUCCESS in 3m 34s (non-voting)\n- http://logs.openstack.org/36211/1/check/gate-tempest-devstack-vm-full/29751 : SUCCESS in 42m 26s\n- http://logs.openstack.org/36211/1/check/gate-tempest-devstack-vm-postgres-full/29457 : SUCCESS in 43m 42s\n- http://logs.openstack.org/36211/1/check/gate-tempest-devstack-vm-neutron/903 : FAILURE in 21m 06s\n- http://logs.openstack.org/36211/1/check/gate-grenade-devstack-vm/15941 : FAILURE in 13m 17s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"b4ce42b12753b467efaa05a28d1ea2a75ac9ffc0","author":{"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},"date":"2013-07-10 11:52:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"62b3c73df82583b4733ae18a64e4d049ce583709","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-10 12:15:25.000000000","message":"Patch Set 2: Works for me\n\nSmokeStack Results (patch set 2):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/98372\n","accounts_in_message":[],"_revision_number":2},{"id":"96cfbabc4ac1ebcfe209365fffa1e131a74ff1c7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-10 12:47:14.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/36211/2/check/gate-cinder-pep8/4781 : FAILURE in 2m 04s\n- http://docs-draft.openstack.org/36211/2/check/gate-cinder-docs/4675/doc/build/html/ : SUCCESS in 3m 54s\n- http://logs.openstack.org/36211/2/check/gate-cinder-python26/4686 : SUCCESS in 5m 46s\n- http://logs.openstack.org/36211/2/check/gate-cinder-python27/4707 : SUCCESS in 3m 06s\n- http://logs.openstack.org/36211/2/check/gate-cinder-pylint/912 : SUCCESS in 3m 31s (non-voting)\n- http://logs.openstack.org/36211/2/check/gate-tempest-devstack-vm-full/30288 : SUCCESS in 43m 31s\n- http://logs.openstack.org/36211/2/check/gate-tempest-devstack-vm-postgres-full/29992 : SUCCESS in 54m 31s\n- http://logs.openstack.org/36211/2/check/gate-tempest-devstack-vm-neutron/1500 : SUCCESS in 17m 16s\n- http://logs.openstack.org/36211/2/check/gate-grenade-devstack-vm/16439 : FAILURE in 22m 45s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"7a00e416067ab99a5dd388e8674c98f8c71dccd4","author":{"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},"date":"2013-07-10 13:16:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7715dc93301090fa54e9dd137a2b27d576ace00a","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-10 14:08:04.000000000","message":"Patch Set 3: Works for me\n\nSmokeStack Results (patch set 3):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/98427\n","accounts_in_message":[],"_revision_number":3},{"id":"7c90f8c6b032c751aa6431f1dae15195cef56678","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-10 14:12:14.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/36211/3/check/gate-cinder-pep8/4783 : SUCCESS in 2m 11s\n- http://docs-draft.openstack.org/36211/3/check/gate-cinder-docs/4677/doc/build/html/ : SUCCESS in 5m 09s\n- http://logs.openstack.org/36211/3/check/gate-cinder-python26/4688 : SUCCESS in 6m 50s\n- http://logs.openstack.org/36211/3/check/gate-cinder-python27/4709 : SUCCESS in 3m 04s\n- http://logs.openstack.org/36211/3/check/gate-cinder-pylint/914 : SUCCESS in 3m 36s (non-voting)\n- http://logs.openstack.org/36211/3/check/gate-tempest-devstack-vm-full/30320 : SUCCESS in 43m 13s\n- http://logs.openstack.org/36211/3/check/gate-tempest-devstack-vm-postgres-full/30024 : SUCCESS in 55m 35s\n- http://logs.openstack.org/36211/3/check/gate-tempest-devstack-vm-neutron/1535 : SUCCESS in 17m 40s\n- http://logs.openstack.org/36211/3/check/gate-grenade-devstack-vm/16470 : FAILURE in 25m 52s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"3b417c5ffe9590d4399e30de4bbfb5956aa40ed3","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-07-11 01:07:19.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"0ec06f19ebb38cedda3d3f7cbc1ef02405f03969","author":{"_account_id":170,"name":"Mike Perez","email":"thingee@gmail.com","username":"thingee"},"date":"2013-07-11 03:00:31.000000000","message":"Patch Set 3: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":3},{"id":"93ee2bf513167a9c317e10152aeb419084cd1d68","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-11 03:00:33.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"d362df238285c2d1796f04d13265c91978bb1fe6","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2013-07-11 05:29:31.000000000","message":"Patch Set 3:\n\nreverify no bug","accounts_in_message":[],"_revision_number":3},{"id":"1f865b713ffaaeac65056f0992bf5434624a8ac3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-11 05:33:37.000000000","message":"Patch Set 3:\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"bb4ed9528423cafae8a393cde8ccc1c76cc3abd7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-11 08:34:59.000000000","message":"Patch Set 3: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://docs-draft.openstack.org/36211/3/gate/gate-cinder-docs/4758/doc/build/html/ : SUCCESS in 4m 37s\n- http://logs.openstack.org/36211/3/gate/gate-cinder-pep8/4864 : SUCCESS in 2m 15s\n- http://logs.openstack.org/36211/3/gate/gate-cinder-python26/4751 : SUCCESS in 5m 03s\n- http://logs.openstack.org/36211/3/gate/gate-cinder-python27/4790 : SUCCESS in 3m 17s\n- http://logs.openstack.org/36211/3/gate/gate-tempest-devstack-vm-full/30943 : FAILURE in 13m 48s\n- http://logs.openstack.org/36211/3/gate/gate-tempest-devstack-vm-postgres-full/30642 : FAILURE in 16m 29s\n- http://logs.openstack.org/36211/3/gate/gate-tempest-devstack-vm-neutron/2220 : FAILURE in 4s\n- http://logs.openstack.org/36211/3/gate/gate-grenade-devstack-vm/17051 : FAILURE in 4s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"458376e4bbc0a61e26f1f9cc7e43a3b0bf37b9ca","author":{"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},"date":"2013-07-11 09:14:49.000000000","message":"Patch Set 3:\n\nre-check, no bug","accounts_in_message":[],"_revision_number":3},{"id":"21ae347cbcb321ce6c6e6efedfc8c698d1ac72d3","author":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"date":"2013-07-11 09:46:14.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\nreverify no bug","accounts_in_message":[],"_revision_number":3},{"id":"20495c195f5576f142272b4a38d93b576dff4c4c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-11 09:46:15.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"b9b0fde95c74f0a054b81adc226975b9b898ae60","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-11 11:34:34.000000000","message":"Patch Set 3: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/36211/3/gate/gate-cinder-docs/4774/doc/build/html/ : SUCCESS in 5m 14s\n- http://logs.openstack.org/36211/3/gate/gate-cinder-pep8/4880 : SUCCESS in 2m 46s\n- http://logs.openstack.org/36211/3/gate/gate-cinder-python26/4766 : SUCCESS in 5m 30s\n- http://logs.openstack.org/36211/3/gate/gate-cinder-python27/4806 : SUCCESS in 3m 51s\n- http://logs.openstack.org/36211/3/gate/gate-tempest-devstack-vm-full/31054 : SUCCESS in 43m 48s\n- http://logs.openstack.org/36211/3/gate/gate-tempest-devstack-vm-postgres-full/30747 : SUCCESS in 44m 47s\n- http://logs.openstack.org/36211/3/gate/gate-tempest-devstack-vm-neutron/2341 : SUCCESS in 17m 20s\n- http://logs.openstack.org/36211/3/gate/gate-grenade-devstack-vm/17136 : FAILURE in 25m 51s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"0c458ff16f71cde2cbddacef0cb37b7ac99e23aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-11 11:34:35.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b3e3c8e6bd6b4184eee331e3133b39bc8103f27d","revisions":{"a1865f4982e62d71ce77d646c81f2e1ae54f7166":{"kind":"REWORK","_number":1,"created":"2013-07-09 11:50:46.000000000","uploader":{"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},"ref":"refs/changes/11/36211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/36211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/36211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/36211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/36211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/36211/1"}}},"commit":{"parents":[{"commit":"e95839466898e5606263721887633b2f66a360c0","subject":"Merge \"Use oslo.sphinx and remove local copy of doc theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e95839466898e5606263721887633b2f66a360c0"}]}],"author":{"name":"Jasakov Artem","email":"ayasakov@mirantis.com","date":"2013-07-09 11:43:42.000000000","tz":240},"committer":{"name":"Jasakov Artem","email":"ayasakov@mirantis.com","date":"2013-07-09 11:50:38.000000000","tz":240},"subject":"Add unit tests for cinder/api/versions","message":"Add unit tests for cinder/api/versions\n\nAdded to the following unit tests:\n  Test action args;\n  Test get most recent update;\n  Test create version entry;\n  Test create feed.\n\nChange-Id: I760e8f419be529cda969e848f151a8cab1053dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1865f4982e62d71ce77d646c81f2e1ae54f7166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1865f4982e62d71ce77d646c81f2e1ae54f7166"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e95839466898e5606263721887633b2f66a360c0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"0038af994b27825af568ef0aac434b46d3fbac94":{"kind":"REWORK","_number":2,"created":"2013-07-10 11:52:13.000000000","uploader":{"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},"ref":"refs/changes/11/36211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/36211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/36211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/36211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/36211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/36211/2"}}},"commit":{"parents":[{"commit":"2f5e26a247665f890cdf26773b5e7070890594d8","subject":"Merge \"Add interface class for backup drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f5e26a247665f890cdf26773b5e7070890594d8"}]}],"author":{"name":"Jasakov Artem","email":"ayasakov@mirantis.com","date":"2013-07-10 11:50:44.000000000","tz":240},"committer":{"name":"Jasakov Artem","email":"ayasakov@mirantis.com","date":"2013-07-10 11:51:58.000000000","tz":240},"subject":"Add unit tests for cinder/api/versions","message":"Add unit tests for cinder/api/versions\n\nAdded to the following unit tests:\n  Test action args;\n  Test get most recent update;\n  Test create version entry;\n  Test create feed.\n\nChange-Id: I760e8f419be529cda969e848f151a8cab1053dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0038af994b27825af568ef0aac434b46d3fbac94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0038af994b27825af568ef0aac434b46d3fbac94"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2f5e26a247665f890cdf26773b5e7070890594d8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b3e3c8e6bd6b4184eee331e3133b39bc8103f27d":{"kind":"REWORK","_number":3,"created":"2013-07-10 13:16:25.000000000","uploader":{"_account_id":8054,"name":"Artem Yasakov","email":"ayasakov@mirantis.com","username":"ayasakov"},"ref":"refs/changes/11/36211/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/36211/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/36211/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/36211/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/36211/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/36211/3"}}},"commit":{"parents":[{"commit":"2f5e26a247665f890cdf26773b5e7070890594d8","subject":"Merge \"Add interface class for backup drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f5e26a247665f890cdf26773b5e7070890594d8"}]}],"author":{"name":"Jasakov Artem","email":"ayasakov@mirantis.com","date":"2013-07-10 13:15:24.000000000","tz":240},"committer":{"name":"Jasakov Artem","email":"ayasakov@mirantis.com","date":"2013-07-10 13:16:19.000000000","tz":240},"subject":"Add unit tests for cinder/api/versions","message":"Add unit tests for cinder/api/versions\n\nAdded to the following unit tests:\n  Test action args;\n  Test get most recent update;\n  Test create version entry;\n  Test create feed.\n\nChange-Id: I760e8f419be529cda969e848f151a8cab1053dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3e3c8e6bd6b4184eee331e3133b39bc8103f27d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3e3c8e6bd6b4184eee331e3133b39bc8103f27d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2f5e26a247665f890cdf26773b5e7070890594d8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
