)]}'
{"id":"openstack%2Fcinder~36302","triplet_id":"openstack%2Fcinder~master~Ie4b46339423e6a1ad3d8ca699a2412b825206c0e","project":"openstack/cinder","branch":"master","topic":"bug/1199453","hashtags":[],"change_id":"Ie4b46339423e6a1ad3d8ca699a2412b825206c0e","subject":"Make String column creation compatible with SQLAlchemy 0.8","status":"MERGED","created":"2013-07-09 17:35:23.000000000","updated":"2013-07-17 14:39:47.000000000","submitted":"2013-07-10 06:46:17.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4849f6d96a868c52d495eea768eacb919619e1df","_number":36302,"virtual_id_number":36302,"owner":{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":2,"date":"2013-07-10 06:46:17.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},{"value":1,"date":"2013-07-10 02:31:49.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":2,"date":"2013-07-10 02:55:52.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":1,"date":"2013-07-10 02:55:52.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-07-09 17:45:47.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2013-07-10 02:31:49.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-07-10 02:55:52.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2013-07-10 06:46:17.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f49c83d77d383a61cf9b98f93669fb746a90a09e","author":{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},"date":"2013-07-09 17:38:06.000000000","message":"Patch Set 1:\n\nNote that this changes migration scripts, which is something we only allow after careful consideration, so please don\u0027t +2 it without lots of thought.  SQLAlchemy 0.8 support is important but so is backward compatibility.","accounts_in_message":[],"_revision_number":1},{"id":"71df22d2d17717c78f002113b0822ad248b71da9","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-07-09 17:45:31.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\nNice quick turn around on that David, Thanks!!","accounts_in_message":[],"_revision_number":1},{"id":"e1f7485fa85b31d93027825937d4926b12afb10e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-09 18:19:03.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/36302/1/check/gate-cinder-pep8/4742 : SUCCESS in 2m 06s\n- http://docs-draft.openstack.org/36302/1/check/gate-cinder-docs/4636/doc/build/html/ : SUCCESS in 4m 11s\n- http://logs.openstack.org/36302/1/check/gate-cinder-python26/4648 : SUCCESS in 7m 26s\n- http://logs.openstack.org/36302/1/check/gate-cinder-python27/4668 : SUCCESS in 3m 05s\n- http://logs.openstack.org/36302/1/check/gate-cinder-pylint/886 : SUCCESS in 3m 29s (non-voting)\n- http://logs.openstack.org/36302/1/check/gate-tempest-devstack-vm-full/29889 : SUCCESS in 42m 28s\n- http://logs.openstack.org/36302/1/check/gate-tempest-devstack-vm-postgres-full/29595 : SUCCESS in 43m 31s\n- http://logs.openstack.org/36302/1/check/gate-tempest-devstack-vm-neutron/1053 : FAILURE in 18m 20s\n- http://logs.openstack.org/36302/1/check/gate-grenade-devstack-vm/16068 : FAILURE in 25m 51s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"6cb5937e1cfc15cbb30388d6c83c6636a2d68674","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-07-10 01:48:30.000000000","message":"Patch Set 1: No score\n\nrecheck bug 1194026","accounts_in_message":[],"_revision_number":1},{"id":"41bf4c7a2d027fbb04f6dfe5adc3faec8033885a","author":{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},"date":"2013-07-10 01:59:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ee6dbd3766a4419df8d760a6b3dc014465f9936","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-10 02:31:49.000000000","message":"Patch Set 2: Works for me\n\nSmokeStack Results (patch set 2):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/98145\n","accounts_in_message":[],"_revision_number":2},{"id":"b65a2874ef2ecdd6475be104ee950635c4a04de1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-10 02:53:34.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/36302/2/check/gate-cinder-pep8/4759 : SUCCESS in 2m 07s\n- http://docs-draft.openstack.org/36302/2/check/gate-cinder-docs/4653/doc/build/html/ : SUCCESS in 4m 16s\n- http://logs.openstack.org/36302/2/check/gate-cinder-python26/4664 : SUCCESS in 7m 18s\n- http://logs.openstack.org/36302/2/check/gate-cinder-python27/4685 : SUCCESS in 4m 51s\n- http://logs.openstack.org/36302/2/check/gate-cinder-pylint/903 : SUCCESS in 4m 15s (non-voting)\n- http://logs.openstack.org/36302/2/check/gate-tempest-devstack-vm-full/30115 : SUCCESS in 53m 46s\n- http://logs.openstack.org/36302/2/check/gate-tempest-devstack-vm-postgres-full/29819 : SUCCESS in 43m 12s\n- http://logs.openstack.org/36302/2/check/gate-tempest-devstack-vm-neutron/1309 : SUCCESS in 20m 42s\n- http://logs.openstack.org/36302/2/check/gate-grenade-devstack-vm/16280 : FAILURE in 30m 39s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"d8912b8c8abcc411205a8d45558889c3a5d26480","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-07-10 02:55:52.000000000","message":"Patch Set 2: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":2},{"id":"7d11f82e2d4c05a03382e4f37a25f4edd24412fb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-10 02:55:54.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"f44428e540183388170d9bf5f225d84878f41d38","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-10 06:46:17.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/36302/2/gate/gate-cinder-docs/4668/doc/build/html/ : SUCCESS in 4m 10s\n- http://logs.openstack.org/36302/2/gate/gate-cinder-pep8/4774 : SUCCESS in 2m 12s\n- http://logs.openstack.org/36302/2/gate/gate-cinder-python26/4679 : SUCCESS in 7m 39s\n- http://logs.openstack.org/36302/2/gate/gate-cinder-python27/4700 : SUCCESS in 3m 17s\n- http://logs.openstack.org/36302/2/gate/gate-tempest-devstack-vm-full/30186 : SUCCESS in 53m 46s\n- http://logs.openstack.org/36302/2/gate/gate-tempest-devstack-vm-postgres-full/29890 : SUCCESS in 43m 05s\n- http://logs.openstack.org/36302/2/gate/gate-tempest-devstack-vm-neutron/1388 : SUCCESS in 17m 17s\n- http://logs.openstack.org/36302/2/gate/gate-grenade-devstack-vm/16346 : FAILURE in 21m 54s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"bff624cc07319b9a5a4d846161d8d4430d9662f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-10 06:46:18.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2},{"id":"22e923ba0fabf3345ef159ec36ba26c1961bd382","author":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"date":"2013-07-17 14:39:47.000000000","message":"Patch Set 2:\n\nFor what it\u0027s worth I think cinder master was already OK with sqlalchemy \u003e\u003d 0.8 due to change I305bfdbcd3c3669c41318ea34115624f729b3909\nbut that wasn\u0027t merged to grizzly.\n\nFor grizzly I\u0027d been patching in Red Hat distros with a less invasive patch to just remove the assert_unicode param:\nhttps://github.com/redhat-openstack/cinder/commit/7c0e9b09c\nI agree with this though to align more closely with nova","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2dbeafb4746c89734d70e15601f24aeb22b0d8f","revisions":{"1d5158ddc5bfc7656a3d2be55efca9ca9ef5e3b3":{"kind":"REWORK","_number":1,"created":"2013-07-09 17:35:23.000000000","uploader":{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},"ref":"refs/changes/02/36302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/02/36302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/36302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/36302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/36302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/02/36302/1"}}},"commit":{"parents":[{"commit":"811d5145cdc4117eb072b355c5022fea4a735aa8","subject":"Merge \"Fix extent size issue when creating thin pool.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/811d5145cdc4117eb072b355c5022fea4a735aa8"}]}],"author":{"name":"David Ripton","email":"dripton@redhat.com","date":"2013-07-09 17:32:08.000000000","tz":-240},"committer":{"name":"David Ripton","email":"dripton@redhat.com","date":"2013-07-09 17:35:22.000000000","tz":-240},"subject":"Make String column creation compatible with SQLAlchemy 0.8","message":"Make String column creation compatible with SQLAlchemy 0.8\n\n3 migration scripts used the convert_unicode, unicode_error, and\n_warn_on_bytestring arguments when creating String columns.  These\nseem to cause problems with SQLAlchemy 0.8.  Nova excised all use\nof such arguments in commit 93dec58156e when squashing migrations\nfor Grizzly, and seems to have no problems with SQLAlchemy 0.8.\n\nChange-Id: Ie4b46339423e6a1ad3d8ca699a2412b825206c0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d5158ddc5bfc7656a3d2be55efca9ca9ef5e3b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d5158ddc5bfc7656a3d2be55efca9ca9ef5e3b3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"811d5145cdc4117eb072b355c5022fea4a735aa8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f2dbeafb4746c89734d70e15601f24aeb22b0d8f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2013-07-10 01:59:32.000000000","uploader":{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},"ref":"refs/changes/02/36302/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/02/36302/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/36302/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/36302/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/36302/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/02/36302/2"}}},"commit":{"parents":[{"commit":"811d5145cdc4117eb072b355c5022fea4a735aa8","subject":"Merge \"Fix extent size issue when creating thin pool.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/811d5145cdc4117eb072b355c5022fea4a735aa8"}]}],"author":{"name":"David Ripton","email":"dripton@redhat.com","date":"2013-07-09 17:32:08.000000000","tz":-240},"committer":{"name":"David Ripton","email":"dripton@redhat.com","date":"2013-07-10 01:59:14.000000000","tz":-240},"subject":"Make String column creation compatible with SQLAlchemy 0.8","message":"Make String column creation compatible with SQLAlchemy 0.8\n\n3 migration scripts used the convert_unicode, unicode_error, and\n_warn_on_bytestring arguments when creating String columns.  These\nseem to cause problems with SQLAlchemy 0.8.  Nova excised all use\nof such arguments in commit 93dec58156e when squashing migrations\nfor Grizzly, and seems to have no problems with SQLAlchemy 0.8.\n\nFixes bug 1199453\n\nChange-Id: Ie4b46339423e6a1ad3d8ca699a2412b825206c0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2dbeafb4746c89734d70e15601f24aeb22b0d8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2dbeafb4746c89734d70e15601f24aeb22b0d8f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"811d5145cdc4117eb072b355c5022fea4a735aa8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
