)]}'
{"id":"openstack%2Fcinder~36834","triplet_id":"openstack%2Fcinder~master~I0396f1252c8faafff3e8b4f9a4aeffb930350a8d","project":"openstack/cinder","branch":"master","topic":"bug/1195788","hashtags":[],"change_id":"I0396f1252c8faafff3e8b4f9a4aeffb930350a8d","subject":"CoraidDriver: Allow volumes in error state to be deleted","status":"MERGED","created":"2013-07-12 13:48:41.000000000","updated":"2013-07-16 19:54:26.000000000","submitted":"2013-07-16 19:54:25.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dada4bbaf295a7817fbd0e99735d3c1529fbe036","_number":36834,"virtual_id_number":36834,"owner":{"_account_id":6798,"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","username":"jean-baptiste-ransy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-07-16 19:54:25.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":6798,"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","username":"jean-baptiste-ransy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-07-16 18:57:32.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":2,"date":"2013-07-16 18:53:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":6798,"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","username":"jean-baptiste-ransy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-07-16 18:57:32.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"date":"2013-07-16 18:53:19.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"date":"2013-07-16 17:41:25.000000000","_account_id":6798,"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","username":"jean-baptiste-ransy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":6798,"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","username":"jean-baptiste-ransy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":6798,"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","username":"jean-baptiste-ransy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-07-12 14:16:13.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-07-16 18:53:19.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2013-07-16 18:57:32.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2013-07-16 19:54:25.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"05d51570c7e75e4f2d53632a98890d0bcba82498","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-12 14:16:13.000000000","message":"Patch Set 1: Works for me\n\nSmokeStack Results (patch set 1):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/99629\n","accounts_in_message":[],"_revision_number":1},{"id":"0a35d377060fa10ccecb450eff10864a04f249fe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-12 14:44:13.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/36834/1/check/gate-cinder-pep8/4952 : SUCCESS in 3m 29s\n- http://docs-draft.openstack.org/36834/1/check/gate-cinder-docs/4846/doc/build/html/ : SUCCESS in 4m 35s\n- http://logs.openstack.org/36834/1/check/gate-cinder-python26/4830 : SUCCESS in 5m 34s\n- http://logs.openstack.org/36834/1/check/gate-cinder-python27/4878 : SUCCESS in 3m 36s\n- http://logs.openstack.org/36834/1/check/gate-cinder-pylint/1010 : SUCCESS in 4m 13s (non-voting)\n- http://logs.openstack.org/36834/1/check/gate-tempest-devstack-vm-full/31626 : SUCCESS in 55m 14s\n- http://logs.openstack.org/36834/1/check/gate-tempest-devstack-vm-postgres-full/31319 : SUCCESS in 44m 38s\n- http://logs.openstack.org/36834/1/check/gate-tempest-devstack-vm-neutron/2985 : SUCCESS in 17m 27s (non-voting)\n- http://logs.openstack.org/36834/1/check/gate-grenade-devstack-vm/17640 : FAILURE in 21m 16s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"9f750d09080610907548e023de35be57eea6eb63","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2013-07-16 17:30:03.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"6ef704498c0547944484a28972dd1743b4f806c5","author":{"_account_id":6798,"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","username":"jean-baptiste-ransy"},"date":"2013-07-16 17:41:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a1eac225074b494ab8188231a7056b3dce279a58","author":{"_account_id":6798,"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","username":"jean-baptiste-ransy"},"date":"2013-07-16 17:47:59.000000000","message":"Patch Set 1: (1 inline comment)\n\nAll done","accounts_in_message":[],"_revision_number":1},{"id":"e5bf13419a718e782e04a28e40952caace99ec9f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-16 18:33:10.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/36834/2/check/gate-cinder-pep8/5039 : SUCCESS in 2m 48s\n- http://docs-draft.openstack.org/36834/2/check/gate-cinder-docs/4933/doc/build/html/ : SUCCESS in 9m 06s\n- http://logs.openstack.org/36834/2/check/gate-cinder-python26/4918 : SUCCESS in 5m 27s\n- http://logs.openstack.org/36834/2/check/gate-cinder-python27/4965 : SUCCESS in 3m 30s\n- http://logs.openstack.org/36834/2/check/gate-cinder-pylint/1080 : SUCCESS in 3m 43s (non-voting)\n- http://logs.openstack.org/36834/2/check/gate-tempest-devstack-vm-full/32707 : SUCCESS in 51m 38s\n- http://logs.openstack.org/36834/2/check/gate-tempest-devstack-vm-cells-full/706 : FAILURE in 17m 38s (non-voting)\n- http://logs.openstack.org/36834/2/check/gate-tempest-devstack-vm-postgres-full/32399 : SUCCESS in 47m 47s\n- http://logs.openstack.org/36834/2/check/gate-tempest-devstack-vm-neutron/4276 : SUCCESS in 24m 38s (non-voting)\n- http://logs.openstack.org/36834/2/check/gate-grenade-devstack-vm/18599 : SUCCESS in 24m 57s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"a25c0c0e71c328acfe6f5bf9df891237f9953f53","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2013-07-16 18:53:19.000000000","message":"Patch Set 2: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"1d42aff4623516fe546fe19fa0d52bd650677149","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-07-16 18:57:32.000000000","message":"Patch Set 2: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":2},{"id":"ab9292262843fedf16eef9deef35e7cdb822539c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-16 18:57:34.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"fd499ea49a8dcb52cf472db4907c1fbf21b08363","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-16 19:54:25.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/36834/2/gate/gate-cinder-docs/4937/doc/build/html/ : SUCCESS in 4m 20s\n- http://logs.openstack.org/36834/2/gate/gate-cinder-pep8/5043 : SUCCESS in 2m 10s\n- http://logs.openstack.org/36834/2/gate/gate-cinder-python26/4922 : SUCCESS in 5m 26s\n- http://logs.openstack.org/36834/2/gate/gate-cinder-python27/4969 : SUCCESS in 3m 39s\n- http://logs.openstack.org/36834/2/gate/gate-tempest-devstack-vm-full/32745 : SUCCESS in 56m 04s\n- http://logs.openstack.org/36834/2/gate/gate-tempest-devstack-vm-postgres-full/32437 : SUCCESS in 45m 34s\n- http://logs.openstack.org/36834/2/gate/gate-tempest-devstack-vm-neutron/4320 : FAILURE in 21m 28s (non-voting)\n- http://logs.openstack.org/36834/2/gate/gate-grenade-devstack-vm/18629 : SUCCESS in 29m 49s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"e2d5326a567be8201442c7a7d371cafaab57dad0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-16 19:54:26.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a042e0732fbac24e412306d337c0742b681cfc2","revisions":{"4751cb5df592c3f854a54846b51a6f9dfa63efbc":{"kind":"REWORK","_number":1,"created":"2013-07-12 13:48:41.000000000","uploader":{"_account_id":6798,"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","username":"jean-baptiste-ransy"},"ref":"refs/changes/34/36834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/36834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/36834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/36834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/36834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/36834/1"}}},"commit":{"parents":[{"commit":"76d8d6c1480ef7b74545107252f47bad858b6ecd","subject":"Merge \"Ignore files created by Sphinx build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/76d8d6c1480ef7b74545107252f47bad858b6ecd"}]}],"author":{"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","date":"2013-07-12 13:41:12.000000000","tz":120},"committer":{"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","date":"2013-07-12 13:48:37.000000000","tz":120},"subject":"CoraidDriver: Allow volumes in error state to be deleted","message":"CoraidDriver: Allow volumes in error state to be deleted\n\nThis fix will allow the delete call to be successfull if volume\ndoesn\u0027t exists on the SAN.\n\nFixes bug 1195788\n\nChange-Id: I0396f1252c8faafff3e8b4f9a4aeffb930350a8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4751cb5df592c3f854a54846b51a6f9dfa63efbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4751cb5df592c3f854a54846b51a6f9dfa63efbc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"76d8d6c1480ef7b74545107252f47bad858b6ecd","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8a042e0732fbac24e412306d337c0742b681cfc2":{"kind":"REWORK","_number":2,"created":"2013-07-16 17:41:25.000000000","uploader":{"_account_id":6798,"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","username":"jean-baptiste-ransy"},"ref":"refs/changes/34/36834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/36834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/36834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/36834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/36834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/36834/2"}}},"commit":{"parents":[{"commit":"5be685218b5dc08b9cd9e024482bbb2ed2eccf7d","subject":"Merge \"Add flag argument to \u0027cinder-manage config list\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5be685218b5dc08b9cd9e024482bbb2ed2eccf7d"}]}],"author":{"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","date":"2013-07-12 13:41:12.000000000","tz":120},"committer":{"name":"Jean-Baptiste Ransy","email":"jean-baptiste.ransy@alyseo.com","date":"2013-07-16 17:41:20.000000000","tz":120},"subject":"CoraidDriver: Allow volumes in error state to be deleted","message":"CoraidDriver: Allow volumes in error state to be deleted\n\nThis fix will allow the delete call to be successfull if volume\ndoesn\u0027t exists on the SAN.\n\nFixes bug 1195788\n\nChange-Id: I0396f1252c8faafff3e8b4f9a4aeffb930350a8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a042e0732fbac24e412306d337c0742b681cfc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a042e0732fbac24e412306d337c0742b681cfc2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5be685218b5dc08b9cd9e024482bbb2ed2eccf7d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
