)]}'
{"id":"openstack%2Fcinder~40035","triplet_id":"openstack%2Fcinder~master~I27640a3c8b255be51bc6396d238098bd25af61ec","project":"openstack/cinder","branch":"master","topic":"bp/user-locale-api","hashtags":[],"change_id":"I27640a3c8b255be51bc6396d238098bd25af61ec","subject":"Sync gettextutils from oslo","status":"MERGED","created":"2013-08-03 00:53:47.000000000","updated":"2013-08-06 09:16:20.000000000","submitted":"2013-08-06 09:16:20.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"146399e539661532c12701d883ab82fc01ee3d50","_number":40035,"virtual_id_number":40035,"owner":{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},{"value":2,"date":"2013-08-06 09:16:20.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-08-05 18:52:09.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},{"value":0,"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2013-08-06 06:58:16.000000000","_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":2,"date":"2013-08-06 08:29:06.000000000","_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},{"value":0,"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2013-08-06 06:58:16.000000000","_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":1,"date":"2013-08-06 08:29:06.000000000","_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},{"value":0,"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},{"value":0,"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-08-05 00:05:36.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2013-08-05 16:53:29.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2013-08-05 18:52:09.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-08-06 06:58:16.000000000","updated_by":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"reviewer":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"state":"REVIEWER"},{"updated":"2013-08-06 08:29:06.000000000","updated_by":{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},"reviewer":{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},"state":"REVIEWER"},{"updated":"2013-08-06 09:16:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"fcf70f582b2121e916696a4571014cdd452f645f","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-08-03 01:13:44.000000000","message":"Patch Set 1: Works for me\n\nSmokeStack Results (patch set 1):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/108263\n\tLibvirt (Centos 6.4) Success: http://smokestack.openstack.org/?go\u003d/jobs/108264\n","accounts_in_message":[],"_revision_number":1},{"id":"6ace73f26ede31a56cc64afa2793fa0d7027e2b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-03 01:43:48.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/35/40035/1/check/gate-cinder-pep8/d81acc0 : SUCCESS in 3m 55s\n- http://docs-draft.openstack.org/35/40035/1/check/gate-cinder-docs/79e0594/doc/build/html/ : SUCCESS in 4m 33s\n- http://logs.openstack.org/35/40035/1/check/gate-cinder-python26/db2e0bb : SUCCESS in 6m 18s\n- http://logs.openstack.org/35/40035/1/check/gate-cinder-python27/bade285 : SUCCESS in 3m 55s\n- http://logs.openstack.org/35/40035/1/check/gate-cinder-pylint/580d8ee : SUCCESS in 3m 52s (non-voting)\n- http://logs.openstack.org/35/40035/1/check/gate-tempest-devstack-vm-full/58a4e97 : SUCCESS in 49m 53s\n- http://logs.openstack.org/35/40035/1/check/gate-tempest-devstack-vm-cells-full/5efc0d9 : FAILURE in 27m 01s (non-voting)\n- http://logs.openstack.org/35/40035/1/check/gate-tempest-devstack-vm-testr-full/cb20743 : FAILURE in 26m 23s (non-voting)\n- http://logs.openstack.org/35/40035/1/check/gate-tempest-devstack-vm-postgres-full/cf1055f : SUCCESS in 45m 22s\n- http://logs.openstack.org/35/40035/1/check/gate-tempest-devstack-vm-neutron/b755ee0 : SUCCESS in 21m 35s (non-voting)\n- http://logs.openstack.org/35/40035/1/check/gate-grenade-devstack-vm/e265a45 : SUCCESS in 22m 06s\n","accounts_in_message":[],"_revision_number":1},{"id":"a93001d81185f2ba073d8dd10ea03a6182bd7041","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2013-08-05 00:05:36.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"4387da8c8f47241fffae9bdd1dea2b90e5958aa5","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2013-08-05 16:53:29.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"fac375c26440c7e175c4d4bb1da89081fc08855d","author":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"date":"2013-08-05 17:03:28.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\nThis is out of sync.  I know it took a long time - sorry.  Please bring in the latest version.","accounts_in_message":[],"_revision_number":1},{"id":"927c14f73ac82592f63ed8c638cd18e96a573e9f","author":{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},"date":"2013-08-05 18:16:53.000000000","message":"Patch Set 1:\n\nAvishay thanks. Didn\u0027t see that gettextutils had been updated today. Will get udpated in the next patch set.","accounts_in_message":[],"_revision_number":1},{"id":"72fc855a92999305600f6335c403775eb7271e3d","author":{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},"date":"2013-08-05 18:26:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"82e01e95d29d5ed4ef3dca0d3bb69c03d9ef2df8","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-08-05 18:52:09.000000000","message":"Patch Set 2: Works for me\n\nSmokeStack Results (patch set 2):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/109170\n\tLibvirt (Centos 6.4) Success: http://smokestack.openstack.org/?go\u003d/jobs/109171\n","accounts_in_message":[],"_revision_number":2},{"id":"d077ddbdc5bbf06b476c69f3a4c7c9c352066410","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-05 19:24:37.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/35/40035/2/check/gate-cinder-pep8/f09178f : SUCCESS in 3m 59s\n- http://docs-draft.openstack.org/35/40035/2/check/gate-cinder-docs/137eae9/doc/build/html/ : SUCCESS in 5m 13s\n- http://logs.openstack.org/35/40035/2/check/gate-cinder-python26/06f4a4f : SUCCESS in 7m 33s\n- http://logs.openstack.org/35/40035/2/check/gate-cinder-python27/f893375 : SUCCESS in 3m 54s\n- http://logs.openstack.org/35/40035/2/check/gate-cinder-pylint/2deab56 : SUCCESS in 3m 58s (non-voting)\n- http://logs.openstack.org/35/40035/2/check/gate-tempest-devstack-vm-full/8f9c7f4 : SUCCESS in 45m 36s\n- http://logs.openstack.org/35/40035/2/check/gate-tempest-devstack-vm-cells-full/cc7a205 : FAILURE in 13m 55s (non-voting)\n- http://logs.openstack.org/35/40035/2/check/gate-tempest-devstack-vm-testr-full/f0c39dd : FAILURE in 32m 26s (non-voting)\n- http://logs.openstack.org/35/40035/2/check/gate-tempest-devstack-vm-postgres-full/9c7286a : SUCCESS in 58m 10s\n- http://logs.openstack.org/35/40035/2/check/gate-tempest-devstack-vm-neutron/d9b1694 : SUCCESS in 20m 58s\n- http://logs.openstack.org/35/40035/2/check/gate-grenade-devstack-vm/46fe5a4 : SUCCESS in 25m 01s\n","accounts_in_message":[],"_revision_number":2},{"id":"1c26b2b4abc4cd2e5f78340f60c25f9bc589a518","author":{"_account_id":4355,"name":"Avishay Traeger","email":"avishay@stratoscale.com","username":"avishay-il"},"date":"2013-08-06 06:58:16.000000000","message":"Patch Set 2: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"d8878d5062a3132cfa793a847911eec6b0732722","author":{"_account_id":2759,"name":"Huang Zhiteng","email":"winston.d@gmail.com","username":"zhiteng-huang"},"date":"2013-08-06 08:29:06.000000000","message":"Patch Set 2: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":2},{"id":"c58262ae21908d327ac0715b198d8915249df88b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-06 08:29:08.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"319dafaa3f8df74e386c6bec5b43f7b6d9c64039","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-06 09:16:20.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/35/40035/2/gate/gate-cinder-docs/9be0e70/doc/build/html/ : SUCCESS in 8m 36s\n- http://logs.openstack.org/35/40035/2/gate/gate-cinder-pep8/547c685 : SUCCESS in 5m 23s\n- http://logs.openstack.org/35/40035/2/gate/gate-cinder-python26/809e434 : SUCCESS in 7m 42s\n- http://logs.openstack.org/35/40035/2/gate/gate-cinder-python27/de3e741 : SUCCESS in 6m 26s\n- http://logs.openstack.org/35/40035/2/gate/gate-tempest-devstack-vm-full/9a9c711 : SUCCESS in 43m 31s\n- http://logs.openstack.org/35/40035/2/gate/gate-tempest-devstack-vm-postgres-full/86155f0 : SUCCESS in 44m 06s\n- http://logs.openstack.org/35/40035/2/gate/gate-tempest-devstack-vm-neutron/7ea65cb : SUCCESS in 21m 58s\n- http://logs.openstack.org/35/40035/2/gate/gate-grenade-devstack-vm/c385738 : SUCCESS in 25m 50s\n","accounts_in_message":[],"_revision_number":2},{"id":"320dd8df7173b0895ef1c2b0663f875be4852704","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-06 09:16:20.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a48a0f11cf1433f676744c07ccce9b322ec2186c","revisions":{"c8d2dbd1606f0f99191c529df7a723502e9d5701":{"kind":"REWORK","_number":1,"created":"2013-08-03 00:53:47.000000000","uploader":{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},"ref":"refs/changes/35/40035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/40035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/40035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/40035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/40035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/40035/1"}}},"commit":{"parents":[{"commit":"1de031a78180375a722a1f9cbd2b2bbde207c9b1","subject":"Merge \"Tidy up the SSH call to avoid injection attacks in storwize_svc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1de031a78180375a722a1f9cbd2b2bbde207c9b1"}]}],"author":{"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","date":"2013-07-23 18:53:38.000000000","tz":0},"committer":{"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","date":"2013-08-03 00:25:52.000000000","tz":0},"subject":"Sync gettextutils from oslo","message":"Sync gettextutils from oslo\n\nThe oslo changes necessary for delayed translation were refactored in\noslo. This patch set brings in the refactored changes, implemented\nunder the same change-id mentioned below.\n\nPartially implements bp user-locale-api\n\nChange-Id: I27640a3c8b255be51bc6396d238098bd25af61ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8d2dbd1606f0f99191c529df7a723502e9d5701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8d2dbd1606f0f99191c529df7a723502e9d5701"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1de031a78180375a722a1f9cbd2b2bbde207c9b1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a48a0f11cf1433f676744c07ccce9b322ec2186c":{"kind":"REWORK","_number":2,"created":"2013-08-05 18:26:21.000000000","uploader":{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},"ref":"refs/changes/35/40035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/40035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/40035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/40035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/40035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/40035/2"}}},"commit":{"parents":[{"commit":"2380feff7cf220ee03ac11605905bd581bf48b39","subject":"Merge \"use \u0027exc_info\u003d1\u0027 instead of import traceback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2380feff7cf220ee03ac11605905bd581bf48b39"}]}],"author":{"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","date":"2013-07-23 18:53:38.000000000","tz":0},"committer":{"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","date":"2013-08-05 18:26:07.000000000","tz":0},"subject":"Sync gettextutils from oslo","message":"Sync gettextutils from oslo\n\nThe oslo changes necessary for delayed translation were refactored in\noslo. This patch set brings in the refactored changes, implemented\nunder the same change-id mentioned below.\n\nPartially implements bp user-locale-api\n\nChange-Id: I27640a3c8b255be51bc6396d238098bd25af61ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a48a0f11cf1433f676744c07ccce9b322ec2186c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a48a0f11cf1433f676744c07ccce9b322ec2186c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2380feff7cf220ee03ac11605905bd581bf48b39","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
