)]}'
{"id":"openstack%2Fcinder~56482","triplet_id":"openstack%2Fcinder~master~I975859699ea882110be6853a9ad6cd8c40253cc3","project":"openstack/cinder","branch":"master","topic":"glusterfs_perms","hashtags":[],"change_id":"I975859699ea882110be6853a9ad6cd8c40253cc3","subject":"GlusterFS: check whether share is writable before snapshotting","status":"ABANDONED","created":"2013-11-14 23:31:36.000000000","updated":"2013-12-03 06:03:04.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0cf83f6bf9241c9d14f8412c69e867a7fecd5ed4","_number":56482,"virtual_id_number":56482,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2013-11-19 22:52:31.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8574,"name":"Qin Zhao","username":"zhaoqin","inactive":true}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-1,"date":"2013-11-19 21:58:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"date":"2013-11-20 15:25:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},{"value":0,"date":"2013-11-19 21:47:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8574,"name":"Qin Zhao","username":"zhaoqin","inactive":true}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2013-11-19 21:58:57.000000000","_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"date":"2013-11-20 15:25:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},{"value":0,"date":"2013-11-19 21:47:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8574,"name":"Qin Zhao","username":"zhaoqin","inactive":true}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8574,"name":"Qin Zhao","username":"zhaoqin","inactive":true}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},{"_account_id":8574,"name":"Qin Zhao","username":"zhaoqin","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-11-15 21:09:16.000000000","updated_by":{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},"reviewer":{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},"state":"REVIEWER"},{"updated":"2013-11-18 07:36:04.000000000","updated_by":{"_account_id":8574,"name":"Qin Zhao","username":"zhaoqin","inactive":true},"reviewer":{"_account_id":8574,"name":"Qin Zhao","username":"zhaoqin","inactive":true},"state":"REVIEWER"},{"updated":"2013-11-19 21:47:06.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2013-11-19 21:58:57.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2013-11-19 22:52:31.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"634fd63ad57d67d53633f8e22bf7dabd48df1915","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-11-14 23:37:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93e0502d8a425d6bebd00e7c2cdaa451afc1fddd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-15 00:12:33.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- gate-cinder-pep8 http://logs.openstack.org/82/56482/2/check/gate-cinder-pep8/4b0c2c0 : SUCCESS in 2m 51s\n- gate-cinder-docs http://docs-draft.openstack.org/82/56482/2/check/gate-cinder-docs/2b0003b/doc/build/html/ : SUCCESS in 5m 44s\n- gate-cinder-python26 http://logs.openstack.org/82/56482/2/check/gate-cinder-python26/b294133 : SUCCESS in 5m 39s\n- gate-cinder-python27 http://logs.openstack.org/82/56482/2/check/gate-cinder-python27/ccb2ce6 : SUCCESS in 5m 20s\n- gate-cinder-pylint http://logs.openstack.org/82/56482/2/check/gate-cinder-pylint/759b45f : SUCCESS in 4m 45s (non-voting)\n- check-tempest-devstack-vm-full http://logs.openstack.org/82/56482/2/check/check-tempest-devstack-vm-full/de7a7e1 : SUCCESS in 32m 03s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/82/56482/2/check/check-tempest-devstack-vm-postgres-full/9066e8b : SUCCESS in 31m 45s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/82/56482/2/check/check-tempest-devstack-vm-neutron/cbbda22 : SUCCESS in 25m 55s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/82/56482/2/check/gate-tempest-devstack-vm-large-ops/ff3291d : SUCCESS in 13m 56s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/82/56482/2/check/gate-tempest-devstack-vm-neutron-large-ops/c931176 : SUCCESS in 17m 23s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/82/56482/2/check/check-grenade-devstack-vm/46053f6 : SUCCESS in 34m 48s\n","accounts_in_message":[],"_revision_number":2},{"id":"32472b359ab77f82c1c1091a7f93d7590af75cfd","author":{"_account_id":7996,"name":"Luis A. Garcia","email":"luis@linux.vnet.ibm.com","username":"luisg"},"date":"2013-11-15 21:09:16.000000000","message":"Patch Set 2: Looks good to me, but someone else must approve\n\nCool","accounts_in_message":[],"_revision_number":2},{"id":"3a0979135737c759a380c1930c7dbe1a980cb931","author":{"_account_id":8574,"name":"Qin Zhao","username":"zhaoqin","inactive":true},"date":"2013-11-18 07:36:04.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(2 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"82e746e76ea1b734182357595746ba209087f0fa","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2013-11-18 23:20:43.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"bea017e219e5c517430771d04bfb8998a3aa4da0","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-11-19 21:11:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"32d0736eace2b9fadc5e0e76810ece068bc550c2","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-11-19 21:12:17.000000000","message":"Patch Set 2: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"d1824672b16a7cbb51fb6c6a3fa8df26860d7382","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2013-11-19 21:47:06.000000000","message":"Patch Set 3: (1 inline comment)\n\nI think we are good after you address my one comment.","accounts_in_message":[],"_revision_number":3},{"id":"f877f40b14bca516bddb5627592d04d31b8104f6","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2013-11-19 21:58:57.000000000","message":"Patch Set 3: I would prefer that you didn\u0027t merge this\n\nI don\u0027t like the idea of waiting for a config error to be caught by the end user when they try to perform an operation.  I\u0027d rather see this checked on init and even better as we discussed in IRC just perform the chmod in the driver and save backward compat and configuration concerns altogether.","accounts_in_message":[],"_revision_number":3},{"id":"6fd878da9f1ab7a9374f111609a78a77a328bbed","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-19 22:52:31.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-cinder-pep8 http://logs.openstack.org/82/56482/3/check/gate-cinder-pep8/7c38004 : SUCCESS in 2m 52s\n- gate-cinder-docs http://docs-draft.openstack.org/82/56482/3/check/gate-cinder-docs/87a7d28/doc/build/html/ : SUCCESS in 5m 20s\n- gate-cinder-python26 http://logs.openstack.org/82/56482/3/check/gate-cinder-python26/9598f04 : SUCCESS in 4m 39s\n- gate-cinder-python27 http://logs.openstack.org/82/56482/3/check/gate-cinder-python27/71cef7b : SUCCESS in 3m 46s\n- gate-cinder-pylint http://logs.openstack.org/82/56482/3/check/gate-cinder-pylint/9c58291 : SUCCESS in 4m 41s (non-voting)\n- check-tempest-devstack-vm-full http://logs.openstack.org/82/56482/3/check/check-tempest-devstack-vm-full/a739d6c : SUCCESS in 53m 42s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/82/56482/3/check/check-tempest-devstack-vm-postgres-full/d49f907 : FAILURE in 37m 00s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/82/56482/3/check/check-tempest-devstack-vm-neutron/42f72dc : SUCCESS in 40m 23s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/82/56482/3/check/gate-tempest-devstack-vm-large-ops/0a67187 : SUCCESS in 16m 29s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/82/56482/3/check/gate-tempest-devstack-vm-neutron-large-ops/d7aca5c : FAILURE in 22m 36s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/82/56482/3/check/check-grenade-devstack-vm/bb862fd : FAILURE in 30m 31s\n","accounts_in_message":[],"_revision_number":3},{"id":"3f6db09b614f4134630140485974a957f57c4fa1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-11-20 15:25:06.000000000","message":"Patch Set 3:\n\nrecheck bug 1251512","accounts_in_message":[],"_revision_number":3},{"id":"c4e4a38ddf520dddbab5798a7fa3db95d2bed2dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-20 17:50:35.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-cinder-pep8 http://logs.openstack.org/82/56482/3/check/gate-cinder-pep8/e196215 : SUCCESS in 4m 49s\n- gate-cinder-docs http://docs-draft.openstack.org/82/56482/3/check/gate-cinder-docs/7fee215/doc/build/html/ : SUCCESS in 7m 31s\n- gate-cinder-python26 http://logs.openstack.org/82/56482/3/check/gate-cinder-python26/6b42498 : SUCCESS in 8m 21s\n- gate-cinder-python27 http://logs.openstack.org/82/56482/3/check/gate-cinder-python27/a93cef5 : SUCCESS in 8m 07s\n- gate-cinder-pylint http://logs.openstack.org/82/56482/3/check/gate-cinder-pylint/101c2df : SUCCESS in 7m 41s (non-voting)\n- check-tempest-devstack-vm-full http://logs.openstack.org/82/56482/3/check/check-tempest-devstack-vm-full/dcc9116 : FAILURE in 1h 30m 20s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/82/56482/3/check/check-tempest-devstack-vm-postgres-full/8fe2255 : SUCCESS in 1h 16m 43s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/82/56482/3/check/check-tempest-devstack-vm-neutron/79ca3ff : SUCCESS in 1h 03m 27s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/82/56482/3/check/gate-tempest-devstack-vm-large-ops/0b0c098 : SUCCESS in 21m 19s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/82/56482/3/check/gate-tempest-devstack-vm-neutron-large-ops/8610b3c : SUCCESS in 21m 49s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/82/56482/3/check/check-grenade-devstack-vm/dce1f3e : FAILURE in 1h 00m 08s\n","accounts_in_message":[],"_revision_number":3},{"id":"fcddcd1cffec3df378a74f32fb6aa14d1b162a83","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-11-20 20:18:41.000000000","message":"Patch Set 3:\n\nrecheck bug 1240175","accounts_in_message":[],"_revision_number":3},{"id":"5e988fb0d0d8f017821bc60eec2d0b030a248879","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-20 21:06:11.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-cinder-pep8 http://logs.openstack.org/82/56482/3/check/gate-cinder-pep8/b548490 : SUCCESS in 2m 40s\n- gate-cinder-docs http://docs-draft.openstack.org/82/56482/3/check/gate-cinder-docs/9db7a23/doc/build/html/ : SUCCESS in 6m 32s\n- gate-cinder-python26 http://logs.openstack.org/82/56482/3/check/gate-cinder-python26/e8bbfc4 : SUCCESS in 6m 32s\n- gate-cinder-python27 http://logs.openstack.org/82/56482/3/check/gate-cinder-python27/fdee91d : SUCCESS in 5m 18s\n- gate-cinder-pylint http://logs.openstack.org/82/56482/3/check/gate-cinder-pylint/d1b3ba2 : SUCCESS in 5m 53s (non-voting)\n- check-tempest-devstack-vm-full http://logs.openstack.org/82/56482/3/check/check-tempest-devstack-vm-full/c050494 : SUCCESS in 39m 50s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/82/56482/3/check/check-tempest-devstack-vm-postgres-full/3ee508b : SUCCESS in 37m 34s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/82/56482/3/check/check-tempest-devstack-vm-neutron/61b3f41 : SUCCESS in 41m 03s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/82/56482/3/check/gate-tempest-devstack-vm-large-ops/7dbe1e6 : SUCCESS in 22m 38s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/82/56482/3/check/gate-tempest-devstack-vm-neutron-large-ops/d318932 : SUCCESS in 20m 38s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/82/56482/3/check/check-grenade-devstack-vm/6c7a37c : FAILURE in 35m 06s\n","accounts_in_message":[],"_revision_number":3},{"id":"4bf62281dc56856556de3459a5c28d6da61260cf","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-11-20 22:28:46.000000000","message":"Patch Set 3:\n\nSee alternative approach at https://review.openstack.org/#/c/57542/ .","accounts_in_message":[],"_revision_number":3},{"id":"a45fd38a5ea4f382975b9ef9217135249f08785c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2013-11-25 22:53:10.000000000","message":"Patch Set 3: Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"8789746b0d42e943fb59bd5be1097b2a647888e8","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-12-03 06:03:04.000000000","message":"Patch Set 3: Abandoned\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":3},{"id":"dc4ebaadc533581ff96053773e59afe95e0ef863","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-12-03 06:03:04.000000000","message":"Patch Set 3:\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b481b35ee400647ff1ab37906229922fb15b654a","revisions":{"ec9d8181c83504e6b7d3d7968ed37e97f8b2b33a":{"kind":"REWORK","_number":1,"created":"2013-11-14 23:31:36.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/82/56482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/56482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/56482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/56482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/56482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/56482/1"}}},"commit":{"parents":[{"commit":"f4dff4d337466f0b63a60f978acaf29d861a6b4b","subject":"Merge \"Remove dup of LVMISCSIDriver in LVMISERDriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4dff4d337466f0b63a60f978acaf29d861a6b4b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-11-14 19:58:21.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-11-14 23:32:36.000000000","tz":-300},"subject":"GlusterFS: check whether share is writable before snapshotting","message":"GlusterFS: check whether share is writable before snapshotting\n\nTest whether the share is writable before attempting to create or\ndelete a qcow2 snapshot.  This ensures that the entire snapshot\noperation can be committed to disk rather than allowing the\nsnapshot info file and qcow2 data to get out of sync if part of\nthe operation fails due to the Cinder user not being able to write\nto the directory.\n\nCloses-Bug: #1236966\n\nChange-Id: I975859699ea882110be6853a9ad6cd8c40253cc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec9d8181c83504e6b7d3d7968ed37e97f8b2b33a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec9d8181c83504e6b7d3d7968ed37e97f8b2b33a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f4dff4d337466f0b63a60f978acaf29d861a6b4b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c6223f0d1fec5f8b5aa539e2926120f86cefefbc":{"kind":"REWORK","_number":2,"created":"2013-11-14 23:37:32.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/82/56482/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/56482/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/56482/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/56482/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/56482/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/56482/2"}}},"commit":{"parents":[{"commit":"f4dff4d337466f0b63a60f978acaf29d861a6b4b","subject":"Merge \"Remove dup of LVMISCSIDriver in LVMISERDriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4dff4d337466f0b63a60f978acaf29d861a6b4b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-11-14 19:58:21.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-11-14 23:38:08.000000000","tz":-300},"subject":"GlusterFS: check whether share is writable before snapshotting","message":"GlusterFS: check whether share is writable before snapshotting\n\nTest whether the share is writable before attempting to create or\ndelete a qcow2 snapshot.  This ensures that the entire snapshot\noperation can be committed to disk rather than allowing the\nsnapshot info file and qcow2 data to get out of sync if part of\nthe operation fails due to the Cinder user not being able to write\nto the directory.\n\nCloses-Bug: #1236966\n\nChange-Id: I975859699ea882110be6853a9ad6cd8c40253cc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6223f0d1fec5f8b5aa539e2926120f86cefefbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6223f0d1fec5f8b5aa539e2926120f86cefefbc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f4dff4d337466f0b63a60f978acaf29d861a6b4b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b481b35ee400647ff1ab37906229922fb15b654a":{"kind":"REWORK","_number":3,"created":"2013-11-19 21:11:38.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/82/56482/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/56482/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/56482/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/56482/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/56482/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/56482/3"}}},"commit":{"parents":[{"commit":"f4dff4d337466f0b63a60f978acaf29d861a6b4b","subject":"Merge \"Remove dup of LVMISCSIDriver in LVMISERDriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4dff4d337466f0b63a60f978acaf29d861a6b4b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-11-14 19:58:21.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2013-11-19 21:09:24.000000000","tz":-300},"subject":"GlusterFS: check whether share is writable before snapshotting","message":"GlusterFS: check whether share is writable before snapshotting\n\nTest whether the share is writable before attempting to create or\ndelete a qcow2 snapshot.  This ensures that the entire snapshot\noperation can be committed to disk rather than allowing the\nsnapshot info file and qcow2 data to get out of sync if part of\nthe operation fails due to the Cinder user not being able to write\nto the directory.\n\nCloses-Bug: #1236966\n\nChange-Id: I975859699ea882110be6853a9ad6cd8c40253cc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b481b35ee400647ff1ab37906229922fb15b654a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b481b35ee400647ff1ab37906229922fb15b654a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f4dff4d337466f0b63a60f978acaf29d861a6b4b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
